)]}'
{"id":"openstack%2Fswift~878597","triplet_id":"openstack%2Fswift~stable%2Fussuri~I650a80cb45febc457c42360061faf3a9799e6131","project":"openstack/swift","branch":"stable/ussuri","topic":"update-url-stable/yoga-stable/victoria-stable/ussuri","attention_set":{},"removed_from_attention_set":{"33787":{"account":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"last_update":"2024-02-07 13:23:21.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I650a80cb45febc457c42360061faf3a9799e6131","subject":"Update url","status":"ABANDONED","created":"2023-03-27 02:09:18.000000000","updated":"2024-02-07 13:23:21.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":878596,"cherry_pick_of_patch_set":1,"meta_rev_id":"146e640f56d8e40a6c01c430610a7a8f666f7399","_number":878597,"virtual_id_number":878597,"owner":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-28 03:56:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-27 03:24:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 13:23:21.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"11b534a1d4d65a4a67915bd6c926a169fcd3814f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-27 02:09:18.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"789a692a4e0d5f1835c47585f408f472dc289f97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 03:24:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/96b3ea924c9d4074b4c9346091d0c584\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3a35628da7954319a8be700e71bfb3d3 : SUCCESS in 1m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ce82c95122f4fe3aac5a1b42eb50caa : SUCCESS in 7m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b6315d1c9ff4cea9194127f3ffd3265 : FAILURE in 4m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b8620993ad774d0497d965498a481e3a : FAILURE in 3m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/47ff69ccb4cf48218e0e0e9f69018907 : SUCCESS in 14m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef43d4fa86564acc9a9072ac102850cc : SUCCESS in 14m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2f3d2d7a4776421da9b5de3aa7cbeeb4 : SUCCESS in 12m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3904c320be134b80b7a3db2d64d1f54b : SUCCESS in 12m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5dfbfa8aaedf43c88fb0f7c81eb0c180 : SUCCESS in 9m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fd915c8fd86546638e7d43c59cc3f9d7 : SUCCESS in 12m 31s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/b9260cb2355443639502b587361a6d55 : SUCCESS in 9m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b7cdca8ea72041a886d40357642fc56f : SUCCESS in 11m 56s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/36c40139be9f403ea0934e8794acd3be : SUCCESS in 15m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/db4cea6edf3348a1a2f34c6d203ced2f : SUCCESS in 10m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d88e7b9ed61d49c6a13f69ffba197446 : SUCCESS in 15m 49s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5de5864a05b747aa973ee19629110dd9 : FAILURE in 28m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fdfb3d6e72e548f881e4aa9f100ab0e4 : SUCCESS in 10m 32s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/53e4701d452247478450e9d2cbf10c71 : SUCCESS in 22m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1e8c00e7042f475fb17c649904fdc68b : FAILURE in 8m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e294b07e4cbb445e81a20aafd359f6d5 : SUCCESS in 47m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/674a54120d8f43088f5ee7326b03a87e : SUCCESS in 28m 17s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f45776098dcd470eb0db1ed33963546a : SUCCESS in 28m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/27e17bfb3e7541f0b885b40669b35141 : SUCCESS in 52m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca43dddf2e38462c92f725efe015412d : SUCCESS in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58d7d482a0144a3b8167807274789b0b : SUCCESS in 6m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/368ebbcc4c2f4bca8b7779dc830a5e04 : RETRY_LIMIT in 6m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f611a6971e91b93ba0fa971cc427e5b87d1cac7","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-28 02:07:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"555c6bbce4ea9fcf305d423c889963984ec96b34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 03:56:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b617a8cea174e2cae08b13d179f0df5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/58d252da8e6b45c8a9325f833378334f : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27e454780fed4d72b89e4f0c7d558041 : SUCCESS in 7m 48s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f1025c0ec1848d38a08dac28b4f73ce : FAILURE in 3m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1391389bf36146b8bf7e5ef33885a8e9 : FAILURE in 3m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8c3981420bd541a4b0a0ab59e3e94d5b : SUCCESS in 11m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/27752335a38b4830b254c719cac53db9 : SUCCESS in 11m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f4e31928eb424c158adea76606cc3205 : SUCCESS in 14m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9b3080c123594fa4bd7d0354a3421850 : SUCCESS in 12m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e80f42f2103248c882e473a66b38234f : SUCCESS in 9m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ff34726abbbc4154aefff361f1084a77 : SUCCESS in 11m 36s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/25d43b4a479c4b4c8c50ab1e4f816f8e : SUCCESS in 9m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5ed242b106ec4e34997012ed73429449 : SUCCESS in 9m 17s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/482310874b0c414490e30195242c57b1 : SUCCESS in 13m 38s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5e2f7dd96ed146c2887280c5e888bfe9 : SUCCESS in 10m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c887f5038db241cbb61560c9bcbf68ea : SUCCESS in 13m 25s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0cfe06127a39459397ab4b72b11daaba : SUCCESS in 10m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0d199b90a3a344eb9691b35b77d1dde4 : SUCCESS in 12m 04s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b37b54bbbee74f7c975cbfdf53e925a7 : SUCCESS in 16m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ee37f9cc4f85442687580564e99fc9e7 : FAILURE in 7m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0f3fb3835fc04acc8a8f877906743b3e : SUCCESS in 49m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/426cead9b4e5444a8f9b2f1185bec36c : SUCCESS in 23m 33s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/fe448cc9cb9d4a169fdf61608b26048c : SUCCESS in 21m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/53c35ee9726644539d64afad598996e9 : SUCCESS in 23m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd5dfb82cd844008bf0ecfaf7aa44945 : SUCCESS in 17m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36d22878d0c14409be9f5848076b367a : SUCCESS in 5m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bd525f5c129e44abae3bb8dca8486406 : RETRY_LIMIT in 5m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"146e640f56d8e40a6c01c430610a7a8f666f7399","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-07 13:23:21.000000000","message":"Abandoned\n\nstable/ussuri branch of openstack/swift transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd9602b833669b5f7eeb9ee011e418ad67f3fb73","revisions":{"bd9602b833669b5f7eeb9ee011e418ad67f3fb73":{"kind":"REWORK","_number":1,"created":"2023-03-27 02:09:18.000000000","uploader":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"ref":"refs/changes/97/878597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/878597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/878597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/878597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/878597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/878597/1"}}},"commit":{"parents":[{"commit":"1721d8586c53609f75070e10e21de9fcf262c0bc","subject":"Merge \"s3api: Prevent XXE injections\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1721d8586c53609f75070e10e21de9fcf262c0bc"}]}],"author":{"name":"Wei LingFei","email":"weilingfei@uniontech.com","date":"2023-03-24 06:11:44.000000000","tz":480},"committer":{"name":"weilingfei","email":"weilingfei@uniontech.com","date":"2023-03-27 02:09:18.000000000","tz":0},"subject":"Update url","message":"Update url\n\nThe OpenStack project is currently maintained on opendev.org, with github.com serving as a mirror repository.\n\nReplace the source code repository address for the python-swiftclient project from github.com to opendev.org.\n\nChange-Id: I650a80cb45febc457c42360061faf3a9799e6131\n(cherry picked from commit 5032b42b10548bda6d27dba81cfec2f95a67d91b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd9602b833669b5f7eeb9ee011e418ad67f3fb73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd9602b833669b5f7eeb9ee011e418ad67f3fb73"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
