)]}'
{"id":"openstack%2Fswift~878598","triplet_id":"openstack%2Fswift~stable%2Ftrain~I650a80cb45febc457c42360061faf3a9799e6131","project":"openstack/swift","branch":"stable/train","topic":"update-url-stable/yoga-stable/train","attention_set":{},"removed_from_attention_set":{"33787":{"account":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"last_update":"2023-12-21 14:05:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I650a80cb45febc457c42360061faf3a9799e6131","subject":"Update url","status":"ABANDONED","created":"2023-03-27 02:09:40.000000000","updated":"2023-12-21 14:05:41.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":878593,"cherry_pick_of_patch_set":2,"meta_rev_id":"8a9807dc883a96e186bd0f02837da6044f01ae03","_number":878598,"virtual_id_number":878598,"owner":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-11 10:42:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-27 03:23:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-11 09:47:13.000000000","updated_by":{"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"},"reviewer":{"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"},"state":"CC"},{"updated":"2023-10-11 09:47:29.000000000","updated_by":{"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"},"reviewer":{"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"},"state":"REMOVED"}],"messages":[{"id":"60b16fcd07b097314f1d89cc9bf2f9d3cb7d71df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"date":"2023-03-27 02:09:40.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"e4a31e8cb5bef0118bf0c630745f550aa3ece60b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 03:23:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff00bc0e2143461f90e491569b82dd17\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8ab3796fbfcd475da538f376dc3c676a : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba3fe107c6134bc98e3714ca6c776c43 : SUCCESS in 5m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/680c9390d086416bb0e0c93214f8cc0b : FAILURE in 3m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a4b5734251b6490bbe9d4d19032d33f2 : SUCCESS in 13m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/756c6a1d382347a1ab3ba2bcb4340d78 : SUCCESS in 14m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/26ad43291f9a48ca95a3046a4c252447 : SUCCESS in 12m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7e5c6b941963413f9fad28038d5892c6 : SUCCESS in 8m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f371b60a6bbd49dabd24edd3b43ac8d5 : SUCCESS in 10m 24s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a7e27960f838481fbdf331bc9a0d950a : SUCCESS in 8m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/52de36c13c524533831eb59d5007f7f6 : SUCCESS in 8m 39s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/51dbe701bfb94029b844fef6d48e211e : SUCCESS in 13m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/11871040c30241938b4383c3c0b8cfd4 : SUCCESS in 10m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/615357cb700742ff8b9ed78d12f05992 : SUCCESS in 8m 54s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0a42cff18eb34302a7a62c6aa58ab919 : SUCCESS in 10m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4b42d10a5cf44a69a346002c5ac45b51 : SUCCESS in 11m 22s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8cb657f2f61f423a9251c07779d9978e : SUCCESS in 32m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2d7808ff6fc94a7c98e5b09691750fea : FAILURE in 7m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9537b5f0a8794c0abc43de18ef0da5ea : SUCCESS in 39m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15841c092e324534a7a34c52feb54303 : SUCCESS in 27m 46s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/eade5e9fb7b643d18fc0ef65377f3bf2 : SUCCESS in 23m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e8d5996de6924b5b871debbc45137162 : SUCCESS in 23m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4a324f1fcb843efb8e3b961256e6b96 : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85bb95f51bc54d2f837ca9dbc18bc048 : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/321a3003738f4ab9a01f96c842d533c7 : RETRY_LIMIT in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55b8257286b1d6808b66688f1130c4ddb89ef501","author":{"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"},"date":"2023-10-11 09:47:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b80aab9562d1c1c3b98d39ff1d178503684d9935","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"},"date":"2023-10-11 09:47:29.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_33801\u003e.","accounts_in_message":[{"_account_id":33801,"name":"wang","email":"wangleic@uniontech.com"}],"_revision_number":1},{"id":"c410f52dc3ace9ccd639981c140e71ab127d9c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 10:42:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/119136ed2ab84e928c5469f4f10f0a16\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/84ab7bb6bf774c6b9f86ebe0680c6c82 : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f121e7c9b31448c997e27bf48069bb40 : SUCCESS in 5m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/370d9d730e8b45b3ab0d3240bad43670 : FAILURE in 4m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/93b566b7b6e240e39a4f2987fae60b13 : SUCCESS in 18m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a4a6ac2f40a4353ab1643dd3caacd78 : SUCCESS in 13m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/742fcf08dede4f558534c2689879d367 : SUCCESS in 16m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bfac40f416384dc2bb18bf120c4edb4c : SUCCESS in 10m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/76c3e4aa20ed4fefa1a3e54bdc4a90c1 : SUCCESS in 13m 09s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/bee81575fa3f45499222e9c5c6386bd1 : SUCCESS in 10m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/29489fe5e9cf48cfb2a4498deb57ecb3 : SUCCESS in 11m 48s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/dd2e3d121dab4383a7c8c66deb546f3a : SUCCESS in 13m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1ed8c8976fad4554ae0dbbd89430f5f3 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7608dde9a98b4c6980ecd56fd842fb1e : SUCCESS in 11m 35s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/28930c19078b4a8088ba5941e73c2f39 : SUCCESS in 10m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/dcb8b685149d419f8c453d61a2ffd8b4 : SUCCESS in 10m 39s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/0f7080428c4c4065801e0a9534d8af6c : SUCCESS in 11m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1ea20724efc344e4b9f68982998b2739 : FAILURE in 9m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8f229b3227604b619bba60c9c51d6a7a : SUCCESS in 36m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c79064162d9a4399b78ff075da0c6dc4 : SUCCESS in 20m 35s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9a037dac85d847559f7dba80f1c72833 : SUCCESS in 25m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/08811fe975364f47861915b8c8a09654 : SUCCESS in 24m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf4ebd536e3a4ab89977b997dae8b7b1 : FAILURE in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03872f1c2cf4425a9498ba2f1f633f44 : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c13eecf936144ea89ea2c6166353e876 : RETRY_LIMIT in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a9807dc883a96e186bd0f02837da6044f01ae03","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-12-21 14:05:41.000000000","message":"Abandoned\n\nTrain is about to transition to End of Life. Open patches needs to be abandoned before branch deletion.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"634b4beb133179f63627c72562c4fd579e426afd","revisions":{"634b4beb133179f63627c72562c4fd579e426afd":{"kind":"REWORK","_number":1,"created":"2023-03-27 02:09:40.000000000","uploader":{"_account_id":33787,"name":"weilingfei","email":"weilingfei@uniontech.com","username":"weilingfei"},"ref":"refs/changes/98/878598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/878598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/878598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/878598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/878598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/878598/1"}}},"commit":{"parents":[{"commit":"70b68dbf65c7bc67795a8dee65f01c14f9eb94e5","subject":"s3api: Prevent XXE injections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70b68dbf65c7bc67795a8dee65f01c14f9eb94e5"}]}],"author":{"name":"Wei LingFei","email":"weilingfei@uniontech.com","date":"2023-03-24 06:11:44.000000000","tz":480},"committer":{"name":"weilingfei","email":"weilingfei@uniontech.com","date":"2023-03-27 02:09:40.000000000","tz":0},"subject":"Update url","message":"Update url\n\nThe OpenStack project is currently maintained on opendev.org, with github.com serving as a mirror repository.\n\nReplace the source code repository address for the python-swiftclient project from github.com to opendev.org.\n\nChange-Id: I650a80cb45febc457c42360061faf3a9799e6131\n(cherry picked from commit 5032b42b10548bda6d27dba81cfec2f95a67d91b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/634b4beb133179f63627c72562c4fd579e426afd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/634b4beb133179f63627c72562c4fd579e426afd"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
