)]}'
{"id":"openstack%2Fswift~878635","triplet_id":"openstack%2Fswift~master~Ib654ba58cdcbf245458816460a15c964dfdb073c","project":"openstack/swift","branch":"master","topic":"p-sharder-log-fix","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-03-27 15:51:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib654ba58cdcbf245458816460a15c964dfdb073c","subject":"sharder: fix \u0027Cleaving not completed\u0027 warning log","status":"MERGED","created":"2023-03-27 09:45:31.000000000","updated":"2023-03-27 15:52:24.000000000","submitted":"2023-03-27 15:51:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878635-p-sharder-log-fix","meta_rev_id":"b3b9dc3d1dc49e28d2f63093e45708b5f7ab67fd","_number":878635,"virtual_id_number":878635,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-27 15:51:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2023-03-27 13:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2023-03-27 13:48:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-27 10:07:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-27 11:15:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-27 13:48:21.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"174f613df9e3f916f1038141639dbd31ae5cdadd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-27 09:45:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfb1d0e416c75ff45628bef5f8822894c8b579da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 10:07:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3374c63c05c243efb62d1f6bfcd5afe4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/48da481a7b78490e86388027a0236756 : SUCCESS in 16m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/532d947e03e149639a80a71b14b5e646 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d8e1bed09194b529211d7478269ecae : SUCCESS in 19m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a0c0654ed63447195dfcefb4dd0129e : SUCCESS in 20m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ff5328675d71faf692ad08143b3fef1214049e09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 11:15:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b87e003de05942c09ebfa9aec5fdae1b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4154618681124c78b884eedeea45ac47 : SUCCESS in 10m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/851677a4b362421fa8e316a457b1d382 : SUCCESS in 56m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e8053c6ce95c44cea73d25bf1498e7c8 : SUCCESS in 40m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12d21dbf31954169a81333a6da9ad400 : SUCCESS in 50m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ff0e0b818044738a36192e2183ac947 : SUCCESS in 30m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40f5ff3e0ea740eda6434407566f2500 : SUCCESS in 14m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49534c37713f4d9ba1e9b542e93d23d5 : SUCCESS in 30m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ffb2dc5af2764688b9e7e6845d4d0aa3 : SUCCESS in 8m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f771383be5684d0f98881949e6abc375 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a6cc4c8667444fb89fcd2cce1b6c00cf : SUCCESS in 27m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4c94a7785af74144b2ae004236831461 : SUCCESS in 17m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2ab81371922143b0b09de9a96f739158 : SUCCESS in 26m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/96c3419584c64bb0bc07d014a7b0ed28 : SUCCESS in 11m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1a0e2114634e43b9865d7ef26ad61c8e : SUCCESS in 29m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/478b1afe11c94aa78545fb312d80fe9b : SUCCESS in 17m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a83e270ad1894300b8623d835018c337 : SUCCESS in 17m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/45d5f31d1dc24d5eb331b080797b07d1 : SUCCESS in 25m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/02a0c439c448430f8df0ecc9db266c26 : SUCCESS in 14m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a1eac604c0674a019e6b386a9f590d98 : SUCCESS in 27m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/066b5f056a92480f9f834205cc85b8aa : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fd5e6b6030124e32aa1ace717f7cf4f2 : SUCCESS in 25m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/783f47c40d944f8783179f79883b459c : SUCCESS in 8m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3b21a4d627624c7bb4e45d0080918bb8 : SUCCESS in 1h 28m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a661db4ffee4e2eab6acbdb45f607ac : SUCCESS in 59m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4b92cf22be3c43c8b51a6b36b773bd88 : SUCCESS in 51m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d52ed49b11d442e4a85a4b1064082c09 : SUCCESS in 52m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a02e134749b4a7da8f0fd5b41d8025d : SUCCESS in 38m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4558a86bcaf4a75a1fd0fddf4008d70 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c2c6e5fd5b734fe7a79c8f27ff14ed36 : SUCCESS in 24m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/218d8ec5fc8d4f138810b23416d04a30 : SUCCESS in 40m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6fe38730fbc8cb502eaf8a36ac08857e000a6817","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-03-27 13:48:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c5d6c945f8b1f9b2a34e18dd2a6c6ba37a4d2ab3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 13:49:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da6f54d45bae712ccddbf3b659fac7d846f03d81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 15:51:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/addffdca21364896aba835cd94daa979\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e40af1edd244da39ed45dfb4958de14 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/7f5a15fa92324700afd216ddd9662f6e : SUCCESS in 1h 04m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/17f0b55ac1e3453eb4ace12f60ce6c2c : SUCCESS in 52m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5523daaa9a694327bffd26ecf66285dc : SUCCESS in 1h 01m 29s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/89488163d99d43c6b62aa6b78c4ff5eb : SUCCESS in 9m 27s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3b2e5bfc257a49f69d3095f2856769a2 : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ac98439532044654b6cf7bf9894bbe76 : SUCCESS in 14m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cb2631136bd04aa1ba0b39aa439af834 : SUCCESS in 13m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/613b21a8ddfb41208d50ecfbd303ff56 : SUCCESS in 15m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f0a86e7dd5f646eb9694673345b0a1ea : SUCCESS in 15m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/07c7f01b4f8744e09f76ca9f2de7d111 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0c767a3dd8604310b52045d0619f271f : SUCCESS in 16m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/991a2cdb03dd40a0a049f29b90cac456 : SUCCESS in 18m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/59ff51ecb24d462ca0131cc68947a446 : SUCCESS in 11m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6e8aa414a3204198bda4374f5923ee35 : SUCCESS in 10m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/87a514e5679547f6a70d943829546a8c : SUCCESS in 14m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c1de5a341a264445a3ba34dc343d75f0 : SUCCESS in 6m 51s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d68dcd7f96af483d8a252b96672fc0c3 : SUCCESS in 7m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d9114215dc644970b103fb816627a556 : SUCCESS in 56m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d13706eb161e4e1bac97495ee81da6ad : SUCCESS in 1h 04m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9620fe3493fb456d9e436fda117cab55 : SUCCESS in 53m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c237119221ce4ffe824d69e235660379 : SUCCESS in 38m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0003a53389b345ff9a19d32cb5171635 : SUCCESS in 18m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/914ffc0e412d4b71bacacce8c2290ab3 : SUCCESS in 4m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a3124692f924a0bbcc6849dfafc547a : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b164c2175fde943469399957a0339151d4c5c911","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 15:51:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b3b9dc3d1dc49e28d2f63093e45708b5f7ab67fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 15:52:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2512eab0407d424698aa4cdff3d43dc3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/526109d4d2aa4156a7636fe206a0a34e : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/92a8ef3b56db44968217f7f4c25ee480 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/043075c236064505a9266531e49de738 : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e15d6880e810cb2419865136e2cbb856acab8e73","revisions":{"e15d6880e810cb2419865136e2cbb856acab8e73":{"kind":"REWORK","_number":1,"created":"2023-03-27 09:45:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/878635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/878635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/878635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/878635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/878635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/878635/1"}}},"commit":{"parents":[{"commit":"6d3d4197151f44bf28b51257c1a4c5d33411dcae","subject":"Merge \"quotas: Add account-level per-policy quotas\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d3d4197151f44bf28b51257c1a4c5d33411dcae"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-03-27 08:57:48.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-03-27 08:57:48.000000000","tz":60},"subject":"sharder: fix \u0027Cleaving not completed\u0027 warning log","message":"sharder: fix \u0027Cleaving not completed\u0027 warning log\n\nFix missing space between the epoch time and \"DB state:\" in the log\nmessage.\n\nChange-Id: Ib654ba58cdcbf245458816460a15c964dfdb073c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e15d6880e810cb2419865136e2cbb856acab8e73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e15d6880e810cb2419865136e2cbb856acab8e73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
