)]}'
{"id":"openstack%2Fswift~878636","triplet_id":"openstack%2Fswift~master~I8c84f6ee15e6999f71b092bbeed414065a22ee8b","project":"openstack/swift","branch":"master","topic":"p-sharder-log-fix","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-08-08 20:05:27.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I8c84f6ee15e6999f71b092bbeed414065a22ee8b","subject":"sharder: fix host logged when send_shard_ranges fails","status":"MERGED","created":"2023-03-27 09:45:31.000000000","updated":"2025-08-08 21:31:53.000000000","submitted":"2023-03-27 15:51:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"878636-p-sharder-log-fix","meta_rev_id":"33f2a7b9abb5eaa61e03d12424bc0d45fadf6b87","_number":878636,"virtual_id_number":878636,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-27 15:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2023-03-27 14:24:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2023-03-27 14:24:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-27 10:08:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-27 10:58:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-27 14:24:56.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"e80f320802ff0e92fee3c4417f2de1192a16ed01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-27 09:45:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b341a0691b7ba7aa4f5ad63eca0fb53169ec9dae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 10:08:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c8b698cb9e64e958e07d97cb9b9386d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41747b57f73d4314aa798e7e6458eff7 : SUCCESS in 16m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28ffcbb20cf3407fb62527cc1e2ff38b : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b722ca1a19a432b8dfba463e07ae428 : SUCCESS in 21m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8f1f972ad7745b9b4a74710bfa8da8f : SUCCESS in 19m 53s","accounts_in_message":[],"_revision_number":1},{"id":"52cf4b1f2349906a1f57596339c5f89688500aed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 10:58:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e5b3618f4484fd2a5f81d5f7ab52c6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66bafd04f9b94e1c80d67cf12ece5f86 : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/e944e5dea05e4f16aff9ebf5e0b328e5 : SUCCESS in 1h 01m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d21e844ca0b34d24b237dda0a907ea28 : SUCCESS in 41m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69b5fc8d550b4a32a0e4c2077302266c : SUCCESS in 1h 05m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e6f2d080aa846389a99d0124eafd51a : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33ad769d8b494667a973ef75047b9d21 : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1121de34f0024223904fe7c684c826f2 : SUCCESS in 19m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0569f4b9785849738ce5aca26781bcc9 : SUCCESS in 8m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a9c7dba109a74702af613348ca83f9b2 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fc741258b9bb48e3bc7d8608ecbc553f : SUCCESS in 18m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c55d0eac1414448c803838752640e290 : SUCCESS in 17m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fef1cd962c064ac8915747904155b4f7 : SUCCESS in 26m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a3c195cee815446aa0725b0723f57f72 : SUCCESS in 15m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/31fa02f50bc548409ce3198926869b2a : SUCCESS in 14m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/21470d5f57e94fc5884dabcbf0db876b : SUCCESS in 15m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/63aae0fd490d4077880db43c92237867 : SUCCESS in 18m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/091155c57d174709accdac6c5e7aedf0 : SUCCESS in 26m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2d566df8ebfc42e79b7f8a35c61ddccc : SUCCESS in 14m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f3fb99923a724dff8c734f41fb1c3d6a : SUCCESS in 15m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/340f64bdc8bf4a3090dc6eecba29e622 : SUCCESS in 8m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/75ffeb7c6776493c9660b43b0e351159 : SUCCESS in 25m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d7cc63e5f8f14474ac9d797c07a2ffec : SUCCESS in 7m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/126e2330bdb447e18527e6649fd31f35 : SUCCESS in 58m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5b4ceb1d3a646be86d152244a16ef86 : SUCCESS in 59m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/50a7a600f80241c381f03c767ca49828 : SUCCESS in 46m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5db0a880276f4fb5804eee1fb3171076 : SUCCESS in 51m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06a671480f124ef693b802e638a79a2d : SUCCESS in 22m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5fb666961dc4fc39434f052f4edef68 : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d6203c326f9d461e99296f01a6ad7926 : SUCCESS in 19m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41679231325847a18f1a32b88b0c6b3f : SUCCESS in 43m 04s","accounts_in_message":[],"_revision_number":1},{"id":"dd9df1390aaad207666ff493bbcedcd65b9459c0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-03-27 14:24:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"46a9e8a2c5e4a9cc35b80703b3fdcc104f59f063","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 14:25:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f1e1c7db589535b7f8e609bc9d02f304c043419","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 15:51:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a3c80ae4b7e4aad8902c7840ab56702\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf47fe18a80e468db6135e27d00f688b : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/2c47de30ee2b478c88717876547ab1b2 : SUCCESS in 43m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a59154125b94441aab7aa2b021bbfb7f : SUCCESS in 37m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a8c6694d23241619381f771bd021b12 : SUCCESS in 51m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a2078eba585f49259e486672b7b5a05c : SUCCESS in 10m 28s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/61d984ed2db74f6695b59e0d1e7bdc6e : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c608b2e5700b4ad987e872882c274dd9 : SUCCESS in 19m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/84939bbe4a5245abb5fa30fb42ef024f : SUCCESS in 18m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fc6c632cb4e14363897bed00a02e79e9 : SUCCESS in 18m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/13cb5cf453f64479a87f91f0f81f67d0 : SUCCESS in 18m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/86f5e39b006249a3864e1b7883adea30 : SUCCESS in 19m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c829205f35fb4230a719e1cf1743af10 : SUCCESS in 15m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/db41a9cf097445b2aa401e0acb72d5b8 : SUCCESS in 18m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a3b3563bc79a4c4696975d8d398a6dd8 : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d5b805dbf5664947904d48d8af3c96a3 : SUCCESS in 16m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e3818881f794461cbd566b2368613544 : SUCCESS in 18m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cb00b359176249eaa37172fd471613da : SUCCESS in 6m 27s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/db598cb4297d427499145b37e1c10676 : SUCCESS in 6m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/103aaf4cce0240c98940627ed777947a : SUCCESS in 55m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c2253b6013942f7878c60a3844b76f1 : SUCCESS in 56m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4bd8d7ad1f8b45dd80d5889b32eefbbf : SUCCESS in 50m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f403dd58e47c4e24a13d8afc18c89b55 : SUCCESS in 54m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffc63680cc0144a7ad98ada1a2f7db07 : SUCCESS in 18m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b717568848c5436bae4b6094241ad8d9 : SUCCESS in 4m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce845a4851714acf8a84b0662c424d8b : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5cb8d81791af9f0bae1a54f508f66fc85094ece4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 15:51:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b1ff09959abc3847369c37691f11b5d74aecd929","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 15:53:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d23cd0a087549858deee0c7715144de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74af0ee9d4564940a5f537007db7bbfe : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ebc4c098bb9f45ff9180f739c2a7166b : SUCCESS in 40s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1e8bfac956364a7b98009b2cfefcc8e3 : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03551947261dc5726a7bc5623f3dcd73b5690b75","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-08 20:05:27.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"33f2a7b9abb5eaa61e03d12424bc0d45fadf6b87","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-08 21:31:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6885b33676dbc4ee0cdfa20798e9a9cf89d14650","revisions":{"6885b33676dbc4ee0cdfa20798e9a9cf89d14650":{"kind":"REWORK","_number":1,"created":"2023-03-27 09:45:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/878636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/878636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/878636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/878636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/878636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/878636/1"}}},"commit":{"parents":[{"commit":"e15d6880e810cb2419865136e2cbb856acab8e73","subject":"sharder: fix \u0027Cleaving not completed\u0027 warning log","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e15d6880e810cb2419865136e2cbb856acab8e73"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-03-27 09:42:01.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-03-27 09:42:01.000000000","tz":60},"subject":"sharder: fix host logged when send_shard_ranges fails","message":"sharder: fix host logged when send_shard_ranges fails\n\nThe ContainerSharder._send_shard_ranges method sets an\n\u0027x-backend-use-replication-network\u0027 header with value \u0027true\u0027, so if\nthe PUT to the root container fails the log message should show the\nreplication ip and port of the container server.\n\nChange-Id: I8c84f6ee15e6999f71b092bbeed414065a22ee8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6885b33676dbc4ee0cdfa20798e9a9cf89d14650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6885b33676dbc4ee0cdfa20798e9a9cf89d14650"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
