)]}'
{"id":"openstack%2Fswift~879512","triplet_id":"openstack%2Fswift~master~Ibb82555830b88962cc765fc88281ca42a9ce9d9c","project":"openstack/swift","branch":"master","topic":"case-sensitive-config","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-04-27 19:02:31.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-04-27 19:02:31.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-04-27 17:51:10.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ibb82555830b88962cc765fc88281ca42a9ce9d9c","subject":"tests for wsgi/daemon config parsing","status":"MERGED","created":"2023-04-04 21:07:58.000000000","updated":"2023-04-27 19:03:38.000000000","submitted":"2023-04-27 19:02:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":10,"has_review_started":true,"submission_id":"879512-case-sensitive-config","meta_rev_id":"43beff45425b5930d2cb2cd331bd853657fd5156","_number":879512,"virtual_id_number":879512,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-27 19:02:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-27 17:51:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-27 17:51:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-04 21:54:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-04 23:14:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-05 11:03:21.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2023-04-10 20:22:53.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-04-11 14:54:51.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"fb7ffbd882130c912e52df248835bb09fc218841","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-04 21:07:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fad1861b2f04f294561064a1e4194e25aa611ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 21:54:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e2055246d5a42dcaafc6ffe0eb8cb02\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81acdd22119c4513a6d85693a76a6fdd : FAILURE in 14m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b1b49a67855b4c90a315bea67c150bc4 : FAILURE in 15m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1b4054c5a324c73b01b6924a566df0c : SUCCESS in 18m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e7fdeb14bcd14739bed995ce851d8776 : SUCCESS in 43m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d35681cc4e216b23a9899a913040fb274128ca3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-04 22:00:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a98f05b560d036f9d8f0fd7e82e275be451d7241","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 23:07:14.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0caa63fc42724a0facd601c4fba3b1cf\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c25711c89ee346299da251fd5648415e : SUCCESS in 14m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d98d65d3c7a443ff942759c90d5a9bc6 : FAILURE in 1h 04m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/11be03898a344a66ad4384ece61b8e26 : SUCCESS in 24m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d57192e93174f4b9d495161558113f1 : SUCCESS in 45m 33s","accounts_in_message":[],"_revision_number":2},{"id":"a3cb3a6c70c8eec614d256e1c7fe5e78ac52c5a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 23:14:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/611084041ac54637bf8a48ea46d10a1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee6b487e3924619b37cf3a888c8ab30 : SUCCESS in 7m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/0227a07e43f64643b34617537d3536d9 : SUCCESS in 44m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/999581bed9f544be97e8bb701bc2e336 : SUCCESS in 42m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f544979540e490d816e688d4dc2919b : SUCCESS in 1h 05m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/469d41bcd0194cbca37cab9fbe05ec66 : SUCCESS in 12m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7141120cae4140578709802946f882a9 : SUCCESS in 24m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85e19779decd40c684456a3da147eedf : SUCCESS in 20m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b476513807944bb0a55c51fea76b46bb : SUCCESS in 9m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/904603f9414945a78c796f9797ab4a66 : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/003df6227d024babb35b4e320bf1f8f3 : FAILURE in 19m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dd61712c9a0045eca1c0f9f719bb4d48 : SUCCESS in 16m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f3bbfc791824438f892ecd6c598c9ec5 : SUCCESS in 15m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9a09a289c2f9400ebe423a2c5fc03818 : SUCCESS in 14m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b1d58ed219c6494fa4811d767ffff3bf : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6a738f2183db411daac85a9fba201413 : SUCCESS in 16m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9e0a83a3453e4bd1bededae0384dcef9 : SUCCESS in 18m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/77e52a1a55bc4cf49e4082e786dffdc8 : SUCCESS in 12m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3d7b728ec7f64682a7e55136ee3f5227 : SUCCESS in 14m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b2f4486dbce146e3b4306b6e9be95bd6 : SUCCESS in 14m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c9290cf6daff4b3c8612268a86935502 : SUCCESS in 6m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5307cd07ab6f48299d629f7caa62a0bf : SUCCESS in 15m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/da8b417c186546d1818ddfed0aa21fcb : SUCCESS in 8m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9586f33b9a8c4c51910d3f94cda0e9e4 : SUCCESS in 45m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f207d2ea3622497cac9c85a96560b07e : SUCCESS in 57m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f6d1a446183140a89294f455844a06bf : SUCCESS in 34m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3a2795bba541470885da3111af5d73ef : SUCCESS in 44m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bda3e66bc5f747c28d2ccffddf79ca98 : FAILURE in 19m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfd52aca99344f6db706a2ffde80e33f : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a1ac232b0efa45558b50f09e7449387f : SUCCESS in 22m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/307d81ab504e449da93523e467d7e9e3 : SUCCESS in 36m 09s","accounts_in_message":[],"_revision_number":2},{"id":"0af9e72409d5851407094d69a29a5fdfc75629cd","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-05 11:03:21.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5b7aa9183457c5e5ae5f661fadb14d102cb7549e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-10 18:01:06.000000000","message":"Patch Set 2: Workflow-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ebaf140316d2ad29812d8ed7cb30a27a239295e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-10 20:04:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"30e9ddda564f5c59412ed68dee89f6aaa23417ae","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-10 20:22:53.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7d1bae2c96edd2c17dd9d26ff2ca524b6090845a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-10 20:29:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dcd485014171cb9d9f9f6e831418587c0be55786","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-10 21:18:42.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e539db955ea83714fd939940a2e5871e32bbcaff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 21:20:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4c701ba83af4563a9239526302dc0ba\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/43cbe295ad4c4b2f9ee8a08bd3a2a1e2 : SUCCESS in 1h 14m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/410ec515162e4cdf974b0c1ff84def15 : SUCCESS in 58m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6aa91cbb6f9f4d3e9ab520f5f3e8cf72 : SUCCESS in 1h 03m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c06e5e13d2024e61949a208e840b6f98 : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":3},{"id":"bdc8f43e79b2585e011bb2f815868c55200d5247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 21:50:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/52f4c40978704154bc25a4f976ea20c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aeacaeeb3a124b719aec6228c841ee97 : SUCCESS in 8m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/cf0ae96ac67649bcb04c2853f1357ed2 : SUCCESS in 59m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/21a15ed322574e4cbc5d85d0a6369c1b : SUCCESS in 32m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e2a8f360b8c4295a9335634338d6e2c : SUCCESS in 1h 03m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3fefc6e66414d5586e7d054253388cc : SUCCESS in 14m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a50f544f906246058d0e47af3afe3803 : SUCCESS in 12m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29818100ce8e48e6826934723cc5a128 : SUCCESS in 20m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5d7ad21680b54bb09d97b2e99ece75c8 : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a7059001eefa43519b9e879f846df4d6 : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e49e2ea34e51448282d4680250534544 : FAILURE in 16m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f1292db95a93487c82e2745d43e94c0c : TIMED_OUT in 1h 30m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/24e95058bb5748b19a9b38c57c52bdd3 : SUCCESS in 15m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/de6dbcf9a9ab468cb88933489d662f1d : SUCCESS in 14m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d21baf9466ad4bdfbd0b4675b94e1efc : SUCCESS in 18m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/539fe70ca45a4c7ab57126d7498e3d26 : SUCCESS in 16m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ad9a86881ec84a7bb774ddfef2847269 : SUCCESS in 16m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9264ae96365b46aeb5e6a6ebad726cf2 : SUCCESS in 14m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/230c319a3c064612990b1ef91534093d : SUCCESS in 14m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/77230685b0904a478ef3b9f24c167cc4 : SUCCESS in 16m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/66bda2d1677a4fb5aeb62ce3d00248ca : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/258eea0e831b4c0e9c490f887e1924a5 : SUCCESS in 16m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ea7c18877cfa48cfa2fa4e5e7a5ede5d : SUCCESS in 6m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/063b10293d064701b2925dacc2fd7fd2 : SUCCESS in 55m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f54efae81a54c398a7e651bd022e927 : SUCCESS in 53m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6e98fde25aef4ef298bdf107ee1bfc28 : SUCCESS in 44m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/92143af93b054b078b56735c3d8db28d : SUCCESS in 54m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0da7afa671064290842868ec549dc918 : FAILURE in 17m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eeb8a9292354095ad0c59a89a4dc485 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b024fdd1840a4e6e867ca0e52545c992 : SUCCESS in 18m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30ae1f749b8147fa9597038776dc8ef6 : SUCCESS in 1h 05m 32s","accounts_in_message":[],"_revision_number":3},{"id":"050ab7cdfd5fd86e74ce0674b0a9eba777df5a4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-11 14:45:42.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"0867d7a40a447b546e09d5ef6c9904cc9d531c3a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-11 14:54:51.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"77be642c8ef36e57bd2ac166d9458221898b16b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 15:05:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/078fcd0292c740d78a610c0ba2129037\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70bd4f824c2044e3848081cf84644707 : SUCCESS in 13m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3fc9fd23ba1c4384bfc6ce32e708cf26 : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/db55ccc9b77246209d8648e035691d0b : SUCCESS in 18m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/89e99963b11c45c4aeabb2c0c2a48c3f : SUCCESS in 18m 22s","accounts_in_message":[],"_revision_number":4},{"id":"ed89c056764c533bbabfcb6ac9ce860f01cddc5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 16:05:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/69992ed4837145e2be960fb0f9ac2751\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198621e8af87446181080b240d246d46 : SUCCESS in 11m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/69129b8b07ca49f782b6775c0de1ab80 : SUCCESS in 1h 11m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6f1f095091af48b598ba2f05fe96345a : SUCCESS in 40m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b9cf4d1055744f180c3e9364cf7fc57 : SUCCESS in 1h 04m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71a49624096f4a81aaaafdec7f9d494f : SUCCESS in 18m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d1e1591c3924bcdb87f07848708ea2e : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5e66020f160440a8161f22a3e5e0a68 : SUCCESS in 18m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7df0d8deec5c43698235a12b7612c701 : SUCCESS in 8m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1c3f641afca54544a4b5222a30f409eb : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fdb876134fd34ca2b461caff3eb71b8b : FAILURE in 16m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c622c7a712034432a48ba30080c44e5e : SUCCESS in 17m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/46e962e1af484d55873719e90e6a2eb6 : SUCCESS in 13m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5024b4173a264e2cba590f401cd52196 : SUCCESS in 14m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6987d389e43a461da56973a4e9045627 : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/321257f75dd843cfb21b66570545f981 : SUCCESS in 18m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/995b66f9254340eda8b23d6ce82d8ad1 : SUCCESS in 15m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5edf722033a74daeb92d9be942efeaf2 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9fbc3e2ce5db49ef9037f7a17cb54c7d : SUCCESS in 13m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3caeb3dadb7f4df29627b0d3e1da42dc : SUCCESS in 14m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3a3e7fd9d34645d6bc74ae420b087747 : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c282bbd6071e438e8d79168ddd91f55b : SUCCESS in 14m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/79968bfa75984c18b442fdfb67ea928a : SUCCESS in 6m 44s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/359177840574482293ae26a33f0631f3 : SUCCESS in 1h 05m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4728cf7dc5b941119f1046d6a28435f8 : SUCCESS in 1h 05m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/85e36424bfb240c38295a88f3667c64f : SUCCESS in 36m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b853a9ec87684a1799b03efdc686ed99 : SUCCESS in 53m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e7b855607eb4bb0845e0998e04a1965 : FAILURE in 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e30a9b2d6e944be2bb0022bc35a7f35b : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b4a1be7aaa9948a4be92c7753fdbd953 : SUCCESS in 20m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b1922ab0bc84a93b9cd6374352ae3d9 : SUCCESS in 40m 34s","accounts_in_message":[],"_revision_number":4},{"id":"a5862130bc9b16e9ce4b30fca1d3f28530f6c1c8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-11 16:10:39.000000000","message":"Patch Set 4: Code-Review-1 -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"794aca6c5c67ecdf1b10ae4bbc39a91c74760fe0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-11 21:43:11.000000000","message":"Patch Set 4: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8febce5c076abd4082466909b34321a6bdb9457a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-14 19:10:15.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2b215197dbd05db33cbdc96ec059748209dd0a66","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-14 19:24:45.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"be577d91215e6759b7ad470dac1a2b4d41fad2ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 19:34:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62c58db78aa84a67ac0e7ee2b4750f67\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e83eca0a99f478595d86f2a5b53d1b5 : SUCCESS in 15m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd466bac4eb04492a5d9b78dffad4cc1 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd94fd3101e04434815afc8b6287eb7b : SUCCESS in 21m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e0a7b873c254da8b2e58d6de5fd694e : SUCCESS in 20m 52s","accounts_in_message":[],"_revision_number":5},{"id":"12e631cc9b6463ddff575b45d57cbdfd70426c10","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-14 19:55:28.000000000","message":"Patch Set 5: -Code-Review Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3893b25624fe8fa252a1fbe5f9a515f3cc09bf02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-14 19:55:34.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6baf48ddaf178ea7d3ef0ecdd3ad1caf495cce8d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-14 20:02:11.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"70f54725342396a69e72cffb9fdaad9f3e9aa69c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 20:17:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70dd8c759ce34a83b2cad8a56ee549cd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0c25cc4abb241d886bff2a765951c08 : SUCCESS in 16m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb7f135431d944f396e99279be3dc1ca : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b096ce3d7e8f4368a99b1cd8af864a4d : SUCCESS in 20m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6542210038284064b5ec558eac28ac16 : SUCCESS in 19m 44s","accounts_in_message":[],"_revision_number":6},{"id":"de9e7549d4e23753ac85c66f350f8ccf7d3b2011","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 21:17:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acfec946513b4a8f8c768467cc28434a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2499f7f4390d4d3eb9b4cd8022d21395 : SUCCESS in 7m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/bd46027eb23147c59355b20edec1e1a9 : SUCCESS in 54m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/252e9ea04bf14df09918df3e0c25960b : SUCCESS in 31m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0e4d6387cb94728b655fc6176d13677 : SUCCESS in 47m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/177aa66641104e38bfec1c018e56e939 : SUCCESS in 23m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/464cd0c9b14e4053864dc64fb086dd44 : SUCCESS in 14m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f1ae508c84649478e2ce485f6e0f2cf : SUCCESS in 26m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c13f048769114975ba58e2a13bcadd93 : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e03097301d19479eb4a2af179e34c5ed : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ea0c67ef71704a25aa2156091583b783 : SUCCESS in 24m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/74a013e7db534337af461b2606c46850 : SUCCESS in 19m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/15736a16841a4d5fbba5b00f0850cb83 : SUCCESS in 11m 45s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a0c3eeb1c38f443f9c7d3c7d2d67b394 : SUCCESS in 13m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de9f89e396ad4268abbcb7f601ff749e : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4b900c40a0c341d3b87456257d53bf4d : SUCCESS in 14m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f310a064d0a8444f8a6237eea484e201 : SUCCESS in 14m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b2c88b2d2a8048638e6a0eb06c312458 : SUCCESS in 10m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/883635c1e65743e6861cae8ba777bf4d : SUCCESS in 23m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7175f3e3f5f647eab866c0c2da4e5aed : SUCCESS in 14m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c156ea761d2045bf9a86c11b24e6050c : SUCCESS in 7m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/61e53e0260e147109030a8487e13ff34 : SUCCESS in 17m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/161a8dd054f5426e96d4ce17e841e9cf : SUCCESS in 8m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e63601dcb6a14e9fad5aed3409ad8642 : SUCCESS in 1h 00m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b6909e732234660800b694fa791f9f0 : SUCCESS in 1h 06m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/631fb841b2ce4acc97ee73c143e7f64b : SUCCESS in 54m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e0295b50c1f4440ca4266cb6c9013151 : SUCCESS in 35m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3115094120f2487585d5c927c35564ae : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc0f8fb63aee419685e75befde3fac69 : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/429c9993222741a3baf50065cb91850b : SUCCESS in 19m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/994f8bacd35f4c71b82322502ea7ea41 : SUCCESS in 52m 59s","accounts_in_message":[],"_revision_number":6},{"id":"de3052fcc05232e23b658d0a89589f59f31dcd24","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-27 17:51:10.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"efaf02cdc606ed83aa2e23a41439776d6614393b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 17:51:45.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9e396a60eb6fc039b3c8018d8e8c76a895eb7008","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 19:02:30.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7cf0f082c8c424493d5ab08a2b23ada\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9b76f1de494115a9bfcb15d41ea6e9 : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/66f77d55007843d08055bc8b17861b05 : SUCCESS in 1h 00m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e0082062527d436793a54599ea15a507 : SUCCESS in 49m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26989272c191452dba1184210c0a02ac : SUCCESS in 56m 29s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5cf3594af7284883b68f7776b022e0bb : SUCCESS in 9m 18s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e1c43fce09224e92a0ae7e6ba7ad40fa : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e9d4396bc54846f8af80e7b70126edf4 : SUCCESS in 17m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0c0ff4d2aef243cea0e66cf7f6ef5d64 : SUCCESS in 17m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/38543be289c4486bba792aa8ca61e85e : SUCCESS in 15m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/579ebd6a1e46417da9be0db07c60cbe1 : SUCCESS in 16m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bf7f33ad1a2247c9932f077fe3894f1e : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e73a24b58035424e9273eda03f113cab : SUCCESS in 17m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0cf780bd27444bfcae387f977f2f70f6 : SUCCESS in 19m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/afce3532c6804e97bbe0dd19119965d5 : SUCCESS in 12m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f4d4e26daf284abeae20674ca8f55b51 : SUCCESS in 15m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/988d838d5190494f9d5d04a38b8bd6df : SUCCESS in 13m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d388417a526a4c1a972c733d3389bb2a : SUCCESS in 6m 30s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f34f3c39569742a085f0b3cce6b591a0 : SUCCESS in 8m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7df12c949b664ea58701412ed3e59a54 : SUCCESS in 59m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/df18b34b3d94402d9628e014c8274f04 : SUCCESS in 1h 00m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9aa4b7a8fb1442fbbe952f6943fa7e1e : SUCCESS in 40m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d17c9289a5a494ca232971c27d58272 : SUCCESS in 53m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdefdd294ab34b8ca2c785435a200219 : SUCCESS in 16m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f048577f46d432fa547616802a265f8 : SUCCESS in 4m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2445290eb6b24176b88be70648e977d7 : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":6},{"id":"7b664093efa71dd1044adf8a225166a0e96a2aad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 19:02:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"43beff45425b5930d2cb2cd331bd853657fd5156","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 19:03:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32ed8c6c71df42c09fcc655229105d6b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37df4f80f0de4b13a2488dd568e3edb8 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/637e11881bf24c77a95f075cf3798de8 : SUCCESS in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1e602ee6f60a448c839b5c19b6a1c067 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c95f8e6c050bbe9244122100ca90e6f335c73ab7","revisions":{"546713c7fe2c9983d6d15ef642b4a65a3eb6a093":{"kind":"REWORK","_number":1,"created":"2023-04-04 21:07:58.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/12/879512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/879512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/879512/1"}}},"commit":{"parents":[{"commit":"91d62b1879bffc8d5667b49c0207bd5a0f182379","subject":"Merge \"tests: Skip s3api functional tests when no s3api user configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91d62b1879bffc8d5667b49c0207bd5a0f182379"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-04 20:42:36.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-04 21:07:43.000000000","tz":-300},"subject":"tests for wsgi/daemon config parsing","message":"tests for wsgi/daemon config parsing\n\nChange-Id: Ibb82555830b88962cc765fc88281ca42a9ce9d9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/546713c7fe2c9983d6d15ef642b4a65a3eb6a093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/546713c7fe2c9983d6d15ef642b4a65a3eb6a093"}]},"branch":"refs/heads/master"},"30acba0338c446bb7ed762d75ad6caef9fd2b055":{"kind":"REWORK","_number":2,"created":"2023-04-04 22:00:41.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/12/879512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/879512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/879512/2"}}},"commit":{"parents":[{"commit":"91d62b1879bffc8d5667b49c0207bd5a0f182379","subject":"Merge \"tests: Skip s3api functional tests when no s3api user configured\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91d62b1879bffc8d5667b49c0207bd5a0f182379"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-04 20:42:36.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-04 22:00:34.000000000","tz":-300},"subject":"tests for wsgi/daemon config parsing","message":"tests for wsgi/daemon config parsing\n\nChange-Id: Ibb82555830b88962cc765fc88281ca42a9ce9d9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30acba0338c446bb7ed762d75ad6caef9fd2b055"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30acba0338c446bb7ed762d75ad6caef9fd2b055"}]},"branch":"refs/heads/master"},"e6049b577184ed0e5ea1a96f91585f84af3329f8":{"kind":"REWORK","_number":3,"created":"2023-04-10 20:04:07.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/12/879512/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/879512/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/879512/3"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-04 20:42:36.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-10 19:57:27.000000000","tz":-300},"subject":"tests for wsgi/daemon config parsing","message":"tests for wsgi/daemon config parsing\n\nChange-Id: Ibb82555830b88962cc765fc88281ca42a9ce9d9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6049b577184ed0e5ea1a96f91585f84af3329f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6049b577184ed0e5ea1a96f91585f84af3329f8"}]},"branch":"refs/heads/master"},"d7d1c1532be4087a4acad113a967f4c3a3f39ded":{"kind":"REWORK","_number":4,"created":"2023-04-11 14:45:42.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/12/879512/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/879512/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/879512/4"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-04 20:42:36.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-11 14:45:32.000000000","tz":-300},"subject":"tests for wsgi/daemon config parsing","message":"tests for wsgi/daemon config parsing\n\nChange-Id: Ibb82555830b88962cc765fc88281ca42a9ce9d9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7d1c1532be4087a4acad113a967f4c3a3f39ded"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7d1c1532be4087a4acad113a967f4c3a3f39ded"}]},"branch":"refs/heads/master"},"74cc6e5d5bdc27a97a85e86ec4f5ca73e030bf29":{"kind":"REWORK","_number":5,"created":"2023-04-14 19:10:15.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/12/879512/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/879512/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/879512/5"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-04 20:42:36.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-14 19:10:13.000000000","tz":-300},"subject":"tests for wsgi/daemon config parsing","message":"tests for wsgi/daemon config parsing\n\nChange-Id: Ibb82555830b88962cc765fc88281ca42a9ce9d9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74cc6e5d5bdc27a97a85e86ec4f5ca73e030bf29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74cc6e5d5bdc27a97a85e86ec4f5ca73e030bf29"}]},"branch":"refs/heads/master"},"c95f8e6c050bbe9244122100ca90e6f335c73ab7":{"kind":"REWORK","_number":6,"created":"2023-04-14 19:55:34.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/12/879512/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/879512/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/879512/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/879512/6"}}},"commit":{"parents":[{"commit":"8e875df65a3158aeb80318c77fd06662c0ed6d52","subject":"Merge \"InternalClient: error if allow_modify_pipeline is True\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e875df65a3158aeb80318c77fd06662c0ed6d52"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-04 20:42:36.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-14 19:51:23.000000000","tz":-300},"subject":"tests for wsgi/daemon config parsing","message":"tests for wsgi/daemon config parsing\n\nChange-Id: Ibb82555830b88962cc765fc88281ca42a9ce9d9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c95f8e6c050bbe9244122100ca90e6f335c73ab7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c95f8e6c050bbe9244122100ca90e6f335c73ab7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
