)]}'
{"id":"openstack%2Fswift~879688","triplet_id":"openstack%2Fswift~master~I5b7ab3b2c150ec1513b3e6ebc4b27808d5df042c","project":"openstack/swift","branch":"master","topic":"breakup-utils","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-04-12 17:29:25.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I5b7ab3b2c150ec1513b3e6ebc4b27808d5df042c","subject":"Pull timestamp-related functions out to a separate module","status":"MERGED","created":"2023-04-05 21:46:08.000000000","updated":"2023-04-12 18:48:25.000000000","submitted":"2023-04-12 18:46:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879688-breakup-utils","meta_rev_id":"747a837811168f010468f6f8a56409ba52650ae0","_number":879688,"virtual_id_number":879688,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-12 18:46:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2023-04-12 07:13:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-12 17:29:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-05 22:38:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-05 23:17:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-12 07:13:04.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"2913143a4c81eb72b5d4d661c3ba7bb1a62a619c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-05 21:46:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b911817d299aa9be1ad1c1844f29f3ea6e368ce0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 22:38:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67902b89befe489bb01bdb39ef80e0b6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/23be4bef799b4aa790b70e9e55e86d8f : SUCCESS in 12m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d912e85b4ac54814abe57e83f1b467da : SUCCESS in 45m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc1273b2f0d245eb8f66b67dba683c11 : SUCCESS in 21m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4725a872182f454a8e370a3bb9001f47 : SUCCESS in 49m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c92b11004a26d6d951d6c5b444d4d92ba914932c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 23:17:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/034e03e1244c4b03a2e1fee4a5a1e198\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53f3d5b4065249c2b345bb20ce22223f : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/3c8ee7cb29084a7e8691dbdc77234718 : SUCCESS in 57m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8c4b4576c7584e0da49767949b3544eb : SUCCESS in 45m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c59bb603510946bea1129227c0ce8c07 : SUCCESS in 53m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f662337ee1c418cae097521e020cd67 : SUCCESS in 28m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/058b06951ac14dc0bbff167b55c9ddb5 : SUCCESS in 13m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3775e2c230a43b796e895b5d5b0947e : SUCCESS in 30m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d62e16a8bd064491af57fe699599cd9e : SUCCESS in 9m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/97b567faa76c45129cf182894912f085 : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7cd37d6179474807bacd0c4a1c356881 : SUCCESS in 19m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0eaa9c0bd0734769a9ea420354a44a5f : SUCCESS in 27m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7013755288514f30a4fe886ffaa04545 : SUCCESS in 16m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8becc0849cfd437191339ee26e9bf0be : SUCCESS in 18m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/27ed14d0b9c148dba3ce20ae3e415b66 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0037e54b23a94a10a64be0979eec3468 : SUCCESS in 28m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f22eb929794043238fa97f897e63cae3 : SUCCESS in 18m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d355dc9d1ece4bf089431c28d1e72f4b : SUCCESS in 24m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a39751fe68464903a37016cb48aa008d : SUCCESS in 16m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/605a5ac469c34d8d84c36e7dde0b49c0 : SUCCESS in 13m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cdbf67c5434b46c5964fe42ce0d32a3a : SUCCESS in 6m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d53c93d9bf93410198a616b286afd2d7 : SUCCESS in 14m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5c5788497c9e4e44a70d215bec6e1d75 : SUCCESS in 11m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/063d806f1ab446c7b7414b1817dd0204 : SUCCESS in 38m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/79e65cf2862a449bbdd52ce0de1c7ed5 : SUCCESS in 58m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/da9b9e4a0db04178a739f8b78b856bb1 : SUCCESS in 51m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b01c4b5ef30a4f51b85a5838200a3c72 : SUCCESS in 50m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/143956fa986a4d3285be513ae1edf4e4 : SUCCESS in 16m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f81d845f8809488b9ddefd5f3af51931 : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d187038c78d43d9b23390382730d364 : SUCCESS in 21m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9474a3fd789b4ffaa6273a3c35a77654 : SUCCESS in 1h 04m 49s","accounts_in_message":[],"_revision_number":1},{"id":"bbe7cc1925e46d5a293bb564b91d93356e20024c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-04-12 07:13:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7efda1ef5953d5d9227a0d0bfcea7b3c83cb8aca","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-12 17:29:25.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"60a203842f02fabdb2a5bbdc28980ef8f7c4d0c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 17:29:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6ba874cb4e416242ee73ad356ea3a8c1a8d9706","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 18:46:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b7643fc3dff444b930314df3180d271\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b699ca16a5fb4c4f95574bfe5f51bf45 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/0754510f14064ccb8f22466e8513951c : SUCCESS in 1h 02m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/59bfca6254294c4797ab23bf2d36ac09 : SUCCESS in 38m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38100ff09e0b448b9b77df2f219351ec : SUCCESS in 48m 19s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2281f19c55da4775a2743a2b9e4eddfb : SUCCESS in 8m 34s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d4d0418a08f44682858fd0271957090f : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d1b4fe7d2ef34850b88aa240f5fc3444 : SUCCESS in 34m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d50f2a99495745a4a1d2a2388497af4c : SUCCESS in 15m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cb86547e5e194cf48530f0041153d272 : SUCCESS in 33m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4280e2dacfa24bb49c375d849367f4b9 : SUCCESS in 17m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ee6f355808284e53af0377e794e3d867 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/253781da653e466dac11eea6fafd119b : SUCCESS in 16m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/50d558bc28d64e2fb5db490199bee843 : SUCCESS in 19m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1b1ef08a62cf41a29e58dda94e216fd2 : SUCCESS in 17m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6daf750fe93b4224bed2b4b8c64ac34a : SUCCESS in 15m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0864900bb9d842149ea9ba6fb7b893b0 : SUCCESS in 14m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1dd8f3e994f14a08973d850a4e931e77 : SUCCESS in 6m 45s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/776e4d2c431e443ba64732ae915c275b : SUCCESS in 7m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/92e3f645ce0a4f279989f5c1439ac79e : SUCCESS in 48m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/10cbdc6b3a664544b358e8b00767ed81 : SUCCESS in 1h 07m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b32c3e0528dc41f8aace2e420c569c53 : SUCCESS in 49m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/85f8f1d2b83e450ab6eefc35ca0b225f : SUCCESS in 52m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e93629a9a25452e9e0bf4813e3f925b : SUCCESS in 17m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d18dfceaf33405e98cb531139ff2621 : SUCCESS in 4m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4262d12f166141848d90248bae11cad7 : SUCCESS in 54m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7b499cb92df2ca7e4f46e9555b8ae6086e1995d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 18:46:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"747a837811168f010468f6f8a56409ba52650ae0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 18:48:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe743a705bb742b88924699a187cdadf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/380a5da75c994c8a8fe5e68cabcea5ba : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/32c5a468996548b197b6240ec09cef10 : SUCCESS in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/3eaf86e9d9aa4044ba6b7c07a96a337d : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c21256d8707aa794d0a46775806cb67bf042bb63","revisions":{"c21256d8707aa794d0a46775806cb67bf042bb63":{"kind":"REWORK","_number":1,"created":"2023-04-05 21:46:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/88/879688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/879688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/879688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/879688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/879688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/879688/1"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-05 21:44:26.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-05 21:45:57.000000000","tz":-420},"subject":"Pull timestamp-related functions out to a separate module","message":"Pull timestamp-related functions out to a separate module\n\nPartial-Bug: #2015274\nChange-Id: I5b7ab3b2c150ec1513b3e6ebc4b27808d5df042c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c21256d8707aa794d0a46775806cb67bf042bb63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c21256d8707aa794d0a46775806cb67bf042bb63"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
