)]}'
{"id":"openstack%2Fswift~879723","triplet_id":"openstack%2Fswift~master~I9dcd5f94f710d05a10848be98fad36e7c24a5c61","project":"openstack/swift","branch":"master","topic":"recon_module_versions","hashtags":[],"change_id":"I9dcd5f94f710d05a10848be98fad36e7c24a5c61","subject":"Add /recon/module_versions endpoint","status":"NEW","created":"2023-04-06 04:08:44.000000000","updated":"2023-04-11 08:53:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9148930f678435e6f1136d4e2ce27425d53048cf","_number":879723,"virtual_id_number":879723,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-11 08:53:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-06 04:36:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-06 05:17:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ebae26a22813e50e6a871d3ad05f557679ee9ea9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-04-06 04:08:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"657adff1ab665d0c8038c6d1bdb9ed7838d4043b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 04:36:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/865034867b63401eb8bcd39c1327a022\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b93a77f186424d298b993653b047aec7 : SUCCESS in 16m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6792c6b074694253a67a62c0680adea6 : NODE_FAILURE Node request 300-0020917368 failed in 0s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/75ce20582367493c9e161db37c950581 : SUCCESS in 23m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0c94e0f5b384d7b80b9054052d65001 : SUCCESS in 17m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3f59488f3fa6e437d6ae08dc4851f597d82fa870","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 05:17:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c53be77f4f744be38acc80695ff1a128\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f064a150d0924be69827633e80e0a2bd : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/d7e3e02f66a54deb92fa326cf9a1ad73 : SUCCESS in 1h 00m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/772d39928b73411b8758ea1f90651031 : SUCCESS in 41m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab7e6975e27543e49d9bbdf902d33b73 : SUCCESS in 57m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2afd7f523db64125b9aea2887f9da44c : RETRY_LIMIT in 4m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc59a502e9b84bd292a5743379875fa4 : RETRY_LIMIT in 4m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ccb836d8bec4c0092466b16f6f5d300 : RETRY_LIMIT in 3m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/68c062e0fbb9422aada532427be82f20 : SUCCESS in 9m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/24021b2eb06048ea884487217b3b3dc7 : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/73b2a907d2ef4e04991caf9c19abf13d : SUCCESS in 16m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0f83f5419a9641aa978799acb1333ff4 : SUCCESS in 16m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/004c971661fc4f3b9cc34f2c9471dc91 : SUCCESS in 19m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f4d38bbc07a44da3b0c6521008fa37f2 : SUCCESS in 15m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2201ec149aae4940b4c12494d6354d53 : SUCCESS in 18m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/86db7e12b85346d48eee2faac8ad8b27 : SUCCESS in 15m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b0f9274fb0d1483a8a1210c262fa6a7e : SUCCESS in 17m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f7c03c6f5fb04751bdb0ec8c7b462b48 : SUCCESS in 11m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2a4fc6238cc94973bfba3713878ef274 : SUCCESS in 15m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/741fcfc95e254d7c9a00ef43947493d4 : SUCCESS in 19m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/93e94a6e8c874a65bb0c6e521d1e2426 : SUCCESS in 6m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b33034522c864479913d08341c734a1e : SUCCESS in 17m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/53acd44bfa844fe0a4143aaa5b1d19d9 : SUCCESS in 7m 06s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e819add0291745299e610e62ac7efc35 : SUCCESS in 49m 04s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fd5279d089a48e0863e9e9f32327d6e : SUCCESS in 1h 01m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bff47ebf78e142029742ef8aa4450d1d : SUCCESS in 34m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad6d7ae052b4436cb5b4ddfb8a4d3a24 : SUCCESS in 52m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9dec72c87eeb4203b0dc87ec64ef3736 : SUCCESS in 18m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca6366c03d3f406aa72782ab2701d321 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55d149e58e9b4ba290775ffdbe36c4c4 : SUCCESS in 18m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4642db2262e94610b8783b4459a84f29 : SUCCESS in 36m 08s","accounts_in_message":[],"_revision_number":1},{"id":"55c886094e9c6e13910cfeacb9d2b120871614e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-04-11 07:32:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"be4626598d372d6e3be78121c953afefb6d63734","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 08:33:48.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8010c943e5ef4a8a8ad06033ebca8f92\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae27c8d376174204ad1efe4e0768d549 : SUCCESS in 16m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b082134dfc84766a7dfab5b800fbd98 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff5ef59bef6149b9ac7378aa23ad05eb : SUCCESS in 20m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e417f66038649aeaefc4f926fd69ed3 : FAILURE in 59m 19s","accounts_in_message":[],"_revision_number":2},{"id":"9148930f678435e6f1136d4e2ce27425d53048cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 08:53:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d449713e5f3c4952a1da20d4c28a5639\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/221c2538ee7140d193c9f52ad0e2abf3 : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/b5f0e98bf3f946789d2e4f0d0100ca78 : SUCCESS in 42m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b79b83e26dd54e93bc188f4470526722 : SUCCESS in 45m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7868e0438a3433e9fc615a2a3cad716 : SUCCESS in 1h 03m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce581a8a72cd4054b239a2290a343475 : SUCCESS in 16m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/974fa5c2c2774dff94e2d8cc2c40ee5e : SUCCESS in 13m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14abaeef0e5443a798fa56b69f0d6d60 : SUCCESS in 36m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b9c6d568d39348228ed2a5428212212e : SUCCESS in 9m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/61bead760880430191f196481e70fafc : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b4b65f8606724adf8d2782aebefbec98 : SUCCESS in 35m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a262fc26d9e417c850e915901935045 : SUCCESS in 17m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/518d9c1f4cb14d1c9457b7246d47c47a : SUCCESS in 14m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/520a599120f44cef94b4d636fc557ab1 : SUCCESS in 34m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6d37e2fdc2d0410e806a42f722fa47f3 : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9eaa0c2674564fc885d8aecf71ba7a0c : SUCCESS in 32m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6829f913a0d84da9bc5b5079c0c82a15 : SUCCESS in 17m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/70eff3aa8ba14d15b0995c5cd1f0687e : SUCCESS in 14m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e933a33569a0491caddab54da3210b2a : SUCCESS in 14m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f2c7927b53b8439f8816e842f89c180b : SUCCESS in 32m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ffd4d349e70f43a0a2ad00cf812d629b : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/04821b29358a438d9b41b05de2be3c04 : SUCCESS in 28m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fe0bdbff0565440e9bbfbca9c450856f : SUCCESS in 6m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/df8f9650d9c44113840f13d48efb0ae5 : SUCCESS in 57m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f82f93c7396479382472d10268332bc : SUCCESS in 1h 10m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/acededfeab7242878f8e6153d723ccb9 : SUCCESS in 1h 05m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d2b309bd767a4d1c92f3e37b4274152d : SUCCESS in 1h 07m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9904fec387ee4c62b0b05cb2a0d76681 : SUCCESS in 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aa53d71d53e4caba10f45e12d39976d : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/50419dc0e2b44ebb8f5a8efe1fbaf692 : SUCCESS in 19m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47cb52232ede4804b47dd76239e0f379 : SUCCESS in 51m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"13213eab92128f247fde19ddfef12476b2574518","revisions":{"3b1680bf5e871ad82ffbe66cf68a71df1061a431":{"kind":"REWORK","_number":1,"created":"2023-04-06 04:08:44.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/23/879723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/879723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/879723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/879723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/879723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/879723/1"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-06 04:04:06.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-06 04:04:06.000000000","tz":600},"subject":"Add /recon/module_verions endpoint","message":"Add /recon/module_verions endpoint\n\nThis endpoint will query sys.modules and attempt to list the versions\nfor all the modules (with easily found versions) loaded in the server.\n\nAs an example, here is a container-server running in my VSAIO:\n\n    vagrant@saio1:~$ curl http://127.0.0.1:6011/recon/module_versions |jq .\n    [\n      [\n\t\"sys\",\n\t\"3.10.6 (main, Nov 14 2022, 16:10:14) [GCC 11.3.0]\"\n      ],\n      [\n\t\"marshal\",\n\t\"4\"\n      ],\n      [\n\t\"platform\",\n\t\"1.0.8\"\n      ],\n      [\n\t\"pyexpat\",\n\t\"2.4.7\"\n      ],\n      [\n\t\"xml.parsers.expat\",\n\t\"2.4.7\"\n      ],\n      [\n\t\"pkg_resources._vendor.packaging\",\n\t\"21.2\"\n      ],\n      [\n\t\"pkg_resources.extern.packaging\",\n\t\"21.2\"\n      ],\n      [\n\t\"importlib.metadata\",\n\t\"version() missing 1 required positional argument: \u0027distribution_name\u0027\"\n      ],\n      [\n\t\"distutils\",\n\t\"3.10.6\"\n      ],\n      [\n\t\"OpenSSL\",\n\t\"23.0.0\"\n      ],\n      [\n\t\"__patched_module_dns\",\n\t\"2.3.0\"\n      ],\n      [\n\t\"dns\",\n\t\"2.3.0\"\n      ],\n      [\n\t\"chardet\",\n\t\"5.1.0\"\n      ],\n      [\n\t\"charset_normalizer\",\n\t\"3.1.0\"\n      ],\n      [\n\t\"requests.packages.chardet\",\n\t\"5.1.0\"\n      ],\n      [\n\t\"dns.version\",\n\t\"2.3.0\"\n      ],\n      [\n\t\"__patched_module_dns.version\",\n\t\"2.3.0\"\n      ],\n      [\n\t\"eventlet\",\n\t\"0.33.3\"\n      ],\n      [\n\t\"netifaces\",\n\t\"0.11.0\"\n      ],\n      [\n\t\"_sqlite3\",\n\t\"2.6.0\"\n      ],\n      [\n\t\"sqlite3.dbapi2\",\n\t\"2.6.0\"\n      ],\n      [\n\t\"sqlite3\",\n\t\"2.6.0\"\n      ]\n    ]\n\nChange-Id: I9dcd5f94f710d05a10848be98fad36e7c24a5c61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b1680bf5e871ad82ffbe66cf68a71df1061a431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b1680bf5e871ad82ffbe66cf68a71df1061a431"}]},"branch":"refs/heads/master"},"13213eab92128f247fde19ddfef12476b2574518":{"kind":"REWORK","_number":2,"created":"2023-04-11 07:32:41.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/23/879723/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/879723/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/879723/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/879723/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/879723/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/879723/2"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-06 04:04:06.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-11 02:55:26.000000000","tz":600},"subject":"Add /recon/module_versions endpoint","message":"Add /recon/module_versions endpoint\n\nThis endpoint will query sys.modules and attempt to list the versions\nfor all the modules (with easily found versions) loaded in the server.\n\nAs an example, here is a container-server running in my VSAIO:\n\n    vagrant@saio1:~$ curl http://127.0.0.1:6011/recon/module_versions |jq .\n    [\n      [\n\t\"sys\",\n\t\"3.10.6 (main, Nov 14 2022, 16:10:14) [GCC 11.3.0]\"\n      ],\n      [\n\t\"marshal\",\n\t\"4\"\n      ],\n      [\n\t\"platform\",\n\t\"1.0.8\"\n      ],\n      [\n\t\"pyexpat\",\n\t\"2.4.7\"\n      ],\n      [\n\t\"xml.parsers.expat\",\n\t\"2.4.7\"\n      ],\n      [\n\t\"pkg_resources._vendor.packaging\",\n\t\"21.2\"\n      ],\n      [\n\t\"pkg_resources.extern.packaging\",\n\t\"21.2\"\n      ],\n      [\n\t\"importlib.metadata\",\n\t\"version() missing 1 required positional argument: \u0027distribution_name\u0027\"\n      ],\n      [\n\t\"distutils\",\n\t\"3.10.6\"\n      ],\n      [\n\t\"OpenSSL\",\n\t\"23.0.0\"\n      ],\n      [\n\t\"__patched_module_dns\",\n\t\"2.3.0\"\n      ],\n      [\n\t\"dns\",\n\t\"2.3.0\"\n      ],\n      [\n\t\"chardet\",\n\t\"5.1.0\"\n      ],\n      [\n\t\"charset_normalizer\",\n\t\"3.1.0\"\n      ],\n      [\n\t\"requests.packages.chardet\",\n\t\"5.1.0\"\n      ],\n      [\n\t\"dns.version\",\n\t\"2.3.0\"\n      ],\n      [\n\t\"__patched_module_dns.version\",\n\t\"2.3.0\"\n      ],\n      [\n\t\"eventlet\",\n\t\"0.33.3\"\n      ],\n      [\n\t\"netifaces\",\n\t\"0.11.0\"\n      ],\n      [\n\t\"_sqlite3\",\n\t\"2.6.0\"\n      ],\n      [\n\t\"sqlite3.dbapi2\",\n\t\"2.6.0\"\n      ],\n      [\n\t\"sqlite3\",\n\t\"2.6.0\"\n      ]\n    ]\n\nChange-Id: I9dcd5f94f710d05a10848be98fad36e7c24a5c61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13213eab92128f247fde19ddfef12476b2574518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13213eab92128f247fde19ddfef12476b2574518"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
