)]}'
{"id":"openstack%2Fswift~880024","triplet_id":"openstack%2Fswift~master~I02717ee093165f907c57c8ad1795f0cb17114ccf","project":"openstack/swift","branch":"master","topic":"recon_module_versions","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-04-11 23:02:00.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-04-11 23:02:00.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"hashtags":[],"change_id":"I02717ee093165f907c57c8ad1795f0cb17114ccf","subject":"/info: Add module_versions to admin request","status":"NEW","created":"2023-04-11 07:32:41.000000000","updated":"2023-04-11 23:09:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"272b2cc6f954299ce2b80cbe76d9d36127617e20","_number":880024,"virtual_id_number":880024,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-11 08:40:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-11 07:56:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-11 08:40:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-11 14:34:48.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"0ef4037a46b2aebf562ce3af53289d5bec9b737e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-04-11 07:32:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1662e570397ba8ff5a5ee23e2b03dce329ce7129","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 07:56:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7b0811dba584b64ab3b783cfd2aedfc\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/494df2cd1a184a58a094a9ff59ff762e : SUCCESS in 16m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf29611de5434ce5926582a4c7d47636 : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/50f651eaa78a495db19d72a6762ce0d7 : SUCCESS in 22m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/43f6aa0df58f4958b5e3f5104ea3eca5 : SUCCESS in 20m 32s","accounts_in_message":[],"_revision_number":1},{"id":"096297826b699b5359a69f431f507a286e957c0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 08:40:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4c1a0c5802f4a3c862ea867ddcebb69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91d9734a8ee74024b46f62965503f7ae : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/24608a8a79fe4ad69b76cf8a7bc59ed3 : SUCCESS in 36m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/562d5fba614f43679273f9cf3f4917a4 : SUCCESS in 49m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8eb09c0ab8d84b6fa0bd536dc354be25 : SUCCESS in 52m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35f8f5980c514aa58e9336d98e62a8ea : SUCCESS in 15m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd7e5e6762bc485cacd9b1c79957d344 : SUCCESS in 17m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2bf3fdf7f974de1a63b10e153ae13f4 : SUCCESS in 40m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7f43daeec12249c99e9e63d7e80e1d64 : SUCCESS in 8m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5ec9bbfe2df548b591c2845d233e9acb : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1588d235515741ff87651514b5f0347e : SUCCESS in 15m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4d82835edbde496da8a29d32b90872e1 : SUCCESS in 16m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8f8d32a4cf5e480db90a7ddc7b87a732 : SUCCESS in 15m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d94c44ce5cfe48599c01e86a14f1bb34 : SUCCESS in 12m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eee4c844ffcd4645886b92504ef75649 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fc4a8af5065e4af3920266ad76801442 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/253cbc80fc6f4c6eb7ac6e18cef310b9 : SUCCESS in 15m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f6ce5b7b4c7b415e8f2bff41e2a85766 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e44475d82f7849b5aad394e9cb2d63f2 : SUCCESS in 11m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7886c32624c44c7a9b9475ad185c1134 : SUCCESS in 14m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9bf1af22127c44789b668ef75e14ed63 : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/64010f0ccb37409b99d27b0db164c197 : SUCCESS in 15m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/95bda6e7999445eea1c7c4e18332c220 : SUCCESS in 8m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d77c6c81dfae4b819d58aa1357f7301a : SUCCESS in 45m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0278d831784249cdbaa4722a9c0ccbf0 : SUCCESS in 59m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7c49dc739f3f43fcb924e13b5e98725f : SUCCESS in 51m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e7860a97847e40068e973ca3786958b8 : SUCCESS in 1h 01m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e66a0758263460bb7617dfcc52e3df3 : SUCCESS in 17m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d73f651bf004e9a9d87477c277ea5d8 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8922fdbc69314e0db0e5ec415af0a9e4 : SUCCESS in 17m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ceb8bfabfb54c8fab9b42b41bcf48a3 : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":1},{"id":"288582684885837a90c788ae2f6abf4019c14697","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-11 14:34:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"92b99a44dd76e99f120a28924583799cb8d12b0e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-04-11 23:02:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"414a676e3715d0e34c7b39875903e4258b6882dc","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-04-11 23:06:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"272b2cc6f954299ce2b80cbe76d9d36127617e20","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-04-11 23:09:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"340d5dc985bc87d2cbf8beaa30025aff0da319fb","revisions":{"340d5dc985bc87d2cbf8beaa30025aff0da319fb":{"kind":"REWORK","_number":1,"created":"2023-04-11 07:32:41.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/24/880024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/880024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/880024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/880024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/880024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/880024/1"}}},"commit":{"parents":[{"commit":"13213eab92128f247fde19ddfef12476b2574518","subject":"Add /recon/module_versions endpoint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13213eab92128f247fde19ddfef12476b2574518"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-11 07:30:18.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-11 07:30:18.000000000","tz":600},"subject":"/info: Add module_versions to admin request","message":"/info: Add module_versions to admin request\n\nThe proxy server\u0027s contructor now calls get_module_versions and\nregisiters it as admin info.\nThis means when you do an admin /info call using the info_sig and\ninfo_expires, the module information will also be returned.\n\nChange-Id: I02717ee093165f907c57c8ad1795f0cb17114ccf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/340d5dc985bc87d2cbf8beaa30025aff0da319fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/340d5dc985bc87d2cbf8beaa30025aff0da319fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
