)]}'
{"id":"openstack%2Fswift~880356","triplet_id":"openstack%2Fswift~master~Ic372ea9866bb7f2659e02f8796cdee01406e2079","project":"openstack/swift","branch":"master","topic":"proxy-py3","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-06-13 23:15:55.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-06-13 23:15:55.000000000","reason":"Change was submitted"},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2023-06-13 23:15:55.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-06-13 21:00:32.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ic372ea9866bb7f2659e02f8796cdee01406e2079","subject":"Fix proxy traceback for GeneratorExit in py3","status":"MERGED","created":"2023-04-13 19:01:17.000000000","updated":"2023-06-13 23:16:54.000000000","submitted":"2023-06-13 23:15:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":80,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880356-proxy-py3","meta_rev_id":"b7da7b6258a4ff200368d6c471df2ea5d7446ef6","_number":880356,"virtual_id_number":880356,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-13 23:15:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":2,"date":"2023-06-13 21:00:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":1,"date":"2023-06-13 21:00:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 20:47:39.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2023-04-13 20:47:39.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-04-13 21:42:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-13 22:01:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 23:10:02.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2023-04-28 14:57:21.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-05-17 21:18:32.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2023-05-17 21:28:59.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"eed8d88cc611e4981eaff46307e68362abc959af","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-13 19:01:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ba4c458d5c56103d719c29fa827aa5f4271a79d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-13 19:01:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"89e224c485ae7d9ea82fd178c21b8a34e1127b1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-13 19:03:18.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"744c5e6a0e16e48f8d183dc42bcaab978285dcd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-13 19:03:52.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d56b21dcaa4aeea2010d6b3eaf53d030c7d88ad4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-13 19:10:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"172326b5f13476336cc75f2f7aac9d78612571a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-13 19:58:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9f65c0f9d415a2e2167852d2e8ebf208eb925856","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-13 20:40:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ebc468abccde2fb6a74d6202d9a68f5ab67fda03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-13 20:47:09.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"cfc7686154acbcf7af79bb1a759c7679068762b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 21:42:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50a34afa3768400cb63d1204e6bb1605\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6362bad1f34d4ec9a3a79c0f1c33235f : SUCCESS in 13m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33482b3137424fdd961be19b6b6c58a1 : SUCCESS in 53m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5d4a0e4ad4d4ab7b7af94aa41e60920 : SUCCESS in 20m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2479e8bc89894794bc364ef70a9a39f6 : SUCCESS in 41m 36s","accounts_in_message":[],"_revision_number":7},{"id":"1dcd6a2f10909c82a9bb585920a34098293feae3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 22:01:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aab09cb7d64e40e8baabb141f36bafa4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/756de373aaab4dd6b88a35888ec0de58 : SUCCESS in 7m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/12f9720dd5d643e588a8eb62fc4fa25c : SUCCESS in 50m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/25696c66f08b4c1fa549860d8a6d7ced : SUCCESS in 46m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae5699bf18a74b238255df51ef693074 : SUCCESS in 54m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12dda9a8ddf148d59e7e2e535197c63d : SUCCESS in 28m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e11d2ee4d1f4e0b8be692b54c001286 : SUCCESS in 31m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/263181b4451046f1b81573e96ff15093 : SUCCESS in 32m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/49cf024be38242c4bed754c30b469d0d : SUCCESS in 9m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/635d59f1358541d3a9e4bfb252523f6f : SUCCESS in 4m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8057da1e868041b8ab136319c9c128a8 : SUCCESS in 29m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e154b8de8ea944ce8a83ab125f3334bf : SUCCESS in 16m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6cf40b8b162649fc862fd2c67476a109 : SUCCESS in 27m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8a9c85881bdc4e458a2b760544870199 : SUCCESS in 15m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f330ddf38bdc480187c1aa95ee63164f : SUCCESS in 18m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8d5a6b1cda4a423c8686850abe82253b : SUCCESS in 18m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9eb5996b000a462c9fe976ed9e04dd32 : SUCCESS in 17m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8bee02888d614985b2d0f18b4296a936 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f255f2cd82104ad29cf1b19aef593299 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a22fcf93ee074e82a2e651f5e0775e7b : SUCCESS in 14m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f3765ac1b87c4807ba7fefb30b320950 : SUCCESS in 7m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1e57b57de81e4472939ce93aab95cf31 : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e0acbb1e0fc9467389978319045d8a48 : SUCCESS in 7m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fb4b78dafcbc4bfca0cbba0d4a3e4839 : FAILURE in 5m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5102329e57fb4451a21aa4f2773e3028 : SUCCESS in 1h 13m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6ac0a0840440469190360dabd88b22db : SUCCESS in 51m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9ab6edea5ed240fe828f24e4323a1284 : SUCCESS in 38m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/487098fd6b4c4ec1b31f32ed67703455 : SUCCESS in 18m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca6e2e75d97b4921bdbd899590ffc48b : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/efb8fbbb2c4d4191a24f16ffc660cb76 : SUCCESS in 19m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d96590cffeb34c9eae8d7834ac146a6b : SUCCESS in 49m 35s","accounts_in_message":[],"_revision_number":7},{"id":"1c2221aef49172cde67d8c2afb49a09509768144","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-13 23:11:03.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"3e3e7947f5f886e9ad77c53b52295a1c752bf85f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-13 23:16:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f1fcf89896244f5508f58c34b7a7b7e895327232","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 23:39:20.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbb2cc82c5b04108bed6c86aab1d906f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/434dd6fa7710444cbc6e39da81245b38 : SUCCESS in 13m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/103661e742d6407aab39c8fef2af8428 : SUCCESS in 14m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/216fb7ad610b4e23ac3df4d405061cd2 : SUCCESS in 20m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fdeb103511784a2ba5245d277c9e2ddd : SUCCESS in 21m 15s","accounts_in_message":[],"_revision_number":9},{"id":"6ba12901d7f8c02bdf3d9c4d0ef9fe06f8fbd35d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-14 22:35:13.000000000","message":"Patch Set 9:\n\n(9 comments)","accounts_in_message":[],"_revision_number":9},{"id":"ad3ae5a31eabdd508ddf56d1d555d6485e570b07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 17:03:33.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/326a2442cc88476e961cc82319b874da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/054e3e5c2d2245f6a1097329ca61541f : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/7887bf6d90f1464e870ed970fed9930e : SUCCESS in 57m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4c576fcc73db4e22b8ce395c3aa58fb3 : SUCCESS in 46m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fecbcb57b88e4915ada72dd54229b155 : SUCCESS in 1h 03m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29c6967a0cca4b5bb0a8ca0273bb61b0 : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ac5ce0e8e524277a417ab5de2fe62fe : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b64ffff6fe744aa98297d3de1b1baa0b : SUCCESS in 15m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d99536c673b5492d85cab464f09cbe79 : SUCCESS in 9m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/168014a0eb644b88b935e5824e2836ec : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7d9ab6a4f8bd401284445cbb742a4fff : SUCCESS in 11m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4eda56549a114983a57e2c250dde8f46 : SUCCESS in 18m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/01affbeda6f24779b03c8f93664c3d05 : SUCCESS in 13m 53s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2ee880c018d64a5da7f10706da14779f : SUCCESS in 15m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9461ddb1a4dd4521a47560d2ab226dbe : SUCCESS in 12m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c97e2596f8a74a69ae6a07e10ae08852 : SUCCESS in 17m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/57ae01e478014ac6b6a6d5bbfe465b9b : SUCCESS in 13m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2c1b18a59ab34a49ae5da0eb3f318317 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f7e9f31ac73d411b9804d5213c81eb84 : SUCCESS in 17m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d7afc424c4954d2c982822925497baa8 : SUCCESS in 14m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/20fc83f717d8451e8e365ce3dce73003 : SUCCESS in 6m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bb99fedb17a14a99a5a5a4763ad93690 : SUCCESS in 15m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/361441ae2bab4bf5ab131c0ae2055ae6 : SUCCESS in 7m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/733883c6c32049cea866cb3b5fb3e594 : FAILURE in 4m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0dc3ce3895f945a59cc45bebec32a7ea : SUCCESS in 1h 02m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2fdac23d815749e782255afb8582ad2b : SUCCESS in 46m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bbc1a46a77b847b98fafdfdca5d10b26 : SUCCESS in 47m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bb3a28120fa4132813e9d58522f1dd5 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37147dabd99d4877b817b3ed120c550f : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b68e88b5c86e4b7da642bcf37e0ddbdb : SUCCESS in 19m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d0d54e2842e414f91c7fce787891a2f : SUCCESS in 37m 48s","accounts_in_message":[],"_revision_number":9},{"id":"6e4cbdff4dd9cccd659c4ba877c78c7db62fa6aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-20 02:47:04.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"4a0ecd20804acc461970c8d00fbcb4f768faeff0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 03:12:05.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f99a5e21a78465bb896b9fc1fcf4088\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d566712073c645429ac466275a01c27a : SUCCESS in 17m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6abf3a5b660a45318cecb3ba0bc85377 : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/189430aa6a644a7b9d34cc9d7330c296 : SUCCESS in 22m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/41d00c7dbbd74a05a5313e2c3c0bc34f : SUCCESS in 18m 43s","accounts_in_message":[],"_revision_number":10},{"id":"56ddf4569ba0b639ab89b23e0b57af194e83d10a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 04:05:11.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23cca90277004d37880eff282f915c8c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f96f0299bbe64d499f52ca47d36d112e : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/de19033981cc472190e6f41045800f05 : SUCCESS in 1h 03m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63aa9b527c254f40a40cfc05a45e6ec0 : SUCCESS in 43m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/885d4200755c4ffb89709408f0ed3ebf : SUCCESS in 41m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/88a8d8e6e50b4d54a56d7b33a5122346 : SUCCESS in 17m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/199c6d211cab462889fd8bb18164bc35 : SUCCESS in 26m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49e6db94ba4a4d809c8b8120e205359b : SUCCESS in 17m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/333e8a7e9db54434896aa0e566f17919 : SUCCESS in 10m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d82915c8438448fbbc4a901635560343 : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5b548dc05e5a4edf9a61f5eb76d60915 : SUCCESS in 17m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/94fc7e636cc64e1080726a30cf9d5d32 : SUCCESS in 17m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a08a55df396c454380408cc5c71b58a8 : SUCCESS in 18m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/665f8c86c5de4d738a19b51550113f4f : SUCCESS in 16m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6026da1979604aeb9bc7ae8d2111161c : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4372b8b579cd4bf48aab0de22908a26d : SUCCESS in 17m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6d5c123c8da745109f549cf242a080d3 : SUCCESS in 18m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/19be82d7cec34d7e8706d4849823adea : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/078400cf1cc745c3be396471622f7121 : SUCCESS in 25m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2b9f2469e91049508e4b5d90526077f6 : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/001067732b7d4bafb3910565270b6a24 : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bb95c9417a09459b94fcc13639046e9e : SUCCESS in 16m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1b13f6a2938d49d68de31f30f855fe0b : SUCCESS in 7m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d1a71569acab470699ed45faf2b7a3e3 : FAILURE in 4m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4b6ca4a49994d738687cd46d38ff272 : SUCCESS in 59m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de29dd4a547c46229942c6eb762d8a47 : SUCCESS in 53m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2a529d275c6b45f5a57d945dc9fa1062 : SUCCESS in 51m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c71e6a873d54702ae6b7e951c1e45cb : FAILURE in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e81fb71d9695434fa596b2c9a5a53588 : SUCCESS in 4m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8376a093dbd4441cb55656ad625363b7 : SUCCESS in 23m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/074bcee9f24e4fcc8eb9a4ee19acc069 : SUCCESS in 55m 18s","accounts_in_message":[],"_revision_number":10},{"id":"5e44d146b42a720b853a8f1cee640f594c13ac53","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-20 22:34:51.000000000","message":"Patch Set 10:\n\n(6 comments)","accounts_in_message":[],"_revision_number":10},{"id":"ae813acdbffd14a33d77eecbdde4dd3e527996ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-25 18:22:45.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"7377d4d971cd972de38c35a5420b8dd0dbd8cd1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-25 18:25:30.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"cd55fd08468e2c7e888220624506e9b9ad6fb9ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 19:35:14.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b2a30f22aa747268fa279161132fd00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e36966ac2a4c8887f91ebba03fbf25 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/bb270ea0883442399bd5438af2d83b78 : SUCCESS in 45m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5702d6a63d5d43b8b2d7e3e805e58fae : SUCCESS in 43m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c52fa7d1c47e4935acd37c2a151820a5 : SUCCESS in 1h 00m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/505c5127cbdf496b96bfb3cd1c5abcbb : SUCCESS in 16m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db0ab2946c8743f18bd6788de1ac675e : SUCCESS in 15m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/349b868972cb47639cb73e5fe1a890ff : SUCCESS in 14m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bfc5db2ebe3946e0ad2c78ee3fa1c58c : SUCCESS in 8m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7b11e3db0dbc4d56b0d79bbf37b94000 : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6aa16cbcf1b54c62b40975dd2e54ddf0 : SUCCESS in 17m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/12a03c2ccdad48b8922724aeb00ab7c0 : FAILURE in 17m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/91ccf16e16a64ddfa366da53b3d8d28e : FAILURE in 15m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0cc17976178a47e7b24055efa263762e : FAILURE in 23m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3a8b056b1f944e568e63aa217ede8fc9 : SUCCESS in 22m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9a307a7ed7c0480e8eeb5021f3331e7f : SUCCESS in 21m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bbe36cdf40c44ef78f106094a8722bb5 : SUCCESS in 26m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d9adfd2433ab4672a5cf8ee6b3fb2994 : SUCCESS in 21m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f5bd3b60e5c94c78a78964fd912cfddc : SUCCESS in 21m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bcb8587ea3694b5d8b46507f04d6fe5e : SUCCESS in 22m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0a2f6029504b4a06aa2baea6ee6533fb : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/114a1f11fadf41499fd8a1434379009e : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fb4e84e0c1124c679a26f5983a26ccc7 : SUCCESS in 7m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0d4b4760026b47edb6956db89d742847 : FAILURE in 5m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c725aca6d5b34e7991a20d7944b1ea29 : SUCCESS in 1h 08m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/256cb44b345143339aa1ed73a8d862a3 : SUCCESS in 45m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5c6963f34f7047ca910cee0e62013ccf : SUCCESS in 43m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6f995a759fa48bea69ac4544436202b : SUCCESS in 15m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee4c3039640544fcaec7f9e87e3990e2 : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c0cf2d55ec7c46eeb3b6f488db69438b : SUCCESS in 17m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82aab5c946974a0e86085846ef977d13 : SUCCESS in 49m 47s","accounts_in_message":[],"_revision_number":12},{"id":"18dd83bf311b599cbd2e40bc37fda6cdc179a868","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 19:40:06.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/232502b71c704f45bd99aa83efd1a608\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e93e3b49561b4faabbd3c7d7231c7b57 : FAILURE in 14m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a10b970ef6f14d9ea4c090c580ccb9aa : FAILURE in 1h 01m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/44552dad98c14b72afd8c1de676269d3 : SUCCESS in 23m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e4917a2ea15452f901f4a9a693b0309 : SUCCESS in 1h 12m 45s","accounts_in_message":[],"_revision_number":12},{"id":"682cbf10a1e8bbf85ca9be764418ed401ed7e6f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-27 21:02:46.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"0a294341e334555dab4e84b903b23cca55c58b48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-27 21:05:55.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"b493dc9d428d54bc22ef4ef42727c6ea5d6e42db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-04-27 21:07:02.000000000","message":"Patch Set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"7971c6bf4955988b8fd4a33172a9adfa8812780f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 22:17:00.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/68d1648e3b944e91babaadbc48d3c8b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e91475d407374022ad9f4d042f3ff79c : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/c2a041c19023420e80f550a9d9b29c1a : SUCCESS in 52m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c46966088cd94503834a0eaade243df8 : SUCCESS in 32m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a199d4efd0e4327aac1f269608a301e : SUCCESS in 58m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/384d5f75005f4e59942289fb88ce2f78 : SUCCESS in 14m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b74b2b38d2bc4413b715d40c3d5414a0 : SUCCESS in 12m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3da3172d8a1a4ca5847591965a980224 : FAILURE in 17m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d4070d0c5c914c1f98fbbbc2663f9b61 : SUCCESS in 8m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/42093b1e04894497977c2703950814fe : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/04df5d428dba4e53a7c169197810fc96 : FAILURE in 19m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/43fab436740244cabbd5ac1b8fe3daa1 : FAILURE in 16m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/06a86f779b5142a3a46b0be2bd7131d4 : FAILURE in 20m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/701cc3e9cb43490ab683296546cf1a69 : FAILURE in 18m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/86c020c9a6e849279b0057b17237e1a5 : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2c536d2297c743eb8dbf125ad6a6e912 : SUCCESS in 20m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8423a048ae7e40a09ca66e79b9d9b978 : FAILURE in 20m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fb633fbb9fc24112926a89be3952fdc6 : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9f31772b337e4150883a2ad43e897a62 : SUCCESS in 20m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/13f270bff17c4301bf3cdaa18e80b019 : FAILURE in 15m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/01ebb1cd1bd34b59ac6a0741c39eebe3 : SUCCESS in 5m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/86efad6c4e594dd7bab4e01ff8b844cc : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fded19e53cb64dcd9841c47263efde52 : SUCCESS in 8m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1cdeb5187aeb4f96b16736608f784add : FAILURE in 5m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f9921f665854fb59008e70a7203936e : SUCCESS in 1h 08m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f79405b59ef447facc0b88903e7bdd8 : SUCCESS in 46m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/38d1777b098b4e9aa2eb60ab21c39f9c : SUCCESS in 53m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85217dc37fc54d93b1d354623fb22bc1 : FAILURE in 20m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/473cb9e8c6084c97a50fc9a944a6b68a : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/be19e8d20604440db7d642639fc2e596 : SUCCESS in 17m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83dbaa6d4a44441eb7963264dfb8a408 : SUCCESS in 36m 13s","accounts_in_message":[],"_revision_number":15},{"id":"93b94552a966b848903e7e8008fc18e2b72a86ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 22:31:39.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/552e105b21334088ad93fc1433333ef3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2afdb3e829240b796c9c1946d799e31 : FAILURE in 1h 02m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f6c889fa3ef942929e330df1dd7f6f5b : FAILURE in 15m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ede6ff195b8846baa976ee71058785a0 : SUCCESS in 1h 22m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6cb563f77264bba9d5932241bd4c420 : SUCCESS in 17m 59s","accounts_in_message":[],"_revision_number":15},{"id":"ab11bb2903aa2d78e6c9d42c4591a049239f0d2b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-28 14:57:21.000000000","message":"Patch Set 15: Code-Review+1\n\n(15 comments)","accounts_in_message":[],"_revision_number":15},{"id":"afc2be9663ed830129c5945e7032411cef1dfa22","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-28 14:58:01.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"4825db638d164771bb920b898f3a5fd99e7f0ad6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-05-01 16:11:02.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"8b3c4e0d36f8f91b96ddb59363b5e759916e55c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 17:18:24.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1a7c09ea73c409eb18a6fa66640528a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f37b4e8ec45483c877ce6742439335b : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/987d670c1d644f2793a6c95e28bebdd6 : SUCCESS in 45m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b003d84aed704a299ed5e1782871818c : SUCCESS in 47m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5038abef2a34a13a767183ab649c185 : SUCCESS in 42m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d32cb868cd8427598a0560f80f8a2ca : RETRY_LIMIT in 1m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56cbdb9cef884c05815f447dcc5dcdcd : RETRY_LIMIT in 2m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de177eb2ce8f473786d44386b327fc29 : RETRY_LIMIT in 2m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b98e71704e994bb588038db73b58b551 : SUCCESS in 8m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8affaeed2ab64645899344b17c47a582 : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/56f426804c5246e4aa4c6455f01a7aea : FAILURE in 11m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c1a9bb7de378436cb85827708cd6a1db : FAILURE in 12m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/088daee613814f3fb52e51ba343c8924 : FAILURE in 14m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7417c3928ab74b4d8eaded4f47f1b0ee : FAILURE in 11m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fb0cf254c903487ebf6ee71fb2c0d0bf : SUCCESS in 14m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/352b74195c134992bcad35c12a4bcb0f : SUCCESS in 16m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/62102a7d0fee4f1f8d226b776bb074bc : FAILURE in 15m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/43109740e79e403684e83e50ce686dab : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/46296227c2eb49188efca404aab2bac2 : SUCCESS in 14m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f631158d88e740dab5813b8913061995 : FAILURE in 14m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6df214dc7ad54befa7810e96fa254586 : SUCCESS in 7m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/881113555a154dca83a0f76966b535d0 : SUCCESS in 15m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e339e2e3eba548359838920590c02f51 : SUCCESS in 8m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b3fbc27ecef34dc5851546527d2c08e4 : FAILURE in 4m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/439ae90c45c04c41a49b5f94a57bcd01 : SUCCESS in 1h 05m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/84a6f7872518495bafb2840702e4157a : SUCCESS in 41m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/91bd18803e8d4ea9b19e761d935e64d9 : SUCCESS in 54m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c485f4d6b5f4ef4aec5d4481ef6495f : FAILURE in 16m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0db8460dfb04e7a96f3773df5c7309c : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d11e5397f9dc4518b4d05face52b1087 : SUCCESS in 19m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aafc00a1b374485eaea2b726cb65d423 : SUCCESS in 59m 30s","accounts_in_message":[],"_revision_number":16},{"id":"18e5f3cfcba572e4c46c97a0d9f254b24ac339b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 17:36:40.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c554c182febd498685df753fa0d44e8f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f439d28b937403ea50f3187517f4057 : FAILURE in 1h 22m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/407d84d70eed484195b0962a1747b4f2 : FAILURE in 1h 02m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b86c87458765466a90ab5a6258f0ce5e : SUCCESS in 55m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc25053b87c1421998bab8a3749509a1 : SUCCESS in 1h 11m 08s","accounts_in_message":[],"_revision_number":16},{"id":"2891393eb4d60e750fb2e6e0cce06bd0eb7a05df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-05-01 18:02:10.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"1848d518274cce33985f62e61acfef80488289a8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-01 18:08:39.000000000","message":"Patch Set 17: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":17},{"id":"65e31112584aa1e1a25e2ad276bbf8f2a05f662b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 18:24:56.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc28c8e9b0cc4c23a510feb34d720817\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5f554bdf2a7410aa47c208b443d61ad : SUCCESS in 16m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c226f58d2b814e25b1a29018d30e262c : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c304b1f7a2d5416d9694b7ae0c81f39d : SUCCESS in 21m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3875d178fc544bad85e468e0be8eac36 : SUCCESS in 18m 50s","accounts_in_message":[],"_revision_number":17},{"id":"e078d1f5e948414ad58ce7f38e2c364631459ae1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-05-01 18:53:16.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":18},{"id":"e51a674ed7790844135c520477d1d8498399a510","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 19:12:50.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f487e17dd9f04d06a790d4965a035750\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb1d463dd8784cf2a5608693b31aaba7 : SUCCESS in 15m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0cb01d1047bd47febd8814f5f25e4ede : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ecf4613f85494c979da960ff52706250 : SUCCESS in 18m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/53665c41eb1c4db5b84884fdb0258e56 : SUCCESS in 18m 39s","accounts_in_message":[],"_revision_number":18},{"id":"f076944c63a0994d7b1391f28412503c376902a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 20:12:58.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca7cb5cd05034c2fb95542528344bd8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1804bdc39a2447c9dbfcd343b883b2e : SUCCESS in 7m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/2b5a71491c4645f4a3c3d1cbe7311c60 : SUCCESS in 41m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/84f48eae64e949758815a058695890e5 : SUCCESS in 43m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e66d910638041aab245a69c6b0826a7 : SUCCESS in 54m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1737409b5204c16a6f9b34c4d82ce0d : RETRY_LIMIT in 1m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c231688ddff46708010a9e53a1a25b4 : RETRY_LIMIT in 1m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a0fa44075d847378c3f42b75d56f9c8 : RETRY_LIMIT in 2m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7dc7776c49944556967c88393fba015e : SUCCESS in 8m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3e465f6b64b54bd9b7d4ea35077b6678 : SUCCESS in 4m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6bb6d0f778a04d3a90f644620fa8f8b8 : SUCCESS in 16m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/84239e53a87a4f589822f8b7e0959368 : SUCCESS in 10m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6a6ad10b7c46416c85f7934421102cfd : SUCCESS in 15m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/015539ad559b429da3b1b5f298ee4683 : SUCCESS in 12m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f67da5da3d444a61b7107c69696299bc : SUCCESS in 14m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/785801e183354a2dae5144a6f74f7339 : SUCCESS in 18m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6e151f12e17747488c0b00fe362bf196 : SUCCESS in 14m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/833da744d7f749ccbaa92c410bba72ab : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8343ff58328c40fa94e55ea21acedfa0 : SUCCESS in 16m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4bbd29e34247470a9345bc8a51ee9238 : SUCCESS in 17m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bd50676df1fd4186ac35e0be1a13c382 : SUCCESS in 6m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1c7723b57664bf291c351f46516a7f7 : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0b927f4ad80b4542a2a3650f35028a93 : SUCCESS in 7m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7b133a9c0cc84e5d94d64eb9966299b0 : SUCCESS in 58m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/05ea57e776f1458b9502c8fec39eeff1 : SUCCESS in 1h 00m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/07c6192e10ca40f7beea4bb176ab4d0c : SUCCESS in 48m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b87caa4814db42d2b20d5e8303693984 : SUCCESS in 45m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee4108bdad5a4c11bfaa3ab4fda57c4b : SUCCESS in 18m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4ae98317aec4ffa88fb252e4ff56d96 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ed5fe6345bb44a5abd7a44b6666e154 : SUCCESS in 21m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff7c72c2e24c48848c8dc61cca0dbc0a : SUCCESS in 56m 13s","accounts_in_message":[],"_revision_number":18},{"id":"9f97397398e3afbdf604d4f083be37ed7c3e06f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-05-02 04:50:47.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"9cd9f145fbe85fda22cf3b06e0c512cb8f54ee99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 06:07:52.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac4645edd32b4784a3ee83edac3e4a7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cb1492898464ff9adaf5a61ec08e2a0 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/808a8ad9be41419cae53a8df630a1aaf : SUCCESS in 1h 00m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c661e9182d864a4a8443e784ab6a77f8 : SUCCESS in 38m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1961487c7c7b4545b46e7a8701b59d03 : SUCCESS in 48m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/088d00805c4646d69a9d0f522e113062 : RETRY_LIMIT in 1m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5530d0b9a5954aa7a59885488ce05884 : RETRY_LIMIT in 1m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2bc743d5e92f45a7a537b7f1ce06c64f : RETRY_LIMIT in 1m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5659bb020abf4793be25ab6380145596 : SUCCESS in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d239d60f784e497ba4adfe578788b9fb : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fb69cb984dae4a98a391ad43b9b9aee0 : SUCCESS in 12m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/72512ef2794a4092aaeb5aa9515ad3ca : SUCCESS in 14m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/32c41dcb1e6d4fc08be97a6b6cfc8e2d : SUCCESS in 13m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c9ecb573c4914dab978765ceec85f68b : SUCCESS in 36m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1b2f6d796eb1448482aba734401b8619 : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3710be56f2994b3fa00bab3cf6d90f66 : SUCCESS in 16m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5fa08414af9d4157855230172e10a47a : SUCCESS in 34m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4dab521978ef4f3fae3ce0019a52a970 : SUCCESS in 12m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/040142bf4c1241ee82bf079e5e049b55 : SUCCESS in 29m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0011bd2713b24378aebcaf31a06dd8a0 : SUCCESS in 13m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1523a4931d7148c3b4bf6a9e4abf5be7 : SUCCESS in 6m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d02e87d9debd4c3db758bb83072e7e9c : SUCCESS in 14m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9678b8b7c8ed44048c240b42de0a8498 : SUCCESS in 7m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6e16815fceec45fe9038a3a0b733c7a0 : SUCCESS in 1h 15m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/94664b986d15406cbacfcd4856015f4b : SUCCESS in 1h 10m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/db1e128a48014627a96951baf147359d : SUCCESS in 50m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/699ba09c9b7c432baf0db03e04a41c3d : SUCCESS in 1h 04m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/540c58280a664faaa49d9964b6808a08 : SUCCESS in 17m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6a85bb03b364d94935337e2595e529c : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5cb4d92fe98b403eba1efc7cc8df2459 : SUCCESS in 21m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9fa8a65380447b8a467db8c7a8fea44 : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":19},{"id":"b360ac78fed802aca6a88c7f3fe1d7dc9c264e84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 06:25:08.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cb44e51c22449ec944306d121493176\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/04e782519d5942b8a300b4d28d3bb812 : SUCCESS in 14m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b5dc281371114e3388e75b955dbad598 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/db7419e8056d4e67a89e2f7ebe334eb6 : TIMED_OUT in 1h 31m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/93e12ae6643042e4a0c5d22c021d7273 : SUCCESS in 19m 30s","accounts_in_message":[],"_revision_number":19},{"id":"3021581915c34872cda9ba3035e3b1503f3cec44","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-02 14:16:28.000000000","message":"Patch Set 19: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"c23a3cb3e7ad946e72dcb745c048a4737c235ade","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-02 20:16:36.000000000","message":"Patch Set 19: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":19},{"id":"ddf91a21c01928d0f244190b59f20a9aa15c01bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-05-02 20:47:18.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"f26e3d034c4e4918b7ac416226e074555456a1c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 21:09:54.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13a9f527cd2542458d48a806572cdd94\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf395e10ca554d5eb9b06141ac5c44b8 : SUCCESS in 15m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b73c3fb2589542fe9a73f1794254d128 : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7905203f9674cbd802cf9765381ea42 : SUCCESS in 18m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/08a6ea5194c6433abac89468d51f5ccd : SUCCESS in 21m 29s","accounts_in_message":[],"_revision_number":20},{"id":"49732a0e3b5690d78ed748fb15eace22a1cfa69f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 21:59:32.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3879d7dbc5c34dd3ad9a50253c023b85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df2ada39ddcd4fbe8125b7d9d7a6bf6c : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/b13e80be44f145dda59b85a3a55a9eb5 : SUCCESS in 1h 07m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/08593468d0bf4d2bb66067772bf9f82f : SUCCESS in 45m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1eb82d5b96654bdcad6b43a460598441 : SUCCESS in 1h 00m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31b1e62cd0e04c48823c22497aac23c8 : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f425e69bc2374ce09b95148360bdb554 : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8fb72d0790d469ebbfc77b873b35361 : SUCCESS in 19m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f8eea38dc41a4045bcadd14738251873 : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e43d7eb36e834edc82549a753d22c9f9 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/719b453d408841a39e6740fc3b4d1c83 : SUCCESS in 15m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/82c0d6940b994746837cdefa5c5eadea : SUCCESS in 18m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8e9ddf8aead04a04821bb7f787a331e3 : SUCCESS in 15m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8523206ba3be404d9e0ee269a3cfcd70 : SUCCESS in 12m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fbefc9b793bc4d5a977bbb8b637cd0f2 : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c50036aef01c4e96b76b867ba6e66610 : SUCCESS in 13m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1fb57b439e76423ca46f1f34d2eb1899 : SUCCESS in 15m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f62c931076fd4748bf334986f5b923a0 : SUCCESS in 13m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4ea3526d58ac48d7a13a62838a3e71d7 : SUCCESS in 12m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3d8869a0d68a48ef98b9c42250d4a99a : SUCCESS in 16m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1ca3742a7a9c489d8ed44e0f0c79c937 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0a54e94beb3946b790016ef46f915deb : SUCCESS in 15m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bac93981bbf24a6a93ff2f6867cd1363 : SUCCESS in 8m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/99f61453534444e98704004836893aa8 : SUCCESS in 1h 05m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ed25c8f92bb4ab095d6d07bc46370c2 : SUCCESS in 58m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/034cebb0b4ca44aab41c109a2992448d : SUCCESS in 50m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/46eb4f36f10b425d92d45b784a84d2b0 : SUCCESS in 48m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9185072fe1704636bb74af61d624caa2 : SUCCESS in 18m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ff6077dcde44965b2c4797f6f5b90ed : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ba8788630db41f4b652fa0ef7ca3410 : SUCCESS in 17m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7dd3d294c0bc46548e42cb5f4a4acf0a : SUCCESS in 46m 02s","accounts_in_message":[],"_revision_number":20},{"id":"a963834c09f2245b6828fabb5a1c77c851700b58","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-03 20:56:52.000000000","message":"Patch Set 20: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":20},{"id":"0edecb9d2e672c781f44304e8fb2c8f14c8bff0a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-17 21:18:32.000000000","message":"Patch Set 20:\n\n(3 comments)","accounts_in_message":[],"_revision_number":20},{"id":"c1ec96e41a07f1a4c991ebcb34ae94a303e40e92","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-17 21:28:59.000000000","message":"Patch Set 20: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"c25f0fd07378087b30d5574c12c201936682cf42","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-17 21:30:52.000000000","message":"Patch Set 20: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"f9d51a11e6d6fd898f7ea317bfa32a059978739c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-19 22:29:53.000000000","message":"Patch Set 20: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"86aea670b4f50e52568662eda84aa9c83fba5f82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-06-13 18:10:37.000000000","message":"Patch Set 21: Patch Set 20 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"16e66e1e68fa9b4c8e032be91139166a65572f39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-06-13 18:13:07.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"18980fb40e68ea24ac3a57ab7591137fad45d11b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-06-13 18:24:32.000000000","message":"Patch Set 23: Commit message was updated.","accounts_in_message":[],"_revision_number":23},{"id":"e07e9b7a69bdbf6c6da0dcba1ad1c28132543b35","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-06-13 18:26:21.000000000","message":"Patch Set 23: Code-Review+1\n\n(22 comments)","accounts_in_message":[],"_revision_number":23},{"id":"c071634b9748f12fc452a35187bc743d24ee8be3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-06-13 18:28:18.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":24},{"id":"d4e379acfad7f55bb0b9ba67b8bfef1b2af0abef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-06-13 18:29:01.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"9ba8d6ddb71ce337aea32c05fc8522ac51a79ef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 19:44:47.000000000","message":"Patch Set 25: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efd116e49d994135ae52d34605328e32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad089379153d4ece857997ba3f692b3f : SUCCESS in 7m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/eb67b06baa934c46908695f0ab969295 : SUCCESS in 43m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f7ea79ee8e654f13ae3e2394176df850 : SUCCESS in 37m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ee92e28f3964b08bb97801521db63cf : SUCCESS in 1h 00m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6965b1a7c3e44179919883c83ea9f3b : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cffb3a246ab64f7f8763add09bd35014 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47b9bb3624234f16b04f305f87273d45 : SUCCESS in 18m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/78d2a13872ab429992cded1e714ee7e8 : SUCCESS in 7m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5f4451da3ee548c7959576c705e26f07 : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f63c7bb3d62b4b24986c5376bdd2c028 : SUCCESS in 19m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9a69354cbadb4c73813eeaf101a532c3 : SUCCESS in 17m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ed70c6fcb24b414d829e41f725e0af35 : SUCCESS in 14m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6636a46f1fd94135b82b171e66126580 : SUCCESS in 13m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2b02a4180d474c5ba9095eb719c76e0f : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6096ef41b399458db905b634aa038127 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1653d99e1273476db7843994df0da3f1 : SUCCESS in 17m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0d415c5ec94a4e1880c4eca8ec2b9f8f : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/451e651ae2f648e2b35cd2722ba54a38 : SUCCESS in 14m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d944d0b34fd545b88d120b875028ef06 : SUCCESS in 11m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/83aace2ca4c248e5bde0a130dafdbc64 : SUCCESS in 7m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eff0aab5183249acb6438c4dab9536d8 : SUCCESS in 15m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/01cbb36b5d0c4765885cbcf7fcd3f95f : SUCCESS in 7m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/466f76c91ac54b2a9819926b15029c2e : SUCCESS in 46m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7747169a26c14b06befadb0a763670c2 : SUCCESS in 1h 08m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9ca0d21a937f471a8f9941da003c18f5 : SUCCESS in 47m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/791d07ac4ff240e89d5dfe1a420f9a7b : SUCCESS in 45m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd7dd82739844ae5b46c95ddbf195efc : SUCCESS in 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e73b698276d4fcebc27aa13dab3f8a4 : FAILURE in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2876b7764f3643a6ae74e730aabc4b6b : SUCCESS in 20m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/823e2d8b3e80440d8192ffca39499d41 : SUCCESS in 58m 43s","accounts_in_message":[],"_revision_number":25},{"id":"74cfd80aeac6055aca7dc417003b3268abf68468","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 20:02:24.000000000","message":"Patch Set 25:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4372379daf0e4819b2533c1241b18169\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d48d29dab01f440e9d1a3fd1eda6a34f : SUCCESS in 15m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8cc62473d6448a9988917d33e5441fe : SUCCESS in 19m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5fc7439206e4b9983a1b8485cf96ea1 : TIMED_OUT in 1h 30m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/461467a6d49d45a88d7e138f418de890 : TIMED_OUT in 1h 30m 50s","accounts_in_message":[],"_revision_number":25},{"id":"772bed8d1330930b52ed30aa2902342e42f484d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-13 20:58:10.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"c0ea6656c6c41a31b42a6b46addd50103c40e299","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-13 21:00:32.000000000","message":"Patch Set 26: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":26},{"id":"2387a744bdf21b135a88d718b7b4aab2f8e2e955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 22:01:21.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d55cb0d46a745d39c849fdf90e1b8f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52ec008aef1f47c8abe221f0d8d0f9c9 : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/3dd249cac78f4aa9b0202100a7855e1f : SUCCESS in 44m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2be6118d8d7044688d58272e2d08f056 : SUCCESS in 44m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9637ddf39f8545f1879e9b0f81a1ca73 : SUCCESS in 47m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39a72732970e4ecf85aaf4d803c09b8c : SUCCESS in 16m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f61c8453d17d48f49d30e714f41df2d5 : SUCCESS in 25m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6268c6581221400fa93d46d0f590cf32 : SUCCESS in 17m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/52ea12fc2c5944899cf891db02c9a517 : SUCCESS in 7m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9b25b0d9443c444d946625cdef93a0cd : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e443d3ba628a4c72befe0ad447f8e6dc : SUCCESS in 17m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a15dfbdfe82b4bb693a4e23a423e888b : SUCCESS in 17m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/358575df599d45049c99cd8775905a16 : SUCCESS in 15m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d85c64aecc1b467886c0f7ab652437da : SUCCESS in 12m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/27c79a4ce415483d8dd34f86d8aa54e4 : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5b039f8b9ff8469785cf6136fd1ba364 : SUCCESS in 17m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/971a0a7e549340a598d89524b53403aa : SUCCESS in 15m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/da35c8fbe4064189a8103932a0e3a4e3 : SUCCESS in 14m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8d95f59726c44dfaae1d30d389229cd8 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e7766bb8e5e643f884fdf8e571958d3e : SUCCESS in 14m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/92dca8dde9954a01b1ddab5f25360634 : SUCCESS in 5m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fe9bc6f4028e4a7ab436271f67bcfac2 : SUCCESS in 17m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/05fd54bfe5514e9fa36ebb18a5a9113e : SUCCESS in 6m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dd7a3d4287c2424bb4e5989609b1faee : SUCCESS in 55m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8127b89ac7684bebb0c9bf34c5126fbb : SUCCESS in 54m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/38e25959f20f42d987e3f9236ca0e232 : SUCCESS in 39m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/962e253870cb414599bdeadf41c005aa : SUCCESS in 40m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e93c92f5fb6a472080fe8147247993cb : SUCCESS in 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e15662eccdb649beb27ff565c51df121 : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/824878722a8a47ccb5ed5475aa920cc1 : SUCCESS in 22m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f289fde565144c6c98c2960a27d9e418 : SUCCESS in 1h 01m 02s","accounts_in_message":[],"_revision_number":26},{"id":"cd5202de7e5782164f473b15e690756d4ca9a43e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 22:01:57.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"28a3b1a07a408ac9d980f3ccf243c69b3e6846f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 22:31:48.000000000","message":"Patch Set 26:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d78e444ac0140f4b2cdbd6894fbaa8c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df6d5fd696164191b2e6f0082cf8dc68 : TIMED_OUT in 1h 31m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd9034dc17f84fc9aa35b607c3d39920 : TIMED_OUT in 1h 30m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/49b9668cc4584afab399d2732698541a : SUCCESS in 22m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bd2bd76eaf84c249471c86fe4ee9707 : SUCCESS in 17m 37s","accounts_in_message":[],"_revision_number":26},{"id":"80b65f0619c949409830416d2fadef1eefb99b8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 23:15:55.000000000","message":"Patch Set 26: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c41e4b6523614127bdc1da63232938aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b2756dc6ad64371a94e1619fcb68c75 : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/0eb1b4abc2654b7b8b22907a899877f0 : SUCCESS in 55m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/227acb2fc9f64466b93430aac7609397 : SUCCESS in 45m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/281fe654efcb40bab4f08b799404af69 : SUCCESS in 58m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cdd9424166ea485cba8a5da4ea7ceb60 : SUCCESS in 8m 58s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2ef15b54d7f84535a2941da1cb958ad2 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5d165b5d774842989d6f1f28d634d84d : SUCCESS in 17m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bb44a3696d0946f3ab9eb2ec1fbd626d : SUCCESS in 15m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f19d346cd89848e5a23f9edc8c483d0c : SUCCESS in 13m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9c56f9756c2f4287803d47d428251ef0 : SUCCESS in 20m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c0b65c11916e4aa5853ded64754a6678 : SUCCESS in 20m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c00142c835ac4320955d6485201caa97 : SUCCESS in 15m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/51228d2b11c943a5ac0c71f30a5c8f2a : SUCCESS in 22m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/700ab977e5d2422ebf3d9147f156f44d : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0862714151514c99b7225a9b43488a95 : SUCCESS in 12m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/47bf971db345497fb0ddc8689d83b07e : SUCCESS in 15m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e4368d60864147128630b0aaec87107a : SUCCESS in 6m 41s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d9d5677d4e074706b7083728e5be6784 : SUCCESS in 6m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c294fb1ae70d4a72afeb510d9a6ddd22 : SUCCESS in 53m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ea532a7105844ba8904ef3d009218158 : SUCCESS in 51m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bbd5b637b57749e4bbae61b01dcc1065 : SUCCESS in 51m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e0fa869f1683424bb86b1b6262bf00b7 : SUCCESS in 52m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00683f0f689c4577adb5dc8ee732abb1 : SUCCESS in 17m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1db1d6eb7e184ef0b472ab7429639fe2 : SUCCESS in 4m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/595ea4e7db13453c8274e36c6079cc54 : SUCCESS in 1h 04m 38s","accounts_in_message":[],"_revision_number":26},{"id":"c5c44ca8ecd4a66d64dd8584830fb817aa463212","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 23:15:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":26},{"id":"b7da7b6258a4ff200368d6c471df2ea5d7446ef6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 23:16:54.000000000","message":"Patch Set 26:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c67a926bf363417ca0933560d1040dab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/735174756936464fad3b6e48796662e4 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/79fd77bd1e0d4f26afbe80582715ff28 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/3d06b5f24ccb4ddfa5e106482dc49707 : SUCCESS in 30s (non-voting)","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"be0c481e44e4f09b3c7fbe37f4b9ee62caa12a27","revisions":{"4dd452fb303f1d0607743d07399e6af5db02a5df":{"kind":"REWORK","_number":1,"created":"2023-04-13 19:01:17.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/1"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"subject":"fix proxy traceback in py3","message":"fix proxy traceback in py3\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4dd452fb303f1d0607743d07399e6af5db02a5df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4dd452fb303f1d0607743d07399e6af5db02a5df"}]},"branch":"refs/heads/master"},"dc3514de34f915d2f0a5a324d7280dbcf9b654dc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-04-13 19:03:18.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/2"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:03:18.000000000","tz":0},"subject":"Fix proxy traceback in py3","message":"Fix proxy traceback in py3\n\nclient when explictly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed.\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc3514de34f915d2f0a5a324d7280dbcf9b654dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dc3514de34f915d2f0a5a324d7280dbcf9b654dc"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5c20aa548747f881321c489b00753a210c710f96":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-04-13 19:03:52.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/3"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:03:52.000000000","tz":0},"subject":"Fix proxy traceback in py3","message":"Fix proxy traceback in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed.\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c20aa548747f881321c489b00753a210c710f96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c20aa548747f881321c489b00753a210c710f96"}]},"branch":"refs/heads/master","description":"Edit commit message"},"73644672464d5ed544798e49ed6db0b1f85e1a0a":{"kind":"REWORK","_number":4,"created":"2023-04-13 19:10:38.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/4"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:10:20.000000000","tz":-420},"subject":"Fix proxy traceback in py3","message":"Fix proxy traceback in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed.\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73644672464d5ed544798e49ed6db0b1f85e1a0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73644672464d5ed544798e49ed6db0b1f85e1a0a"}]},"branch":"refs/heads/master"},"b0cdc65579189841fb0285abbf3beec3c814c37c":{"kind":"REWORK","_number":5,"created":"2023-04-13 19:58:05.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/5"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:57:57.000000000","tz":-420},"subject":"Fix proxy traceback in py3","message":"Fix proxy traceback in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed.\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0cdc65579189841fb0285abbf3beec3c814c37c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0cdc65579189841fb0285abbf3beec3c814c37c"}]},"branch":"refs/heads/master"},"1a49e6941a35735869dad195f01e103dc310784c":{"kind":"REWORK","_number":6,"created":"2023-04-13 20:40:53.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/6"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 20:40:47.000000000","tz":-420},"subject":"Fix proxy traceback in py3","message":"Fix proxy traceback in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed.\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a49e6941a35735869dad195f01e103dc310784c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a49e6941a35735869dad195f01e103dc310784c"}]},"branch":"refs/heads/master"},"f614942baa7a1b764c20783b02f2aa3858400b7d":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-04-13 20:47:09.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/7"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-04-13 20:47:09.000000000","tz":0},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed.\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f614942baa7a1b764c20783b02f2aa3858400b7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f614942baa7a1b764c20783b02f2aa3858400b7d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"77f4e65e3b17f2314728a5bcffe31d65972240b9":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-04-13 23:11:03.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/8"}}},"commit":{"parents":[{"commit":"c78a5962b5f6c9e75f154cac924a226815236e98","subject":"Pull libc-related functions out to a separate module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c78a5962b5f6c9e75f154cac924a226815236e98"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-04-13 23:11:03.000000000","tz":0},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed.\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77f4e65e3b17f2314728a5bcffe31d65972240b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77f4e65e3b17f2314728a5bcffe31d65972240b9"}]},"branch":"refs/heads/master","description":"Rebase"},"ad3ab7827f4fe98ea0771246f3cc07dc09e245f4":{"kind":"REWORK","_number":9,"created":"2023-04-13 23:16:31.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/9"}}},"commit":{"parents":[{"commit":"c78a5962b5f6c9e75f154cac924a226815236e98","subject":"Pull libc-related functions out to a separate module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c78a5962b5f6c9e75f154cac924a226815236e98"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 23:16:23.000000000","tz":-420},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed.\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad3ab7827f4fe98ea0771246f3cc07dc09e245f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad3ab7827f4fe98ea0771246f3cc07dc09e245f4"}]},"branch":"refs/heads/master"},"2ef580c14fb734c266b21241bc701c3ad0343a36":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-04-20 02:47:04.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/10"}}},"commit":{"parents":[{"commit":"4b6f54d063c33e13ced7d6dd5208921680f56ca6","subject":"Merge \"ssync: Round-trip offsets in meta/ctype Timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b6f54d063c33e13ced7d6dd5208921680f56ca6"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-04-20 02:47:04.000000000","tz":0},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed.\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ef580c14fb734c266b21241bc701c3ad0343a36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ef580c14fb734c266b21241bc701c3ad0343a36"}]},"branch":"refs/heads/master","description":"Rebase"},"838df397123f4ce3efba9c9ecd26c58184e0f3f3":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-04-25 18:22:45.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/11"}}},"commit":{"parents":[{"commit":"9f7527973997baa5e685ea57d1563232234bc872","subject":"Merge \"Assert ChunkWriteTimeout errors are logged\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f7527973997baa5e685ea57d1563232234bc872"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-04-25 18:22:45.000000000","tz":0},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed.\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/838df397123f4ce3efba9c9ecd26c58184e0f3f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/838df397123f4ce3efba9c9ecd26c58184e0f3f3"}]},"branch":"refs/heads/master","description":"Rebase"},"ecb4da3546406ae78ce2b4651079dd4c1e06f4b0":{"kind":"REWORK","_number":12,"created":"2023-04-25 18:25:30.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/12"}}},"commit":{"parents":[{"commit":"9f7527973997baa5e685ea57d1563232234bc872","subject":"Merge \"Assert ChunkWriteTimeout errors are logged\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f7527973997baa5e685ea57d1563232234bc872"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-25 18:25:23.000000000","tz":-420},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed.\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecb4da3546406ae78ce2b4651079dd4c1e06f4b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecb4da3546406ae78ce2b4651079dd4c1e06f4b0"}]},"branch":"refs/heads/master"},"e708e06a92ea5d25d81997c81244ea061ba20a50":{"kind":"REWORK","_number":13,"created":"2023-04-27 21:02:46.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/13"}}},"commit":{"parents":[{"commit":"9f7527973997baa5e685ea57d1563232234bc872","subject":"Merge \"Assert ChunkWriteTimeout errors are logged\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f7527973997baa5e685ea57d1563232234bc872"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-27 21:02:37.000000000","tz":-420},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed.\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e708e06a92ea5d25d81997c81244ea061ba20a50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e708e06a92ea5d25d81997c81244ea061ba20a50"}]},"branch":"refs/heads/master"},"e800d52bba504566d03d7f7c9ead851963ce7994":{"kind":"REWORK","_number":14,"created":"2023-04-27 21:05:55.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/14"}}},"commit":{"parents":[{"commit":"9f7527973997baa5e685ea57d1563232234bc872","subject":"Merge \"Assert ChunkWriteTimeout errors are logged\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f7527973997baa5e685ea57d1563232234bc872"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-27 21:05:48.000000000","tz":-420},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed.\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e800d52bba504566d03d7f7c9ead851963ce7994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e800d52bba504566d03d7f7c9ead851963ce7994"}]},"branch":"refs/heads/master"},"86152d473d9e15c1d0931c8a868afb051c9ad97c":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2023-04-27 21:07:02.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/15"}}},"commit":{"parents":[{"commit":"9f7527973997baa5e685ea57d1563232234bc872","subject":"Merge \"Assert ChunkWriteTimeout errors are logged\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f7527973997baa5e685ea57d1563232234bc872"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-04-27 21:07:02.000000000","tz":0},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed. We have done it by killing all running\ncoroutines in the ContextPool\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86152d473d9e15c1d0931c8a868afb051c9ad97c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86152d473d9e15c1d0931c8a868afb051c9ad97c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8382dc6914d50a4fd1bf917b3e5dc77f6fbfa404":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2023-05-01 16:11:02.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/16"}}},"commit":{"parents":[{"commit":"f99a6e5762896c7789d168bc49d8cdcb47903264","subject":"Merge \"Log (Watchdog\u0027s) Timeouts with duration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f99a6e5762896c7789d168bc49d8cdcb47903264"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-05-01 16:11:02.000000000","tz":0},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed. We have done it by killing all running\ncoroutines in the ContextPool\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8382dc6914d50a4fd1bf917b3e5dc77f6fbfa404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8382dc6914d50a4fd1bf917b3e5dc77f6fbfa404"}]},"branch":"refs/heads/master","description":"Rebase"},"839c5daa74478f4307d5c5829b124317d19d5dd8":{"kind":"REWORK","_number":17,"created":"2023-05-01 18:02:10.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/17"}}},"commit":{"parents":[{"commit":"f99a6e5762896c7789d168bc49d8cdcb47903264","subject":"Merge \"Log (Watchdog\u0027s) Timeouts with duration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f99a6e5762896c7789d168bc49d8cdcb47903264"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-05-01 18:01:49.000000000","tz":-420},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed. We have done it by killing all running\ncoroutines in the ContextPool\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/839c5daa74478f4307d5c5829b124317d19d5dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/839c5daa74478f4307d5c5829b124317d19d5dd8"}]},"branch":"refs/heads/master"},"7aebf3d867edb2586c5daf51194785c56649e8c0":{"kind":"REWORK","_number":18,"created":"2023-05-01 18:53:16.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/18"}}},"commit":{"parents":[{"commit":"f99a6e5762896c7789d168bc49d8cdcb47903264","subject":"Merge \"Log (Watchdog\u0027s) Timeouts with duration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f99a6e5762896c7789d168bc49d8cdcb47903264"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-05-01 18:52:56.000000000","tz":-420},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed. We have done it by killing all running\ncoroutines in the ContextPool\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7aebf3d867edb2586c5daf51194785c56649e8c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7aebf3d867edb2586c5daf51194785c56649e8c0"}]},"branch":"refs/heads/master"},"5ca09e7eaf998833fe51487c82370503b0ac7e30":{"kind":"REWORK","_number":19,"created":"2023-05-02 04:50:47.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/19"}}},"commit":{"parents":[{"commit":"f99a6e5762896c7789d168bc49d8cdcb47903264","subject":"Merge \"Log (Watchdog\u0027s) Timeouts with duration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f99a6e5762896c7789d168bc49d8cdcb47903264"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-05-02 04:50:41.000000000","tz":-420},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed. We have done it by killing all running\ncoroutines in the ContextPool\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ca09e7eaf998833fe51487c82370503b0ac7e30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ca09e7eaf998833fe51487c82370503b0ac7e30"}]},"branch":"refs/heads/master"},"8062f9006f77183ede4f841089ebed8d45462877":{"kind":"REWORK","_number":20,"created":"2023-05-02 20:47:18.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/20"}}},"commit":{"parents":[{"commit":"f99a6e5762896c7789d168bc49d8cdcb47903264","subject":"Merge \"Log (Watchdog\u0027s) Timeouts with duration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f99a6e5762896c7789d168bc49d8cdcb47903264"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-05-02 20:47:02.000000000","tz":-420},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed. We have done it by killing all running\ncoroutines in the ContextPool\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8062f9006f77183ede4f841089ebed8d45462877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8062f9006f77183ede4f841089ebed8d45462877"}]},"branch":"refs/heads/master"},"44bedebcf2fa89202869e14a1d000d576d3328d7":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2023-06-13 18:10:37.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/21"}}},"commit":{"parents":[{"commit":"cc26229eefa24b33d54bb1420895be685705d4d5","subject":"Merge \"container: Stop pulling alternate stats when getting policy index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc26229eefa24b33d54bb1420895be685705d4d5"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-06-13 18:10:37.000000000","tz":0},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed. We have done it by killing all running\ncoroutines in the ContextPool\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44bedebcf2fa89202869e14a1d000d576d3328d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44bedebcf2fa89202869e14a1d000d576d3328d7"}]},"branch":"refs/heads/master","description":"Rebase"},"ef8469af83c3d34b91452b19b3dfa1895351b673":{"kind":"REWORK","_number":22,"created":"2023-06-13 18:13:07.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/22"}}},"commit":{"parents":[{"commit":"cc26229eefa24b33d54bb1420895be685705d4d5","subject":"Merge \"container: Stop pulling alternate stats when getting policy index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc26229eefa24b33d54bb1420895be685705d4d5"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-06-13 18:13:00.000000000","tz":-420},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed. We have done it by killing all running\ncoroutines in the ContextPool\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef8469af83c3d34b91452b19b3dfa1895351b673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef8469af83c3d34b91452b19b3dfa1895351b673"}]},"branch":"refs/heads/master"},"119104293f728a64a0054e2a76900fcb663d0d09":{"kind":"NO_CODE_CHANGE","_number":23,"created":"2023-06-13 18:24:32.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/23"}}},"commit":{"parents":[{"commit":"cc26229eefa24b33d54bb1420895be685705d4d5","subject":"Merge \"container: Stop pulling alternate stats when getting policy index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc26229eefa24b33d54bb1420895be685705d4d5"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-06-13 18:24:32.000000000","tz":0},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed. We have done it by killing all running\ncoroutines in the ContextPool\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/119104293f728a64a0054e2a76900fcb663d0d09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/119104293f728a64a0054e2a76900fcb663d0d09"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6d709db1b5bb791b348f226266e2be1701fff223":{"kind":"REWORK","_number":24,"created":"2023-06-13 18:28:18.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/24"}}},"commit":{"parents":[{"commit":"cc26229eefa24b33d54bb1420895be685705d4d5","subject":"Merge \"container: Stop pulling alternate stats when getting policy index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc26229eefa24b33d54bb1420895be685705d4d5"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-06-13 18:28:09.000000000","tz":-420},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed. We have done it by killing all running\ncoroutines in the ContextPool\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d709db1b5bb791b348f226266e2be1701fff223"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d709db1b5bb791b348f226266e2be1701fff223"}]},"branch":"refs/heads/master"},"c50f20301833f43798161212fa62025474aebd1d":{"kind":"REWORK","_number":25,"created":"2023-06-13 18:29:01.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/56/880356/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/25"}}},"commit":{"parents":[{"commit":"cc26229eefa24b33d54bb1420895be685705d4d5","subject":"Merge \"container: Stop pulling alternate stats when getting policy index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc26229eefa24b33d54bb1420895be685705d4d5"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-06-13 18:28:56.000000000","tz":-420},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed. We have done it by killing all running\ncoroutines in the ContextPool\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c50f20301833f43798161212fa62025474aebd1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c50f20301833f43798161212fa62025474aebd1d"}]},"branch":"refs/heads/master"},"be0c481e44e4f09b3c7fbe37f4b9ee62caa12a27":{"kind":"REWORK","_number":26,"created":"2023-06-13 20:58:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/56/880356/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/880356/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/880356/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/880356/26"}}},"commit":{"parents":[{"commit":"cc26229eefa24b33d54bb1420895be685705d4d5","subject":"Merge \"container: Stop pulling alternate stats when getting policy index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc26229eefa24b33d54bb1420895be685705d4d5"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-04-13 19:01:07.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-13 20:57:52.000000000","tz":-420},"subject":"Fix proxy traceback for GeneratorExit in py3","message":"Fix proxy traceback for GeneratorExit in py3\n\nClient when explicitly closed before finishing the download.\nleads to a 499, but the shutdown logging for proxy in py3\nneeds to be fixed. We have done it by killing all running\ncoroutines in the ContextPool\n\nChange-Id: Ic372ea9866bb7f2659e02f8796cdee01406e2079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be0c481e44e4f09b3c7fbe37f4b9ee62caa12a27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be0c481e44e4f09b3c7fbe37f4b9ee62caa12a27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
