)]}'
{"id":"openstack%2Fswift~880676","triplet_id":"openstack%2Fswift~master~I4d4f20b92b85255ac8bf66f2c830e691e64bbe47","project":"openstack/swift","branch":"master","topic":"p-memcache-error-timing","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-05-18 11:52:53.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-05-18 09:43:27.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I4d4f20b92b85255ac8bf66f2c830e691e64bbe47","subject":"Memcached: emit memcache timing metrics when exceptions raised","status":"MERGED","created":"2023-04-17 18:13:42.000000000","updated":"2023-05-18 11:54:05.000000000","submitted":"2023-05-18 11:52:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":61,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880676-p-memcache-error-timing","meta_rev_id":"38c12e5f396669be0501d33a5dc7f40e0997e75a","_number":880676,"virtual_id_number":880676,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-18 11:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"recommended":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2023-05-17 16:23:57.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":2,"date":"2023-05-18 09:43:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":1,"date":"2023-05-18 09:43:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-17 18:15:51.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2023-04-17 19:13:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-17 19:27:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-27 11:19:35.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2023-05-17 16:23:57.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"d381f62b336248a11dd53168e07379a22083cad6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-17 18:13:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"519e8a685a665a2ae554cf058b22828f068e3ae7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-17 18:14:01.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"549a541abac2a6bb645d5f15cd105d20627573c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 19:13:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b8a1de8b21d472f966fd67c32cac6da\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47e7bb0d82224713a2d9365c4e385cf9 : SUCCESS in 16m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f37454915fb14b3885420168fa16ebdd : SUCCESS in 15m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7933f7415324938be483e385fc2d5a3 : SUCCESS in 18m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/94b644756b614eba9239585b057728de : SUCCESS in 57m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4235d3a715b045e17ec65806d585499d1b16e9df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 19:27:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7d505a468824f968f8c3980c5fb0332\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50465186c1dd45dabcd4f976a6e0c260 : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/209f88b6e6764080aaf9d5fd9f49a0aa : SUCCESS in 46m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/72c3344bda114b8e831c646ed7fa20a3 : SUCCESS in 37m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f9e61e1b1c04976a6e24635d91373e4 : SUCCESS in 48m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/265218fa7b5441ad9e49e947ebf85fff : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87f02ae7e8a44e1b85f26852e17d3d53 : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/345ca91b13d7405e91fcc898ef3ca060 : SUCCESS in 18m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0fe323433ffa4878b7d446dd28409d9f : SUCCESS in 9m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/08e485d8423c4fc79a84aef545f8723c : SUCCESS in 6m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ec125760dea641cfa2300496bde82f89 : SUCCESS in 16m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e355b1fb801343ccbe42278947ad9eb5 : SUCCESS in 15m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/769b44bba8c643b6932fc94964cd1c8f : SUCCESS in 15m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e29e369ed6d24aa9803040a8ff53e5ec : SUCCESS in 14m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fb29cd262adc437984f8d497ca2c0994 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/35778d729e4f4d26b867aab52a5bc706 : SUCCESS in 16m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/655224682c74463392443391e3be17cc : SUCCESS in 17m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2c5cfdda13eb4ee3ae19d4f7fcf60ad6 : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/35775bcccf5e4ede8a26be650e3d96db : SUCCESS in 15m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f812ddad988e4d3c88ea850ea244ae6a : SUCCESS in 16m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/69adc4a8a0434f29b2c6aecfa8512586 : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4c7cacfcbfa1450a8541e2b6897a6028 : SUCCESS in 16m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/125f1449cb8541e6a06bde0ccf88fb1b : SUCCESS in 8m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1887d5a1d7ae45db9a0f8c520d2850c2 : FAILURE in 52m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f9415b20f0db435aa79b4d652a8c9800 : SUCCESS in 1h 07m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a633bcf6daf844c9a087fbaf207abfbf : SUCCESS in 47m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f13d34237bb544afac0ce8adb1beb72a : SUCCESS in 53m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6bda457e7d2740b9a1a470d6983384c1 : SUCCESS in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6436b95684164c2194228c24758c217d : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4a18f5b9ca9940aea5e4cc86dbb40890 : SUCCESS in 19m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/caa8c8a0f9e24b4a8e1e872e41896c04 : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":1},{"id":"2a25a89b943c0373ab75750716d3eaff47520105","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-04-17 22:24:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9f065595f0c09368895452eb5ecd7e5c411ef8e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-18 16:55:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"546faa671452c16283556d8cb142916dc7ab1c2f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-18 16:55:38.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"2fb8a4a82eac05d35065dcfc0a76952fe90539ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 17:19:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d9c1de81129452db913fa5eb0ff53d9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e461d20fa12d4ecfb687a9c80b325e2c : SUCCESS in 16m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b08b480ff6c478fb82039d797f21c07 : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9c3a777f2374998902999aa39f8723d : SUCCESS in 20m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfda34b5cc69461eb1b8c9ded600b586 : SUCCESS in 22m 09s","accounts_in_message":[],"_revision_number":2},{"id":"1610d01e987b1e3b6085a02230b5dc9629ff32dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 18:45:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e28f9b619f004b5d92913cb17b6540d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/852ddf9d0e6e42a19545e6d3bd963898 : SUCCESS in 7m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/1afb090912d24d4aa06b7a2186a18c12 : SUCCESS in 1h 00m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e177e0ca097246f6b01bdab8ca3ceff8 : SUCCESS in 51m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/edad2d37c98d42c59355be9ec9889021 : SUCCESS in 1h 12m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9afd30c052534bc18d48e33694904f15 : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2b8684e32cf44239fb8f365151f5dc5 : SUCCESS in 15m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8d743d7b3ca41869cc3e1d48ff5dff4 : SUCCESS in 19m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3991f50d07414966a3f116cb5258a2d9 : SUCCESS in 9m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8315b4e848624bb2a6548b39fbfd500f : SUCCESS in 6m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9609dd12508947ada495bcf71f1f4d6b : SUCCESS in 16m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/97d6562122a84a308f360c520de88064 : SUCCESS in 19m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b322123a252c4c109dd17752c8f389fe : SUCCESS in 15m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/09adf5b2b521402eb49c53084a459dbd : SUCCESS in 15m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/af51b2a7e8564958891e6251d2faede6 : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b162e6ebf74247e695283f9906b9eb15 : SUCCESS in 22m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e1677b9e818f4ad2ad56a2c0a0e2b753 : SUCCESS in 17m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a851308b28a844e8b5d6198f8ed25166 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9dfbaf1a711e4736ae85000bc2c1aa99 : SUCCESS in 13m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/98bfa288bb0e4446af0e52ef268b4633 : SUCCESS in 14m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/99fad91d0e704385840cf5f2f5d58e37 : SUCCESS in 7m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8db501e8df6543b3b3919a9f8908dbe4 : SUCCESS in 16m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/efe7880dab494379a4b71d50dd711d54 : SUCCESS in 9m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/05ad2564db7542ee85778ec90aeafb2b : FAILURE in 1h 00m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f2b6c96b3d643659637b3d1c46248a5 : FAILURE in 1h 02m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/acfefd16b1a44c16a8d53544d0bac381 : SUCCESS in 1h 09m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d29adf7cd15a4e1a9e9fcf880d754e86 : SUCCESS in 1h 23m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83d0973bf5c1473fbc07aa6d9f029f0c : SUCCESS in 17m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a71160395af4784ab5a808f9f5355b1 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10588bd32d8f4f87a35223a37ac1120e : SUCCESS in 24m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9e2533f7eb54c81a449ea977889faae : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":2},{"id":"9333d7d9c5086c1546272f9ad99c1215c2fd3a07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-04-20 05:49:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"efd800328423d3fc8ed8968d2ed4078be32543ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-04-20 05:58:51.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8ab37c1b16ec5cb13bef6f4202f1d4504e4bbeea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 06:24:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58e5daa35be54defac06574eea45aa06\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a376761d0bf240bc88c297ff22960b31 : SUCCESS in 15m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ceadf68a45d549bbb07b9b9985ae8202 : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d365bb247ad141ebbc5d39173145010a : SUCCESS in 18m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/780b27d5e48d4de186ddbfb1797d5ea3 : SUCCESS in 22m 48s","accounts_in_message":[],"_revision_number":4},{"id":"5f86726e49d0a7e22b1ae36cb8f83db7ab69474d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 07:16:23.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ff2848ea17d4dbf94cf51241c69955d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e437b73d934d37a5c767308d68e0de : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/462edd46583249b78d2af436d776a73e : SUCCESS in 49m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ce211d2816a74f7aa8c3b33eb428459f : SUCCESS in 32m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/355d377650cf4a00bf4040fb9fe972b8 : SUCCESS in 52m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29c7623956a248ed8696d9feb982f575 : SUCCESS in 17m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45d3114f2f8d4f45af8e711702938ca7 : SUCCESS in 17m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95862db32ea44c859d352d61147c0a7b : SUCCESS in 18m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87c93d5c59bf45049112da9a74550c23 : SUCCESS in 8m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f800e542113441d5a2e5e369e4c700bc : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/154792ad65664e08a699232ed999da7f : SUCCESS in 23m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cd15ac93c08f4d089615388cf88b9899 : SUCCESS in 20m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5c14e38329fd4afc8f06711b2eca51d1 : SUCCESS in 22m 29s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3cd5208cdb0646ad86303ba474ae2fca : SUCCESS in 22m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/12bcd779eea2424daad9441d68acb6d3 : SUCCESS in 20m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7f1c2f8776ae457e9dc3a0cc9b7cd05c : SUCCESS in 23m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bfc2c4c653724aa78ca16c25708c27b4 : SUCCESS in 21m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e04f33ab4e254fb2bc04789486d75a7e : SUCCESS in 20m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/50478a38d87048cebcda90ebfd7b9638 : SUCCESS in 14m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a23d13820a3c4cb09ac101c49eb61416 : SUCCESS in 13m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8697cf3a06964488b6b1db2e3a6ffb97 : SUCCESS in 8m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f0fe99acb3364c338dad4e26b16d69da : SUCCESS in 16m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/24ccc734912348e49a10a19a9a84397a : SUCCESS in 6m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/97fd98e858d6403683b856696a85c9e4 : FAILURE in 1h 05m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d539eac7de084039899a3a6ffde37cc8 : SUCCESS in 1h 03m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c4af340335c9400b9edc04fbea76a6db : SUCCESS in 55m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b33f4c3a6c8f4a21b5c799f2661f7b1a : SUCCESS in 34m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91a7aa4a930c470e912f8a68b07f7879 : FAILURE in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86a5064d253749dc94edd2338a7dbe5b : FAILURE in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/09affdd03ec549aab4e06db292c6592a : SUCCESS in 24m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2181861bac64b53a4076518f8bd173b : SUCCESS in 51m 03s","accounts_in_message":[],"_revision_number":4},{"id":"3ba18e674cb709739515714e34be355348358538","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-04-21 05:54:25.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f301488c61400e7a3e8803eb08ed64a9727a5ab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 07:09:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c066616bd335447f908b8369ef6c9831\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4940e983e3c34bff936fbfd8da7ceaf3 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/16aeb41ee55249069fbaefbdf738fe89 : SUCCESS in 48m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3e403225dbc9462f8f39cf31ebbacb7b : SUCCESS in 41m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0d362e816f44c7ab0dcf75ed8a1af67 : SUCCESS in 48m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/634bc47f837646958e0bf05c5e558aad : SUCCESS in 26m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd666c794747451897d9f6feafffac2c : SUCCESS in 25m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df294e0faf5943cbbd0030527da8c7ea : SUCCESS in 25m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/60be71a59b94433ba558019b6c69751d : SUCCESS in 7m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a9b0533fa94642b0a65ccf6031344c3b : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b6c948c3ef38418585ade351e5569c5f : SUCCESS in 16m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dcc172a02ad74be29962a242aeeac11d : SUCCESS in 15m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7cf8afa2227d42d3af2f30ff0819a6f5 : SUCCESS in 12m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/045f6bce0ade4fda99e6d51ba5361547 : SUCCESS in 22m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/72d62c6dca8742559d48804d06f68c2e : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/95db49d13c4f412ea7927c7053c81d7b : SUCCESS in 15m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c6fa573fed76407eba64698c057859aa : SUCCESS in 16m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/19357dd7adf6491ab782dfbf8e382826 : SUCCESS in 22m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e4e8ba256cc147ac837c2a56867dddc9 : SUCCESS in 23m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/09fead05d3354c3fbc4c1620db7710d8 : SUCCESS in 15m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ffdad2123dc041028fc9150c380e7025 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2750d8a363c04020882733e19cf89411 : SUCCESS in 22m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/140b541ed71f44e2b6c84f8fbc44480a : SUCCESS in 7m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5fc5304cf46c490c8c88bf6645ad78cb : SUCCESS in 1h 12m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ffe43d8aa2174dbe911ef450fa066075 : SUCCESS in 1h 02m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2dc99a2ac65140cd884bfb548ceb1dac : SUCCESS in 50m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f171b19d62d140d18f1deaf507886812 : SUCCESS in 54m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/269d4e96ce1747d7a72f699717b31e22 : SUCCESS in 26m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88412e86290f4200bb052db8e9ffb8d9 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d6a9351ee06b414abb3fff2fc49c309c : SUCCESS in 28m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f7d11d95be044aeb507e1b16cc14a57 : SUCCESS in 42m 42s","accounts_in_message":[],"_revision_number":5},{"id":"2ced5e56071379b122bab6f67d360de00aac3bd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 07:15:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e7c3d28f8c1484f976c932ed0dd77f7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/634a0314efdc4b4bb564c7617f8ee3b1 : SUCCESS in 14m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b979a4b8f89a47949eaf828b1486cea7 : SUCCESS in 1h 12m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/862ad360f77b4dff854aa6a85297034d : SUCCESS in 1h 19m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e312b44d7095493cab4a20abba8f429b : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":5},{"id":"b16f3e265bcf80b25c55f827f268dc05f51c5970","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-04-21 23:40:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"994fa32b034ca0f65034777fbace2718d2057fc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-04-22 23:21:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"0c66e0fe1600ff8462ae373376c99abf8c1c4e7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-23 00:10:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f66ac89ae44443508a79e3a64e89274f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d532d4eb7d1a4b15816ac8e6c595ca1e : SUCCESS in 14m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f1597c166b6b4a699159e1f7d05eb7e4 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b98bfdff3ba44451a9374d4ac7be7410 : SUCCESS in 21m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/49a7d65fdd714282b0b690f2a9620e62 : SUCCESS in 47m 25s","accounts_in_message":[],"_revision_number":6},{"id":"38d7de2c68faf076ef41c850cef1007ec329e72b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-23 00:23:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb917ca5b52b454cb2513b91d574b165\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4afec82182a4d058dfe20b337ddf885 : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/335cad0b4a914e9584b2752faacb40c5 : SUCCESS in 41m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e5753b21636c420290d44450fb850197 : SUCCESS in 45m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9005e3aa9d6a4870b0a63fbf797307ab : SUCCESS in 57m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/714c8db75e3146f4a7aa8c4e1da3ad5a : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8deac0451351499a9f284b6c9e69ebd1 : SUCCESS in 21m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5becef01cf914e9da3de0e0e05994943 : SUCCESS in 21m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/71d30106e83648e285132f832c2e335d : SUCCESS in 7m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/df82588df48e479c9aac19d377bf6bb4 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a0d17c560a0c497cb5ec8cd5f7b62113 : SUCCESS in 18m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ad0e951e38841a0ba2ef11a40d79427 : SUCCESS in 16m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/78038aeab9a94a8c953fba9b7f33c2ad : SUCCESS in 19m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3001ccca021141cb8780a79d76352803 : SUCCESS in 18m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ba0f64a053314db5bc2bd5898ca690d5 : SUCCESS in 17m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a81c1a0a95ec44cca9b98e0712b2a3f4 : SUCCESS in 20m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/16ba1725f3a94c6ba410ef17e617d166 : SUCCESS in 21m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bf9d00bebe5447248e719df440719558 : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a7272c1aa90a451aa352c48244392f2b : SUCCESS in 14m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/062c62b342d0421ca8e86d911186c67f : SUCCESS in 17m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4c140675c8744452a54381cabb43424d : SUCCESS in 6m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/42a65c644b7c4e81ac90ae78bd6f225c : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/195fd14136c1408f90b0d7fb9efb555d : SUCCESS in 7m 06s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/94994e00d0f2443c841163afa6a8c082 : SUCCESS in 51m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/167c78d3b8f845c9892737d77cff2d8c : FAILURE in 54m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/efc8b105fd0e4452a86183581cdccaad : SUCCESS in 41m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/553acff8f8b44387bbf9efd58136e0b9 : SUCCESS in 42m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62ad465c6cf24496b42f3ed4228e4902 : SUCCESS in 21m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c1c24ec178a46178e9b5d43a7ef8ef3 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4830faae23074b75bd89a5420021409d : SUCCESS in 23m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9897d368686b4a7cab3531e9e61d9c6b : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":6},{"id":"80f803ab0f05f370cd89c2574135f8caa0a098a4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-24 15:09:16.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"46996e6cf92970de04645ae437b70f8b063851af","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-24 16:29:57.000000000","message":"Patch Set 6: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ebe4074e17bdbb853b671b8ce15e9f6217935f9a","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-04-25 05:17:24.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6905dbd539d32eafae1043d2efd4daf6d1e1ccc6","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-04-25 05:17:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"566f3f44b2e414e96a2f99aa444543daebd8edc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-04-25 05:18:23.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"9b3449b6a0e0d5bcc4a588d7e5c3fe17b6cf972f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 06:14:48.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6acd5d8eed584f2fb5583ca1665d2ee0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41cfd750385b46ea8be11714948f489c : FAILURE in 15m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf018fc933fe4829a26f60e663f587de : FAILURE in 14m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f00b0395babb4321b0bbcd345b39160a : SUCCESS in 54m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/02b52ee362c14355aea52fd8328ae497 : SUCCESS in 18m 55s","accounts_in_message":[],"_revision_number":7},{"id":"37ff4df79df4520bf16b60970b8fefb4da9b8d92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 06:28:21.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9afe3b98894402c96b26f2c7a54861a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/920b644d12ba4db49ad5e3026d6bbc37 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/9fd4e4028f334ef08fab2bd6cee80341 : SUCCESS in 59m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/04302f5c800e41ef89554fe807b5151c : SUCCESS in 47m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b85af92a70184bbe8fa3ae8687572257 : SUCCESS in 1h 02m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/681343d0cb864711b4cd2e08df59b2be : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d822f289f274545b2cb9c1f68536702 : SUCCESS in 30m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/249aa8c3af9d4761892c401fc9f595c4 : SUCCESS in 19m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3b7e6a7041e545a4a2a804ae78578068 : SUCCESS in 8m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6c87d4a213d04ead9fc1e6163f9d3483 : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8cee4d11b2da4e3dace0ab5df1edb5ff : FAILURE in 17m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/54fd6ab6c8c94830a6b0eddeb5a32230 : FAILURE in 15m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/570a1b7f33bb4da882f53971292c7f34 : FAILURE in 16m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ff82c67b7236472a967bebd03622ed93 : FAILURE in 16m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/93aa337366bb4f99a6fb77830a9b91d9 : SUCCESS in 13m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/88862b0952db47df9825fe3b3f9896a5 : SUCCESS in 17m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/27cbc702035f40e49a9be53ccd137bf5 : SUCCESS in 16m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c848685e6a11467b9996497a2b313cb5 : SUCCESS in 13m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5d6611943730457cb299841b6daa713f : SUCCESS in 14m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/76a665b8ef6741f29d1f809ac8c96ca2 : SUCCESS in 14m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/698ee53b60634f85bd5a1c24879affa3 : SUCCESS in 6m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f43cedb54324bcfa0a7a99279ca2183 : SUCCESS in 14m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d580e888bd5a4a89b7ec0fbb54101a02 : SUCCESS in 7m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0d10cd4a48b14e708d76cfbfe95f0287 : SUCCESS in 1h 04m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b3a638ea122f409895a94322c8a048e5 : SUCCESS in 55m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/168466f753384db89c4ebe6e669b2877 : SUCCESS in 50m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f23d39eb8edd4b52b5376c3c68d22896 : SUCCESS in 53m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf70e603b6ca45449d198adebe26b033 : FAILURE in 16m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6508596a6b2f48b09c36eb351e81ceff : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/282d1e39e85944e29603c27ae84e2a99 : SUCCESS in 39m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6da4a1cc4fb4ac8aafa4cf14bda5c7a : SUCCESS in 56m 14s","accounts_in_message":[],"_revision_number":7},{"id":"ae848cc6373df60d4415511c49f46286028a05d8","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-04-25 22:08:46.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"31fe6c1034d86bc9564e3e8d6eee752fde3d4029","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-04-26 01:28:54.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"a716cea87cd5d0dd7bb7e5e2acf83efa0bf051eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 02:46:25.000000000","message":"Patch Set 8: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bedfc38dd524806b776b3e414588b24\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e153ea62ca0402680b325ac68830744 : SUCCESS in 6m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/393ba1ecaba543baaf386e4c81e35fd9 : SUCCESS in 41m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/04d5be85cca243f68d34bf8975a68f9b : SUCCESS in 28m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6f3455654f7419da2b8d075c7eebcc9 : SUCCESS in 1h 02m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d4f4b80efc448a8873303e034488071 : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84d004cacfb74aea9ce2fde9c855b2cb : SUCCESS in 14m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a3ddc9395dc451596d9593ffab7b184 : SUCCESS in 13m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bc4e6ff944d3462fa6dd32bb5677d272 : SUCCESS in 7m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d8279d31e93f4b8780d2428904efc5c7 : SUCCESS in 4m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a25a731884749be8c75739238a541c2 : SUCCESS in 15m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a985c8c45184af0ba05911e11678f44 : SUCCESS in 17m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d73e9c8779884a6c9c91193749e73fa5 : SUCCESS in 12m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/97e912f9f02f436382469f08ba97596d : SUCCESS in 15m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/97f689160eb6497e960ad3d820572052 : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7847f3409360459a940d6fc26585e886 : SUCCESS in 16m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ef005d3ac5fa41d2b8a296d96543a97e : SUCCESS in 14m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f4bf3dfedc2b4629acc1fbb638894b35 : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fc85401c47fc41c394f9e7ba607c51cd : SUCCESS in 12m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6dab0e01ccda4c98a18a2548a30d8f95 : SUCCESS in 14m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d3399df55e2a4fbf9ce21cb1b4d47426 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aef4d752dc6045c3a3b89a41c0f12e7b : SUCCESS in 16m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6795a01c3bb7479c95926f7b00586a54 : SUCCESS in 7m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1c1521d0c7504b0591671f2718344c36 : FAILURE in 1h 07m 09s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f2ebe7e4e0f4897b44221699d6feeab : SUCCESS in 1h 16m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de94016747fa4058af582eda105ed742 : SUCCESS in 53m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/55f439e75eae4cd4bebd27af39bf9ffe : SUCCESS in 53m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba3bbc47e55d428e9627ee42f88b9e2d : SUCCESS in 17m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1343b5560646460b93ffd934122b062b : FAILURE in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6dae7fba4b3648a09da89e22b13b9adf : SUCCESS in 19m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13336048ab0143a8969d92e2d2014a85 : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":8},{"id":"49f25bec8175fcb4f335e9466bff0cad52a6270e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 02:50:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7803001ccb147f2928f2cfd60eadc14\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fa61a9f8294479c80eae1f2f57b3d11 : SUCCESS in 14m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d6251c285fdc45329ec31b168359319d : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/46f689538ade41b2bcc14393105d663c : SUCCESS in 24m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/342cb37c36964a2c9b702dc59fae0c63 : SUCCESS in 1h 20m 23s","accounts_in_message":[],"_revision_number":8},{"id":"7aa2316dd503367142fd14842fd53e7c220b4083","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-04-26 03:36:02.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"b5b4c5e8eae3f380f4a101329ebfdc9125064830","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 04:00:05.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f36676a9f5394c90ac760320ae54744a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a761fd32818d4b9a9418cd83630a303d : SUCCESS in 16m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e14a7235e18440d296f6aa7333e31d0c : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f1add4c5dc243ffb6c2023eff6c8470 : SUCCESS in 21m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed4cab8076f54be4a778429b850dae3c : SUCCESS in 22m 55s","accounts_in_message":[],"_revision_number":9},{"id":"5f4ee153a841a0d1243b23a6658f70020ac7978e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 04:44:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c9b0d4d5bb342e990ecf11988848971\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e98539fe13f45e29bd1fa8e0a85a256 : SUCCESS in 10m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/2b2f472ac6354323929bbad8ba238298 : SUCCESS in 51m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/65686da8f5ad40c6935c83b0fb306db0 : SUCCESS in 37m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da0f16d28112438190e0c4426d5d515f : SUCCESS in 57m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb04a56fe0d44b91bb6b27f98917d634 : SUCCESS in 25m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c3d0d342a69473fac4ecee339197284 : SUCCESS in 24m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6f62fdfd44b4d4d8b35d6159649a313 : SUCCESS in 27m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4b08a337148f4833a4270668fdb51ed1 : SUCCESS in 8m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bda26babc9674217a46272ddb8221287 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7d5cfc4ff0a64c3bbda688bf688b7008 : SUCCESS in 25m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/905db998a60a411f99d1e1ac7e0f37f9 : SUCCESS in 25m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f83cc622f3aa409185d6c59cf8b55891 : SUCCESS in 15m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2473c29447d44ac5911a8bd340d03899 : SUCCESS in 16m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/22fa646c6019473f9cd604dcd1d074f8 : SUCCESS in 24m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c22703f764554d92bfdbbb15f72e7780 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bbc1d3178f194f608b1d746a86aa50a2 : SUCCESS in 17m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1c1b4acf5b374f32be848138e85a8a38 : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/787001042fda4725b09467bb5c09ae7a : SUCCESS in 15m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9f238471491e438ea15bd3c0e15ae9b9 : SUCCESS in 13m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7b3dee739d6c4993b3dc539b0f305c10 : SUCCESS in 8m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/af81538e6bfa45248e5209279f7b08f8 : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6d0681998f0948deac1fab641aa0c321 : SUCCESS in 8m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/66b1035276e24424b0c6ca33e7bce44c : SUCCESS in 53m 10s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5339f8c9b7554fad9331ebd6b1875df5 : SUCCESS in 1h 01m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd59ead89b4f4b2382831ba63547b48a : SUCCESS in 48m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb34ef8b018244e58da92a3913d41ef2 : SUCCESS in 50m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cac84781c2b4b42988de276269257b5 : SUCCESS in 32m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bb88e0982414c509e148137ff7c27a4 : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0dc7e346b2cb437997f536b5e05ce6d0 : SUCCESS in 24m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b85abfc107848d1bd1f445e64c4b177 : SUCCESS in 52m 32s","accounts_in_message":[],"_revision_number":9},{"id":"07fec925d777c1d878aa60fe3d9640ccf93ddca1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-26 14:28:58.000000000","message":"Patch Set 9: Code-Review-1\n\n(18 comments)","accounts_in_message":[],"_revision_number":9},{"id":"5a02b295e829dbc7f8630a1d885cfadd69874c73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-04-27 05:48:21.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"5c07a8315781e6c343eeb8ac655e6a01657b7597","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-04-27 05:51:02.000000000","message":"Patch Set 9:\n\n(18 comments)","accounts_in_message":[],"_revision_number":9},{"id":"f789b6899d0828211b5043c03b20633d732b2ac6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 06:42:36.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60f95f161f9745df819e4c2544943527\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c82a4b2f634478b8aa77a1a7ff9079d : SUCCESS in 12m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a73bfa31e4f8488aa827f931c889090a : SUCCESS in 52m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a793d9e6498140248aa4fbb7bebc95a0 : SUCCESS in 50m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b86db9fd76e345daaec1e7acbede1eeb : SUCCESS in 19m 08s","accounts_in_message":[],"_revision_number":10},{"id":"0f1ff1699fb2709422826b86912d53bc67f0ddaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 07:02:50.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/101535abf46e4efc83aef90b3f696822\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bde28ef5a9244d13a478269713fa88d5 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/24b7f3caa11449dda1699c6e6f334dde : SUCCESS in 1h 01m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/85a81df4326e424d97f7b36b7ab53ac3 : SUCCESS in 51m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2e841bcd0ac414ba80f6f904c8e0928 : SUCCESS in 1h 08m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53db51ae821e4a56afa6a3f7e6fddf8f : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/723137bb73204c1fa70cbbe366e7ddc4 : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68cf33526e7e40efa979d84dd88af532 : SUCCESS in 19m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c6d49fcc7a75423e8682595559e843dd : SUCCESS in 8m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/31947a36d02c4c9096d6ee0094083927 : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/12e13a57aeb24e0f883715b974c45425 : SUCCESS in 22m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f3e37d676844480487514a0080a8736c : SUCCESS in 16m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f923a6260dc54bc2bcf8fba1175756d6 : SUCCESS in 13m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a69787c2ece84ee5a3fa9b159e6afb99 : SUCCESS in 19m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bdeb7492023c47e7b394e19ec37f7ea4 : SUCCESS in 20m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7dec4054c05446f3992bf8b00bef48cf : SUCCESS in 14m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0433742b4b4845dfa5707ab095689ff9 : SUCCESS in 23m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ca595166e9504e66b729ca6f1af8cb9f : SUCCESS in 12m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/672dadc9ca504f5cb818e71d0792907c : SUCCESS in 19m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b51e32edb485470da1c31873ebb77d96 : SUCCESS in 18m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c1f55b00694f4383ba661fd75e642f83 : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/320ec9a73f164eb2882dabd7daf26e70 : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c0e1370687dc47f7b9615cfd7d11af09 : FAILURE in 6m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/66d85dd22602475d99b8c70db32bab7c : SUCCESS in 58m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e01324f08c5e40299b6743e33f4033e1 : SUCCESS in 1h 00m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f11d20bdeb7476a8a5008d823939d96 : SUCCESS in 55m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e611c2455f6c471b9d6408f48f7f0251 : SUCCESS in 55m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25ba3fe0cef9414b9701c0b04a17a488 : SUCCESS in 17m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bdbc4bdd71440a0bf209899721de995 : SUCCESS in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/087bef1ead4f4ec0bb8c695ea399cf07 : SUCCESS in 21m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3137a7a816f4b218297f490015f393d : SUCCESS in 54m 14s","accounts_in_message":[],"_revision_number":10},{"id":"141d493a0b4bfe8bcca685996745464d7f3b8e2e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-27 11:19:22.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"a43c6c49385edf2c47a19247a4850d42c5aa0ef2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-27 11:19:35.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"1fee05cb9b64fe107e9411358a1aaf578de69b35","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-27 11:22:42.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e189be9c175e0758906664143936f1817dcb7af1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 11:50:29.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56f2c9b9079643768b0c761a3f424916\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f31e8cd4de2540e58cdcf63e12c3ca47 : SUCCESS in 23m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b7439edec38423c82c48725fed311a6 : SUCCESS in 24m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5bf4e0ec655d4e979a5297fb27e3eb8e : SUCCESS in 25m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/58223c8d8a484f0182cdd1f1594d86a5 : SUCCESS in 24m 27s","accounts_in_message":[],"_revision_number":11},{"id":"edeff5c1d7791fac226d0411ff6450961481a300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 12:40:28.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65a696ecd66d4e70a16348d6e11fb5fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5467fcc9a01941a09823227eb83b4284 : SUCCESS in 7m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/49b6d4d01ff8422095869d1851d4be8d : SUCCESS in 1h 02m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/82c1f70168954af78215416ddba10171 : SUCCESS in 41m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6144f8e8e4fd49a1ab3641ebeab38952 : SUCCESS in 1h 01m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/896521db4919495cb10bb541f23a1ece : SUCCESS in 15m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/745e58b3d52543bb8a3fde2b8c616f6e : SUCCESS in 24m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/503ac78943ec48319e0d438d961c4582 : SUCCESS in 24m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9926006618414f5b9e8660c4f81c0684 : SUCCESS in 9m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7069fea6e07d42828ea267ea32654732 : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fc640ffefd51499ea9de089369a158ed : SUCCESS in 18m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/737ee6d122874287b8e74a4d58cd2229 : SUCCESS in 15m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/87aec7abedfa409a942ed09968d36451 : SUCCESS in 16m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d7ce884d76114494af7420ee75dd586b : SUCCESS in 17m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b5fcb28cb9a24e95bc618d07aea89db9 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e5e106531550401b9f27574f9c11ca92 : SUCCESS in 17m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bfc8882631ea42fabed3c5084e3a2337 : SUCCESS in 19m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5ceb96fac5a04f8abaff237117b06d9e : SUCCESS in 13m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f8377a5183554403bc1fc754868dc68a : SUCCESS in 13m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ee896e9e35dc48e89ca7ebf64d46e836 : SUCCESS in 13m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/21305d04e4094d58a231746c9f1fbc20 : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4ebdedf63e234eb3a1cc7308360fa902 : SUCCESS in 14m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/306a14cf039945d9b94e97db63c1df63 : SUCCESS in 7m 25s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0338d937234846cb816656b8aaffad2d : SUCCESS in 57m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c5af680bdd374af3afe07899a464be89 : SUCCESS in 1h 02m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/814ddff222af42f3954d55545f6c031b : SUCCESS in 47m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae33bc6ea9164705bd91ab4de8a0ec04 : SUCCESS in 55m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0dc4813fb4d74707ab223b6034e9489a : SUCCESS in 16m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ae3656769df40e49e636f5492ac9065 : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05982e4549004a438c2931351d2c9a54 : SUCCESS in 18m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/151f9620d7a1499b92db36486d33467c : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":11},{"id":"bd4823dbf2143b1144427fa9f96701e256c1f9aa","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-04-27 21:52:02.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"d9405d1be5a68e8e989556451cf91f52d5ff2f4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-01 21:26:29.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"3c83ccbf5b78596bf4dcb57406ae168e2d2e0120","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 22:35:08.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d2cd6fbed2749e198366b6aaa03c0e9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1cc6ff098686421d95b2ca68e66f43a7 : SUCCESS in 14m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9251bff4d6bb4138b94d8a196985821a : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9049e07f77649319f1825cbe80bf6fb : SUCCESS in 1h 08m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ace1646f8a4c4108bd3c6a746f57f382 : SUCCESS in 22m 34s","accounts_in_message":[],"_revision_number":12},{"id":"ded9629fc18b9a2b1d77868a481c17f1a3e5bd93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-01 22:36:42.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f378fbfd52e54f92b79e04d577f0f5e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bca53f25e7a14731bec2f1126efbd45a : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/1c154ca4f5b34746a3250af4e932e4a8 : SUCCESS in 57m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0a32b6f003ab4d54876fc81801f72422 : SUCCESS in 44m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b4b872a17974ffd81392b86c911b4c6 : SUCCESS in 1h 02m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99d628619f5143d9a17f88590ae4d392 : RETRY_LIMIT in 1m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01495777c9044cccbc0dce1e8fc6fe77 : RETRY_LIMIT in 1m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6cc32c5dc5604970aea3746a67f8ff8f : RETRY_LIMIT in 1m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5ab3e0a956a748698035303e022744a4 : SUCCESS in 8m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/253d5db6d8c24f57b81b786693c639e7 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/33b8e37efe5b43cda234a7bbb9dd189f : SUCCESS in 33m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe0ea1b8e5ce416b8d523f12e4ef505f : SUCCESS in 17m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1c0483411c32434d9022e879d0ba24d9 : SUCCESS in 13m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f8aebe5cd18648c3a02afc97e44a37b1 : SUCCESS in 14m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/640e6f066fd34cceac22de6aeff7b15c : SUCCESS in 32m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d6315a0148634adb8ea048dccb6d4faa : SUCCESS in 17m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5861d3b53b5845f6a962b9f12bca9482 : SUCCESS in 14m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aec6fa4a00fa4c5487af01e23cd5bce1 : SUCCESS in 31m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5e1302b1e63247399d51b34f07e04c61 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1ddc23275bde417a90b647d5274bd4c5 : SUCCESS in 39m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fe25f810624a4b4eb9fde204f4950e2b : SUCCESS in 5m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8b7244f55edd46e49879cdd426d69b13 : SUCCESS in 14m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/174a1001aef24fff9851b38bc783b88d : SUCCESS in 7m 09s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2e4dad1dd38348afb85598393059ce65 : SUCCESS in 50m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/59e6587144f64456a767fe3174afd7b4 : SUCCESS in 1h 05m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4c146290661e4d838ec9d24242557579 : SUCCESS in 38m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/215ac294768c4be2b6166cef1a0e9c94 : SUCCESS in 31m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/408fe0f882614f2f90b918c7d72dbc57 : SUCCESS in 16m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40b323b910ff4e96a0117610b6393bbf : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fce263a7653748e7a3df514cd48cc0d7 : SUCCESS in 19m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb8d34cca1464b4da5c08b1d5c3de0de : SUCCESS in 37m 15s","accounts_in_message":[],"_revision_number":12},{"id":"a967e90b53896a978b7f3d52da791deef01da46e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-09 18:37:15.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"22d02207287cc3af6e851b2601f0c74d3e3841c4","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-09 23:02:39.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"676975c19c8c1d0bf49d46c0dfab499c2d14b67c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-10 09:24:34.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"5e32e6755fba6f4db5d21f98e5868ae11d71e197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-10 20:25:51.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"d9e9188a72d7ea4d8fc1e7fe5fed1a108543616b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-10 20:28:11.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"74cf31d0850afa65b67d49a9230b78bf28eba1f6","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-10 20:28:25.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"536ae2f75900b15186898ec5276d7d314ad36901","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 21:24:08.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0b85db9f315434cbea24ddf61190643\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ebb62d3e3954e3797b8965aab3f6be7 : SUCCESS in 17m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b086e4530b35402495374d276f5a8333 : SUCCESS in 15m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e372a0089a3e4f72b5d5fb4130dabb4f : SUCCESS in 20m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/112e5a80f1da4f92af52a884a4b98810 : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":14},{"id":"113a7085637e1c86bce2e9a17054d60b952c9ebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 21:35:25.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fed1fb01bf3f4fd0908a9c65fb067218\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37158a1aee5b4023b1527bd9106cb145 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/81604e152dfd4e18b5cb514758d15a2c : SUCCESS in 45m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cc564bf975b94acc9d34ecd65fe574ce : SUCCESS in 29m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37bb4680427e4395b7747d7efbb4fbc6 : SUCCESS in 51m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1d199b24ccc489380a5986b1d760c8e : SUCCESS in 16m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b49ba18366f47d4823814657115d4d2 : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0e213064de74bb9bb97952febbbdf60 : SUCCESS in 16m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/97507853467948a9b17c88f6481c9da4 : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/07db12369eb140cd9384b28e2dd7dfe1 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/32efb4d3ab974157888a5d08e8638879 : SUCCESS in 20m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/282a267272684884878e7736b93e43d4 : SUCCESS in 19m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/817fdf2d506b4f4095e03126a14a1a1b : SUCCESS in 15m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f971f933ffe6404ebdc3266baba454d6 : SUCCESS in 19m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/268abec631a34f4aa08d7776fea3f396 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/42d88540bb0349509511daca55b87ccb : SUCCESS in 17m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/24591d35a3704c6f9825200fb8e4682d : SUCCESS in 20m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/be1d4d6bd72a4800b07a01cb1ced79fb : SUCCESS in 12m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dc9e3fc818164c0aa2b427696c93ad70 : SUCCESS in 14m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c33654110e9b4e71a1dedcc8b2bcea45 : SUCCESS in 15m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b3d7a712f0e440ec93b6a8ce7e7dec94 : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/049ec003391b493a88080f57491a6b97 : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/43ab207a69984dcc94c472a56eccb8ed : SUCCESS in 6m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ad853655f391480484a3e6cf47d171af : SUCCESS in 54m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fefba35aa35548f7ba7f6ed9cf11e4a7 : SUCCESS in 56m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/11c3859c05c74cb4b860a20cf526f756 : SUCCESS in 1h 01m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f0d153dea2c746049da71c6e13c60ffd : SUCCESS in 52m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c87f80fb4c8f42408a0b8c4999d4d3da : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd43a3524ab44f9a834e36c22f241ee8 : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0b86fb991b694c17b1ce731186e32d04 : SUCCESS in 17m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f58a732371c4febab4707ff6a5e387b : SUCCESS in 37m 06s","accounts_in_message":[],"_revision_number":14},{"id":"490edb02704fd8398842a2dc22a5a230d22f6c8a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-11 10:47:46.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"05620665989f2c2f75efc6fe5eed09f85a485550","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-17 16:23:57.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"bc6e8ee12dc15e245ac144d882615b0b0baef617","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-18 09:43:27.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"14f41f39401cf8164209889bd47bcfd2b7c44377","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 09:44:03.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"5381ee322cd81d82e08fdfceaf8f64a97945ec22","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-18 11:33:43.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"830a93d07387de77ced902bcab3bcd0a2db60bed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 11:52:53.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d0742b12df24f6eb6ae1ab40ec166fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b23f6381c5b44937be6cf872336dab5a : SUCCESS in 8m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/ffe2bfc1fd46455098ccefb152a655e6 : SUCCESS in 1h 06m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/26b2c53366654dd990bf5352397ff642 : SUCCESS in 52m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1425c8a4c8364b4997295cd1d9b497c0 : SUCCESS in 54m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/56805fec1d38422e8836dfb795988972 : SUCCESS in 9m 39s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/6ccee183f77848de9a7e719f2dc5c979 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3bf12b8ad94641da997a40008df77ace : SUCCESS in 15m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/08e51ec1d46d4d2bb85c0b7067464b65 : SUCCESS in 30m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/026a8c3b827f48499c16a30b394a9bd2 : SUCCESS in 13m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f6cd0f92bf954d6f87e80eb834bc3e8f : SUCCESS in 17m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5b03f7e65c56411e9a90853b303465e3 : SUCCESS in 18m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cad5ef96da74489b9e32112f26149f22 : SUCCESS in 17m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6d13b8d376204ed99dc7b9943248c7be : SUCCESS in 17m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9ab8e977c5144dffbacb1e450d5457ce : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5274aff858444d98b65e774615e1eea0 : SUCCESS in 17m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7532145a4a08491c8bfeb89d2988003d : SUCCESS in 17m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f57a8fc38b9b4a2d8337d7ff619ad227 : SUCCESS in 7m 30s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/af01d6f9d01443ddab0256810bc06004 : SUCCESS in 8m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/31180f68dc6048d4a2b2e0bb7ddeebff : SUCCESS in 59m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c0bd3054dbc4642a38490961ab53840 : SUCCESS in 58m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f528fc99814041bbbd45c4f9f3679967 : SUCCESS in 49m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/633978cab3334aebbd92e36ced9516d9 : SUCCESS in 59m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77e78e4c69b24c32914f4f85b666c79e : SUCCESS in 19m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b5d84973c9f45dba9980fcf51b74078 : SUCCESS in 4m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/757888c7fa5b44cc9551299ce4a2c7ec : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":14},{"id":"edcf8818c5b4566124512d0712938ea45ba06f07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 11:52:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"38c12e5f396669be0501d33a5dc7f40e0997e75a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 11:54:05.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57d6cf3f48e242279d1d2c31c5226f16\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5949aa1864154a449b0443358581ecf9 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3bdc738730a84ea3b2079186adb0ad6a : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0027b55d40f34d86ac3a5049af26f1fb : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"7b4642567ae2c35dd0e8d07ea628a80d04ccc820","revisions":{"4db3bd29d04ef7bc7726a464872b8a60f9999417":{"kind":"REWORK","_number":1,"created":"2023-04-17 18:13:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/76/880676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/880676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/880676/1"}}},"commit":{"parents":[{"commit":"2e89e92cb77bd5341a1a290bdfc9402b02f71ab6","subject":"Merge \"ssync: fix decoding of ts_meta when ts_data has offset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e89e92cb77bd5341a1a290bdfc9402b02f71ab6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:08:32.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:08:32.000000000","tz":60},"subject":"WIP: memcached: emit timing metrics when exceptions raised","message":"WIP: memcached: emit timing metrics when exceptions raised\n\nChange-Id: I4d4f20b92b85255ac8bf66f2c830e691e64bbe47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4db3bd29d04ef7bc7726a464872b8a60f9999417"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4db3bd29d04ef7bc7726a464872b8a60f9999417"}]},"branch":"refs/heads/master"},"8c72d27fb047f85ed08b78db705a3a0c720b2011":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-18 16:55:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/76/880676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/880676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/880676/2"}}},"commit":{"parents":[{"commit":"4b6f54d063c33e13ced7d6dd5208921680f56ca6","subject":"Merge \"ssync: Round-trip offsets in meta/ctype Timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b6f54d063c33e13ced7d6dd5208921680f56ca6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:08:32.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-18 16:54:57.000000000","tz":60},"subject":"WIP: memcached: emit timing metrics when exceptions raised","message":"WIP: memcached: emit timing metrics when exceptions raised\n\nChange-Id: I4d4f20b92b85255ac8bf66f2c830e691e64bbe47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c72d27fb047f85ed08b78db705a3a0c720b2011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c72d27fb047f85ed08b78db705a3a0c720b2011"}]},"branch":"refs/heads/master"},"ddbfb99fd0cb1fc8b046859df1c0f78c066aee4a":{"kind":"REWORK","_number":3,"created":"2023-04-20 05:49:46.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/76/880676/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/880676/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/880676/3"}}},"commit":{"parents":[{"commit":"31d6ebb44f80d4dd0f2d3e78615b094458eeadad","subject":"memcached client: log user provided keys in exception error logging.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31d6ebb44f80d4dd0f2d3e78615b094458eeadad"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:08:32.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-04-20 05:49:01.000000000","tz":-420},"subject":"Memcached: emit memcache timing metrics when exceptions raised","message":"Memcached: emit memcache timing metrics when exceptions raised\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I4d4f20b92b85255ac8bf66f2c830e691e64bbe47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddbfb99fd0cb1fc8b046859df1c0f78c066aee4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddbfb99fd0cb1fc8b046859df1c0f78c066aee4a"}]},"branch":"refs/heads/master"},"77262dec3ee9b66ab7735c22353b6f273e498fc4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-04-20 05:58:51.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/76/880676/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/880676/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/880676/4"}}},"commit":{"parents":[{"commit":"2a46355f87e9be5c5b168c55818f64f8d18aafea","subject":"memcached client: log user provided keys in exception error logging.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a46355f87e9be5c5b168c55818f64f8d18aafea"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:08:32.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-04-20 05:58:39.000000000","tz":-420},"subject":"Memcached: emit memcache timing metrics when exceptions raised","message":"Memcached: emit memcache timing metrics when exceptions raised\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I4d4f20b92b85255ac8bf66f2c830e691e64bbe47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77262dec3ee9b66ab7735c22353b6f273e498fc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77262dec3ee9b66ab7735c22353b6f273e498fc4"}]},"branch":"refs/heads/master"},"d530786159171928bda72f5eaf09953858b29a92":{"kind":"REWORK","_number":5,"created":"2023-04-21 05:54:25.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/76/880676/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/880676/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/880676/5"}}},"commit":{"parents":[{"commit":"4e5a7fbef0a9328c1fe70530bd83a6f0c8469a9c","subject":"memcached: log user provided keys in exception error logging.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e5a7fbef0a9328c1fe70530bd83a6f0c8469a9c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:08:32.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-04-21 05:54:14.000000000","tz":-420},"subject":"Memcached: emit memcache timing metrics when exceptions raised","message":"Memcached: emit memcache timing metrics when exceptions raised\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I4d4f20b92b85255ac8bf66f2c830e691e64bbe47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d530786159171928bda72f5eaf09953858b29a92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d530786159171928bda72f5eaf09953858b29a92"}]},"branch":"refs/heads/master"},"42d7b35f91c7e45d684e02680938c95ae7f5e6a5":{"kind":"REWORK","_number":6,"created":"2023-04-22 23:21:41.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/76/880676/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/880676/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/880676/6"}}},"commit":{"parents":[{"commit":"7834edf3dd99b0b588a95975ed75cbd48b3fe1ee","subject":"memcached: log user provided keys in exception error logging.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7834edf3dd99b0b588a95975ed75cbd48b3fe1ee"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:08:32.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-04-22 23:21:25.000000000","tz":-420},"subject":"Memcached: emit memcache timing metrics when exceptions raised","message":"Memcached: emit memcache timing metrics when exceptions raised\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I4d4f20b92b85255ac8bf66f2c830e691e64bbe47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42d7b35f91c7e45d684e02680938c95ae7f5e6a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42d7b35f91c7e45d684e02680938c95ae7f5e6a5"}]},"branch":"refs/heads/master"},"0801bef3db667b4e12054aaf803d7ea83a583b4a":{"kind":"REWORK","_number":7,"created":"2023-04-25 05:18:23.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/76/880676/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/880676/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/880676/7"}}},"commit":{"parents":[{"commit":"365b208bc3d685028b21c41998a120b636a86ede","subject":"memcached: log user provided keys in exception error logging.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/365b208bc3d685028b21c41998a120b636a86ede"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:08:32.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-04-25 05:18:13.000000000","tz":-420},"subject":"Memcached: emit memcache timing metrics when exceptions raised","message":"Memcached: emit memcache timing metrics when exceptions raised\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I4d4f20b92b85255ac8bf66f2c830e691e64bbe47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0801bef3db667b4e12054aaf803d7ea83a583b4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0801bef3db667b4e12054aaf803d7ea83a583b4a"}]},"branch":"refs/heads/master"},"bf1b03297f28b4c457e8d60a6847313ee6552c5a":{"kind":"REWORK","_number":8,"created":"2023-04-26 01:28:54.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/76/880676/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/880676/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/880676/8"}}},"commit":{"parents":[{"commit":"365b208bc3d685028b21c41998a120b636a86ede","subject":"memcached: log user provided keys in exception error logging.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/365b208bc3d685028b21c41998a120b636a86ede"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-04-17 18:08:32.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-04-26 01:28:28.000000000","tz":-420},"subject":"Memcached: emit memcache timing metrics when exceptions raised","message":"Memcached: emit memcache timing metrics when exceptions raised\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I4d4f20b92b85255ac8bf66f2c830e691e64bbe47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf1b03297f28b4c457e8d60a6847313ee6552c5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf1b03297f28b4c457e8d60a6847313ee6552c5a"}]},"branch":"refs/heads/master"},"bdcd5e53618610141dca2e4d6691d5544d2386f1":{"kind":"REWORK","_number":9,"created":"2023-04-26 03:36:02.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/76/880676/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/880676/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/880676/9"}}},"commit":{"parents":[{"commit":"365b208bc3d685028b21c41998a120b636a86ede","subject":"memcached: log user provided keys in exception error logging.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/365b208bc3d685028b21c41998a120b636a86ede"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-04-17 18:08:32.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-04-26 03:35:51.000000000","tz":-420},"subject":"Memcached: emit memcache timing metrics when exceptions raised","message":"Memcached: emit memcache timing metrics when exceptions raised\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I4d4f20b92b85255ac8bf66f2c830e691e64bbe47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdcd5e53618610141dca2e4d6691d5544d2386f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdcd5e53618610141dca2e4d6691d5544d2386f1"}]},"branch":"refs/heads/master"},"4fe4b3c0dabf1477f8d267ff5736d2c4dde499bd":{"kind":"REWORK","_number":10,"created":"2023-04-27 05:48:21.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/76/880676/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/880676/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/880676/10"}}},"commit":{"parents":[{"commit":"365b208bc3d685028b21c41998a120b636a86ede","subject":"memcached: log user provided keys in exception error logging.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/365b208bc3d685028b21c41998a120b636a86ede"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-04-17 18:08:32.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-04-27 05:48:01.000000000","tz":-420},"subject":"Memcached: emit memcache timing metrics when exceptions raised","message":"Memcached: emit memcache timing metrics when exceptions raised\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I4d4f20b92b85255ac8bf66f2c830e691e64bbe47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fe4b3c0dabf1477f8d267ff5736d2c4dde499bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4fe4b3c0dabf1477f8d267ff5736d2c4dde499bd"}]},"branch":"refs/heads/master"},"0b6dd8da887d23504ada332aca63f77813b63103":{"kind":"REWORK","_number":11,"created":"2023-04-27 11:19:35.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/76/880676/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/880676/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/880676/11"}}},"commit":{"parents":[{"commit":"365b208bc3d685028b21c41998a120b636a86ede","subject":"memcached: log user provided keys in exception error logging.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/365b208bc3d685028b21c41998a120b636a86ede"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-04-17 18:08:32.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-27 11:19:08.000000000","tz":60},"subject":"Memcached: emit memcache timing metrics when exceptions raised","message":"Memcached: emit memcache timing metrics when exceptions raised\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I4d4f20b92b85255ac8bf66f2c830e691e64bbe47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b6dd8da887d23504ada332aca63f77813b63103"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b6dd8da887d23504ada332aca63f77813b63103"}]},"branch":"refs/heads/master"},"4d19482b9c25ee3f4a8399210273740f884b2ce6":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2023-05-01 21:26:29.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/76/880676/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/880676/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/880676/12"}}},"commit":{"parents":[{"commit":"9fb860880d755291377c28702679c4d5deb55a6f","subject":"memcached: log user provided keys in exception error logging.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fb860880d755291377c28702679c4d5deb55a6f"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-04-17 18:08:32.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-05-01 21:26:04.000000000","tz":-420},"subject":"Memcached: emit memcache timing metrics when exceptions raised","message":"Memcached: emit memcache timing metrics when exceptions raised\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I4d4f20b92b85255ac8bf66f2c830e691e64bbe47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d19482b9c25ee3f4a8399210273740f884b2ce6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d19482b9c25ee3f4a8399210273740f884b2ce6"}]},"branch":"refs/heads/master"},"06fec1d2822f17d41efce0f14dc23a37909af0a9":{"kind":"REWORK","_number":13,"created":"2023-05-10 20:25:51.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/76/880676/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/880676/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/880676/13"}}},"commit":{"parents":[{"commit":"9fb860880d755291377c28702679c4d5deb55a6f","subject":"memcached: log user provided keys in exception error logging.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fb860880d755291377c28702679c4d5deb55a6f"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-04-17 18:08:32.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-05-10 20:25:40.000000000","tz":-420},"subject":"Memcached: emit memcache timing metrics when exceptions raised","message":"Memcached: emit memcache timing metrics when exceptions raised\n\nBelow new metrics will be added:\nmemcached.[method].timeout.timing\nmemcached.[method].conn_err.timing\nmemcached.[method].errors.timing\n\nAlso, MemcacheRing method and time spent in the function will\nbe logged when Timeout or other exceptions are raised.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I4d4f20b92b85255ac8bf66f2c830e691e64bbe47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06fec1d2822f17d41efce0f14dc23a37909af0a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06fec1d2822f17d41efce0f14dc23a37909af0a9"}]},"branch":"refs/heads/master"},"7b4642567ae2c35dd0e8d07ea628a80d04ccc820":{"kind":"REWORK","_number":14,"created":"2023-05-10 20:28:11.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/76/880676/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/880676/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/880676/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/880676/14"}}},"commit":{"parents":[{"commit":"9fb860880d755291377c28702679c4d5deb55a6f","subject":"memcached: log user provided keys in exception error logging.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9fb860880d755291377c28702679c4d5deb55a6f"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-04-17 18:08:32.000000000","tz":60},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-05-10 20:27:58.000000000","tz":-420},"subject":"Memcached: emit memcache timing metrics when exceptions raised","message":"Memcached: emit memcache timing metrics when exceptions raised\n\nBelow new metrics will be added:\nmemcached.[method].timeout.timing\nmemcached.[method].conn_err.timing\nmemcached.[method].errors.timing\n\nAlso, MemcacheRing method and time spent in the function will\nbe logged when Timeout or other exceptions are raised.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I4d4f20b92b85255ac8bf66f2c830e691e64bbe47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b4642567ae2c35dd0e8d07ea628a80d04ccc820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b4642567ae2c35dd0e8d07ea628a80d04ccc820"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
