)]}'
{"id":"openstack%2Fswift~880677","triplet_id":"openstack%2Fswift~master~I3086e3453b13411304e431e83d6379e255c1777c","project":"openstack/swift","branch":"master","topic":"p-memcache-incr","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-05-31 02:09:04.000000000","reason":"Change was submitted"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-05-31 02:09:04.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-05-31 02:09:04.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-05-26 21:08:29.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I3086e3453b13411304e431e83d6379e255c1777c","subject":"memcached: handle failed incr+add+incr better","status":"MERGED","created":"2023-04-17 18:13:42.000000000","updated":"2023-05-31 02:13:55.000000000","submitted":"2023-05-31 02:09:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"880677-p-memcache-incr","meta_rev_id":"84ceeced6f3d509a0cb008b315971def2950b521","_number":880677,"virtual_id_number":880677,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-31 02:09:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-26 21:08:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-30 16:22:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-17 19:14:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-17 19:30:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-18 02:05:39.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2023-05-23 07:15:10.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2023-05-24 05:58:01.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2023-05-24 15:01:34.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-05-26 21:08:29.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"61996b10f85dfc670405d18375cf6dc85704e91e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-17 18:13:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff553ad57b074d7abae1e9c0ce8313107e38f06b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-17 18:14:07.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"f5a12ce483b9667d93f8aa962084f07bbc39b6bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 19:14:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc2d2b813f0c4359b13d5627176408e4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3527ddc7db1c4a6ca571681fdea22b4a : SUCCESS in 16m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d6060d898a7a4bdd955c426133db517e : SUCCESS in 53m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7586369fe82a4313aaf48f57ec272778 : SUCCESS in 21m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b45f2390c62c47d1b0fd0e90b066b0bf : SUCCESS in 58m 11s","accounts_in_message":[],"_revision_number":1},{"id":"8ec0caa8f80dc096ef0cd920acc7db1ebab88e04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 19:30:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35ceb074c5a847cea12bc1622d962fff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29c825666a4845fd965e6b294048b2b0 : SUCCESS in 7m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/700887a7ef5e46a7945a39604da02683 : SUCCESS in 1h 08m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0d705f9d5b854701b9b43ee5989888d6 : SUCCESS in 33m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7f5bc72310d42f1b702a12c6c23dbdc : SUCCESS in 1h 06m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3511ae0a0e884e70a50db0e6ea8c46bf : SUCCESS in 13m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/126bc7617115437aa1b3c7cae435060d : SUCCESS in 14m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed00e3c31ad04f8b83a9522e757440c8 : SUCCESS in 28m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/979494eda5d147dcaf376b79aaa2b9a3 : SUCCESS in 8m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/91f60bf7bf344e879279d940f86d2430 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/607ce2968e26420283548f55ed52e977 : SUCCESS in 14m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/27566eeca69e4446bfd4614fa909ae99 : SUCCESS in 15m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/129e62d72a0c49f89dbff185293dab9b : SUCCESS in 16m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4adce8e1c4e243b891ab6773dc366210 : SUCCESS in 12m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bc54bd74464a48d8a6f7a1fe08d3dcd6 : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bd8e85cb65874ed1bba8bdd2b7aca770 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cee5c430385d4502bae22d64b2b2a0b2 : SUCCESS in 15m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/828ef32ab1064c8d82a0a1d4636eb9d2 : SUCCESS in 13m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c486d8b0445f4132aa650642337400a9 : SUCCESS in 15m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f9ef4b3ba4a347d4929e7cb1ce04ea79 : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d3e750f483374af6b3e820b5d6963618 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8591f396b680462b93b8e4ae7a27417a : SUCCESS in 14m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4c2fe4f4ca9d42db835f5523fa0a46d6 : SUCCESS in 6m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e61c4151dbe84e1cbdc3dd6d141e953d : SUCCESS in 1h 06m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/84d46c1c5b894e5dbf5fbae019586f8b : SUCCESS in 58m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5a59c1fef654b38b20a3bf7dde00f3b : SUCCESS in 56m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/03b69e3a2f1b42f0b2ab44292e622fcb : SUCCESS in 37m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6382d6e9ec847c1acf77d4727ddd13e : SUCCESS in 14m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fdc24a440894166a9c0875e67f67a82 : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d81a6413501f42beb5673631e2e3cdfd : SUCCESS in 19m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a615456acb504304afa3495dd19fd118 : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":1},{"id":"29af0a7f67073aafcda6d9331dd86a8339e788b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-09 18:33:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4f6899a87a0db5312d59e779a6fcbd0fe64ed1f1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-09 18:35:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2499677b17f2b764a24f2fd849c3467e1c9bac6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 18:55:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73af293aff3b4db0a823edf97b5f982a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99fae11279274023bc833439e2147282 : SUCCESS in 16m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15a2893d7b1d47f58a36d5f8689a1fdc : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/21fcd43ccd7942b298f9548096cb761d : SUCCESS in 19m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e59853d3c2d4cac8d1320c131ee6b35 : SUCCESS in 20m 30s","accounts_in_message":[],"_revision_number":2},{"id":"4c97a3b3f8a41c9ea9aa391f00045d26cf3c92c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 19:43:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3737915366b547e1a11b83710d2475de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24b0ba5e859240c3abe948d0f3a56634 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/c067b32a2c9b4a3383f9f41086064811 : SUCCESS in 39m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/14429ac7c35b404da872c436f79395af : SUCCESS in 52m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e697d58ddbde4cdf8bf2647a0e8b1a9b : SUCCESS in 42m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c0a328bf3acc433486b8f11ed056bef0 : SUCCESS in 17m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/10748986ee2342d090e139966063fda1 : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09254ce2b27b437da65c8015ceb143b4 : SUCCESS in 18m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/119d0246337d4acd8b6b97a99a480eb2 : SUCCESS in 7m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/97cd615182d14b17bd1c9c76487869c2 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e9a2acef0e648b6a7b9c6a9e73d1652 : SUCCESS in 17m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bfe6a635386949dcbe8187361aba16a9 : SUCCESS in 13m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c22722eb2111415e94eace068c9ddc96 : SUCCESS in 16m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ed2527a6f33f4a048e095094a3f75397 : SUCCESS in 14m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/039de0e023b144809def0f073636b387 : SUCCESS in 17m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/99e134f85fea4bf2a7d47bff7e0db8fe : SUCCESS in 18m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/812f8212fc1544d281ef0c3c57022fc2 : SUCCESS in 17m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/06dc7739d3ef414296096d37222ee96c : SUCCESS in 14m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ae5cfa8eff974afc91a0f35b42921a33 : SUCCESS in 11m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/10e81cf29a5c43389e3cfbc152c44374 : SUCCESS in 14m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d42daffdd6d94bdeadfe3650c5135208 : SUCCESS in 6m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/712dba59e3fd48e4890e2312dc09ca5c : SUCCESS in 13m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e83e866b9b824a2c96954f5dbd4b93e4 : SUCCESS in 6m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1bb069458e694703bd5c771a8e161eb1 : FAILURE in 1h 00m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/22906e7fc36c4b908a9f8ff465e26e46 : SUCCESS in 1h 04m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/056809d580064e559c120425eb026dc6 : SUCCESS in 57m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/81373978909b4991aacf08598982edba : SUCCESS in 41m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be10417a090e49d7973a592129db4024 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58095124657b4040a153dedadcc2c002 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/553aee86e48843579f3c660a12a05801 : SUCCESS in 22m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9822595fb0424687be101ea687840ec1 : SUCCESS in 36m 21s","accounts_in_message":[],"_revision_number":2},{"id":"8b1a0460794481286aadb82d0100c06af4180d9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-10 09:53:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"55e1f8564d7755c4fed653479dde120decf873ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 10:14:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7febcc8babb64635b2a9ba529258799b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c46e25308448483eb96e2a364567c42b : SUCCESS in 15m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/343e71e9e64242afa6a270ce65c635de : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/05471f43beb24aefb2238aa6832a0ec2 : SUCCESS in 18m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/733e0021bced492485603c0c10e938d3 : SUCCESS in 20m 13s","accounts_in_message":[],"_revision_number":3},{"id":"f12da81ab2b8ac2a26d2d9853ed5582411b6fb62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 11:05:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/114db57e58944713ae455baae7cde988\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/402bffc0216d4264a92babafa4eb1c9e : SUCCESS in 8m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/d369595515ef4f1a90c882f92328fc22 : SUCCESS in 59m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/44e367d0b9634c878951b22076f01a9e : SUCCESS in 45m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b1f14b3774b4d2997aa36fded65a249 : SUCCESS in 1h 00m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a57528d1f24749d39ee60c0e7deb0fab : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b46b1ce7bb6342afae1159974090d67f : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae6315da40744b7e9ea9983497ea8d9c : SUCCESS in 18m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5d3865fc04434846bf23b79d3eed2288 : SUCCESS in 9m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4b44404841a443f59145a642d8b3199c : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/39c992c07d5547df97b11de683e78cc5 : SUCCESS in 15m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d05f9f0acf6d45a28aff7f2d833b5381 : SUCCESS in 13m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7d1e64fb09d4401898c1cc44c43eec30 : SUCCESS in 14m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ba76203cbf484bb89f2afa070302ceb1 : SUCCESS in 16m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/58960465e6d3460e8b5a97f76de4997f : SUCCESS in 17m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dbb8990f123c4c068541433d388e7c29 : SUCCESS in 19m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/072c0669c861464aa1055c3b5cc7c2c6 : SUCCESS in 18m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0d1b8de6b0124835b54619c523b5b888 : SUCCESS in 13m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/aa730a3867c34075aad97bba79b6ad0c : SUCCESS in 16m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a9d172169991438684b1b2b733ab486e : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/10106985818645de9b9ba557da22faf3 : SUCCESS in 7m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/66ae32976e684440bffe7e212dcce4d6 : SUCCESS in 16m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/464cacb646414affb42a46aed57941c8 : SUCCESS in 7m 07s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c76ef811cf1b411fa214f0a5c2fb2bb3 : SUCCESS in 50m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e8d5a66a1f924f2e890015e9e88c824c : SUCCESS in 1h 02m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6ed111cb785a4c26b3f988913418b966 : SUCCESS in 49m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/33c03c4ffb144ad59af22aded36c169e : SUCCESS in 53m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40e6f700f1fd41b79d3795488b9d0a68 : SUCCESS in 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b9c11589c6f4d6cb9a551fda69fe21a : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a8b414512676480b999cd829e36cf600 : SUCCESS in 17m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bee73ed8492641e5a6fc2ef57d10b5e5 : SUCCESS in 52m 32s","accounts_in_message":[],"_revision_number":3},{"id":"411f4bb6f7dd958e8dc081be23aef503680e8bcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-11 10:50:55.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"5e1dca32c07cb014ed3f0db0e2f99684b18d36bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 11:58:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a18cf6c0fe2640a7a70f4ff61036a6a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/590f47683d0c4c6b80c3d1f192d997f3 : SUCCESS in 10m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/7b97edbe51d243879b67dc53cf1629e8 : SUCCESS in 49m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/12ed9f48ede6452199ea169314fefcf8 : SUCCESS in 49m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85cdd5703481486c838237f2f1536786 : SUCCESS in 1h 00m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b8f7c733cbe14b31a0db4f27a08c9630 : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0db6792f2dfd4621a67bc244ffca2cd1 : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1718c6d331542758ed740890d7e6d54 : SUCCESS in 18m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a24ef222eab1465891a94486bb0f0ddb : SUCCESS in 8m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d3ee6dba8a644584983d6d9059ae8198 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0e60a4eb0cc54d949502c231585bedea : FAILURE in 31m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f0eb0d64bf6e475ab85f128b43dea682 : FAILURE in 33m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/29d4158127db4c28acde1ed5a9e7e6d9 : FAILURE in 12m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2d8b0d586cb94b06abead0201b93a893 : FAILURE in 29m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bf58e3633f30498f87b2ae29665a026f : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b0256447f6474450b3b1158eda8201f8 : SUCCESS in 16m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/76ce9b15c95a48c1ab92c3ae4175e7de : SUCCESS in 20m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/db759518f2b14b2d9eefb7cc32ba66b1 : SUCCESS in 13m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9789242f1a26443fb988862553c1b4ee : SUCCESS in 18m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a0ff14b8f7bc4829bc528086b4e2dc22 : SUCCESS in 13m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2c96213ec56146658753125a9839ea9f : SUCCESS in 7m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9610beadb3854a9f89d8c4dcd9f797cc : SUCCESS in 16m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2728f766ee5646a6b328164c2183e7c8 : SUCCESS in 6m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f872b3095e9348379057c990dbd3137f : SUCCESS in 48m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f919ec56f71c4b30a17368f08991d4f5 : SUCCESS in 1h 01m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/003b53ae1fe7438d8a64da9af753491d : SUCCESS in 40m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/41c383aefaa4407c89f4de78917cce49 : SUCCESS in 51m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/150739133f5a4cc5b5b3afd908930471 : FAILURE in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61ff2b26fd084544b7f179edd72ee8f5 : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ceffaac6a77b4af6bf753871b8a9c4c4 : SUCCESS in 25m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42923fef51514a54b5c6984ef06082a8 : SUCCESS in 56m 07s","accounts_in_message":[],"_revision_number":4},{"id":"25c2c7cbad69cd029e609d02613b838e58f8e97a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 12:13:58.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74eb6824dad8419db536d54e18e378f1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01c5a7cd448d4693b31cc93a5f589277 : FAILURE in 1h 16m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08a9f61975404f5791811c011b89ba13 : FAILURE in 12m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b64ca6f72e845f589f4c487672a706f : SUCCESS in 1h 20m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e77475abfde6440f87f2b2c6bfc168ac : SUCCESS in 1h 13m 21s","accounts_in_message":[],"_revision_number":4},{"id":"db8ac484104f29a97e86878c46779651a443c924","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-11 13:00:10.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"bc9c8f8de40b03de553af2ae2bf69e015d33ad05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 14:13:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22a16c71331c4c6c98d118d4a334bacf\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6d6c26e28e84d9e86c652bca9752315 : SUCCESS in 15m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e967352e2c94899997395595f7e73e3 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0031f49132eb468a85f47dc4506063cb : SUCCESS in 1h 11m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8b5ba9f905b4b35b87dd7afe46fb1a1 : SUCCESS in 19m 03s","accounts_in_message":[],"_revision_number":5},{"id":"fcbe8303b9cd34bdf8d293d0d2d7ae8798ec7890","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 14:24:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8689dc42c2744bf58d69b71ba887ecca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feb3ed2805c8467a803a068d90545ca7 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/e002c37b247b4018b6c030082ae51bb4 : SUCCESS in 58m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/294cf114ef494f4cb49c14e73ca2e5a1 : SUCCESS in 48m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e05ef5c5598c4e41bde778d34d9b2496 : SUCCESS in 59m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32ebf5395ef84f31b5c7af51546633de : SUCCESS in 16m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13a16e1bc7624fffbcafae48e5a19533 : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5dcb537a8aa48ff8d0a683867e4b903 : SUCCESS in 19m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/851e4ce42f794066883a3f7ef67ac235 : SUCCESS in 10m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a64ca673592c4437a513ee2e640e5502 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5ad2cd838af94205ad97478608b565f3 : SUCCESS in 18m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4bb7122535e6424984191fd736ca10fc : SUCCESS in 16m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/67e0f32310074e728e413e1555a8e5ac : SUCCESS in 15m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b4be1f3b3dd744a59eedfe687652c304 : SUCCESS in 15m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bc65fd8e6a014a9faa656dd68811434a : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/107531921cf74058b103fa4a23834dfb : SUCCESS in 23m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/71e1a0ceff3349fda22407e939580bff : SUCCESS in 16m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6d4483e1bc2c49ed9f8bff3cc889559f : SUCCESS in 13m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fdebc33e4cba4a5588543758e1728e2a : SUCCESS in 15m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4465bf79856141519db5c3bcbc598998 : SUCCESS in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/949d54c6a9324d0f87555c3dbbed04a7 : SUCCESS in 7m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f384ebfca62b4cb3b9bcbfde2db167db : FAILURE in 6m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/31cc365577924c0db1b51181afd6711a : SUCCESS in 7m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f7c582bfd0954fb18d2c2c18939c71f7 : SUCCESS in 58m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c09d33ebb514b26a2a0e3967078a8d3 : FAILURE in 57m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/804095d1f0a047d69a4f89684a2caeac : SUCCESS in 47m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f68c8339875c4ef3a032647b75f77079 : SUCCESS in 42m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a3c891f5c5c409eabacd38101434114 : SUCCESS in 17m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3625e9e47a04e70a8a707b1c17fcefe : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/63843629aa26456692cc6e001d345e46 : SUCCESS in 23m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/306e19a6368e41569c02ecb285fbdf8e : SUCCESS in 51m 56s","accounts_in_message":[],"_revision_number":5},{"id":"52311405499f6d864aec8242a4f9252b5fa1a4da","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-17 16:27:54.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fb8ddd25ad93dd599e03fd87430189bae7c6819b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 17:22:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5938d63a0ac841c080b5d73632e6384f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25638bb61d5340ce86cf7e659528598c : SUCCESS in 17m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0cd1bc72e0de4434b1664f5a65cc1e90 : SUCCESS in 53m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2c21c939f7b416da74690078628c1fe : SUCCESS in 21m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0c1826d17cb42a5a5a7b07732b6d133 : SUCCESS in 23m 35s","accounts_in_message":[],"_revision_number":5},{"id":"723b91aa0b771112567b365a03d1c48052c054b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 17:40:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44daf9b6ade941149ba722a11df6d9ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/249a0cf106534ee38ae3fd143b783b6b : SUCCESS in 10m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/725cee93a97543b09d0e3cb1e8182143 : SUCCESS in 51m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c38fc9e688594ff7abed4e1e807cc156 : SUCCESS in 54m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00adaee5eb634ef993433fd10a58e579 : SUCCESS in 48m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f1bccbd223c41bfa5cdccedf4fc77af : SUCCESS in 20m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f8cf22ef7cb94478980f802fe281cfff : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f97c3a3e1ae453a8b78d0c9b4dfa21b : SUCCESS in 21m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ee9723b9d232450591fac8ba44e9c759 : SUCCESS in 8m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0b0a3750802b426e99f9c6bca2eeca8a : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1ce75751e8de435da74d09ec70fec999 : SUCCESS in 14m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bf61c0734c9d409bb903dd6add29162a : SUCCESS in 22m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a0af4d7133c44b2b805a83b867ad7461 : SUCCESS in 12m 53s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/83c4e9a06ca24541a633b60062806a06 : SUCCESS in 14m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/76525c12f3884aaaa1ef9767ace8b0d4 : SUCCESS in 21m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1dd664c575804ef39a662ab674b1fc61 : SUCCESS in 17m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5bd91753516b44778caa1fadf59b2177 : SUCCESS in 16m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/26c5cec06fe14fefb7a787ea2f62c538 : SUCCESS in 13m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/92c770ae9cbc4c908b143dbeaed59f92 : SUCCESS in 16m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a08d28688dce447ba66abe464cd2a6b2 : SUCCESS in 14m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d36ac0d785914484835675451d9781ff : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1180a3bd765c4ac88c6cfc755bf2e637 : SUCCESS in 14m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/23ac90c15de84ee8aa5a96615ac9726c : SUCCESS in 8m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0aebe8c311424c378b887d5a51c726f3 : SUCCESS in 1h 04m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7d8e615d2eb142c0b145913fdd7727b6 : SUCCESS in 1h 02m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/efacabf9895b436a94fcc4a038c19917 : SUCCESS in 55m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e400bd39845d4b3c8061a44adeff13c8 : SUCCESS in 43m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/774bb7df82f34e15952310726834cefd : SUCCESS in 21m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eda9747bfb940dd9597f7dcf67b05c5 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fe21229203844c62af99869b198393fe : SUCCESS in 19m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f939c305af6f418ba11513948b6b8bc6 : SUCCESS in 49m 30s","accounts_in_message":[],"_revision_number":5},{"id":"f7c078fd9dbf4a0189462824ccd54386bebd2cda","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-18 02:05:39.000000000","message":"Patch Set 5: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"eaf7e211bb7a97a5bc512eb6550fcdcee036795c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-19 10:11:52.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"02e271acc30602b6e8e16c9b3741625bbc6db2d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 10:28:38.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3ff5025706a460f9d3022f55e69cdfa\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c54c1daf297c49109c44f226d8474d68 : SUCCESS in 15m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/77cbfec3bf644744aa7d5823c9459f44 : NODE_FAILURE Node request 300-0021255915 failed in 0s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc2923d990044bcdb9d2317dd47f4b46 : NODE_FAILURE Node request 300-0021255916 failed in 0s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/636cb63ba3ad41fe82918da770fa8415 : NODE_FAILURE Node request 300-0021255917 failed in 0s","accounts_in_message":[],"_revision_number":6},{"id":"a20ed13699e3597c89ef5731194174baab803b7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 11:28:41.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b07760fadcf421db8c882ac961e656a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb6ba8968b4461b9c193fbf8ae9dbb6 : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/d6ca4633dbc645509b677834c50b6115 : SUCCESS in 1h 08m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0515745c707e4abfb199937f106eb307 : SUCCESS in 45m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11d5a8c8c834433faa23239ee7e8cba6 : SUCCESS in 1h 07m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07492cccc3224855827326024693383b : SUCCESS in 27m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b96b699f47b4b88bd31b1ae521137d9 : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d278eaa609e4f20be9be8f48b416f24 : SUCCESS in 17m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/859fd53a3661428b9434d7b11e826c07 : SUCCESS in 9m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b57990ed22354dc7a6e808bab37d6b3e : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/525569e0f1ff40649e691d4b3038111b : SUCCESS in 20m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dba70a36c8eb4aae8366fe05c1c4f945 : SUCCESS in 17m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5c457ca8ae1f44b5b36aa1b3fff0047f : SUCCESS in 28m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/97aa7999e53b4760962627efd2969795 : SUCCESS in 25m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ed42919086134dcaa2341f37a02f2660 : SUCCESS in 28m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/94ee990640ea481a8454f48f21944e15 : SUCCESS in 26m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3bc7c0c9e13247299868f1baefa091f0 : SUCCESS in 18m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/69b7e76e9b324909a2bb3280d6595b18 : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7e735b55719e46cfa544829f8eb7eb20 : SUCCESS in 16m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/40c62e17bd23462993dcd55dc9ad0c74 : SUCCESS in 16m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6fecdfcb61644d85932e1b52b707e293 : SUCCESS in 6m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e2f92b948c1b48cbbf11db73596a9dc3 : SUCCESS in 15m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6d478c8827ca49fa94aff6ff3daf4fbf : SUCCESS in 7m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/85331f05645347408457e68fd18a5ba8 : SUCCESS in 49m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6009700450242d4b5bb3c143b569470 : FAILURE in 1h 03m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3002778383264729b99b4b2265067ef8 : SUCCESS in 38m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c07f954c85f24f4fb67344e1e8ca260c : SUCCESS in 45m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bf61752164e42d1a9382c8fc0425367 : SUCCESS in 16m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56687331b39e496c98c2ecd902cf145f : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a0ee2ea156944bc3930ef1b19d977bf2 : SUCCESS in 25m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b78c30f7e6f4e8cb1f47118705f8d37 : SUCCESS in 54m 11s","accounts_in_message":[],"_revision_number":6},{"id":"d56a6e4487c7f85042a364a4b1570f7371b17602","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-19 13:59:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8732f0e332d0c127f422cb41b4e32ea75c6f3143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 15:21:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0b20d258255494b84d82a120112320e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbd6b5fb962448079ff738330913e7d4 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/271e4132b80f485ebb994d14513145d2 : SUCCESS in 58m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/98c5514f4f2a46bba6dfceeed969de60 : SUCCESS in 42m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/833d3b2975b94ec2a966c1cb9af31dc4 : SUCCESS in 55m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bcff34595ae343d8bb900c4dfb1e9cc6 : SUCCESS in 16m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c121f7db996c458ab9a9d2086bcbc572 : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ada37daa955242d483f6ff63887465c7 : SUCCESS in 18m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1a43110db8264a40b8ca6890a37deaf0 : SUCCESS in 9m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4251a087de6e4f23aa01b4a8828de3ba : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/90adfbd4a7bd4e89a3f5a24163aaad23 : SUCCESS in 18m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3cbf5adcc54e41a4a09fb88ffb01b69f : SUCCESS in 15m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bf538056bef6437493180535fc67d023 : SUCCESS in 16m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b7b8e8eb240b49f089b5ea9599adb48a : SUCCESS in 16m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ed41bb2f07aa4f9ea33786bb08ed0acb : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/95f94526b58041e98d9f363c8f625053 : SUCCESS in 20m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/97a97ab6150d4f4c9e9834ec42ec3b1c : SUCCESS in 15m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3d62335637ec45be9926696dfdde4d68 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/83271ba7f20248898fb7c1f206cf8234 : SUCCESS in 13m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/aa9289d5f5154d119e2854eada2e2564 : SUCCESS in 19m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ce0732b6ff294101aca4a11f0c3983b3 : SUCCESS in 8m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8aae9f5d9f3b4085b1d3a8f8a914f947 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fc307867b53f4852a34f79c55dca1842 : SUCCESS in 8m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/85d0131f8db24cdca797bc0b0d524ce5 : SUCCESS in 51m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/122f6765decc4c5bbc2d86c2f4bcf8f5 : SUCCESS in 1h 16m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7527e1b155ff403e87970d611eaf5844 : SUCCESS in 45m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/54bd2dfad1a14fb9b4ac1677100b6c2c : SUCCESS in 1h 00m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fcaca3e74d445588968883e7dcfce18 : SUCCESS in 22m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c934dc4baaa34168a3a2ae2bba2873fa : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/feb633717b0c48a78ef1134e1d13dd10 : SUCCESS in 27m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fac14699fc04184ac5f76dde28e95fc : SUCCESS in 42m 58s","accounts_in_message":[],"_revision_number":6},{"id":"078817f0b8832dc63b7b9199db0c55498a812048","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 15:33:16.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cb85bd5ccb54e8b9278601afcefc4ef\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b297033b1d44b78ab599f7a47634487 : SUCCESS in 15m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/206109f0d4574fd49fd30692160e3436 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/619096dc31944758b90481fa4da52fb3 : SUCCESS in 21m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2bbae8340414d158662d2b7d3b08625 : TIMED_OUT in 1h 31m 09s","accounts_in_message":[],"_revision_number":6},{"id":"8337cc8bfdb9ae1d408ec90ea111460ef8ec4bd9","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-22 14:00:32.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4db4b463f72d23de8b6cd51b2cacdfb8b03fa5f5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-23 07:15:10.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"3ed314f6e1b0037ce5801f33bdcaa6fb4448313f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-23 13:06:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e768587ea6d257c10682302c710e1add6a0d5a5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-23 13:09:00.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"5fb9b6af5624bd7790c77cad9e7824a9f8d5a9fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 14:16:28.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbe7035448ab4cffbbe029679a7225ee\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0956f7f299204bfdbaee3cb710d66edb : SUCCESS in 43m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1f89dbeada2428cab965a89ce309c25 : SUCCESS in 1h 05m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fa4438fde554f3e87c3040b3dab59b5 : SUCCESS in 19m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f220589904274c11b4994419dbc381eb : SUCCESS in 22m 53s","accounts_in_message":[],"_revision_number":7},{"id":"0905c7542ac0ded40349c5931a96ed5f97ca05f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 14:33:01.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba3538e5b2844684b368fb088002c799\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0d494747ad743528f018d765777230c : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/db70aed570c24705825df4cd93d69313 : SUCCESS in 55m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/31760e86e25948748f2bdd50fb3e07db : SUCCESS in 50m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26c3328d4b0443ca9767a59efbe84b09 : SUCCESS in 56m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93c609e9520b4cb59ddf819419f1c7c9 : SUCCESS in 28m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cae838b4aff3435ca5e818e9b7be32b5 : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32c9f3ba1e314e43be61f1d23af41c20 : SUCCESS in 17m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2e5f220f9d1441c78f08c54738c90c3e : SUCCESS in 8m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/65659a2bd7b24ee4bfb1e4bc4abc7cb8 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9af47905f64841b0b5032f1290b9c0a6 : SUCCESS in 17m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5763b75817124c6581896650686cb761 : SUCCESS in 13m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fd6a633f7ec24f7a8fae11f9b80d2432 : SUCCESS in 15m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/abce7807211148a1829b24955e986eac : SUCCESS in 12m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8bcc9a730c1f4b5b814c64231c3c739f : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2993084a177d4b98afaea2ea5322f2e5 : SUCCESS in 17m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5d94880c275b490992b8f9ef474790a4 : SUCCESS in 16m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a0f96d6b305d40529d8ffb19485be02e : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/afbe02fd04084cbaaffdf54979b4044e : SUCCESS in 13m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d3ceb5215f914fdaa111689c50142b1f : SUCCESS in 15m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0147c6f8458543dab241c0f7110f8dbc : SUCCESS in 6m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/55c18e59e5194f93a250504666e8e331 : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b3c17cb17a834376867ca2a3d57c96ce : SUCCESS in 6m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8d78bbad0ac643edbe411d1797450c04 : SUCCESS in 1h 03m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1336b62b06cb498984e10fe9990dabdf : SUCCESS in 1h 04m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/97e2968da8d046eba03abb64922ab3ea : SUCCESS in 56m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/42d8f510ae9040b2b1eac3b414273bbe : SUCCESS in 45m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcc10e02e140414090a5ddccef89ac37 : SUCCESS in 17m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3763b130011344868950138676e150d3 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2dc26dba35864cb58365bc915e3acf48 : SUCCESS in 18m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/405568728f8944de8fc9f6fd94b067d4 : SUCCESS in 50m 54s","accounts_in_message":[],"_revision_number":7},{"id":"87162a4dec4ff0830302c2db4f23a30ea9f37f9f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-24 05:58:01.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5762b569d38823e0ec10b143a5ecac5788d36238","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-24 15:01:34.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b652ec5068ab1c773f28f736c91671ae5ec43729","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-25 11:53:04.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"7446428c9465e5cc178e47d49cbed9d922f087da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-25 11:54:50.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"227af7e4d14c33dbdc6d452c5c4cfb1fc0b10e87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 13:10:03.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4f595866a114828982b5e971a581b03\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8abfba192ee345019df73ef5b7c88142 : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/4b4e029fe70542fca0080e08eb92edcf : SUCCESS in 49m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0ec89f0260e14c2faa7310fd70c54615 : SUCCESS in 56m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3deaf36489fb458193c1886df9d14f0f : SUCCESS in 1h 07m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed489bacf4894a8b81749b9fd4f0ac33 : SUCCESS in 18m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c5dce102fd84d80b7626e3a09e4e4fb : SUCCESS in 17m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d71368170b54a858ac933028b6cb6b5 : SUCCESS in 19m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5519608672e2426bab0e26e3609c7f7b : SUCCESS in 8m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3952c22b50b9407bb6d9f7998ca43543 : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1205c659ebef4d19a16fed7219df475a : FAILURE in 18m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b652a53ec93f4d60a931d7f651eced90 : SUCCESS in 13m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0a55ee5598214a079166743233560293 : SUCCESS in 13m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b23733baf37a41e5b37e842646a47ab4 : SUCCESS in 15m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/95b1db97ef974971a7c4a7073d6d3951 : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ee24a833539b4d039ff11728893543a1 : SUCCESS in 17m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9bc687d482a943a98860840ec3eaa3b9 : SUCCESS in 17m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5a838271c7c641c9864591e35ebf9e5f : SUCCESS in 18m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4b8201f231b2443184c544394e6f06e8 : SUCCESS in 15m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0148389134bf45be8a77305021a6d1ec : SUCCESS in 17m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/00d7f5f7c4e9485ba6c72a91f3da0190 : SUCCESS in 7m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/84b23e85897b410ca24bf4257d122eed : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/53ee9742246a4263a545282f7e39df65 : SUCCESS in 8m 09s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/62954ad230ce42438d1d5de56024c74e : SUCCESS in 55m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0fac2b1466b54165b3911d211191ee0d : SUCCESS in 54m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/989bc8534d944ede84f1cda1142fe56f : SUCCESS in 47m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2c034aa84a454d6f8210847a0d0e1218 : SUCCESS in 48m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47cb9e3fa69049e5ae714e01833be552 : FAILURE in 17m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6283e4d399194874bbc58ff48c3efe63 : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f0aab714d4f4c13b21f02b7482fdec5 : SUCCESS in 24m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d0650e0ae9e4e96b8412201ff260170 : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":8},{"id":"08370ed03b512368d9110520696ef48b39313704","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 13:27:11.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83418223d55e4ed8b11890f990ddf7f6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/791b477ed70440a9a25510f8be935501 : SUCCESS in 1h 29m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be4332e273bd46479360f1bd5406fd52 : SUCCESS in 20m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/405b5d8b310143689a584e7882d4f925 : SUCCESS in 1h 25m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a5aa5934e14483d9d1d6a0c4317565c : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":8},{"id":"526a2d1709e2e51b964801f604866ddd759375c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-25 15:20:58.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"5412403411d294bd963c93426168dee27e650f33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 15:46:09.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc4a570b29ba4545893c20f6f13dbd37\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/247df214807d48f2a7d5e0a137ae54f2 : SUCCESS in 18m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb37ac9d9dd345fcba6b4078fbba54f3 : SUCCESS in 18m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/60e8b92799154dd0926ef9621a222050 : SUCCESS in 20m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/76cbec7884e145948a79d40479f264c2 : SUCCESS in 20m 29s","accounts_in_message":[],"_revision_number":9},{"id":"9823273b4a5e3239b63b829b1ff49eba1b68a660","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-25 15:50:10.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"de28417bb86c6d30bd2b18c30d072c49873fe378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 17:20:18.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f6bfd1ffdf648eb916b2537deebfb02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c451945aa5854a8aa5621629c0f07025 : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/3ebb2ddd87f84d82ac0fab319da3bd12 : SUCCESS in 1h 06m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fd1b87bb435a43268386ea4f2363a398 : SUCCESS in 57m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ffa0104eb4649cc9228e33b33757ca6 : SUCCESS in 1h 05m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85bbff346a894aaf8fdc284219eb503f : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4bbcd3fbdc864d73bbd071a91f0106e6 : SUCCESS in 20m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb42634b83b248eba254c3f32bc3b811 : SUCCESS in 19m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6d05bc7354e54d31af63ebf11916d89a : SUCCESS in 8m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c5897fe50f4f4467aa8395f85709bb03 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5da361f92a1145d280b194c0f872f766 : SUCCESS in 18m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9d1eb06dee3c4f01af8b9373e579ec00 : SUCCESS in 16m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ea560b504a5342c793224535969af18d : SUCCESS in 14m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/42b0ef3a0bca4b7d8e10f7f655745e24 : SUCCESS in 14m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/62fdb781e79e42ab9712a6ae42f1f6c5 : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f212597e96bb400aaabd261024426776 : SUCCESS in 17m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6dfd39dac5594be59528cac62619d300 : SUCCESS in 18m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/01a2156c3b324fadae8c710670aa1914 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/88cd84162cdc4c2597ffc4c3b405026d : SUCCESS in 14m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c520ddabaebf4d889bf5140070591794 : SUCCESS in 15m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0db8b4a8d40b4095ad3b2a7867542873 : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0da108a9273c43bc8f11c5208357b41e : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/33d65f8c962f437895d4e0c06404d232 : SUCCESS in 8m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ea22a89f6cc84675a30295a233b93bba : SUCCESS in 1h 00m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cbf588585e2b41d3b8dd9b963366b708 : SUCCESS in 1h 01m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a9409b08a7b34dd4bb0b6f2cf025ee09 : SUCCESS in 50m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9ff20be14eaf4190a7143159b223b647 : SUCCESS in 53m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb833c6d7cc64512be5f32d12e362ef9 : SUCCESS in 18m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a7f634c5b21404086ba1e37ef9690a2 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf00234d6ff74773b8cdd7d62426c4db : SUCCESS in 20m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5368914e08fc435481711b49396624c2 : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":9},{"id":"0b8c8decce18fb2b00d1cf1b1c7edebfd03c3bfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-26 15:03:39.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"37914f02006bdd45c0189219dedd84d95298e4a8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-26 15:05:04.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b19b74715be64ff7014eccba6ed539534e0299f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 15:26:02.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8c0c579926b497ea868708a2f21454d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1bea44c65ff548478636e24ba1bc8dd4 : SUCCESS in 15m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e1dccb6c2244964bf25036bf1ec3e3e : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/33a6c11719aa4ab78a655568ba7363d8 : SUCCESS in 21m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c59e946e329648f09a69fe043fe5a6a3 : SUCCESS in 19m 48s","accounts_in_message":[],"_revision_number":10},{"id":"02f4ed89b4eb2f8f8aea90df9d849c576e5c6742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 16:16:44.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3175d13b3f37448dafe885e328904808\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32728b385b9d4b2081b3a1ae01fab929 : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/bb2522b95b9f46f1b13da5bc24ce8850 : SUCCESS in 49m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/387b7547155742ecbde9c9e79b3f89c9 : SUCCESS in 42m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a46e7f238aa94519bcf253db708889af : SUCCESS in 52m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/547fe67c2b104296aa4cb67e6a54cc26 : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4dec7b6ecd414258858eebe0febc8c87 : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d245df130414bc09d862269656fb339 : SUCCESS in 28m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2c92d6b5c7df40008a3926f1f58275d4 : SUCCESS in 9m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bd093e35eba94714baf3063b7316a26d : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f0777d5c87a54a628e6c3f9f2ce293cc : SUCCESS in 20m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/172d7af49daf407bb93a7441f6248948 : SUCCESS in 24m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6a587186605648cc82daa17d2a37c505 : SUCCESS in 16m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/56c5a8d936814ab6a81b4b3503ba0609 : SUCCESS in 15m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/25da7c630f5b4c21a3306381e762130e : SUCCESS in 18m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4960c79304134843bc803eb29830fdbc : SUCCESS in 26m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/24fb318076094c139aac827c43138493 : SUCCESS in 26m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a256b20e78164439ae6fbfa5c65b0955 : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/92b8a095510a445293d24c253d4f33dc : SUCCESS in 15m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c33bbcc99c1b48c99acfd8eac47a37ab : SUCCESS in 24m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b301cbfb24fc460bb254e1041e3090f9 : SUCCESS in 8m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/69cf6711161b45f2b0ed72cff37601f9 : SUCCESS in 24m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c214b108c9a641a9ac0bf1d970f79aa7 : SUCCESS in 8m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c73467e702d7499c905b5af9561c453e : SUCCESS in 1h 08m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f30f91798d349668c9995c23c07a273 : SUCCESS in 1h 11m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7d635c208fb54e8f8ad03c7089092f89 : SUCCESS in 45m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/275c35fa3d014e5986008e07a67a901d : SUCCESS in 49m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ff56bb04bee4abdae6c0fc67b8dddd4 : SUCCESS in 25m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cba85f4279314e049a9d00fbe9f27d82 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/99f3a92f018a4fbfa90b1ad66d0cf9ef : SUCCESS in 27m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62be49e26b2141b3b23e1a9653632d02 : SUCCESS in 54m 57s","accounts_in_message":[],"_revision_number":10},{"id":"5e7ae563a050e280d48e3326931481368d76063b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-26 21:08:29.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"afcc585b0180ff20cbbef5e9b46763b60c49f27b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-30 16:22:56.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"6dcf05a2f418d19cb9bf4044869224732c1e50b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 16:23:42.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"d827ce99a3b781c2aa506c965aa5d245c7e871c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 17:37:36.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47e8c14197da4bbf83daf2d661639694\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cdc225613b940329a60cbf090f5337e : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/d2f89be8c1ff4a22bf23cb2688035959 : SUCCESS in 48m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8a3c21ce3fb344a9ac5ca34477da98de : SUCCESS in 45m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2ff5e9d034f4b10b059b6342b556ae7 : SUCCESS in 1h 01m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e5e5f998884d4e639be475451ec30fef : SUCCESS in 10m 06s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/942505c2e5734694b52027a0f4867eee : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae9a12589ac24b2b84ea0018a292675f : SUCCESS in 15m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f0b4c1078d5f4cd495b0a95ef82ad554 : SUCCESS in 17m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5afb5373f56d4b6bb846fe780697e4b9 : SUCCESS in 26m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d779f7c0ef254628853fccffea08c30c : SUCCESS in 15m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/73d6bbeb2c8b49d286304d26834f6650 : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3c9f9588d78241ac8ca0912d09748243 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4fefe957278848e3868969a081d54729 : SUCCESS in 17m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8d1a2b7c29d045aabba423045eedee08 : SUCCESS in 13m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7e3e3203463346dc9b712a3b3ee3378b : SUCCESS in 14m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/17f541ffa0fc4f7bbaeac0049efac1f4 : SUCCESS in 26m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/02cc1d2323344f4f8c23d7af29d8210b : SUCCESS in 7m 59s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f92452ced0a74f9e82788045281be68b : SUCCESS in 8m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ab1a71d400ed4379aca70759685e87b3 : FAILURE in 53m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/99f2b50061a8492a9420e0a49b9521e2 : SUCCESS in 57m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d7967600a6a643d9b143ff81cb2d4e21 : SUCCESS in 47m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c80f3891941e41a18ac14e9879d0548a : SUCCESS in 1h 06m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fae5b31e57d43bda500e15f6cce59a0 : SUCCESS in 17m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f160030f7f4f4388169cd8d0187c71 : SUCCESS in 5m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de056ac297be410ca0b5871fc7c1c0dc : SUCCESS in 1h 06m 35s","accounts_in_message":[],"_revision_number":10},{"id":"a5c4c9a228eb8ec85157d7dbad2edeb9cd1d734d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-30 23:51:50.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4924240e7c7992db7dddd9a79157b45493b4ec71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 00:13:06.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e82700a9f9f45f184e356dd17dc0584\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a5b53cebf4549efa6081dc65b094edc : SUCCESS in 16m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8592a6c385264d8caec6a0d24e342154 : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f41a3e909bc4986ae8670a86bb98a31 : SUCCESS in 19m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/af84cd1e00a2471b87877d80d318c447 : SUCCESS in 20m 19s","accounts_in_message":[],"_revision_number":10},{"id":"8dc4221c6a86aa68c49a9883c4e7c5284f546dad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 01:01:15.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67c691875c5d415ba55eb14931372a79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3004190def34deba770f852024f57c6 : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/8abd48a3754a4500a4780b4c4439670f : SUCCESS in 1h 02m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1395d66125f54337bad0e598b4c9feb0 : SUCCESS in 51m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60bb1f7e0b5c4314974e2b408af1d581 : SUCCESS in 1h 02m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/843692f4925c4334856b46a9870847a0 : SUCCESS in 17m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36714587506149789797b45f9e9c7218 : SUCCESS in 13m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7105ef46f1f4fc089c09ed3116e2404 : SUCCESS in 17m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ba581172b294d778e9ceaf4761d267f : SUCCESS in 7m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/afda49c7e7444b48a60dcc6134c3e589 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/06a8279870d449ddb838433be920c14a : SUCCESS in 18m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/67d6bef6333e4059bbc998e070892f69 : SUCCESS in 14m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d067b5ed69d44dab9ba98a7bb1d25e3e : SUCCESS in 17m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/73202f0a5d3e4a54a385818637201012 : SUCCESS in 16m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d8c86d0e01364b15bc1278ab47c0120f : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b890e30ba34c4bf68b23fb1df688fb18 : SUCCESS in 16m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6389773d218c4211ac53486adf65e220 : SUCCESS in 19m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f15a8b1b56104d469744ca2459094bbb : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0f09a01155f7469c8cd02e41ae97321a : SUCCESS in 14m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8385f04263cd4b01a3fb86c9b7f43f91 : SUCCESS in 16m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/573b6130b7144aacaa1ae1c23ae03afd : SUCCESS in 6m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f3f7cf0851e7488bbe614a1d973a9fc4 : SUCCESS in 17m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d33f918ff45d45ae997cd3d90b1b6633 : SUCCESS in 7m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9e183bcc1a4a4de19c11ca0141933c81 : SUCCESS in 50m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6373cdf4bfb4cf5879f3a54a4c24cd5 : SUCCESS in 59m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7fc6603835724b4d9e67ad0d1bb92ec3 : SUCCESS in 45m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/51771c4ff0ef48f0880b4a94ee08adcf : SUCCESS in 46m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f6d506f56194f9a948e2948cc8c3948 : SUCCESS in 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/937696d3b7aa4a62a8beaebd12cd0e83 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ffb1f5314b1b4c259130a4ded57eee37 : SUCCESS in 22m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17bef4e0001c4a47b7cdf32a69dc33ef : SUCCESS in 1h 01m 02s","accounts_in_message":[],"_revision_number":10},{"id":"e70d19b86373e60dcfa066672634a3416ff13df6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 01:01:48.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"56b6d4e25530d089d7c7e81b23861b14355efeea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 02:09:03.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e61380574184172857e6dbf0dccb9eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91448efdae1f417682930a84d49ddac7 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/1624912998864dffa614177b4ae645fc : SUCCESS in 44m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7f40f60ebb5f4b7a8d89ab91b153b3a6 : SUCCESS in 58m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd567a149c4e48b591c630837df65da7 : SUCCESS in 1h 01m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9affb54fb53a49a79b732462d92e201e : SUCCESS in 8m 24s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/73ea741941f24989a320be49b4cf9247 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/52027727edb744c0b16591696a29e89a : SUCCESS in 24m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/70d5bed5f2ab4a47aefae7c5b8e01f2b : SUCCESS in 22m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9499ac1baaac49799b9fe6bbe79245cc : SUCCESS in 23m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6f9c2972db1f49309054e577b93cafc3 : SUCCESS in 22m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/432a90bd865143f6a79c8198a428fcb8 : SUCCESS in 24m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b770002f1d8b42dc9b7011dd5f27ede7 : SUCCESS in 22m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1f09d8755c2b49aba395d8a3976420d4 : SUCCESS in 18m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/12ad645bac124c5fab02c1257fb1fe94 : SUCCESS in 22m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/971c69c3c64f44e1856702326e2e65a6 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7e9b19dfa79d4f2d95efc3e03c200542 : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a95de590113f4827a251072980ea3ea3 : SUCCESS in 6m 44s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6b6432e347cf47fbbdb1e6c06f2766be : SUCCESS in 7m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/714057e014544cc1afb8833b65659ea0 : SUCCESS in 48m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3d68bf7e46e7494cb53243eac0466fff : SUCCESS in 1h 02m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/45b6ef5c624444a983d3588ebdc1d82d : SUCCESS in 51m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1fabb02aba544c9699b436668a3a2077 : SUCCESS in 37m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdbb8f6eeaf84705aad05772a6827356 : SUCCESS in 12m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25428f90c40f488b875e510e1603ccd9 : SUCCESS in 4m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1de5fd8bfe2a4224b1242e5be744542d : SUCCESS in 45m 08s","accounts_in_message":[],"_revision_number":10},{"id":"53c0641c3c5e90b7834391d98e6a2fdb5f58cb71","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 02:09:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"84ceeced6f3d509a0cb008b315971def2950b521","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 02:13:55.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3812b548d76141e3b4a8c4402529f92e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e74dcfb512145979ce21d8bef421465 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/05490d3d14624633b0d3c1917d45b7bd : SUCCESS in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0aa92136dd4f4607be24eceab939d189 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"49a5293cbd7d731eba816aebec47ffe65066dc51","revisions":{"a056edd8489498db14900b651df4761ed2834f4a":{"kind":"REWORK","_number":1,"created":"2023-04-17 18:13:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/77/880677/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/880677/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/880677/1"}}},"commit":{"parents":[{"commit":"4db3bd29d04ef7bc7726a464872b8a60f9999417","subject":"WIP: memcached: emit timing metrics when exceptions raised","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4db3bd29d04ef7bc7726a464872b8a60f9999417"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:10:48.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:10:48.000000000","tz":60},"subject":"WIP: memcached: handle failed incr+add+incr","message":"WIP: memcached: handle failed incr+add+incr\n\nHandle the case when:\n  * incr fails because key is not found\n  * add fails because key is found\n  * incr fails because key is not found (e.g. the key expired since\n    the add)\n\nChange-Id: I3086e3453b13411304e431e83d6379e255c1777c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a056edd8489498db14900b651df4761ed2834f4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a056edd8489498db14900b651df4761ed2834f4a"}]},"branch":"refs/heads/master"},"8b788737c465475686caec3138eddce28f57e509":{"kind":"REWORK","_number":2,"created":"2023-05-09 18:33:19.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/77/880677/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/880677/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/880677/2"}}},"commit":{"parents":[{"commit":"6dc58f3882f20e31cb6c91ebafc66b887d9590c5","subject":"memcached: log method and time_spent for exceptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6dc58f3882f20e31cb6c91ebafc66b887d9590c5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:10:48.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-09 18:24:37.000000000","tz":60},"subject":"WIP: memcached: handle failed incr+add+incr better","message":"WIP: memcached: handle failed incr+add+incr better\n\nImprove the exception handling for the case when:\n  * incr fails because key is not found\n  * add fails because key is found (another actor added it\n    concurrently)\n  * incr fails because key is not found (the key expired since\n    the add, presumably because the TTL was short relative to\n    the execution time of the incr method)\n\nPreviously this resulted in an error log and a traceback. This patch\neliminates the traceback and adds the TTL to the log message.\n\nIt\u0027s not clear that this failure scenario is due to a server problem:\nslow execution of the MemcacheRing.incr code path due to the green\nthread being suspended could cause the second incr to be delayed more\nthan the TTL. This patch therefore prevents this particular failure\nscenario causing the server to be error limited.\n\nChange-Id: I3086e3453b13411304e431e83d6379e255c1777c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b788737c465475686caec3138eddce28f57e509"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b788737c465475686caec3138eddce28f57e509"}]},"branch":"refs/heads/master"},"8784dbfdf6c19843c8669342bd6a37cc8cabe49b":{"kind":"REWORK","_number":3,"created":"2023-05-10 09:53:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/77/880677/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/880677/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/880677/3"}}},"commit":{"parents":[{"commit":"6dc58f3882f20e31cb6c91ebafc66b887d9590c5","subject":"memcached: log method and time_spent for exceptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6dc58f3882f20e31cb6c91ebafc66b887d9590c5"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:10:48.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-10 09:50:20.000000000","tz":60},"subject":"WIP: memcached: handle failed incr+add+incr better","message":"WIP: memcached: handle failed incr+add+incr better\n\nImprove the exception handling for the case when:\n  * incr fails because key is not found\n  * add fails because key is found (another actor added it\n    concurrently)\n  * incr fails because key is not found (the key expired since\n    the add, presumably because the TTL was short relative to\n    the execution time of the incr method)\n\nPreviously this resulted in an error log and a traceback. This patch\neliminates the traceback and adds the TTL to the log message.\n\nIt\u0027s not clear that this failure scenario is due to a server problem:\nslow execution of the MemcacheRing.incr code path due to the green\nthread being suspended could cause the second incr to be delayed more\nthan the TTL. This patch therefore prevents this particular failure\nscenario causing the server to be error limited.\n\nCloses-Bug: #2019095\nChange-Id: I3086e3453b13411304e431e83d6379e255c1777c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8784dbfdf6c19843c8669342bd6a37cc8cabe49b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8784dbfdf6c19843c8669342bd6a37cc8cabe49b"}]},"branch":"refs/heads/master"},"78312820e356d8fc3a12a75e7cd6b56c80f59cfb":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2023-05-11 10:50:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/77/880677/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/880677/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/880677/4"}}},"commit":{"parents":[{"commit":"7b4642567ae2c35dd0e8d07ea628a80d04ccc820","subject":"Memcached: emit memcache timing metrics when exceptions raised","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b4642567ae2c35dd0e8d07ea628a80d04ccc820"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:10:48.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-11 10:50:38.000000000","tz":60},"subject":"memcached: handle failed incr+add+incr better","message":"memcached: handle failed incr+add+incr better\n\nImprove the exception handling for the case when:\n  * incr fails because key is not found\n  * add fails because key is found (another actor added it\n    concurrently)\n  * incr fails because key is not found (the key expired since\n    the add, presumably because the TTL was short relative to\n    the execution time of the incr method)\n\nPreviously this resulted in an error log and a traceback. This patch\neliminates the traceback and adds the TTL to the log message.\n\nIt\u0027s not clear that this failure scenario is due to a server problem:\nslow execution of the MemcacheRing.incr code path due to the green\nthread being suspended could cause the second incr to be delayed more\nthan the TTL. This patch therefore prevents this particular failure\nscenario causing the server to be error limited.\n\nCloses-Bug: #2019095\nChange-Id: I3086e3453b13411304e431e83d6379e255c1777c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78312820e356d8fc3a12a75e7cd6b56c80f59cfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78312820e356d8fc3a12a75e7cd6b56c80f59cfb"}]},"branch":"refs/heads/master"},"df857563dfe8bc19440c9628a8a2c875ed9753e0":{"kind":"REWORK","_number":5,"created":"2023-05-11 13:00:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/77/880677/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/880677/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/880677/5"}}},"commit":{"parents":[{"commit":"7b4642567ae2c35dd0e8d07ea628a80d04ccc820","subject":"Memcached: emit memcache timing metrics when exceptions raised","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b4642567ae2c35dd0e8d07ea628a80d04ccc820"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:10:48.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-11 12:59:43.000000000","tz":60},"subject":"memcached: handle failed incr+add+incr better","message":"memcached: handle failed incr+add+incr better\n\nImprove the exception handling for the case when:\n  * incr fails because key is not found\n  * add fails because key is found (another actor added it\n    concurrently)\n  * incr fails because key is not found (the key expired since\n    the add, presumably because the TTL was short relative to\n    the execution time of the incr method)\n\nPreviously this resulted in an error log and a traceback. This patch\neliminates the traceback and adds the TTL to the log message.\n\nIt\u0027s not clear that this failure scenario is due to a server problem:\nslow execution of the MemcacheRing.incr code path due to the green\nthread being suspended could cause the second incr to be delayed more\nthan the TTL. This patch therefore prevents this particular failure\nscenario causing the server to be error limited.\n\nCloses-Bug: #2019095\nChange-Id: I3086e3453b13411304e431e83d6379e255c1777c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df857563dfe8bc19440c9628a8a2c875ed9753e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df857563dfe8bc19440c9628a8a2c875ed9753e0"}]},"branch":"refs/heads/master"},"763653f55ee619c769715f1fa08ae873e4414752":{"kind":"REWORK","_number":6,"created":"2023-05-19 10:11:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/77/880677/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/880677/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/880677/6"}}},"commit":{"parents":[{"commit":"f714f43c90ec6666edaf1ebedec9294d7b6d304e","subject":"memcached: encapsulate common command parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f714f43c90ec6666edaf1ebedec9294d7b6d304e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:10:48.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-19 10:11:19.000000000","tz":60},"subject":"memcached: handle failed incr+add+incr better","message":"memcached: handle failed incr+add+incr better\n\nImprove the exception handling for the case when:\n  * incr fails because key is not found\n  * add fails because key is found (another actor added it\n    concurrently)\n  * incr fails because key is not found (the key expired since\n    the add, presumably because the TTL was short relative to\n    the execution time of the incr method)\n\nPreviously this resulted in an error log and a traceback. This patch\neliminates the traceback and adds the TTL to the log message.\n\nIt\u0027s not clear that this failure scenario is due to a server problem:\nslow execution of the MemcacheRing.incr code path due to the green\nthread being suspended could cause the second incr to be delayed more\nthan the TTL. This patch therefore prevents this particular failure\nscenario causing the server to be error limited.\n\nCloses-Bug: #2019095\nChange-Id: I3086e3453b13411304e431e83d6379e255c1777c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/763653f55ee619c769715f1fa08ae873e4414752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/763653f55ee619c769715f1fa08ae873e4414752"}]},"branch":"refs/heads/master"},"dd32b876fb8133ad48fd761d8456592be5f6defe":{"kind":"REWORK","_number":7,"created":"2023-05-23 13:09:00.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/77/880677/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/880677/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/880677/7"}}},"commit":{"parents":[{"commit":"f714f43c90ec6666edaf1ebedec9294d7b6d304e","subject":"memcached: encapsulate common command parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f714f43c90ec6666edaf1ebedec9294d7b6d304e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:10:48.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-23 13:08:45.000000000","tz":60},"subject":"memcached: handle failed incr+add+incr better","message":"memcached: handle failed incr+add+incr better\n\nImprove the exception handling for the case when:\n  * incr fails because key is not found\n  * add fails because key is found (another actor added it\n    concurrently)\n  * incr fails because key is not found (the key expired since\n    the add, presumably because the TTL was short relative to\n    the execution time of the incr method)\n\nPreviously this resulted in an error log and a traceback. This patch\neliminates the traceback and adds the TTL to the log message.\n\nIt\u0027s not clear that this failure scenario is due to a server problem:\nslow execution of the MemcacheRing.incr code path due to the green\nthread being suspended could cause the second incr to be delayed more\nthan the TTL. This patch therefore prevents this particular failure\nscenario causing the server to be error limited.\n\nCloses-Bug: #2019095\nChange-Id: I3086e3453b13411304e431e83d6379e255c1777c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd32b876fb8133ad48fd761d8456592be5f6defe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd32b876fb8133ad48fd761d8456592be5f6defe"}]},"branch":"refs/heads/master"},"3dcfd3774e4479735b5809925e077bf9f936f898":{"kind":"REWORK","_number":8,"created":"2023-05-25 11:54:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/77/880677/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/880677/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/880677/8"}}},"commit":{"parents":[{"commit":"f714f43c90ec6666edaf1ebedec9294d7b6d304e","subject":"memcached: encapsulate common command parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f714f43c90ec6666edaf1ebedec9294d7b6d304e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:10:48.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-25 11:54:38.000000000","tz":60},"subject":"memcached: handle failed incr+add+incr better","message":"memcached: handle failed incr+add+incr better\n\nImprove the exception handling for the case when:\n  * incr fails because key is not found\n  * add fails because key is found (another actor added it\n    concurrently)\n  * incr fails because key is not found (the key expired since\n    the add, presumably because the TTL was short relative to\n    the execution time of the incr method)\n\nPreviously this resulted in an error log and a traceback. This patch\neliminates the traceback and adds the TTL to the log message.\n\nIt\u0027s not clear that this failure scenario is due to a server problem:\nslow execution of the MemcacheRing.incr code path due to the green\nthread being suspended could cause the second incr to be delayed more\nthan the TTL. This patch therefore prevents this particular failure\nscenario causing the server to be error limited.\n\nCloses-Bug: #2019095\nChange-Id: I3086e3453b13411304e431e83d6379e255c1777c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3dcfd3774e4479735b5809925e077bf9f936f898"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3dcfd3774e4479735b5809925e077bf9f936f898"}]},"branch":"refs/heads/master"},"fec1deb4746dfe8e21baf06823a643ead8507722":{"kind":"REWORK","_number":9,"created":"2023-05-25 15:20:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/77/880677/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/880677/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/880677/9"}}},"commit":{"parents":[{"commit":"f714f43c90ec6666edaf1ebedec9294d7b6d304e","subject":"memcached: encapsulate common command parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f714f43c90ec6666edaf1ebedec9294d7b6d304e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:10:48.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-25 15:20:23.000000000","tz":60},"subject":"memcached: handle failed incr+add+incr better","message":"memcached: handle failed incr+add+incr better\n\nImprove the exception handling for the case when:\n  * incr fails because key is not found\n  * add fails because key is found (another actor added it\n    concurrently)\n  * incr fails because key is not found (the key expired since\n    the add, presumably because the TTL was short relative to\n    the execution time of the incr method)\n\nPreviously this resulted in an error log and a traceback. This patch\neliminates the traceback and adds the TTL to the log message.\n\nIt\u0027s not clear that this failure scenario is due to a server problem:\nslow execution of the MemcacheRing.incr code path due to the green\nthread being suspended could cause the second incr to be delayed more\nthan the TTL. This patch therefore prevents this particular failure\nscenario causing the server to be error limited.\n\nCloses-Bug: #2019095\nChange-Id: I3086e3453b13411304e431e83d6379e255c1777c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fec1deb4746dfe8e21baf06823a643ead8507722"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fec1deb4746dfe8e21baf06823a643ead8507722"}]},"branch":"refs/heads/master"},"49a5293cbd7d731eba816aebec47ffe65066dc51":{"kind":"REWORK","_number":10,"created":"2023-05-26 15:03:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/77/880677/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/880677/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/880677/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/880677/10"}}},"commit":{"parents":[{"commit":"f714f43c90ec6666edaf1ebedec9294d7b6d304e","subject":"memcached: encapsulate common command parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f714f43c90ec6666edaf1ebedec9294d7b6d304e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-17 18:10:48.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-26 15:03:15.000000000","tz":60},"subject":"memcached: handle failed incr+add+incr better","message":"memcached: handle failed incr+add+incr better\n\nImprove the exception handling for the case when:\n  * incr fails because key is not found\n  * add fails because key is found (another actor added it\n    concurrently)\n  * incr fails because key is not found (the key expired since\n    the add, presumably because the TTL was short relative to\n    the execution time of the incr method)\n\nPreviously this resulted in an error log and a traceback. This patch\neliminates the traceback and adds the TTL to the log message.\n\nIt\u0027s not clear that this failure scenario is due to a server problem:\nslow execution of the MemcacheRing.incr code path due to the green\nthread being suspended could cause the second incr to be delayed more\nthan the TTL. This patch therefore prevents this particular failure\nscenario causing the server to be error limited.\n\nCloses-Bug: #2019095\nChange-Id: I3086e3453b13411304e431e83d6379e255c1777c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49a5293cbd7d731eba816aebec47ffe65066dc51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49a5293cbd7d731eba816aebec47ffe65066dc51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
