)]}'
{"id":"openstack%2Fswift~880783","triplet_id":"openstack%2Fswift~master~I7ab605d48972e8dc06e630d160c745baeea91355","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-04-20 03:13:58.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-04-21 17:30:09.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-04-20 22:21:36.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I7ab605d48972e8dc06e630d160c745baeea91355","subject":"docs: Clean up cross-domain doc formatting; call out CWE-942","status":"MERGED","created":"2023-04-18 21:19:52.000000000","updated":"2023-04-21 17:32:08.000000000","submitted":"2023-04-21 17:30:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880783","meta_rev_id":"603751667e6ee38edbad8d2780c0fe02900a8ccb","_number":880783,"virtual_id_number":880783,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-21 17:30:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2023-04-20 03:13:58.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2023-04-19 11:05:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2023-04-20 09:27:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-18 21:42:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-18 22:23:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-19 00:52:19.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2023-04-19 11:05:15.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2023-04-20 03:13:58.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"5b94a31ae5765fd11ea4bbb0448ad2c21dab05d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-18 21:19:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efedf418dfa46c65c67b44155f8398fefab9535f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 21:42:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38c5b39af93141bd9bf5ee9cd2a5445e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/026c05c35d294c87b9f656ceb56513df : SUCCESS in 16m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e5872d7389a748409f68dcff0d165f0f : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d87e586bd044bda8d1adf42743e1af1 : SUCCESS in 18m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a859ebabe55423889435839a1d1ca14 : SUCCESS in 20m 51s","accounts_in_message":[],"_revision_number":1},{"id":"64820798e1d1bedd951b3342f18cb41730a01a51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 22:23:18.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94a586594e7f43eaa4717533b830e721\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf85345ba71f4e00885363e9230e73ac : SUCCESS in 7m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/5795fa8c284b4cbaa8c2e6da1ff93ad0 : SUCCESS in 41m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/01f26a31957b4ee992e4ceaddea97140 : SUCCESS in 34m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23fbb4ad0905426aa7edf41d846dadaa : SUCCESS in 58m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/24ca8d54df7a4c80941c7c2886fb1f0e : SUCCESS in 13m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e383887522384b2a93d40dfc9c5a1a4b : SUCCESS in 21m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7d7751ba710453da9ee3d9d06353134 : SUCCESS in 15m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/718b27ad28ec4b928f52b8c046d63d41 : SUCCESS in 8m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a997d6ff1f514703b86bbf4409a97d81 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0eebee9fa597485798527d261e0d1dd2 : SUCCESS in 18m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2eb969d24d594dcdac639ca55e663b02 : SUCCESS in 18m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/531f9c23ba5e4dbb864376fda6872914 : SUCCESS in 19m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/edeac43220f44186bc48da0fc4c1e083 : SUCCESS in 14m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bb0df4bafc7e4296a89763423498cd5b : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/562559dd1ecc4ffca56840673346018c : SUCCESS in 19m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9d1aa73cd1dc4908b4fca8ef413b06c1 : SUCCESS in 18m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9c7a014bb20b4d79aa6152fc17202990 : SUCCESS in 13m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5f4fe699455a4ff9bb35c0a96bf45d81 : SUCCESS in 12m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/79733062c1e549a1864da613cb44eced : SUCCESS in 16m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ffb4ac9e15f448ef93b80738a8098da7 : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b2d293df0a0f49129c75b62e7d26e2a0 : SUCCESS in 15m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5b4bf06d77354aa08fd97fe494faf319 : SUCCESS in 6m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4796d02929774fad9da2c7310d711839 : SUCCESS in 51m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f40d2cd4836f45ebbde2c1a0f172b09a : SUCCESS in 46m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e744c25e65f6450c901e9a6b2c8a6185 : SUCCESS in 55m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8b33f97a745c43eabc3723cf9b126835 : SUCCESS in 33m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdb15846910448978b108f2fb9f0045e : SUCCESS in 29m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0777d2e547974cc29e84e6cd4cd0c92d : FAILURE in 3m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01e439215e7843929824a860e3a7523a : SUCCESS in 16m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fcdfc25f26a4dadbc5a736aa1dea0f4 : SUCCESS in 35m 06s","accounts_in_message":[],"_revision_number":1},{"id":"9b7c328b735e35b0e61492ff4dc0b513976d4f5b","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-19 00:52:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6edf7ce7cb3b243c63fa915a5f42dec2dc24a83d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-19 11:03:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d07ff62ec1a91d0f8d373c8c98a89726157f6ced","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-19 11:05:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d092a1ceec66f99827a7d6f7faef3cc40649d71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 12:14:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/54c3b2a241ce42bbb208b3b1553316b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15372b56e16341b4b8d1c20e7023aeb5 : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/4c8e141b61ba47f8bdaaebfd063a521e : SUCCESS in 45m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/905e2dc0e33746b095355c1462e20422 : SUCCESS in 38m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2f9c900f3ff4623ac98b28a6b0b1ec9 : SUCCESS in 47m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61046045a74a47f691719695a1f71b1d : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37bb0003567d46d6a725dc7cbc08a54c : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae5ea247088041278ac59c891174a12b : SUCCESS in 19m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/246efed4e44d47aeb880279281296769 : SUCCESS in 8m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d2b897b3549e4969addc8eac9b08ebc8 : SUCCESS in 7m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ba620f117335491c8bb85b2a72c3454e : SUCCESS in 17m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7c04ea5bbca940429d40a1c4c11f39a3 : SUCCESS in 15m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/edda06f856c4437984bbbce0a9ec1c80 : SUCCESS in 14m 17s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8803335bbadc46a2842f084e71455979 : SUCCESS in 30m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f92cbd77b32c4730966d2303a525f2cd : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a716f79e107f47239e2b4bdafdb41362 : SUCCESS in 29m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e2cb663468a94a1dbd78ca364d7c71e9 : SUCCESS in 18m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6348ebcf3f2642729499190aae3ed67d : SUCCESS in 25m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/47226561a21e41fb9ce061e3f91d1fe1 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9aaacb161f0045238b0c9c574c493a0f : SUCCESS in 14m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c0e363f3cc0240ef9f11f80dac7bb86a : SUCCESS in 8m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd471a9dec214fda934c58b928258038 : SUCCESS in 15m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d3349fdfa5ec4d07b1a35855099d6fbc : SUCCESS in 7m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3b6b7143bb0f4177be4f7c76a6a4355a : FAILURE in 56m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/54cc5e7cf9a64a418e1b41f46ccef402 : FAILURE in 59m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/50aa415ec3b94317beae94b6ad02d22e : SUCCESS in 50m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1596c04a2b3841f394fb917b1678eb23 : SUCCESS in 54m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abafce060c6549a697c03d093adb8438 : SUCCESS in 23m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5244b0d4c86841f7be90b8ba628cb8fd : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/274269029d774d54b7a427f2c8c2e212 : SUCCESS in 20m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc02c88f39e84585afdabf1cfc0352ee : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":2},{"id":"7526714626786874353b8ae75363283ea9eeb675","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 12:43:23.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fce6ef02e100432c9acbe6217ffd80d7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3ff7d8319b64740a83b9bc38cf7bed6 : TIMED_OUT in 1h 30m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5ccceab18f8416e903b9085482aaa2e : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b97126046576487fb298f70b871264e3 : SUCCESS in 21m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cbf196c8a4c4b7aa6d5419433bfb1b2 : SUCCESS in 19m 45s","accounts_in_message":[],"_revision_number":2},{"id":"336b88eab25fd500e20d2eca89cd2a8e3cb1c9fa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-19 18:03:23.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4703e6e6a15953598498126eaad2b75a25f567c9","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-20 03:13:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0c2768335a2b80b0abd143fa6effb66ca4f81430","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-20 09:26:53.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"1dbc61dc87d8047a394e08a4cc21536505f24f49","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-20 09:27:02.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"42fad25400253715952717cc32a2951db2ca6804","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 10:58:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/27e6a896ed5c42189dffb23dc6b4a440\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dc2854a2eeb42f5b97db8c4d86a6510 : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/0cea1b2238834554a4aaba6499239972 : SUCCESS in 58m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/68ea96d2ea024c8a9f21beda399c0634 : SUCCESS in 46m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15e7aefe225b4911993dcecdc29bfad4 : SUCCESS in 1h 00m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15237d34d2f745dfbd960de34be0c735 : SUCCESS in 24m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b39c94a97878440faaf6c29b5c56b0c0 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb78767667124dc49c91fbd02195a7f5 : SUCCESS in 18m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dda6330e574943eb81e41f8be5d54bb5 : SUCCESS in 8m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b094659f68d54428869006a2179cd25f : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/be7c02f8da6740c9a4a8fecc125a9de4 : SUCCESS in 17m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/df4ded09f02548bcbf471607ce2d8c6b : SUCCESS in 19m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/62c10b2f25e645519335ac677e0b4fd9 : SUCCESS in 23m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2e58382f1ed0415e82478f202e97ffe8 : SUCCESS in 24m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1ed8f05257c64d819892c26375343439 : SUCCESS in 23m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/572f4a39fff24ad683e04a608b85ed3e : SUCCESS in 17m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e7122c6791b54961af39b1a30864d303 : SUCCESS in 17m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/de20fb4f889f47c5aadaa6f2185bc06b : SUCCESS in 13m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/43c5796bd244423f9133656131841b30 : SUCCESS in 21m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2563f728964546c7b4834df5928cf6e2 : SUCCESS in 23m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3845aceda41440caad7e888fb637f03f : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d3dde865e40b420298049b97ff1a5df0 : SUCCESS in 16m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b57391b3807c4dcd9ed6d0f25132396d : SUCCESS in 7m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cd0f9593461348fc8596a264fbe4e301 : SUCCESS in 54m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/46962a4e56aa4fa087e6c0f5d3cbe405 : FAILURE in 1h 03m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/910df3a26d124c2796ad5f0ec7631405 : SUCCESS in 59m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/85b7a614d3d14067ac57cb8ad71faf11 : SUCCESS in 53m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/854d83be4ee04681ae60011bd82642cd : FAILURE in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f9a5a3682b4168b7cefd5907d29d92 : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2b78b9ea08424d8f955fd8d26418283b : SUCCESS in 24m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/accb57e7c7e247e9a14670a1a303e5ee : SUCCESS in 39m 12s","accounts_in_message":[],"_revision_number":2},{"id":"548d1e053fcd29ec862288d82f12165e7e92813c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-20 22:21:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"33fcdd6f825935e7a1f259848551033b6f6a6f81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 23:35:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e19cac5926042efa6f00606e72d8c81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c18e3d5a21a461c99328c92e0c593b5 : SUCCESS in 8m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/df49dea6c3774975aff9a68ff9eec628 : SUCCESS in 1h 00m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/219fe88f79ae427b8622494b908ff0e8 : SUCCESS in 31m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af433edbec8a438fba204081ae443e51 : SUCCESS in 45m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb93674517ff4fecb38ad3587abd02b9 : SUCCESS in 17m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0302ca7aea142e5b1a38d485fa1bfc3 : SUCCESS in 21m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/202ac4b5881745a09010f9075601c402 : SUCCESS in 14m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/29a23e05c2d742baa8066088bd66cb5b : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/deed6ffb60ae4ac7b3c7144d99b10f0c : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4fe4841002ed46b99f75cf06cb9c26e4 : SUCCESS in 18m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/00e1051567d74c8dac7dd6031f3d3edd : SUCCESS in 20m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8fa1e6370ded45d8b7a4c61f4ca7b174 : SUCCESS in 16m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2062385acb2c4c519c7f60b4788f2b76 : SUCCESS in 16m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4b884b1d4a2c4067b7aaf50586c2934b : SUCCESS in 20m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9a51bcad82cc4deaa0d570637da3146c : SUCCESS in 15m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0716645545f540deac4f05897a4b4120 : SUCCESS in 21m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7e67e5276e324c18b0565b01f2c2419a : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5db9eec5d5a24d6c85b1d280c58c954f : SUCCESS in 16m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c3e83f92c19541e5a7c80949b6b12894 : SUCCESS in 18m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/64ba8bb1b24c4c2cb1b36daa4f067186 : SUCCESS in 6m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/890af59fac1d47debe78f8da9990af77 : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/88c9c76147294903acee218fd5c51111 : SUCCESS in 7m 23s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1c5b52c4c90d4d4185fa8aeb13ac96ee : SUCCESS in 1h 02m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e46b3f851dac4ee7b861f7ab458143f6 : SUCCESS in 1h 07m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/61f5a8beb95a4a16bcf25ce4a791cc98 : SUCCESS in 41m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b5d37a97a2b540c481b9aefcb1f645c4 : SUCCESS in 42m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fed3ce92da5244d6a03b97ac0861b0be : SUCCESS in 19m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b789e64f390547c0869c229ba4cb4092 : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/69ad1b88885b45e4aac907d6a8f94b59 : SUCCESS in 19m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/262303c0db1a47c49da729bc75ad8d71 : SUCCESS in 54m 18s","accounts_in_message":[],"_revision_number":2},{"id":"6210c428c068ec6dd690f987645aea52116c353f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 23:36:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5673349e1baf1162b41a391ed0166726e786e718","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 23:39:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb3ef7aa87cf4717a66116c1b39dec9d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d536bf1d421740ea99c1451feaf7fe9f : SUCCESS in 14m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0b2b67a4f5f46c693dedb6a7cc25cf2 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a09c990eb63945f4bb8c39e6f1b1e5ac : SUCCESS in 16m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ee5236cc5c34f10b73baab32151dffe : SUCCESS in 1h 15m 38s","accounts_in_message":[],"_revision_number":2},{"id":"a7fd6af46927ea5ac15e6ccbf1bd6c59e2d0b8b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 17:30:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1da8846b82864a8ebaaf2132f5c8c957\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f8f4e616deb4fdeaf045faad737b6e5 : SUCCESS in 8m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/d22f5135473c4b3b9a0ef8f8b7cc7ee9 : SUCCESS in 48m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/406ec6a2e7104ab3a23247dbcdccee91 : SUCCESS in 47m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd3c86a6f4bb4af3a95a4027ae616c20 : SUCCESS in 1h 04m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3492014de87c4d54abc9e1650b1a811e : SUCCESS in 8m 07s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d8f5ae74f31e442eac5b34b315e544fc : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/02b2a20547e54cf9a61d2e676ab12806 : SUCCESS in 16m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dc5152c2d1ec48aaa8c814d5707e2f68 : SUCCESS in 15m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/17f357d3f9bf4b42ad53e3ec21aad225 : SUCCESS in 14m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2a9b7ac8518a43ee9a0b8484eddefe2b : SUCCESS in 15m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/85ca220d3d6f42dd87006415bbcfa889 : SUCCESS in 23m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eb423acdcae9447491eb8ca207f32b00 : SUCCESS in 23m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/81be06a7e19e48568ee7c3c950127cd2 : SUCCESS in 18m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/68a6867d6ce84bf0b7f64e9cf868d268 : SUCCESS in 14m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/95b348a3e8fa4d17824e77999bafc42d : SUCCESS in 15m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e51c821944a64b208af568acb272299b : SUCCESS in 16m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b5ca18438ed54de7bf62b9281f7d82ce : SUCCESS in 6m 41s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/35d1382751c5423ba676d5394f11874e : SUCCESS in 7m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/40568db95b134bba8b23c224d38f9088 : SUCCESS in 39m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f077e830ab4a431b92d388964930e67e : SUCCESS in 1h 01m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/23df524ef3f941449e078a6cfa802f83 : SUCCESS in 52m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4f04f18898d943d396c6ae5973331493 : SUCCESS in 40m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39c925359fdc4d41b9f2976313147dfc : SUCCESS in 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4841061fe40a4a829f240efaabeb996e : SUCCESS in 5m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80eb9e740c3f437bbff991becc7f4631 : SUCCESS in 47m 49s","accounts_in_message":[],"_revision_number":2},{"id":"92d6cc34bdfce01afb949cc15a18dc19c0b18788","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 17:30:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"603751667e6ee38edbad8d2780c0fe02900a8ccb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 17:32:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc3a1bfd92cb4380b05aa1a0e757de0d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f4fb6d74c640a4bb895b7534fb2ba8 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2b5ecf76c7cd4938815703a7d8fc9b10 : SUCCESS in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/fb55cca195c14697b5c615fa7d231ef0 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ed1f5193e5e69cb19eafa55d6b4c29a4e3575b00","revisions":{"4c51d505884dc0e8a6993a39a0177333637f8213":{"kind":"REWORK","_number":1,"created":"2023-04-18 21:19:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/83/880783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/880783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/880783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/880783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/880783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/880783/1"}}},"commit":{"parents":[{"commit":"4b6f54d063c33e13ced7d6dd5208921680f56ca6","subject":"Merge \"ssync: Round-trip offsets in meta/ctype Timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b6f54d063c33e13ced7d6dd5208921680f56ca6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-18 21:19:31.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-18 21:19:31.000000000","tz":-420},"subject":"docs: Clean up cross-domain doc formatting; call out CWE-942","message":"docs: Clean up cross-domain doc formatting; call out CWE-942\n\nChange-Id: I7ab605d48972e8dc06e630d160c745baeea91355\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c51d505884dc0e8a6993a39a0177333637f8213"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c51d505884dc0e8a6993a39a0177333637f8213"}]},"branch":"refs/heads/master"},"ed1f5193e5e69cb19eafa55d6b4c29a4e3575b00":{"kind":"REWORK","_number":2,"created":"2023-04-19 11:03:38.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/83/880783/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/880783/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/880783/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/880783/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/880783/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/880783/2"}}},"commit":{"parents":[{"commit":"4b6f54d063c33e13ced7d6dd5208921680f56ca6","subject":"Merge \"ssync: Round-trip offsets in meta/ctype Timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b6f54d063c33e13ced7d6dd5208921680f56ca6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-18 21:19:31.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-19 11:03:27.000000000","tz":60},"subject":"docs: Clean up cross-domain doc formatting; call out CWE-942","message":"docs: Clean up cross-domain doc formatting; call out CWE-942\n\nChange-Id: I7ab605d48972e8dc06e630d160c745baeea91355\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed1f5193e5e69cb19eafa55d6b4c29a4e3575b00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed1f5193e5e69cb19eafa55d6b4c29a4e3575b00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
