)]}'
{"id":"openstack%2Fswift~880911","triplet_id":"openstack%2Fswift~master~I07bd992868f9d8799d2dfa5e840bd6f1cf557723","project":"openstack/swift","branch":"master","topic":"otel_tracing_tracing_namespace","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-10-24 05:06:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I07bd992868f9d8799d2dfa5e840bd6f1cf557723","subject":"tracing: More unit tests","status":"ABANDONED","created":"2023-04-20 07:15:13.000000000","updated":"2023-10-24 05:06:53.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"610b7569bf4e36ec712be1a34852d5ddc8c4092e","_number":880911,"virtual_id_number":880911,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-26 08:36:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 07:42:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-20 08:17:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4739139aa4fdfde450d33ea32341fed497507e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-04-20 07:15:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cb62fd817bf59e5bf6f29fdccdea2575b0500d7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-04-20 07:16:30.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49d08118a46dff9129ce1cfa10da38a8ff2f0edd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 07:42:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49bbb1ac4cd9480d9dbb9e9484b4884d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a0862cdb77c944b994217112d445eb57 : FAILURE in 21m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32774d00f1894621aa987513c698690d : FAILURE in 16m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/46887819707c4ba580e863c923668722 : SUCCESS in 25m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fed09dcf760463ba2672c3aa1b7bea2 : SUCCESS in 19m 38s","accounts_in_message":[],"_revision_number":1},{"id":"e82a7e70b266222c159862f103de67575f13c4d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 08:17:05.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95a703f5b88f4770936680ade632ea52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c6d47d4471f48b88953bc7f3f6dbf1c : SUCCESS in 8m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f4925df93a149939a28caf1527348b4 : SUCCESS in 30m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bfc85d39d89745609edcff7664dcb4f3 : SUCCESS in 24m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c2b86ddcc7044178ffaa5fa730a0711 : SUCCESS in 25m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/81312df9f9fd4b388925eabfcff0c6e2 : SUCCESS in 5m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3c72f3947eba4fceaceef0941b5af8da : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b69f78235fa3488bb57b1a0aa729853a : FAILURE in 3m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/53851469bb3542bb9afcaa3f2c4904c3 : FAILURE in 25m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4007ddd0976a4a3d81f7c3a0a88ee286 : FAILURE in 23m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7d21ffe689084bd0bf399381a11fb365 : FAILURE in 27m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/517c9ae1c8c3493ea34440609653bc7c : FAILURE in 3m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/78211eec8530408498ed1041b40217a9 : FAILURE in 2m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4880e12882404b00a8ab30409ab14ecc : FAILURE in 3m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bce949a1865547e990ff780bc68707ce : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fe3ea40b8a174cd3943158bc825ab3d2 : SUCCESS in 26m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5a3d6ddb5e554d10b7ce1fadfe82751c : SUCCESS in 27m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b7434fc3b42540e6b474cb3a70dc8473 : SUCCESS in 51m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6611eaa1636b4da68247cbba9306cf44 : SUCCESS in 52m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b4883946bb347b5b9886ec97dc1fa15 : FAILURE in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbd4771039d94806a8fc37df4c285d16 : FAILURE in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2cd59cc78b9b4eefaf91018c7d13cc18 : FAILURE in 7m 12s (non-voting)\n\nWarning:\n  Comments left for invalid file test/unit/common/middleware/test_account_quotas.py","accounts_in_message":[],"_revision_number":1},{"id":"fd3536b18045f2d7f87910276ee85af5998ff121","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-04-21 07:03:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"aa830e7fd486d8965a6cdece1ecb3a048b492a8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 07:27:56.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4d9d01390e248c6842995a333453650\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d48638fd0e244939a8c7bf08315edc6 : FAILURE in 16m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28712e412b0c49e39872748b87dd3a63 : FAILURE in 16m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4cd508b37914c1284847c36b57ec98b : SUCCESS in 18m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/66b4974ce56747d4a247ef6633da4696 : SUCCESS in 22m 48s","accounts_in_message":[],"_revision_number":2},{"id":"017c9a25ce0483f574d2edce038dc73a6c111733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 08:05:22.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b77a9fd43c9b41ae95a3849e4e8909f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa671740f0bf457bb0a5300398b4eca3 : SUCCESS in 7m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e44c9ff9c2c4ffebb57636b5f8d2f61 : SUCCESS in 27m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/039553a7bf4144a7acd43e2ed26e7a6f : SUCCESS in 28m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3fac1db42b234dd89b3449644e13b8d8 : SUCCESS in 27m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dd8907d44fa748548b61dd5dd892710e : SUCCESS in 5m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e5fff11b13a14a88b8bfed041c4c5c5d : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2ca448f687594344a99c4d8989f12d14 : FAILURE in 5m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6b460b057e8f45bab6289e89fce8e619 : FAILURE in 27m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d5aad7ba75ea4521964b41d4d2e0529e : FAILURE in 16m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3d91de4edf6c4879b375d80b387c61f9 : FAILURE in 17m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a1c4f157a68f47eb8d6bc072e0ca3af7 : FAILURE in 5m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a310870da0cd4412b4b84b940274c798 : FAILURE in 3m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4b108ae64f1c45baa37ecdb566e0d681 : FAILURE in 3m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f6fbfc0bc2f54a85b22e4989a53390de : SUCCESS in 24m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/67174e037b5c4582bc92571a0193353c : SUCCESS in 16m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/223b2e3e5cc5481db95f6c8c2dec75a7 : SUCCESS in 25m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d93a87467e1e4fd087bf77df54012dc0 : SUCCESS in 48m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f90a13021cb2442a98edd40a84dfedea : SUCCESS in 55m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/268b73f7855e4d648826bc70d1b9a04d : FAILURE in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8066ec5c03914341ac888e805362e334 : FAILURE in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d9861fa2ecb449cf97764251724e1929 : FAILURE in 7m 36s (non-voting)\n\nWarning:\n  Comments left for invalid file test/unit/common/middleware/test_account_quotas.py","accounts_in_message":[],"_revision_number":2},{"id":"3bf1085b6c49de876896f65d465f22dd77724292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-04-24 09:55:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"819d686998b2e360b0ffc5e9b192a5f21e6b9561","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 10:45:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9013c34de9b249dd9a0ab9c61d8198a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d567b30ad949b5ac7759dcd92b3a66 : SUCCESS in 8m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b688157cc88640b6a431bf961caaef58 : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a18284b28ef49ffbf5940da3ce3cb39 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5df7e95f7f544e5a4016f1d4d89fe44 : SUCCESS in 15m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7ab02ece21ce456eb358fec288d434e9 : SUCCESS in 7m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/62d67ae9a1a047188380ea25b6b2e38a : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/251c6fd6b5e24a4e998096ea694de59c : FAILURE in 3m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c7f3f32a19d94c14b28df485ae2cc427 : SUCCESS in 20m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b72d0646d6074e19b6dd8d744d865c32 : SUCCESS in 29m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a5ce28b05145409f8daf2fcd1a815fa9 : SUCCESS in 14m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/be7a82173dc2421a8111c780676807bf : FAILURE in 3m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1f69860919a04c06a6b39c447946f98a : FAILURE in 3m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4c71d105aedc42e0a32c26217fede4ce : FAILURE in 3m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/52919e5aba5748f1ad33436db3d927a0 : SUCCESS in 25m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/462909e5c856405a8bde17869787bd8e : SUCCESS in 13m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2d2413fbf6a043ba8bf0a9a0b99d03ec : SUCCESS in 16m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/338d4ef347f84a69b3944c6cda8673dc : SUCCESS in 35m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8bc7ebd66ecc42ac9453de1c724a7dda : SUCCESS in 46m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/908021e8d0fc4b0484e306a676fe4670 : FAILURE in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/201bdcfa12b94027adb8b2b941c3d1a5 : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/07864fb88c034ee2a848cf615ea70e86 : FAILURE in 7m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5a5d80dd87ecd6742be086ec536245cec89f318f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 11:29:09.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4ae9c29de9949d6b1b7575c1b078dc2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/269c86925fc64a7d8c4c60922f09b341 : SUCCESS in 1h 21m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce4ae1e309344790ae7da644ec2151fb : TIMED_OUT in 1h 31m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f004135f14e04b9cac9899cd4f1e2e26 : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2325e6b7ba9e445b9607377ace29be8d : SUCCESS in 21m 57s","accounts_in_message":[],"_revision_number":3},{"id":"50865c65ff24d7c1b563d90d51d922d0724bce39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-04 07:27:03.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"be3ebe64ab8cfe1eea33606dd977956badea13fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 07:51:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63778b1c46ce4cb4bc300b2fc64063df\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e287f01aed2743a4b633c59bdf37a8a0 : SUCCESS in 17m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04450937b36d4efb8dd18a03bf18554a : SUCCESS in 17m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/33868361d22c40408a94ab85da853672 : SUCCESS in 20m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/730667537f264f7ea79c423903318637 : SUCCESS in 21m 45s","accounts_in_message":[],"_revision_number":4},{"id":"b878b39fbe49f5bfe25735824cab02a34ce12163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 08:51:52.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc9f900af2274216afb00d12d582c2eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/136f633ec3a6406fb9e135c79b07ca96 : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/528925928dbf48cd8d17d7239f612ce5 : SUCCESS in 43m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e9fd20a5796d498289ecacf77da48e6d : SUCCESS in 45m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/264cf70749b14616a3fc18a41ea9286f : SUCCESS in 1h 04m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d17eb1dfe10f4d8dac75b475a5df3a39 : SUCCESS in 17m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9dd69bb2e7e497597774119240f7858 : SUCCESS in 27m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7eafd43103b42f4850cce01b990fbaa : SUCCESS in 18m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5903fef88be6432a8557bf1e7c645f70 : SUCCESS in 6m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6ad038393a914013ba8db37cf3e16936 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/efe848d14fcb4cfcb8b8910573aa219d : FAILURE in 2m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/edbc9aa017ea493098fbc07ded8d624e : SUCCESS in 16m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cb41138755d84259a35f8fd24f9b96b1 : SUCCESS in 26m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ce0f09a451864a2eb7f0d10a4a4214ba : SUCCESS in 25m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/644690883c164e689fcc905286569f7f : FAILURE in 3m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f69117c5d0174d57954d894236696e53 : FAILURE in 3m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fe6c59f27683439181c85b0705093ea4 : FAILURE in 3m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f39e089e2b0f482b906e9eae1ac1d829 : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9b1dac99bbd84555adbec4d786e72305 : SUCCESS in 18m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/76538a3e944c4727b032a0855903d154 : SUCCESS in 14m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8c35147bb1cc494cad0e1f9148d9409c : SUCCESS in 8m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/954e29c0364e4d9093af85220119278c : SUCCESS in 17m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/782a94208d4947a3b512ab8131c103a0 : SUCCESS in 7m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cf1446ea29c6416aa2d24d34fef6db4b : RETRY_LIMIT in 5m 10s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f101d9c9fee4fc9a1a09d98ee9a5b7b : SUCCESS in 1h 17m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5846087585054b3e85b885bf05df0824 : SUCCESS in 1h 01m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9ec7d37e85224ad4abc917f610ad62ba : SUCCESS in 48m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89caa4db10bf47dc8945cce7d82b3f72 : FAILURE in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/574032c89b6d48dabae1f8d5bfd18a22 : FAILURE in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ba0304656e0d4e52bc289c476377bc74 : FAILURE in 7m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db91b2a6927444b5aa9d0aeed8a82bb3 : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":4},{"id":"eda20399073a484e5605c10eb494124106e5d28b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-05 07:16:06.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a7927eece281e8e216d1b6c262f47bd6e1221750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 08:50:13.000000000","message":"Patch Set 5: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/117fc79df8414e0b9fa44fdc25ba81c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0959d3d3be304182a86169f5906e0810 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/a5cf17e230854910b573770fd1993478 : SUCCESS in 57m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ab5461cfd8b840bd95fe3cb3a0d4aa6f : SUCCESS in 38m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67b32dbc5019437692b648eb2d8925bb : SUCCESS in 1h 00m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e976cc15785d465fb64ecee89a3e6f82 : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e5e767736f548309fc3f87616039837 : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a74782b4c5245c8a7abc072fdcc36fc : SUCCESS in 20m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6484c1f087854eca9bae0a555761436a : SUCCESS in 6m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/634c4c2e234e4f8ea0280a369632d3e3 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/95486eb6802f495fbd3fe811e814e652 : FAILURE in 3m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c5ed35daa9c44db7b9bd937933b43840 : SUCCESS in 17m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d26d96430f8345fdb6b5151b9d5402e5 : SUCCESS in 36m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/eb50395414614668b52b879d552da659 : SUCCESS in 34m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/49ea06d6ae354e57a9b4b1a6e913564f : FAILURE in 3m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f34997a04f9c4e71848bffee0f3a853c : FAILURE in 3m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2abfbf026d3b418397f0044ead35818e : FAILURE in 4m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a81651267f584dbab0081b6bba259e5d : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d48b6d06f7604337a459f96464f24899 : SUCCESS in 14m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/90b3391ed2cb4c87bbacf1fee7e6f56f : SUCCESS in 14m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f8d90fe6f97349b495ce05d4e0a7c4f0 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/386765cedc9b4cd4a33616cf34413b19 : SUCCESS in 16m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/54c9520b2cba4a998d7c80ae0b5e7519 : SUCCESS in 7m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cdd2695f3e874c1b9add6c99e6278715 : RETRY_LIMIT in 4m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/41c2f21f842a4696851c4b1b5995a52d : SUCCESS in 1h 16m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8773fa69acb413fbd5348a638677345 : SUCCESS in 50m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/afa8ae7d2fc743d19497f899e2e73369 : SUCCESS in 44m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99308b52a9ee47069b5f84e437e14eda : FAILURE in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08718679bbf54b68bbf78ac0a4c5f8f1 : FAILURE in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/33b1eb77b6d045e2b9aeca536d3151f4 : FAILURE in 7m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2b5dd64dfe44fefa6a89df3829428d5 : SUCCESS in 55m 36s","accounts_in_message":[],"_revision_number":5},{"id":"249db4e30736b949a557bc25c270dfc1cddba7be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 08:50:15.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7033e5ca5774778839a91ae723c8035\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49ce48898eca4f03bc8ada46c931be0b : SUCCESS in 1h 14m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95df53bedc9c46559fe32904aae214b2 : SUCCESS in 1h 27m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d3ede1568444cc58f540acaff21e4b8 : SUCCESS in 24m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f14ecb4027bd42c4b405b8e2f309949e : TIMED_OUT in 1h 31m 17s","accounts_in_message":[],"_revision_number":5},{"id":"4723325a37f41c069649eae28bfa8f1d476a7348","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-08 06:23:03.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"df76db6b850be005dab327a065e80da44213a570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 07:45:38.000000000","message":"Patch Set 6: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5257308687fc4d76a749b7f248f5ac18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32db2a67ca2d4346992b338d564ed818 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/5873882e5d254c90b9f17e66f6db6a2a : SUCCESS in 57m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/86c20e3fe84646b69c016a60b992a27b : SUCCESS in 36m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/463529187cd3405f81e2bc8541616476 : SUCCESS in 51m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2dbe99adecd40658b14706426f17b8b : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b32dd28ea0e4c71b2bf875f017b0de3 : SUCCESS in 23m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8eab5db3aeff4625b36cfe5eb7b42bac : SUCCESS in 24m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ceda0857c1f14a25a6e2783a5210bd1a : SUCCESS in 5m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dcf84d4ddcee4fd28e5bf7b4e7279a70 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6aa93ed2a4ba46ccadb35888b346eb76 : FAILURE in 3m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b0dde4dd4bb6494b958d41dacd232a7a : SUCCESS in 22m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ae61eba3346c47e897933e079a5ae1dd : SUCCESS in 17m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2e524a4a2dcc43509e6926e8f414c421 : SUCCESS in 16m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d4878955f24348aebacb091c57f5545b : FAILURE in 3m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/40b664f6eec14253b29a19a2e1c607a1 : FAILURE in 3m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/efa6c610fdf24b609f98b9df70cb8562 : FAILURE in 2m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5d0ab4a151de4990a64d605477cd2d18 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2cafeb404b60478cba8d8955244b7051 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/095354a184684453bcae2bf8275b54be : SUCCESS in 15m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eeadb2fe17f74c4e974c840ed0f6b203 : SUCCESS in 7m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b30b2fbf84e446569e14c4e84ceaea97 : SUCCESS in 17m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8be105bc93a449dda50cc816e886667a : SUCCESS in 7m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/835e10a80cf2422ebcf1c1d0edea44db : RETRY_LIMIT in 3m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b41b4c35495e4af7b1d87bfbeb15b569 : SUCCESS in 1h 17m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c64fa19cf230472f92fca9e53e6424c1 : SUCCESS in 55m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d25e301e108a4e018252b2b5b71ebf0e : SUCCESS in 52m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd8af05b58e94f48883d5d9083ad4a6d : FAILURE in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ee944fb618d466cafb835e9c1e69c39 : FAILURE in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/64c3d998d21a48a1bb2b07ffc219d77b : FAILURE in 8m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4355c5ba7d5646ce96276b4d937b4c3a : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":6},{"id":"aa9b8fd0e3e9ca8768c4fd8fd52d19baf0561ca4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 07:57:37.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c8780bec51047f0ad437ca34b85ff2b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbc1347a31fb4ffcb14ed36f1d8103f7 : SUCCESS in 14m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/838e029b9da747ac91c7743ce17be610 : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b7b7d7ac69647aea3dd5295dce228fd : SUCCESS in 21m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a23aacde5484d0db0c5ae0ffb3e5b31 : TIMED_OUT in 1h 31m 07s","accounts_in_message":[],"_revision_number":6},{"id":"15f00c54413db81f3c6b608f75268a5a710ddea1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-10 07:01:14.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"5b54dff6d941cf9823e3442ac9e0569ab94f899d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 08:09:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/499d3dedb7844cdc82227be9db88ea81\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a271505ab0f74045a9abe66af91c4323 : SUCCESS in 16m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a5c588fd2e44974897c99981728c5df : SUCCESS in 1h 05m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/51119cb16b5a4d6c9d9fb14080f508f9 : SUCCESS in 20m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6955e67be474259b1710a7f3be50ba6 : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":7},{"id":"4a1360c56c559b830588eb0dac8987f7f226710b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 08:13:43.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cf40d9726d7407592442aea6507380b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee93180491ba446ab1f2655e4ec03ed7 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/d5c0825ca5e04fbabe3022d37508e107 : SUCCESS in 1h 02m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/94d6bbc4c94a4f99bb550ceca8917b2c : SUCCESS in 55m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e8374b8db304a6587c7d7d5efe0c7d4 : SUCCESS in 1h 02m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/defb778fa6d0494ea3c71327da51cd1f : SUCCESS in 17m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/093a293258714c129ad554ee98badb48 : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9dea352e1f1840a7930d8a5727525952 : SUCCESS in 19m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1275a8bc1b2b4be3b439d4d2812b2b5d : SUCCESS in 6m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/09831cd404a348b78c0946122eb6d3d6 : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/299a9739d3c846a68d7b6dc62c655f23 : FAILURE in 3m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6bd44bc709544bcd864fc0353411a379 : SUCCESS in 17m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/25168f2f1b994e42b781c7679196a6f1 : SUCCESS in 21m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9b750c5fe6cd4eb3a319b4070bd8d702 : SUCCESS in 21m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5079d319e7aa4657b3b58962bf003da6 : FAILURE in 5m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7ab7f1c2967d48158cc1abff6f88412b : FAILURE in 3m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e46f185fea5845c3bffba548d4afab7c : FAILURE in 3m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f2a8cc309fc54726b985bbdcd54e2f67 : SUCCESS in 14m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e6f7ad2ad22f4b54af394e6997b07bd0 : SUCCESS in 15m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/69c638ad38504f31ac5ecd760c1304c3 : SUCCESS in 17m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2b57fab970c04ba3a263895f05dd5d24 : SUCCESS in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d1452714ca1044d1af2a676af88d5412 : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/950574fa74e04b7c9b6418215e336e6b : SUCCESS in 6m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9530ae9b2dff47a39ca1c1df09afced3 : RETRY_LIMIT in 3m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0f9bd4a3d864b87a6cc56f53578a63f : SUCCESS in 1h 07m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/644114897aaa44ff854dfd3227c124b8 : SUCCESS in 55m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/11e194e996f5416386a64460434bf6eb : SUCCESS in 57m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a3e1acba568478c9e46c9c2d7af6fad : FAILURE in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/995ab2070fb54048afa9bec1764dafe6 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9605061b12524524a35a503e7714e399 : FAILURE in 8m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/536c65e6121145758460287c3543e961 : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":7},{"id":"57d310ba00c3fe20065c5b2b35c8b96d98284eec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-12 05:48:28.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"d691f053ba9020400c37806b6ceb6a838b5bed03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-12 05:53:20.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"faed705a6293bd71615d951e6a242050f10c356b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 07:18:14.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/836f9296fd0a4cbd944ee5d772c1fb61\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b4268cd513e4fcda5e6c4bde23200a1 : SUCCESS in 17m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b9d9866a4b3f47b685c0c7f8216b3171 : SUCCESS in 1h 13m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/01b97b496bea45e5a24b1a19eee4ff31 : SUCCESS in 18m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/69c4fa3e9be24c40b10c991ed2fad2a4 : FAILURE in 1h 22m 17s","accounts_in_message":[],"_revision_number":9},{"id":"f5fbbeb45f1b14795d78d658de111fd826b11b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 07:21:38.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/058b6d135ee94d619fcb1340af9dd283\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95ac56f7fde947f2907de84710f0837d : SUCCESS in 9m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/b960e5408a5f4dfb9724199e752aed5a : SUCCESS in 1h 08m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bde53b89719e4760876af10f9f70882c : SUCCESS in 48m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ea25fbde340494082af0fda661ae25a : SUCCESS in 1h 00m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e585133645e244e392689184301bcae6 : SUCCESS in 12m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b98bc262c4f47c2a2505f6165f4d0f0 : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f325d672d61a4e9e9dc194d53c32954c : SUCCESS in 20m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dd3d26ccb6954a58a7fe41981d42bbf3 : SUCCESS in 6m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3079dea5d6fa463f94dfeeb569284f1f : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/947e66e501314748a966d1fd79e1ea8b : FAILURE in 2m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/086ed838b67d4a739dfb0811e2478c13 : SUCCESS in 20m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4b08a1f58bf64fd297dc8087e5e8d85e : SUCCESS in 17m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9b32fcd4262e4f42bb4a947a776c7882 : SUCCESS in 17m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/49463181262e4ad190b7b9d550fdca45 : FAILURE in 3m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bfb70fe53fbe42bb9198caf99e0f1b93 : FAILURE in 2m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2a7f42cd93cb402ab9245ccf3178c1fc : FAILURE in 2m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ab1cba5e54954365be65da9ec0f45925 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/07669b302a924a78a997be7b5441e823 : SUCCESS in 16m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/178665dc520d4f048edb6d2aca7464da : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dcc113369ff14f2cbb1f620042a36853 : SUCCESS in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4d920775f0dc4b00a0c9d70e46d8c75a : SUCCESS in 15m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ac85dc1230f945c2b00595bf22f48e68 : SUCCESS in 6m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3e7ccda9ca8046bcb287251a1330d44c : RETRY_LIMIT in 3m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b161cd847bf49f1ad01247cbc127917 : SUCCESS in 1h 20m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/16b49e50f58d433f8ec85c3035dc3e86 : SUCCESS in 40m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ce882d59938c44e3831e0fbd89c7e112 : SUCCESS in 57m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e41a856fffdd4b5188c1e5b288495ba3 : FAILURE in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfeda493cd49415b94ff745235911923 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3e17470876a941c888a08f8e3d980d38 : FAILURE in 6m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32401a87cf044753b11196e2e932846f : SUCCESS in 43m 20s","accounts_in_message":[],"_revision_number":9},{"id":"fb7b02349e7329e92ca3d2ae45fd4fe43220de25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-26 06:29:35.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"505887445a124b0e6261c2add7504fbbabd501fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 06:54:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bb294a2e8224f29bf15f43dd2b87619\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a070e05911c34a73b65f444871c9e076 : SUCCESS in 15m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7301931ef01a45ad86c2c27a36fc3f9d : SUCCESS in 18m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7aa2c4d620aa45c7a1d0d43739a12ed2 : SUCCESS in 19m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0aae4441c1304fdbab6836c523723353 : SUCCESS in 19m 24s","accounts_in_message":[],"_revision_number":10},{"id":"3c370d0a1fa9f1bd4a2aa55210b97cc5eb49ca73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 08:36:00.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f170980b38b4445d97073435edb5d2bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31585b463a3745ef9ab145c53e718c00 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/8414a2787db743ce8f96519c2adab234 : SUCCESS in 47m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b52708002fcf49cea104b0f38122abb8 : SUCCESS in 42m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/887350ecd5db4fe5b023d560460e6efd : SUCCESS in 52m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74e56fab6303423e845cb12a0da86569 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61ce5777e5c14b14b3e1efebf75537cd : SUCCESS in 25m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2b5ed875c4f49489d57245ec827a0e1 : SUCCESS in 18m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d9c43e25b8c9476aa39f383f185cd848 : SUCCESS in 6m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/68b5f5d58d014e6a8dcdea397cb408c6 : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3a6ed95d2c2c45ee8ba9966c467f6ba5 : FAILURE in 3m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dbaffd24c66b4127b6dd91b4bcbe16f7 : SUCCESS in 22m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5b0b4c268e1c44e08aa3a2fd7dada586 : SUCCESS in 20m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bf5c2316f8eb41d4b5a43429f0416d4e : SUCCESS in 14m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2185eb83d9c74aff9d4c1d183956b33b : FAILURE in 3m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/71f999ba7ec94509be84658ed73d42cf : FAILURE in 3m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c6b4b6054fc942b0bcdb957d0b5f7723 : FAILURE in 3m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/63cc052d66f744a498c2747ac4fd05f4 : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d2c30e334b554a729d69127f265a055c : SUCCESS in 15m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3541d92e1ba74f96aff456366493ae41 : SUCCESS in 18m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f51bcf9e5016459a926f7689015d450a : SUCCESS in 7m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0d30c17490ea4350ba994b3d2250a14f : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c7a5008b43ee4f518ab5dbbef80f6f73 : SUCCESS in 7m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3320e65775ff404188a53cb4739b9483 : RETRY_LIMIT in 4m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb5a2fc3e5c84d28a07c9656c6462f3e : SUCCESS in 1h 08m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5312539a4e6a4f1baacaa0e7a12752fc : SUCCESS in 45m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/13884c67fbb74c1dbc1dd439e9781294 : SUCCESS in 58m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8150f8dc4e7244bd84097401db72d222 : FAILURE in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f231ff27331a4432b473ee1aff40d5a5 : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9dfa05cad1db4eabb9d0ba78f190436e : FAILURE in 7m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd8f50e526e948b38adc12d27e24ca1e : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":10},{"id":"610b7569bf4e36ec712be1a34852d5ddc8c4092e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-10-24 05:06:53.000000000","message":"Abandoned\n\nsquashed down","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"7bb2100d0ef1b4a67927081f98f5bbc6064d6aed","revisions":{"4862ee7f503c52542ffb607331bcbbe030583472":{"kind":"REWORK","_number":1,"created":"2023-04-20 07:15:13.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/880911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/880911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/880911/1"}}},"commit":{"parents":[{"commit":"224a9682318f14defef5ebc9cad677398538bcfa","subject":"Tracing: Add overview documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/224a9682318f14defef5ebc9cad677398538bcfa"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-20 07:11:38.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-20 07:11:38.000000000","tz":600},"subject":"WIP: tracing: More unit tests","message":"WIP: tracing: More unit tests\n\nAdded an activate_tracing(env) unit test helper to help activate and\ninterrogate tracing. Which is being used to test middlewares we\u0027ve\nalready targetted for instumentation (more then just importing the\nmixin).\n\nChange-Id: I07bd992868f9d8799d2dfa5e840bd6f1cf557723\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4862ee7f503c52542ffb607331bcbbe030583472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4862ee7f503c52542ffb607331bcbbe030583472"}]},"branch":"refs/heads/master"},"cc1d1f12fbe274f93599b6e8ede7afc86595c0df":{"kind":"REWORK","_number":2,"created":"2023-04-21 07:03:48.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/880911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/880911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/880911/2"}}},"commit":{"parents":[{"commit":"224a9682318f14defef5ebc9cad677398538bcfa","subject":"Tracing: Add overview documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/224a9682318f14defef5ebc9cad677398538bcfa"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-20 07:11:38.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-21 07:03:08.000000000","tz":600},"subject":"WIP: tracing: More unit tests","message":"WIP: tracing: More unit tests\n\nAdded an activate_tracing(env) unit test helper to help activate and\ninterrogate tracing. Which is being used to test middlewares we\u0027ve\nalready targetted for instumentation (more then just importing the\nmixin).\n\nCurrently working through SLO delete unit tests.\n\nChange-Id: I07bd992868f9d8799d2dfa5e840bd6f1cf557723\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc1d1f12fbe274f93599b6e8ede7afc86595c0df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc1d1f12fbe274f93599b6e8ede7afc86595c0df"}]},"branch":"refs/heads/master"},"c104e74a5f26e3d4267f4846ab839eaab5422545":{"kind":"REWORK","_number":3,"created":"2023-04-24 09:55:55.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/880911/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/880911/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/880911/3"}}},"commit":{"parents":[{"commit":"224a9682318f14defef5ebc9cad677398538bcfa","subject":"Tracing: Add overview documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/224a9682318f14defef5ebc9cad677398538bcfa"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-20 07:11:38.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-24 09:55:10.000000000","tz":600},"subject":"WIP: tracing: More unit tests","message":"WIP: tracing: More unit tests\n\nAdded an activate_tracing(env) unit test helper to help activate and\ninterrogate tracing. Which is being used to test middlewares we\u0027ve\nalready targetted for instumentation (more then just importing the\nmixin).\n\nJust finished the SLO unit tests.\n\nChange-Id: I07bd992868f9d8799d2dfa5e840bd6f1cf557723\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c104e74a5f26e3d4267f4846ab839eaab5422545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c104e74a5f26e3d4267f4846ab839eaab5422545"}]},"branch":"refs/heads/master"},"74e7e79e4910272aac6b7d45d057ba7840e05b98":{"kind":"REWORK","_number":4,"created":"2023-05-04 07:27:03.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/880911/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/880911/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/880911/4"}}},"commit":{"parents":[{"commit":"bc546092790fd8363222ac675318bd686443d568","subject":"Tracing: Add overview documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc546092790fd8363222ac675318bd686443d568"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-20 07:11:38.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-05-04 07:26:39.000000000","tz":600},"subject":"WIP: tracing: More unit tests","message":"WIP: tracing: More unit tests\n\nAdded an activate_tracing(env) unit test helper to help activate and\ninterrogate tracing. Which is being used to test middlewares we\u0027ve\nalready targetted for instumentation (more then just importing the\nmixin).\n\nWorking through controller/obj.py unit tests.\n\nChange-Id: I07bd992868f9d8799d2dfa5e840bd6f1cf557723\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74e7e79e4910272aac6b7d45d057ba7840e05b98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74e7e79e4910272aac6b7d45d057ba7840e05b98"}]},"branch":"refs/heads/master"},"6b0e1961cc3b1f1d62cce51c30ba691efe228f58":{"kind":"REWORK","_number":5,"created":"2023-05-05 07:16:06.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/880911/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/880911/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/880911/5"}}},"commit":{"parents":[{"commit":"bc546092790fd8363222ac675318bd686443d568","subject":"Tracing: Add overview documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc546092790fd8363222ac675318bd686443d568"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-20 07:11:38.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-05-05 07:15:27.000000000","tz":600},"subject":"WIP: tracing: More unit tests","message":"WIP: tracing: More unit tests\n\nAdded an activate_tracing(env) unit test helper to help activate and\ninterrogate tracing. Which is being used to test middlewares we\u0027ve\nalready targetted for instumentation (more then just importing the\nmixin).\n\nWorking through controller/obj.py unit tests.\nAlso added a proxy error_limiting nodes test.\n\nChange-Id: I07bd992868f9d8799d2dfa5e840bd6f1cf557723\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b0e1961cc3b1f1d62cce51c30ba691efe228f58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b0e1961cc3b1f1d62cce51c30ba691efe228f58"}]},"branch":"refs/heads/master"},"073cd594c545ee9932d52d6ff65b17cbf834b6ac":{"kind":"REWORK","_number":6,"created":"2023-05-08 06:23:03.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/880911/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/880911/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/880911/6"}}},"commit":{"parents":[{"commit":"bc546092790fd8363222ac675318bd686443d568","subject":"Tracing: Add overview documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc546092790fd8363222ac675318bd686443d568"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-20 07:11:38.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-05-08 06:22:02.000000000","tz":600},"subject":"WIP: tracing: More unit tests","message":"WIP: tracing: More unit tests\n\nAdded an activate_tracing(env) unit test helper to help activate and\ninterrogate tracing. Which is being used to test middlewares we\u0027ve\nalready targetted for instumentation (more then just importing the\nmixin).\nAlso added some request_trace middleware tests\n\nChange-Id: I07bd992868f9d8799d2dfa5e840bd6f1cf557723\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/073cd594c545ee9932d52d6ff65b17cbf834b6ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/073cd594c545ee9932d52d6ff65b17cbf834b6ac"}]},"branch":"refs/heads/master"},"58deca696fd8a2489df754ad1b2c1a517f696300":{"kind":"REWORK","_number":7,"created":"2023-05-10 07:01:14.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/880911/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/880911/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/880911/7"}}},"commit":{"parents":[{"commit":"bc546092790fd8363222ac675318bd686443d568","subject":"Tracing: Add overview documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc546092790fd8363222ac675318bd686443d568"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-20 07:11:38.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-05-10 07:00:46.000000000","tz":600},"subject":"tracing: More unit tests","message":"tracing: More unit tests\n\nAdded an activate_tracing(env) unit test helper to help activate and\ninterrogate tracing. Which is being used to test middlewares we\u0027ve\nalready targetted for instumentation (more then just importing the\nmixin).\nAlso added some request_trace middleware tests\n\nChange-Id: I07bd992868f9d8799d2dfa5e840bd6f1cf557723\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58deca696fd8a2489df754ad1b2c1a517f696300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/58deca696fd8a2489df754ad1b2c1a517f696300"}]},"branch":"refs/heads/master"},"52e67e7d13c24e581b52e176f9a4d6bffddc9871":{"kind":"REWORK","_number":8,"created":"2023-05-12 05:48:28.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/880911/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/880911/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/880911/8"}}},"commit":{"parents":[{"commit":"bc546092790fd8363222ac675318bd686443d568","subject":"Tracing: Add overview documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc546092790fd8363222ac675318bd686443d568"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-20 07:11:38.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-05-12 05:47:10.000000000","tz":600},"subject":"tracing: More unit tests","message":"tracing: More unit tests\n\nAdded an activate_tracing(env) unit test helper to help activate and\ninterrogate tracing. Which is being used to test middlewares we\u0027ve\nalready targetted for instumentation (more then just importing the\nmixin).\nAlso added some request_trace middleware tests\n\nChange-Id: I07bd992868f9d8799d2dfa5e840bd6f1cf557723\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52e67e7d13c24e581b52e176f9a4d6bffddc9871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/52e67e7d13c24e581b52e176f9a4d6bffddc9871"}]},"branch":"refs/heads/master"},"e171c4266780f9188b87d9d05224f0e0c877223a":{"kind":"REWORK","_number":9,"created":"2023-05-12 05:53:20.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/880911/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/880911/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/880911/9"}}},"commit":{"parents":[{"commit":"bc546092790fd8363222ac675318bd686443d568","subject":"Tracing: Add overview documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc546092790fd8363222ac675318bd686443d568"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-20 07:11:38.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-05-12 05:52:50.000000000","tz":600},"subject":"tracing: More unit tests","message":"tracing: More unit tests\n\nAdded an activate_tracing(env) unit test helper to help activate and\ninterrogate tracing. Which is being used to test middlewares we\u0027ve\nalready targetted for instumentation (more then just importing the\nmixin).\nAlso added some request_trace middleware tests\n\nChange-Id: I07bd992868f9d8799d2dfa5e840bd6f1cf557723\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e171c4266780f9188b87d9d05224f0e0c877223a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e171c4266780f9188b87d9d05224f0e0c877223a"}]},"branch":"refs/heads/master"},"7bb2100d0ef1b4a67927081f98f5bbc6064d6aed":{"kind":"REWORK","_number":10,"created":"2023-05-26 06:29:35.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/880911/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/880911/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/880911/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/880911/10"}}},"commit":{"parents":[{"commit":"4da0296b01a3d14821b82ec17a74cf00c8d76510","subject":"Tracing: Add overview documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4da0296b01a3d14821b82ec17a74cf00c8d76510"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-04-20 07:11:38.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-05-26 06:29:10.000000000","tz":600},"subject":"tracing: More unit tests","message":"tracing: More unit tests\n\nAdded an activate_tracing(env) unit test helper to help activate and\ninterrogate tracing. Which is being used to test middlewares we\u0027ve\nalready targetted for instumentation (more then just importing the\nmixin).\nAlso added some request_trace middleware tests\n\nChange-Id: I07bd992868f9d8799d2dfa5e840bd6f1cf557723\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bb2100d0ef1b4a67927081f98f5bbc6064d6aed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bb2100d0ef1b4a67927081f98f5bbc6064d6aed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
