)]}'
{"id":"openstack%2Fswift~881035","triplet_id":"openstack%2Fswift~master~Ia2b3cc4f719347705dec15f1fc97d982e72bfb34","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-04-20 19:36:14.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ia2b3cc4f719347705dec15f1fc97d982e72bfb34","subject":"CI: fix lower-constraints job","status":"MERGED","created":"2023-04-20 17:27:21.000000000","updated":"2023-04-20 22:06:17.000000000","submitted":"2023-04-20 22:04:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881035","meta_rev_id":"f082b7de20b110b7fe2bd85f6ca43cf9cd244e20","_number":881035,"virtual_id_number":881035,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 22:04:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-20 20:52:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-20 20:52:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 17:37:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-20 17:37:36.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-20 17:37:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d503500568287a1aaa200a312073cf2ac99bbea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-20 17:27:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46aac6035f19a8dfbdd0b92780ff8c9d2d973061","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-20 17:37:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0da1e580921a6ebb7c6efeb73bf95b971f4ec4d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 17:37:33.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@master\", line 677, column 31","accounts_in_message":[],"_revision_number":2},{"id":"84808b0ff833232f152d0ce5c2bc86182052cd57","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-04-20 17:37:36.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@master\", line 677, column 31","accounts_in_message":[],"_revision_number":2},{"id":"3faf18abd4bf3897dec61c6fe152ec9824d91b0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 17:37:40.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@master\", line 677, column 31","accounts_in_message":[],"_revision_number":2},{"id":"541eaad6ab282447dc84842c14305e85fc345e6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-20 17:39:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6ca1a6827b7a445cae6b592a3a304ed7e7d32d87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-20 17:52:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"926531ccba7921e45712549452ff416a78a811c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 18:58:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/038d7a74c3f743d9905edbee9f3946b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/128a15e66dd4460193f8ec61037e24ab : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/5796bb4ef5a742269fa067dd9e8575e0 : SUCCESS in 55m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bfb1412fb2e747429e87463b95a72c8c : SUCCESS in 46m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8013f44ff09d4fc1af1eade0e81de82f : SUCCESS in 1h 00m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66a0bde306894a9d85794fb8383f8323 : SUCCESS in 16m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/602bcaa9cd224bd6ad906fb692fea822 : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2315537cb464d6c84beab02001229d8 : SUCCESS in 13m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b033c6ad11244ee5bd6e22447b2abd90 : SUCCESS in 9m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1760376dc0f24fbdb0a7bc3fbcba78c2 : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/84f65483edec48b9b3db317c5a076422 : SUCCESS in 16m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a6ec647fc8cf4b12a3a9dc0ee0c943d4 : SUCCESS in 17m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/28d7293d4c8f483481af22c4c87704ae : SUCCESS in 14m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/54233048658a4f92a83528a17b027ec1 : SUCCESS in 17m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d20ef80e673e4d548759572de095ee80 : SUCCESS in 18m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1caa0afa53eb4f57a670510c1a736ac0 : SUCCESS in 18m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4bbd0c03a8e84637b55548feddfd8788 : SUCCESS in 17m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a24678bc8aed4429ad0f82dd7c824b67 : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a9b10b9028c241a1a43af23b06b426ea : SUCCESS in 13m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/699a57d94d654669bce6d1e5fe0d9917 : SUCCESS in 17m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3a6d3a5351e14a4487df540600d6bb6e : SUCCESS in 5m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b311b1cf7ca14fd0bcfad94e9ddf1fcf : SUCCESS in 15m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/18eb17cee8c34d31afd19769faf3c078 : SUCCESS in 7m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/87689f7f9abd4823af22d9c57d2b8790 : FAILURE in 51m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9157914f92644c6c86a2b5837c2d973c : SUCCESS in 52m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0ce82a6dcbdf481aa9e4b1cc3476c18c : SUCCESS in 54m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9818d46fd6104c46babd77d8e44dc229 : SUCCESS in 51m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52c14e27a4b742d49d73d5b254c8d5e7 : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a71ac1a126b1497dbf0378018a3dd3d7 : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bd0877fa5fec48248f6462bca4c962d8 : SUCCESS in 19m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8087def41f6e488dad208138c27a7f72 : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":4},{"id":"78786c9cf3fb21aa0910a1ccd01ac0e7953d4c4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 19:00:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc0c29a67883423aa9d2846d8763dfc3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d83d4c7e241488b8006d55983f63d4f : SUCCESS in 14m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c48415e72d754b13b76ca9ec3a75c4b0 : SUCCESS in 1h 02m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/318d693aef8f42a08cdea4a8e749556b : SUCCESS in 24m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3751f3a9da654292960ce508ca922cfb : SUCCESS in 1h 06m 30s","accounts_in_message":[],"_revision_number":4},{"id":"2edbf2dc1045bea3ff95071e37a3ba3086276e3e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-20 19:36:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a0949a98cecf0d6b5df3f007ebcf3c234b77b998","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 19:57:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58452b7517ea4247b3011cdb05b50367\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/085cb19507674b878ebe96df4886b122 : SUCCESS in 13m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71fe1711e17b4ae497c755b90a0e5f89 : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/506baab58c064d68aafad2ff74833881 : SUCCESS in 19m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee71b42110c74ddbbc361e21d3d54bbf : SUCCESS in 18m 02s","accounts_in_message":[],"_revision_number":4},{"id":"9d9d309b43b0a0337e4e45e0331be4372ac7d5e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 20:50:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be50f0d76f6e4775a56f8243bc52303e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e88deb625fb4c32a7ae1d4eb096f664 : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/7d8e35d2c077477b9d2cb6812655a56b : SUCCESS in 55m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/62e85c3a6ba04acba4f31117db52d9cd : SUCCESS in 47m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da52206890b841b2a36637c39574789d : SUCCESS in 59m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9316632adb4543939621e45a43d0bb34 : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/76c7af1d30fe408695502f78f3eac157 : SUCCESS in 16m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7262bc109b134eea93296ca4b6579e6e : SUCCESS in 20m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/28c6b05ea3ef4274aff8b7e30b111109 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/43560a7da5ad46f896f54de385c454ef : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83895c0efda34ac2849a75d29836e2d6 : SUCCESS in 16m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/58a8a3482980435fa893d8356599a4bc : SUCCESS in 16m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5e208131d7fd4f7c94d5a9c282ff3ff7 : SUCCESS in 15m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/089a66756e14422b8baddb7ec674b100 : SUCCESS in 15m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/979a422e0e904c7b8e1be985031dd755 : SUCCESS in 19m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7e3dd6e8ad4248b4b1c898b8f46bcdff : SUCCESS in 15m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fceadbd489ea4bcba2069fd063fd4018 : SUCCESS in 16m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c9b17e446c0a4c37a4d0f812550f3cdb : SUCCESS in 14m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f2c6f5b38f5743d89a0ee5911111d841 : SUCCESS in 16m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5e68214ae466470b80927f0a1530a4d9 : SUCCESS in 14m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/06c6953dc6954b61a81e35f64ebcc4b5 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c741e166a7214ad8b44ba6aaccaa290b : SUCCESS in 14m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b619e72e353f439a8e68d2f195916c74 : SUCCESS in 7m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a9b6f2bcdb1a412f98cb90b2fc000cdc : SUCCESS in 56m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f14686198bc481cad1f212ea288af7e : SUCCESS in 1h 01m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/63715b5b861946758b14da13b6432cbd : SUCCESS in 58m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b4a7e423bc674505aa3d8291337ab453 : SUCCESS in 58m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4997c34dc2d4d2e911ef7f702b7d292 : SUCCESS in 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b13b25eeb18b464cb503abd3aee9a0b3 : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a829a9e787c4c7dad6adf3be286adc8 : SUCCESS in 19m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbd5f1ccada14c7b884ee21640a9bde9 : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":4},{"id":"1cbf2af6e5d1fd2802090d5f09c9bc26dda2e70d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-20 20:52:30.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"647803ee9591e722d3c6a452b70383d428d1c3c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 20:53:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ae828a0ec85305529f9250b6e3e74f4f99ca4e20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 22:04:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7f28361fa3e4d09a6c7769341ae7f1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9978439c1afd47c89c3fb5a51bb18aa9 : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/ff19a37960a74aab87f24be4fecc6076 : SUCCESS in 51m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5b440f0881ec4639be2455e1c7fb15da : SUCCESS in 43m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fc988f7a6d745d2b579af4fc0fc5fe4 : SUCCESS in 52m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e326f7681b26428388ea4e04f01431b8 : SUCCESS in 8m 37s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ea81708b4cd64990a5c95661715f032f : SUCCESS in 7m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c8c147ac21994e379f271c407c463f3f : SUCCESS in 34m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0f11f26a30f64fecbc3830461d899a0a : SUCCESS in 34m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/432f2a8c4f5d43ee8a8167af100c598d : SUCCESS in 13m 46s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/72dd7585f7d040df8102a4b4a3cf060e : SUCCESS in 32m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8c3bb17cbcaf4113b0309f768123e3ad : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/453b0943f3ea420dae5d08acb542df17 : SUCCESS in 17m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2b42c38740a3401cae747f482eea6251 : SUCCESS in 16m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/386ce41b4d354057b2116faf5c2b5f72 : SUCCESS in 29m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9550daceaa2340d0bf83b28def900875 : SUCCESS in 15m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/53ff706e2e954465b4fac16f23f8c520 : SUCCESS in 16m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ee269af8105b4663801a93516d112bed : SUCCESS in 8m 19s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9c8ddea3ee9049a28f62ba269de62687 : SUCCESS in 8m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/035cc1d1bf09426d9e3f0634dd435ecd : SUCCESS in 57m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ee728efaf4540538c67132e0fa403d2 : SUCCESS in 54m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/356f5247dfaa443693f402e3b957bda0 : SUCCESS in 48m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9ba6dcbab53c49e0a1f10c905684f92c : SUCCESS in 45m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/469c320481a642a9850649b9c3d71d56 : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/634a26babbdf4467a7e7741836b785c1 : SUCCESS in 5m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a92a28b691ac43669174799ec7ee536b : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":4},{"id":"38d41c17bc172d81fa25ec451ccf31d44aa44c01","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 22:04:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f082b7de20b110b7fe2bd85f6ca43cf9cd244e20","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 22:06:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f695fc1754184b87a6f5a5c6cd7952e0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b16e35d72d435085087072d2bbdbc6 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6d42fa6cf25a483da9d8082fe7fd3746 : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/772ee76f9bd847cfb4934706b303efb3 : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bfc13862792e9e574f1967525f420789b78c44e1","revisions":{"9ffb2c92cf0a7dba6765ce3cd69bdeb446be7674":{"kind":"REWORK","_number":1,"created":"2023-04-20 17:27:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/881035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/881035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/881035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/881035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/881035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/881035/1"}}},"commit":{"parents":[{"commit":"4b6f54d063c33e13ced7d6dd5208921680f56ca6","subject":"Merge \"ssync: Round-trip offsets in meta/ctype Timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b6f54d063c33e13ced7d6dd5208921680f56ca6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-20 17:27:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-20 17:27:13.000000000","tz":-420},"subject":"CI: fix lower-constraints job (hopefully)","message":"CI: fix lower-constraints job (hopefully)\n\nChange-Id: Ia2b3cc4f719347705dec15f1fc97d982e72bfb34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ffb2c92cf0a7dba6765ce3cd69bdeb446be7674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ffb2c92cf0a7dba6765ce3cd69bdeb446be7674"}]},"branch":"refs/heads/master"},"dd15501f8c5d835886dfbc036349fd1fb977ceb4":{"kind":"REWORK","_number":2,"created":"2023-04-20 17:37:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/881035/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/881035/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/881035/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/881035/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/881035/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/881035/2"}}},"commit":{"parents":[{"commit":"4b6f54d063c33e13ced7d6dd5208921680f56ca6","subject":"Merge \"ssync: Round-trip offsets in meta/ctype Timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b6f54d063c33e13ced7d6dd5208921680f56ca6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-20 17:27:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-20 17:36:57.000000000","tz":-420},"subject":"CI: fix lower-constraints job (hopefully)","message":"CI: fix lower-constraints job (hopefully)\n\nChange-Id: Ia2b3cc4f719347705dec15f1fc97d982e72bfb34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd15501f8c5d835886dfbc036349fd1fb977ceb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd15501f8c5d835886dfbc036349fd1fb977ceb4"}]},"branch":"refs/heads/master"},"d5af6a3127a9eecaf7bea3a1a50598305d914065":{"kind":"REWORK","_number":3,"created":"2023-04-20 17:39:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/881035/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/881035/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/881035/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/881035/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/881035/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/881035/3"}}},"commit":{"parents":[{"commit":"4b6f54d063c33e13ced7d6dd5208921680f56ca6","subject":"Merge \"ssync: Round-trip offsets in meta/ctype Timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b6f54d063c33e13ced7d6dd5208921680f56ca6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-20 17:27:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-20 17:39:03.000000000","tz":-420},"subject":"CI: fix lower-constraints job (hopefully)","message":"CI: fix lower-constraints job (hopefully)\n\nChange-Id: Ia2b3cc4f719347705dec15f1fc97d982e72bfb34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5af6a3127a9eecaf7bea3a1a50598305d914065"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5af6a3127a9eecaf7bea3a1a50598305d914065"}]},"branch":"refs/heads/master"},"bfc13862792e9e574f1967525f420789b78c44e1":{"kind":"REWORK","_number":4,"created":"2023-04-20 17:52:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/881035/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/881035/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/881035/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/881035/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/881035/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/881035/4"}}},"commit":{"parents":[{"commit":"4b6f54d063c33e13ced7d6dd5208921680f56ca6","subject":"Merge \"ssync: Round-trip offsets in meta/ctype Timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b6f54d063c33e13ced7d6dd5208921680f56ca6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-20 17:27:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-20 17:45:55.000000000","tz":-420},"subject":"CI: fix lower-constraints job","message":"CI: fix lower-constraints job\n\nSwitch from focal back to bionic so tox gets installed with py36. This\nensures that we get virtualenv\u003c20.18.0 so we can still create py27\nenvironments (support for which was dropped in virtualenv\u003d\u003d20.22.0).\n\nNote that this also means we can drop our tox\u003c4 pin, as newer tox does\nnot support py36, either.\n\nChange-Id: Ia2b3cc4f719347705dec15f1fc97d982e72bfb34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfc13862792e9e574f1967525f420789b78c44e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfc13862792e9e574f1967525f420789b78c44e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
