)]}'
{"id":"openstack%2Fswift~881141","triplet_id":"openstack%2Fswift~master~Ia87caebc9a3c70602209daeda97fc588f4bafc76","project":"openstack/swift","branch":"master","topic":"fix-nonlocal-parts_iter-patch2","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-04-24 14:19:26.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-04-24 19:49:03.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-04-24 19:49:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia87caebc9a3c70602209daeda97fc588f4bafc76","subject":"Extract some closures to methods","status":"MERGED","created":"2023-04-20 19:49:39.000000000","updated":"2023-04-24 19:50:05.000000000","submitted":"2023-04-24 19:49:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"881141-fix-nonlocal-parts_iter-patch2","meta_rev_id":"c3dc36602be98063f89a7a28bb58fbc28d711ff7","_number":881141,"virtual_id_number":881141,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-24 19:49:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2023-04-24 15:51:54.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2023-04-24 14:30:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2023-04-24 14:30:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 20:43:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-20 20:59:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 23:44:15.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-04-24 14:15:27.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2023-04-24 17:29:15.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"897f14f5f3f1e7f1245fd350fef7afee322ca1e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-20 19:49:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1c303eaec4748a5cfec4e07f4a689b5b7dbd347","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 20:43:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3dbf4c82e4245e28bf69065582f8340\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d815895d9ef4d06bdcfeb1b16775c19 : FAILURE in 51m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29e6f188c15d43beb48ff9dd5f35b89c : FAILURE in 13m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e31069bffc4443cbae9e3f31607fad34 : SUCCESS in 51m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f9b3fe067e84ede971872d911f38e8d : SUCCESS in 22m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a71a937c7bceaa45550e45aea20fe085eaca0f1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 20:59:14.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/72c6f84bf4ba4feda489401eb541482c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce33c6af2f6e49ffaa6831d268fb811e : SUCCESS in 10m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/5a7732ac4690422897cbba3440fc2eb1 : SUCCESS in 1h 02m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/462e1757a2634754be2dae0e4deb34e0 : SUCCESS in 52m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1816995bed04798b02117e77649d93c : SUCCESS in 58m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c87a5ac1ae24c0082977d4c6dc12c5d : SUCCESS in 28m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c744f6af450468fbe709fa57e321ff4 : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6afe9b5280124cb3b53b864f4b5a5d00 : SUCCESS in 15m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ba3014e443a94a95940f2a91a5e6d0a6 : SUCCESS in 9m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1e70b1b1b990454a805d7ba25ca7d62d : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/57b686406542469db2bd82cbf28643ff : FAILURE in 14m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e510fa7b305c404385e1f26e3e38d80d : FAILURE in 17m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e74988318d974ae29e19803c411a4f8d : FAILURE in 14m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/53f331ecfd884b358c5b84baf81b5379 : FAILURE in 15m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/03ea247d8206485e821cd28152a6f48b : SUCCESS in 14m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/07f1ed9b869542a9b51ff96d6df8722c : SUCCESS in 16m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bc4432f6d45649d79c4e2467a84b0e2c : SUCCESS in 17m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9af34afc395049698436b7e342512c90 : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a72b03c41af74e4594b7aac9044be43d : SUCCESS in 14m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/436cfc13ebfc44568b3c37b689dc9dd1 : SUCCESS in 14m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f19d9495db6a4a4f959b3403873e7aa2 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6ba93f71c0e3489b9264e5209abdb5c1 : SUCCESS in 16m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ecd560192dd24820b47cbb4db08b7c57 : SUCCESS in 7m 06s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7456e3a838d543b6acf708de9caada71 : SUCCESS in 54m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8139af699c5c4afc9146990c2dea4f9f : SUCCESS in 50m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9ee10da31834d79bb4e2414b985f068 : SUCCESS in 51m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f9d4f78b8ee64c70a8c4fc1e6aba4fb0 : SUCCESS in 1h 00m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbbdaf4f949c4b9981e0b933e188bf8a : FAILURE in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce8b14a7ebde4a6484125624c19d8883 : FAILURE in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4635a83b0f3c4fedad11037e21e8486b : SUCCESS in 19m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e30cd4c3a472458fa880c8503ae8978b : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1a34817444612304945aef4528b038b9972c178d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-20 23:33:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"41027c38e95185e5a25b5dd55794004450703b1b","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-20 23:44:15.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a8bfa75361a2a76667f91f62c03ef33f0be2320","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 00:25:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c9171085f334f38a961498dffeff364\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51d6f675ad2e4b48a40e8fcdf812d5ac : SUCCESS in 14m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48ac25e135fa4f35a901a202e74b12ad : SUCCESS in 50m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5abb0e5a4134745b294ee4b53c42411 : SUCCESS in 18m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/30dbdcd75ebb4985a46372329124f300 : SUCCESS in 17m 38s","accounts_in_message":[],"_revision_number":2},{"id":"124c81ec5ac4dbebe7c8e1c202ff9fb92d93968c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 00:33:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4941f13321e4e6085686aad0a6acdb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32b02802f8d74fbd8f7681af73344327 : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/af7e0883fee24859b5f94b2ca294025f : SUCCESS in 42m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6dae7e87f95d460096957292e0a15a18 : SUCCESS in 47m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9301041a20144f8bc0546422a38b83f : SUCCESS in 47m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac3efae2e20b40999478ea2fa04f518d : SUCCESS in 21m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5450287b3a045518765dd2edb1702bf : SUCCESS in 24m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef9bc8b9a51845dca2e27571277abcc1 : SUCCESS in 24m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ed2c0176db7d421b97c7c1064f60eb75 : SUCCESS in 7m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c82bbcc0f7b749b69469937ca129a728 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0431a9d5a7dd4df292802782b8433331 : SUCCESS in 16m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/505f351ac678407ea977d46f43d03919 : SUCCESS in 16m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c0b456e573054fc8a36e3f5b86303468 : SUCCESS in 15m 46s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/75ce2b456fa54ab9a3270c26ca4d2d9b : SUCCESS in 15m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d1a3ac01f4ab445d97d2e6b030521e0c : SUCCESS in 20m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eb10c2f67a8d443e820910c2f9bf5ec7 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/898be840bbac4a2cbfe9b41c15e2ae80 : SUCCESS in 17m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9553aedecf394f15abad1ef7e334a871 : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8e341b0e982640f8ab26c1f9458da140 : SUCCESS in 16m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7dddc70bb8224fc680065b70f83f0bd3 : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/023d000d6d04496b8c0ed9cde5eaf352 : SUCCESS in 6m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f04d550810d54e25a1be786b99713d28 : SUCCESS in 14m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5a7572ec03ac48e78fd34095583e317a : SUCCESS in 7m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fc1d54e0f3bb4af8b1057cf8a7c6f964 : SUCCESS in 53m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/dfebbca7cec74c7e88eedf3065ac3ac2 : FAILURE in 57m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dec8d3171f074f7f93af94241224bd49 : SUCCESS in 43m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3f9f474e192d4fe8a5617cdcbadf09d0 : SUCCESS in 42m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1565927264b24bc990626a737b3495dc : SUCCESS in 16m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48543d879836410482188e076a772d30 : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9bd3033044f0467d8cc96ed6b692ef97 : SUCCESS in 23m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bac76e5caed472a83921fa13ebc2e85 : SUCCESS in 51m 51s","accounts_in_message":[],"_revision_number":2},{"id":"237b401535e7082902f1f62f122fd1e191143bac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-24 14:10:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5ec70d36e681c97fd844fc95a4cedc38b00a6532","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 14:10:23.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 881141,3","accounts_in_message":[],"_revision_number":3},{"id":"9bd6dd743c5be6ccd85d2f172959c7a34f92b2c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 14:10:34.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 881141,3","accounts_in_message":[],"_revision_number":3},{"id":"7abfa6855854d9135f91c49b5eebd8d9a703b4fa","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-24 14:15:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"817e4cd53c192b46060a3b451f6db4805904ed5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 14:16:31.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 881141,3","accounts_in_message":[],"_revision_number":3},{"id":"a669ebd4daa461155c23a5b0c64c9d23a7e4579b","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-24 14:19:26.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"65af7372ada5357a0113ed9041d8257119d38f4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-24 14:26:55.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"bfcbfab95b1913f4fa699a6706fe3883bf20573e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-24 14:28:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fd1958454541427b8f084616989b344cf11910a2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-24 14:30:25.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"50d4b842d7c9047f72de30aa77241b18bfcc737f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 15:43:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdbe14fe254c46c983b4a873fc8b9f80\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/923412959167498b95bf33df40b14691 : SUCCESS in 14m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92e538e9d6a942f58cb419fd6aaa9bc3 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/502dd10f87a142d0a2ab36d3130c41cf : SUCCESS in 1h 14m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb2f758f1a854e37b38a74bd2657fb58 : SUCCESS in 22m 01s","accounts_in_message":[],"_revision_number":4},{"id":"eeea399a58c529f13372dd094706de334e4127fa","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-24 15:51:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"78c71c3342586ef7bf24c8cde9eadf2edad01d3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 16:40:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5e00f5c2ef14110bef46c30597051fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b2fd58061542d2a290a2149e8de08c : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/bbafe2d6be9040b69812432ba76ae53c : SUCCESS in 1h 03m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9a855953ab9f4fa0963f422c61be92a0 : SUCCESS in 49m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d072d64929414eefa949a81ba442e78f : SUCCESS in 57m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f37a81c6cf7149d4bd4e7cd1fa2b4c14 : SUCCESS in 16m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25b9974ea7974d16b724c31d0541bc0c : SUCCESS in 15m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a30f48c0f20248649212bb5b76af686c : SUCCESS in 15m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4403bd814bba446a874a6aba79c78436 : SUCCESS in 9m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c2e4abe4f376432aa544bafa74ec03e2 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2325fe41a66241b1808f6505b5c9c441 : SUCCESS in 28m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d0a2ad43bc3d44099744a042f9518be8 : SUCCESS in 18m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb9fadf3dcab499992a899792bea3d96 : SUCCESS in 28m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4ee1ee3e24a047969083c4dc0cb01bce : SUCCESS in 17m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cef9a765c3ba4472833549e975180710 : SUCCESS in 21m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7edf8dc8192f4080959bef7f34a19168 : SUCCESS in 28m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9091202b41474840ac3d3a5336932f04 : SUCCESS in 30m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9e69adc7d61e4c228830afc63f01a919 : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/abfece28520d48848edec6c62dfe3d27 : SUCCESS in 12m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eaea7add274d4a758a78f55ef96b5a6e : SUCCESS in 15m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/588e8b32da8a445e9edd82cd55482ab0 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eeb847f5f4af466fa48d843c93f13f99 : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/efcedb0300bd4ecc836874428e81e5fc : SUCCESS in 6m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8235ce1f0abe41dd927ef2b431cb4a24 : FAILURE in 1h 01m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/14783617077f4ab897eca565bcf6a15f : TIMED_OUT in 2h 01m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/99fe1891e6ea478f8935a41ac2c3da78 : SUCCESS in 52m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/adb4576aad3a4cc8af40ce9c661e4c0d : SUCCESS in 50m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b82bd1cd00d4b4393919b82317ce780 : SUCCESS in 22m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27044d8d20d74c3a93a078ff41d29881 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30d2305c228540a7b58f85f9002364e5 : SUCCESS in 18m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46ce6fd0076e458e8a62038a48cf5a6a : SUCCESS in 53m 15s","accounts_in_message":[],"_revision_number":4},{"id":"8c4b2a2126ea31637e75982f11b38b57d1aa662a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-24 17:29:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a00ad6922f44e8cba8c8335b8bb26e0b965e6fb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 18:33:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90bc3ccaf6ac493f9ade6878ef71771c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3423469b7644b499d49fcde4473589a : SUCCESS in 14m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/079e80bcc595405f99521cde9770e32c : SUCCESS in 56m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ae0e522d85d4407a6ed76b66fb12655 : SUCCESS in 18m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4de890d0098b4ae2ad45c23eafe7ec09 : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":4},{"id":"f2c917204c6d6a20848f75d620f25dc9ce05b45b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 18:37:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b564bc68f8b44373b1abd7672593b12a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67d2b819bac44de59e33e0d4cfbd6329 : SUCCESS in 7m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/e85ef0b191dc4222966372690d9e13db : SUCCESS in 1h 02m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c68ca2e4bca94aceb95f4e3aca3fbdb8 : SUCCESS in 36m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7ec516b7d1c4aa39703daa59bcbc144 : SUCCESS in 1h 02m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9359e0a9157d4615962b1e5d900519e1 : SUCCESS in 17m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9012c5fa6824cc78338667f6e1ba06e : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3dc7fcf4859a4167b4dc143e7952ab67 : SUCCESS in 18m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c6f51066f8ad4d3baf46ab232f55fe94 : SUCCESS in 7m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9ed7a00874064c099f780f232417c658 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/00b56f59825947f388f5f8f7f4d6d48e : SUCCESS in 22m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/64e9ea47256642b397cf4174561f240c : SUCCESS in 16m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e2e15e5233d14d518fdbd1a8cca1181e : SUCCESS in 14m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5d98ddb66e304e8485b70714cd6e7360 : SUCCESS in 23m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d35df73e4ad8475b8ac1ba5be419e30d : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/751923b2f0284e5e8e21ecd3873de86d : SUCCESS in 17m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dec99ee13bef42e2af3e7de9f33f8fbf : SUCCESS in 23m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c7831b523bb640d9a4f5d90c26a95f53 : SUCCESS in 14m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2e06980be67d4b15a1a65ed711c4a7ed : SUCCESS in 14m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b456cf6b32114d06a6a52baef075020c : SUCCESS in 14m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/226caf4925664173a448fc64dd1542df : SUCCESS in 7m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/39c08c737aef46798ad80e67c768051a : SUCCESS in 15m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee0a302fca924bdf877e161d57ee0f20 : SUCCESS in 7m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a4e40ec33bb04f00ac66cbd4b1d88c5f : SUCCESS in 55m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c876a7df4ea14ed29cdd343313b1fff5 : SUCCESS in 1h 04m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/06910d6e86844c58b3b90303fa1cd3dc : SUCCESS in 50m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ef7864a9a7a7469d910b7af9ddfa76a4 : SUCCESS in 47m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7568be20699f43c7bbc73b3f35364fcc : SUCCESS in 18m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b48cdbbc93474852af258542c4059531 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4aaf7dd875c64fb88bf1309ecce74e20 : SUCCESS in 20m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c63b4ecd17ac40a087fa84a4f625e896 : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":4},{"id":"fe10fa291c866fe7a502624b12720ba1dbd3e83a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 18:38:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"16065565bb0f0312f3600ab57db8a59a0cf24842","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 19:49:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89e118750c5f408daba6e441e387b009\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac42b2cad79a495280aaea1212e1d663 : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/62dd6abce4a24a06b1ab81752ca182e1 : SUCCESS in 45m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/202a4afbd8be4d22a75f4a8d357d5b66 : SUCCESS in 46m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9f67533e43a4fb499271e31b7851278 : SUCCESS in 59m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7ffdc78734a54711973cc4aec750c52d : SUCCESS in 8m 03s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c4c7f6ec20c5482daa64928c9b1a1e0d : SUCCESS in 6m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a6cd268e184447f09d79fd28ba924964 : SUCCESS in 16m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8c438e81662544e6883372efa153dcb7 : SUCCESS in 18m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/89528917e21e4e3c8f015cebbbac4415 : SUCCESS in 15m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/edb4930075e148dc8df0b16e68f5e48c : SUCCESS in 18m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8b715022024f475f8450e13fa53a48d1 : SUCCESS in 18m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f345fec1f9914e82b2aaf48439f92af8 : SUCCESS in 17m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/95b7ef0a181745c18faeba3ee72df490 : SUCCESS in 17m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c9d4d5d326184d22848545f3d18d9506 : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/54c70087716748f8aa4193018a0b4bd6 : SUCCESS in 14m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8d3a5700226c4d279f39f4963921e6f7 : SUCCESS in 16m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b00f97f41d484d7f927a2e15b5216143 : SUCCESS in 6m 48s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f5e92efda7cb4c659d95a2488a1d24ff : SUCCESS in 7m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/828c627baa724a75ba5944763267d87c : SUCCESS in 56m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cafe69a748fb4bf1814ea3907ad6b79d : SUCCESS in 1h 04m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dc7a2eefac964a9abd528bb21e7ae0b8 : SUCCESS in 45m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad91ba0d42f24ac9a32e2a12ab985fa3 : SUCCESS in 54m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f634f16469ed42ecb59cd165fd0ea3db : SUCCESS in 17m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4743cd8838b4408a7e23fb459538184 : SUCCESS in 3m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7cd5e14429a494fb36c391443e2cad2 : SUCCESS in 40m 30s","accounts_in_message":[],"_revision_number":4},{"id":"d8fe38657833142cea268c4a065a4b28cb6edc9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 19:49:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c3dc36602be98063f89a7a28bb58fbc28d711ff7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 19:50:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afb2cbdd896e4088a8b62a9fb97c37b4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0261a6e524e4bf1a1e649992e990c3b : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9a8065b49ec94532bff397092149e5ec : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/4bf4909a8c854b33b265b4600c5ebf6c : SUCCESS in 30s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"74fc18d90df38efd12a75e64dc9645156a7f65a6","revisions":{"9d3edcf7774a9aafe4d64c96793402a469b5553c":{"kind":"REWORK","_number":1,"created":"2023-04-20 19:49:39.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/41/881141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/881141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/881141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/881141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/881141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/881141/1"}}},"commit":{"parents":[{"commit":"4b6f54d063c33e13ced7d6dd5208921680f56ca6","subject":"Merge \"ssync: Round-trip offsets in meta/ctype Timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b6f54d063c33e13ced7d6dd5208921680f56ca6"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-20 19:49:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-20 19:49:26.000000000","tz":-300},"subject":"Extract some closures to methods","message":"Extract some closures to methods\n\nChange-Id: Ia87caebc9a3c70602209daeda97fc588f4bafc76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d3edcf7774a9aafe4d64c96793402a469b5553c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d3edcf7774a9aafe4d64c96793402a469b5553c"}]},"branch":"refs/heads/master"},"e4ee6749707cb004297ee04af16c1547073cbb9d":{"kind":"REWORK","_number":2,"created":"2023-04-20 23:33:38.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/41/881141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/881141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/881141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/881141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/881141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/881141/2"}}},"commit":{"parents":[{"commit":"cf728cfb9763d96d244dbc7d7009bef70edc0351","subject":"Use instance attr instead of non-local","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf728cfb9763d96d244dbc7d7009bef70edc0351"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-20 19:49:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-20 23:33:17.000000000","tz":-300},"subject":"Extract some closures to methods","message":"Extract some closures to methods\n\nChange-Id: Ia87caebc9a3c70602209daeda97fc588f4bafc76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4ee6749707cb004297ee04af16c1547073cbb9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4ee6749707cb004297ee04af16c1547073cbb9d"}]},"branch":"refs/heads/master"},"44183e660818929860672dd0aa234f3e3245fa87":{"kind":"REWORK","_number":3,"created":"2023-04-24 14:10:03.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/41/881141/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/881141/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/881141/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/881141/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/881141/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/881141/3"}}},"commit":{"parents":[{"commit":"cf728cfb9763d96d244dbc7d7009bef70edc0351","subject":"Use instance attr instead of non-local","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf728cfb9763d96d244dbc7d7009bef70edc0351"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-20 19:49:26.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-24 14:09:53.000000000","tz":60},"subject":"Extract some closures to methods","message":"Extract some closures to methods\n\nChange-Id: Ia87caebc9a3c70602209daeda97fc588f4bafc76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44183e660818929860672dd0aa234f3e3245fa87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44183e660818929860672dd0aa234f3e3245fa87"}]},"branch":"refs/heads/master"},"74fc18d90df38efd12a75e64dc9645156a7f65a6":{"kind":"REWORK","_number":4,"created":"2023-04-24 14:26:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/41/881141/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/881141/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/881141/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/881141/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/881141/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/881141/4"}}},"commit":{"parents":[{"commit":"0a2bfe981f640acb20a41ff88d56a651fbf7cb35","subject":"Use instance attr instead of non-local","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a2bfe981f640acb20a41ff88d56a651fbf7cb35"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-20 19:49:26.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-24 14:26:13.000000000","tz":60},"subject":"Extract some closures to methods","message":"Extract some closures to methods\n\nChange-Id: Ia87caebc9a3c70602209daeda97fc588f4bafc76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74fc18d90df38efd12a75e64dc9645156a7f65a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74fc18d90df38efd12a75e64dc9645156a7f65a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
