)]}'
{"id":"openstack%2Fswift~881142","triplet_id":"openstack%2Fswift~master~I1eaa501827f4475ddc0c20d82cf0a6d4a5e98f75","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-04-21 17:30:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1eaa501827f4475ddc0c20d82cf0a6d4a5e98f75","subject":"tests: Fix test_cleanup_ondisk_files_commit_window","status":"MERGED","created":"2023-04-20 19:53:00.000000000","updated":"2023-04-21 17:31:18.000000000","submitted":"2023-04-21 17:30:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881142","meta_rev_id":"2b5f02d485606cc873eac8cec7bebc56cf922235","_number":881142,"virtual_id_number":881142,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-21 17:30:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2023-04-20 23:35:28.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-20 22:51:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-20 22:51:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 20:52:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 23:35:28.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"019a29b611a1f1edb2d113b99fd967222c984df7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-20 19:53:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4fdce055f3d9cac06062eea5ec722a008eb3db8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 20:52:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a659fe0d198490cb9c4930ed1d25646\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa1a1b7557f843c7998fc6181395ba72 : SUCCESS in 7m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6dd90d1f678f4b3584f96d5c8f050cdb : SUCCESS in 27m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3098d31fba7d4549b26ebc753b59ab0f : SUCCESS in 26m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c837c832049640f9b4c5169642de12ea : SUCCESS in 22m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b41d6690ceca40f7ac8e674b62e71c60 : SUCCESS in 8m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bc399a91c19e41c3b6ce5dd9d56c2ab0 : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/84e988a80d1f4265baf02b78ba639347 : SUCCESS in 16m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0bbb3d27bbac405da865fa3187230c3a : SUCCESS in 17m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/71873203aeca4a7aa131f125b4289e78 : SUCCESS in 16m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8319f98bcd3e4014bb86cab43357a9ea : SUCCESS in 19m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c8a79d30f10c4a3da1b47d1a84c09dfd : SUCCESS in 19m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cc5a2defad8a4c5c82c950f990273020 : SUCCESS in 16m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/835fc7b8a0e14b6a9001abaa153f6595 : SUCCESS in 17m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a334410ae52f42a3abc9c5477fe00050 : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/14f0bb851e5341ce92d9d6c983f0d073 : SUCCESS in 13m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9ce4484071ed4b8cb1446fc86a1d16e5 : SUCCESS in 15m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d697282df7cd4cb58b47b493788041a9 : SUCCESS in 54m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/697e91fb8335475688e0c0e2c6b8be2c : SUCCESS in 37m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b75cb319227b4f81a77bc084e6d78903 : FAILURE in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da8ce5c21b1c4882945f2b87d728e636 : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a986f1d74da34197a4060d479f4de831 : SUCCESS in 17m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99133ce723daf1dc64959f691a5fd399aeaa2765","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 21:11:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/407f40a73f3f4bbb97345a4d4c1120c5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/44f9a59148a94e928368d74f7b2cd10f : SUCCESS in 1h 16m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1911347ba994522bca59293748a92bf : SUCCESS in 57m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1441b0f263384691a05912044bd57bb8 : SUCCESS in 17m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d71af97fbc349ba935d555d40e7f187 : SUCCESS in 1h 06m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8f998fdc220cbe9ac2d4c87fd102fc68811cafbb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-20 22:06:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ceb9c0dcc0fde5bd8671bf497d18e1f80769ffe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 22:29:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eac53d3421cd4f9cbfed76aff6f91866\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3b6ffd34e1243fcb2965d484b0a7369 : SUCCESS in 15m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/681ce8a737094ed0aad18d334d4eb74a : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b95b81a0b964ed1baa68522513aad52 : SUCCESS in 21m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/96e5ebc46e8640438b2d1ff73e9deadc : SUCCESS in 16m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2a8a9a328b912ec9ecaafa84d7f62d14cfe8568e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 22:50:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/807dff7e33df4641b2c9e6eb6add465a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da96e61a90d747b7867dc64726db9e19 : SUCCESS in 8m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a9a63c6ea30450eba1a7de179c09059 : SUCCESS in 15m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67bb02c2cb5a4f34846cd915184b1214 : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/337caebd9407437784b08a70cea46f2f : SUCCESS in 20m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c7159dd03cce458abff2f465a64cd180 : SUCCESS in 9m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ba8aa4e905e044d582be608eb5f28e51 : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2ee441e39dd14f92b04cd5c26cefdf16 : SUCCESS in 18m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/923b941dd5ca4a72a122dc8423f04ef5 : SUCCESS in 17m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c1156637c4864f489d9343c8071f72c7 : SUCCESS in 16m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/32d863dca9a1435d8a25a73550a8aef2 : SUCCESS in 17m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/52a5a96ea787420bb0df8512d015e911 : SUCCESS in 19m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d33849f50418475895ef3caba76172bc : SUCCESS in 16m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2b19a391fd0442bebec507d61463669e : SUCCESS in 19m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a7b763be541341f2aae9c4cef23a1ce5 : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/707686c74cc940e7bb36c084fe23b700 : SUCCESS in 17m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2c4817d06eae4f7fa1f9c0b26e5f8c9d : SUCCESS in 17m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/22b02111cd3b49b1b4309d6ac0cf5f89 : SUCCESS in 32m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/18a3c96626a244b8901d4516ba10b0a1 : SUCCESS in 42m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31bbd5870885438ea4d4cfd1f025159b : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ac92df849084458bcc93bf2467008a0 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d79c0b44ee0d43659cb48fb1f6a5468e : SUCCESS in 20m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aaf7df051028d9c38a10df06b198303e9eb12674","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-20 22:51:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"776acc4656e57b6ad0741fefe5c2f1402a616991","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 22:51:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"952a1619ff834214805e3f653401c9de2b5afbe6","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-20 23:35:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"21f97e6a0f522a945bbb64d61a48004206d459cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 17:30:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/810c9e32d9214421b590563f97381492\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82b7e36419824b87aa007e94e7630447 : SUCCESS in 8m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5b0fa560928344fdb6d898936234064a : SUCCESS in 10m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c0c7cfc92fad41edb508c138a4dd1c31 : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1aac5617bd9149dd94ad79c9be901d43 : SUCCESS in 19m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4a1d16440ba34b12a5af47786c274b2a : SUCCESS in 17m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fe97a74b64cf48eeb4accab58591f8d3 : SUCCESS in 17m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3757bcec10e7412083518cea99b23200 : SUCCESS in 16m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/126d59e0e9724ed69be6b144cc2a9597 : SUCCESS in 18m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5ade1b1199f24b76af83841feab6ccbf : SUCCESS in 18m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fae9d2c310f148ba97ffb8b8c7ebaee5 : SUCCESS in 15m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3896e2ce27434a57838a705f7b8c7b9b : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5005b9a0b22a4023834e9c8f589f4541 : SUCCESS in 19m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a5a30d75510e4206aa2e7f0f09c06b78 : SUCCESS in 15m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/201aeeb2c2fe489484e200df138d9b9d : SUCCESS in 7m 10s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/73134b600ada4d25a491fbcb4d2052b2 : SUCCESS in 6m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/210011ce47d64148b88373ec6e59700a : SUCCESS in 36m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c59fc01106c4d1e8fb6d5ca618c346b : SUCCESS in 40m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71cec71933ae4c11bd138eb4cbeffd62 : SUCCESS in 18m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/082b1deebbed4dd9ba8e7cd3d41b93a4 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"bbf7d1ac5eb2690a7097c00e0073721aca0f5446","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 17:30:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2b5f02d485606cc873eac8cec7bebc56cf922235","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 17:31:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f35737b94d304bc7ae2dcd721f03b9b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c141a8c972c1493f8f4fcad41738be82 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/36b5a823e5344e7d9cb62ff8535066ab : SUCCESS in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/4ea1eba064c8446e8b0575438ef59d24 : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"105727bc8961e1cd263d0edd00c7f9ce46dacd2c","revisions":{"105727bc8961e1cd263d0edd00c7f9ce46dacd2c":{"kind":"REWORK","_number":1,"created":"2023-04-20 19:53:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/881142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/881142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/881142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/881142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/881142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/881142/1"}}},"commit":{"parents":[{"commit":"4b6f54d063c33e13ced7d6dd5208921680f56ca6","subject":"Merge \"ssync: Round-trip offsets in meta/ctype Timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b6f54d063c33e13ced7d6dd5208921680f56ca6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-20 19:51:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-20 19:51:06.000000000","tz":-420},"subject":"tests: Fix test_cleanup_ondisk_files_commit_window","message":"tests: Fix test_cleanup_ondisk_files_commit_window\n\n`much_older` has to be much older than `older`, or the test gets\nflakey. See\n\n- test_cleanup_ondisk_files_reclaim_non_data_files,\n- test_cleanup_ondisk_files_reclaim_with_data_files, and\n- test_cleanup_ondisk_files_reclaim_with_data_files_legacy_durable\n\nfor a more standard definition of \"much_older\".\n\nCloses-Bug: #2017024\nChange-Id: I1eaa501827f4475ddc0c20d82cf0a6d4a5e98f75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/105727bc8961e1cd263d0edd00c7f9ce46dacd2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/105727bc8961e1cd263d0edd00c7f9ce46dacd2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
