)]}'
{"id":"openstack%2Fswift~881152","triplet_id":"openstack%2Fswift~master~Ic0574a8bc061a3226d2e7e51aa6d4e497bd2905f","project":"openstack/swift","branch":"master","topic":"value-error-test","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-04-21 15:11:49.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"Ic0574a8bc061a3226d2e7e51aa6d4e497bd2905f","subject":"failing test","status":"NEW","created":"2023-04-21 01:02:03.000000000","updated":"2023-04-21 15:11:49.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"60fa8bc8ec24a63410489e7000f347a953997c3b","_number":881152,"virtual_id_number":881152,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-21 02:02:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-21 02:02:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b458972013fef5dbca1050a004618b114b070eeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-21 01:02:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d93a164a120c502559ff809b28cc49b809918a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 02:02:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c48245883e644ac98d76e694cc44ba06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cb3ea1ece14477bbd37733b04751a61 : SUCCESS in 8m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/48b2a5348e614291aa2344ca379d68b3 : SUCCESS in 22m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4102c08fa5084ea5860dec34c6179f24 : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9485900298cf4d8a9fee677160855172 : SUCCESS in 22m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/53392171410f4791b25a635223869c38 : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5af0cf92f1e24be98334d20a21465a68 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f505eb67a375469a9ae36c1971550fc4 : SUCCESS in 17m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a253152fef694ad883ad11a5b25dfb10 : FAILURE in 21m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d14963b46aaa4ebf8b82f09fe1be16e2 : FAILURE in 21m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/461edee1920143b08d30ae9c9816e69e : FAILURE in 16m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3698e313d10e41d599ac5c1afbe5c739 : SUCCESS in 22m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d8a92dd80004ad7ae6a098b6a91aaba : SUCCESS in 20m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ae9bc37ec37c466ebe450a8034118dfc : SUCCESS in 22m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/398f259c6e2440548f4a4fe801d7a103 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/19387ced680f40509771bf8c214e140a : SUCCESS in 15m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/74bd11c0391e405697db89146681f617 : SUCCESS in 21m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/42f055961d1c4084a7b57132878ddaf1 : SUCCESS in 38m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b6125919520c420082bfff0b8c5898af : SUCCESS in 54m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/392f9acba6c54ede8409ee4cdabb83bd : SUCCESS in 18m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/204c4f84fca24c8e82ebf3374929487a : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/acfefc8ba2914cc0993f167a4dc34cb9 : SUCCESS in 23m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8fbc66d84222f7bf5c1a06eaea8a69dc0cdcfc81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 02:34:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2339587767e741519f1be0348ac2f53b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b3a63c2554842e28d421f6b9d6b21df : FAILURE in 13m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bad54049c05e4b669b5c1d57ccf9bf75 : FAILURE in 15m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d12dc131a3014696bb1cee416891f051 : SUCCESS in 22m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1f7aa77de3d4427a5fc5b4832d81908 : TIMED_OUT in 1h 30m 52s","accounts_in_message":[],"_revision_number":1},{"id":"60fa8bc8ec24a63410489e7000f347a953997c3b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-21 15:11:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34a4b267df50d01412d6d05a03b1e999da2e56b2","revisions":{"34a4b267df50d01412d6d05a03b1e999da2e56b2":{"kind":"REWORK","_number":1,"created":"2023-04-21 01:02:03.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/881152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/881152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/881152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/881152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/881152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/881152/1"}}},"commit":{"parents":[{"commit":"bfc13862792e9e574f1967525f420789b78c44e1","subject":"CI: fix lower-constraints job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfc13862792e9e574f1967525f420789b78c44e1"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-21 01:01:56.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-21 01:01:56.000000000","tz":-300},"subject":"failing test","message":"failing test\n\nChange-Id: Ic0574a8bc061a3226d2e7e51aa6d4e497bd2905f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34a4b267df50d01412d6d05a03b1e999da2e56b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34a4b267df50d01412d6d05a03b1e999da2e56b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
