)]}'
{"id":"openstack%2Fswift~881584","triplet_id":"openstack%2Fswift~master~I37cf3d78db08177f5d9aae079b6a860ba7bdeca0","project":"openstack/swift","branch":"master","topic":"pipelineTimeout","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-05-01 04:56:41.000000000","reason":"\u003cGERRIT_ACCOUNT_35790\u003e replied on the change","reason_account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I37cf3d78db08177f5d9aae079b6a860ba7bdeca0","subject":"debug proxy ec pipeline client timeout test","status":"NEW","created":"2023-04-26 16:04:14.000000000","updated":"2023-05-01 04:56:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"8721dd382bcbedf93b62eb06e9bfc10e9f010666","_number":881584,"virtual_id_number":881584,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-27 17:04:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2023-04-27 17:10:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-26 17:04:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-26 17:06:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-01 04:56:41.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"CC"}],"messages":[{"id":"7d79e51fd05482445b8514171e17870c5da99d18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-26 16:04:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e7e924684d2e85ddff89408f24480be0cf97f1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 17:04:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9684fd5ba7974e7dadc7da2d776caa19\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ddd231ce90d743a28375d97bab308451 : SUCCESS in 17m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b628e9a4611c42d3ac276f58510fddec : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3888a8ab07548959cd78667d0d540f8 : SUCCESS in 23m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f9d9be1d5e14e6282d3106e84f20afd : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":1},{"id":"41b656bdc0456517c39efcaae566ac5ada749989","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 17:06:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38af731719f74fba90f853295a095660\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/483f247fdae64ffcaa92bc50038a616d : SUCCESS in 8m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20329986ed984dbd8819ce8b61c2c179 : SUCCESS in 17m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a640db7ccb04a19903e51db4e96896e : SUCCESS in 17m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e2651d8ff7b4b6fa97ed9120937e912 : SUCCESS in 19m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f84234a777c42a48cb9a710eb46ee6c : SUCCESS in 8m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5ea3b6b5bf374aa9bfa8885acadfa267 : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b9ed43c6f8814216aee14a2676a66c17 : SUCCESS in 38m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/14f666da176c4672873582aa7a800a53 : SUCCESS in 15m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/24a9b307c38247fc8ae1c0bc40ea3bb1 : SUCCESS in 14m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2888a0ab7e9a4c348ed677fcf0aa8ebd : SUCCESS in 14m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a6a37aa907cf46aebb6004d93cbb78eb : SUCCESS in 32m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6c7bd5203ccd45fa82de39ce66483591 : SUCCESS in 38m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/62f7ce9517fb484eb28ee063bc71df2e : SUCCESS in 39m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e5c24382a10a43aa867a43428e691942 : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/270fb04ac03449c8a8fa7108049daa14 : SUCCESS in 15m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/50cef3796a534fdf92df2a7e86dad679 : SUCCESS in 15m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cd22cdf65f244accb5bc791c92203880 : SUCCESS in 1h 00m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5d45293070854528b89c596a411640f5 : SUCCESS in 1h 01m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/067d1f13e52d4b00a9d762d749ad3e3a : SUCCESS in 17m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/513ebd9d23064e4297119a122f178667 : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/586ab8032af4406a965c84e23ae6da91 : SUCCESS in 19m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b73ace71124bc275186ace8567ecf1ae46a4efa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-27 15:51:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"bc496e392d6f52411ee5a4a5385f04fd34823aa9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 16:16:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eae05a985de647d89a8963ac91762479\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cdd42173467248c2ad4488947f0fa093 : SUCCESS in 20m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e14a64abfd0d41ac8e452a812f09d65f : SUCCESS in 18m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b2b77ec16fe4b6391acc4de755655e6 : SUCCESS in 22m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed82bc3632624abc809b87b1cccb2a3e : SUCCESS in 23m 27s","accounts_in_message":[],"_revision_number":2},{"id":"3199662ec95795e7a0d8099b234d394b0c3197a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 17:04:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d76b871011642158ff402d578d0f0b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/077c31bff1f3403e976ecd022dae82c7 : SUCCESS in 8m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd46da04932f4d7eba0460a7c27288a2 : SUCCESS in 33m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2fba6dc59a94a5a85adbbc4f9231fec : SUCCESS in 16m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c87e8ccad869468285b12c08f2eeafdf : SUCCESS in 16m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ee78a14b72644aee9382a0110aa742f6 : SUCCESS in 10m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2d640fab01c449f9965423765551fe96 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/526cddd5aca34f35a90ddc85308dea1e : FAILURE in 18m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/46834c33a1654c5aa9be599d33d130f1 : SUCCESS in 15m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/965f5d7ecd5c4883b3d38e53fc1cecd3 : SUCCESS in 20m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0534f1d1998743e691d21cf24ec955b2 : SUCCESS in 17m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b61fffbd9ee24fd099d7c43cebfd60c5 : SUCCESS in 29m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/141744d0c5fc4ca3a91618a46d3fbe06 : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5181094f14614c21bba2a73966752a58 : SUCCESS in 19m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4ddcf7823c4c4a5f8d8a28472c9e0a09 : SUCCESS in 28m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/787a7b66c76843e28b3fcc0dc75e7a80 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/22698edd653b4e4e9aa829f7829d3c14 : SUCCESS in 15m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a2e70d74f544e89a2a0eb3e28b8e835 : SUCCESS in 47m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b0d117fbf65840d09c77ef95578aa222 : SUCCESS in 50m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58d16c4a9c3a4bb3abca8705b86219cc : FAILURE in 21m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55308889718246338399409cce65dd4c : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ab439eb87e542a59c89c02665563926 : SUCCESS in 20m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4df9582ba7a443d53fb008d4ba36bdb0d54b7f05","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-27 17:10:35.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8721dd382bcbedf93b62eb06e9bfc10e9f010666","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2023-05-01 04:56:41.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1c527cd464c01119fa640e75abd93f01137aec8a","revisions":{"6d4f953d0d1a602d5a126b62244418de951bee33":{"kind":"REWORK","_number":1,"created":"2023-04-26 16:04:14.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/84/881584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/881584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/881584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/881584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/881584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/881584/1"}}},"commit":{"parents":[{"commit":"86f7a37362631a5414a2ea919baeee823ac91973","subject":"Unit test for keepalive timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86f7a37362631a5414a2ea919baeee823ac91973"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-26 16:04:05.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-26 16:04:05.000000000","tz":-300},"subject":"debug proxy ec pipeline client timeout test","message":"debug proxy ec pipeline client timeout test\n\nChange-Id: I37cf3d78db08177f5d9aae079b6a860ba7bdeca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d4f953d0d1a602d5a126b62244418de951bee33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d4f953d0d1a602d5a126b62244418de951bee33"}]},"branch":"refs/heads/master"},"1c527cd464c01119fa640e75abd93f01137aec8a":{"kind":"REWORK","_number":2,"created":"2023-04-27 15:51:53.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/84/881584/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/881584/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/881584/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/881584/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/881584/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/881584/2"}}},"commit":{"parents":[{"commit":"86f7a37362631a5414a2ea919baeee823ac91973","subject":"Unit test for keepalive timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86f7a37362631a5414a2ea919baeee823ac91973"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-26 16:04:05.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-04-27 15:51:42.000000000","tz":-300},"subject":"debug proxy ec pipeline client timeout test","message":"debug proxy ec pipeline client timeout test\n\nChange-Id: I37cf3d78db08177f5d9aae079b6a860ba7bdeca0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c527cd464c01119fa640e75abd93f01137aec8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c527cd464c01119fa640e75abd93f01137aec8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
