)]}'
{"id":"openstack%2Fswift~881597","triplet_id":"openstack%2Fswift~master~I32a161e555179ca34d306ac37e4097611853e36b","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-04-26 22:37:23.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-04-26 20:20:38.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I32a161e555179ca34d306ac37e4097611853e36b","subject":"CI: Use [tox]requires to pin virtualenv","status":"MERGED","created":"2023-04-26 17:56:32.000000000","updated":"2023-04-26 22:39:18.000000000","submitted":"2023-04-26 22:37:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"881597","meta_rev_id":"2c58231025936172302418f925dc8af3976021f5","_number":881597,"virtual_id_number":881597,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-26 22:37:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-26 20:20:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-26 20:20:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-26 18:11:21.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-04-26 18:30:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-26 19:54:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d518ee898a458c824c6229cee43111f52d5f618b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-26 17:56:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac7b92c67fade2706591a7ff09228d0745e45e33","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-26 18:11:21.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"42df5ea98287298a901a5ed5670dcbb6299896fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 18:30:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18c0f40a89964a8d87bcef4b6c0e6563\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c87aa6a32cf4876b2b284df1e148b15 : NODE_FAILURE Node request 300-0021069382 failed in 0s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5589643c7344574bfbf6e672bd21242 : SUCCESS in 22m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d38807ecc8054de8b3bc285f55863285 : SUCCESS in 32m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1274b820bb2f4fdca0789ff6ed692b17 : SUCCESS in 25m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c50fbd07204b936e0e3058d6519a3b456cfda69e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-26 18:42:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"76f04739485eef0b1a3160ce6ade9901d097d387","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 19:44:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c7c9277ebdf44218440bb3c18f25fbc\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b45b2722a122438194ee7e74186de093 : SUCCESS in 20m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f721e5c9b71c49a7bc6865ef68adae92 : SUCCESS in 25m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c08121ac71dd4e168447732332b0caee : SUCCESS in 1h 00m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce7944771d3d4b76bf537d8e0a51300b : SUCCESS in 26m 11s","accounts_in_message":[],"_revision_number":2},{"id":"fe5e34db81628d0df686af9b808c42da420ee96e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 19:54:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4806e443eb444629849f13379d6337d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/da865deb38f44dd29e22c44abbb974e9 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9138acbada63433bb6ab93bf86bd1390 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa11c4f14d1c4930add78d408d628b9e : SUCCESS in 5m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/8f392981bf9e411888f191eab6778618 : SUCCESS in 1h 06m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/20d72423fda74a8a8229f35322dc15c9 : SUCCESS in 46m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e82130aa26d4841a11c759cbacb8dcf : SUCCESS in 56m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66b68515d57345f2abc6ea2c9640fc13 : SUCCESS in 16m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc6fd79ffef1453483b197f592d28be3 : SUCCESS in 14m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a46c46c62b9e40929a66b19841d661e1 : SUCCESS in 21m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f0c174be560e4d76a48dfd5eabff7814 : SUCCESS in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/91c7da4f87d64a44951fb8f04d9cf270 : SUCCESS in 4m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8527e6f9e5a04b439095e6402497d428 : SUCCESS in 26m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1d1536a681094bb1ac67a51abc52476f : SUCCESS in 15m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9074d882c15d49b98c996b0b325481c1 : SUCCESS in 12m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/80c70e0077eb42eab8575e6fab728a47 : SUCCESS in 17m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1dc89c9d31ae453db9120489a50e0ef5 : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4c43b5991c9547c485371279ec5dc0d6 : SUCCESS in 18m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e9dacdfbdacb47e6b0208a44d49b4b63 : SUCCESS in 17m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c879d11a87b14ec28b01286738b68a08 : SUCCESS in 13m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5df133882e984592b1bbd79ead1f8206 : SUCCESS in 14m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4a2e7451769e4e8cb5ba68e4ca12810e : SUCCESS in 14m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a49c2a727b0048d1b87447103bb1a625 : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6b79a7fb24ac4a4db779e2f1141843ee : SUCCESS in 16m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/97e9d5e292ca42589bafa45bcdde933a : SUCCESS in 7m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3611f68188a843ffa21e1201723c26e6 : SUCCESS in 55m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/179280543bb748c2a3a352cca770714d : FAILURE in 1h 02m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/544a0ad2b232449896410956855dfc13 : SUCCESS in 56m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9d7b46393ec44f09907da749b78d00b8 : SUCCESS in 50m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62e74dbcc85244dda9724376c148891f : SUCCESS in 17m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d7e3de5eeb1409893e447df17a02c0a : SUCCESS in 5m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a3b14d2800ad43d384b0b862020f3a83 : SUCCESS in 27m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e3ee7ca5cd7403cbdb4534c64fc37c8 : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":2},{"id":"a411e8fa8f2fd690b7d05ced6890f9c97fd9bf33","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-26 20:20:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4b3db74514312bdd25507e4d45ec7bf3d2a8145","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-26 20:25:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"241e4015cdf2ab70eaf8efecfa5e629d51aaaf6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 21:30:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac6813b866cd43e4a4f4e70e92f5b06d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/badd645e726f4a80b3bb1b17452d2cc4 : SUCCESS in 21m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af923e03d4784b0594deda28235c4949 : SUCCESS in 1h 07m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c6c622892734b678051f872bd568a04 : SUCCESS in 30m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8091f01528454a688732ed538378729e : SUCCESS in 26m 01s","accounts_in_message":[],"_revision_number":2},{"id":"59a930232782ff0df72cdcbd6e22096c06421def","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 21:30:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/803ce4d803334c3a807e810ab5ff99ef\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ba5da7c7c22840f28ebb05461791eaae : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fc4add315114a5c94359c5a2823fd1d : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ac809948ac7454abcbc0855aaec2446 : SUCCESS in 5m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/55c4afc09fcf4c5fa0f8dde30711c237 : SUCCESS in 53m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/93f8dc307664431c8a1bf8de0f448f5c : SUCCESS in 53m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0e7a6bf47b74a819190cfa6e98e3249 : SUCCESS in 1h 03m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/182a4e2bd9a34935b44b7d3e000d615f : SUCCESS in 15m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/897434dfe92a49c3995a166e295c99ce : SUCCESS in 15m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e52f3088273344ceb9d2fbbf7f7ec437 : SUCCESS in 20m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d4052183a5f9438fa132a28c00ac1941 : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/945245c1773043d8b3ab4a3b7f31dcd7 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5f31b1a9a31743caaae2a662dc4797d8 : SUCCESS in 28m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/21e90979d4ef4a5b8176cf189ecfa4e1 : SUCCESS in 16m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8c774be4228644a3ae6472b0946169f0 : SUCCESS in 23m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/386f9994b8ec4ec692f4fbda7c63bdf7 : SUCCESS in 26m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a7e4a747f62b48c590cdc49b564f88a6 : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cfdc0b21c4e34ecd8849fb9d150018e9 : SUCCESS in 16m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a15238f6345a42f4b150c59b4bb5deba : SUCCESS in 17m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8da740abaac748ca8aa84e25350af2ba : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e3dc6d0d1cc040ad89b04ea9b8fb7375 : SUCCESS in 15m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1103e2dfc5c845ed933b444b8bb896dd : SUCCESS in 26m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/92e4851a84e046c6bd4cb6f004418b64 : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/daba40f581544dde8dbe5740644297a4 : SUCCESS in 17m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9a7454efe9274ffb9db5a25041c96e0d : SUCCESS in 10m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/24ca91b4021c46918113e58ed09a5e6b : SUCCESS in 53m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d79cfa3ff5b94e9d95239be6c65d7ffc : SUCCESS in 1h 00m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/478a37d04754439a8ca7115a16fd34ab : SUCCESS in 1h 00m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f641ab64dd94955bbb31b4fa4a69beb : SUCCESS in 54m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc398e9c7f31461ab51feb9757101b21 : SUCCESS in 17m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d936032599ef44d8a217f3ae016048b7 : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b1499860533b4c03851495fb3e3638ab : SUCCESS in 30m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6fe5cd6cecec47e8ac9f0030db8254f7 : SUCCESS in 51m 48s","accounts_in_message":[],"_revision_number":2},{"id":"8e6cb9ebc68d7723707800190febef980371b773","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 21:31:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3900b065255e4c8a398fdf061fd76810696c5b18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 22:37:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/465e5e869f9440d4ac945efb89ffb1da\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/15d45d59e8e043b8b6fda4ddf25650c9 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3290d43ad3c4801ba87df31bf5075ef : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06a968cdf5f74c93aa78ee034ebed4b5 : SUCCESS in 4m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/c623b79e2a234a6e9a74e860f52950c3 : SUCCESS in 47m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/25fbc7bcc24741009979648e10b94928 : SUCCESS in 47m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8511313caac940f19b8cb1e221592266 : SUCCESS in 53m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4f2f4d7543194328b966b0b13b270f14 : SUCCESS in 9m 54s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3888ecbd1f874876a79114c6062bfb04 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b37863069b62424a938911f26be46497 : SUCCESS in 17m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cd823adf64ff4b51acd2d18bfad1670e : SUCCESS in 27m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/43abb0d09d36429db2730c24287f53df : SUCCESS in 16m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cc38eda1e0fc4013951d3ed154a287e4 : SUCCESS in 15m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fc7018cea72f47f98f34590356bb24a2 : SUCCESS in 19m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f7418ee7bf9a428fab43a3a2821c6dfd : SUCCESS in 28m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b65750b49da24d1ba0f6558cd751a4f2 : SUCCESS in 30m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/418d77e9bd544452a8b0a58787e670e3 : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7b58a5f5f5a74083bb427020be7899f5 : SUCCESS in 25m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7a45709e1ff341f392069410e635f94d : SUCCESS in 26m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/279ba4017e1b41fc8ef2ef4982a09cba : SUCCESS in 7m 12s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/03fe7b00ffad4968a80855f213949bb1 : SUCCESS in 7m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ad4304f3e8b440ddabc232044f3ae579 : SUCCESS in 1h 03m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6aa7de5649334270b312811bae95cad6 : SUCCESS in 1h 01m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce84c09ad19544498d0428f344e79c16 : SUCCESS in 50m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/057bf1011c30448abdce2f0c74dbb08d : SUCCESS in 51m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ade61730e68457181a57a5a7b162b20 : SUCCESS in 38m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30b8695327494f0f8e80fea375c6aeb5 : SUCCESS in 4m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0354c91eb88048afa73ef34622bb22a7 : SUCCESS in 55m 47s","accounts_in_message":[],"_revision_number":2},{"id":"df858d98357873d92c178c3a805fc8ec36bef36d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 22:37:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2c58231025936172302418f925dc8af3976021f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 22:39:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9658d2ccd7f24560bfa1bf91499e624d\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/780c9990d71e4f81ae97a2c7f8429324 : SUCCESS in 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a8607839374abf9fb001e692174662 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/630ad3e16a5f4269bcc364f8efd54212 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6650a6d308914fa487dcfc8c6b0198da : SUCCESS in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1b0e83a2845f4e06b63a2600cc3b834c : SUCCESS in 28s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0518c60235abc2405f532f80a9e1b998955957a9","revisions":{"a63d6fbf10e2d8b7bb8333670645c7b5249534fe":{"kind":"REWORK","_number":1,"created":"2023-04-26 17:56:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/97/881597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/881597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/881597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/881597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/881597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/881597/1"}}},"commit":{"parents":[{"commit":"0a970d073684c06a9382d673c03d476d4b2c773b","subject":"Merge \"Extract some closures to methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a970d073684c06a9382d673c03d476d4b2c773b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-26 17:54:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-26 17:54:05.000000000","tz":-420},"subject":"CI: Use [tox]requires to pin virtualenv","message":"CI: Use [tox]requires to pin virtualenv\n\nUsing an antiquated distro to keep us on an EOL\u0027ed version of python\nthat *just so happens* to have had support dropped for it around the\nsame time virtualenv removed the ability to create py27 environments\ndoesn\u0027t seem like the right way to address the problem.\n\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/881035\nChange-Id: I32a161e555179ca34d306ac37e4097611853e36b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a63d6fbf10e2d8b7bb8333670645c7b5249534fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a63d6fbf10e2d8b7bb8333670645c7b5249534fe"}]},"branch":"refs/heads/master"},"0518c60235abc2405f532f80a9e1b998955957a9":{"kind":"REWORK","_number":2,"created":"2023-04-26 18:42:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/97/881597/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/881597/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/881597/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/881597/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/881597/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/881597/2"}}},"commit":{"parents":[{"commit":"0a970d073684c06a9382d673c03d476d4b2c773b","subject":"Merge \"Extract some closures to methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a970d073684c06a9382d673c03d476d4b2c773b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-26 17:54:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-26 18:40:03.000000000","tz":-420},"subject":"CI: Use [tox]requires to pin virtualenv","message":"CI: Use [tox]requires to pin virtualenv\n\nUsing an antiquated distro to keep us on an EOL\u0027ed version of python\nthat *just so happens* to have had support dropped for it around the\nsame time virtualenv removed the ability to create py27 environments\ndoesn\u0027t seem like the right way to address the problem.\n\nMake sure we also include the tox\u003c4 pin that we\u0027re using project-wide\nso we don\u0027t get some nonsense about \"Multiple top-level packages\ndiscovered in a flat-layout\".\n\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/881035\nChange-Id: I32a161e555179ca34d306ac37e4097611853e36b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0518c60235abc2405f532f80a9e1b998955957a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0518c60235abc2405f532f80a9e1b998955957a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
