)]}'
{"id":"openstack%2Fswift~881642","triplet_id":"openstack%2Fswift~master~If15c3fee706d295de678f812839a5d0759089ccf","project":"openstack/swift","branch":"master","topic":"tox4","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-06-25 23:07:22.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"If15c3fee706d295de678f812839a5d0759089ccf","subject":"Support tox4","status":"MERGED","created":"2023-04-26 19:00:34.000000000","updated":"2024-06-26 01:34:38.000000000","submitted":"2024-06-26 01:32:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881642-tox4","meta_rev_id":"99aa4d8333e7dd94a43d82f07800f101f84bd485","_number":881642,"virtual_id_number":881642,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-26 01:32:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-25 23:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-25 23:07:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-26 19:59:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-26 20:08:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"75494c742802a425d6186e9f17164c7d8e513612","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-26 19:00:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"219ae225d3f69f56554b312ae9bc105363745312","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 19:59:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51a4e1946d6c4a219ad3a79f39aefb18\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/58ce280febb34a67933c91f88c3666e2 : SUCCESS in 18m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f305837645e14a1d850fe819d0ab1849 : SUCCESS in 17m 22s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7ff7d054f134c3ca28f21adda4802b2 : SUCCESS in 25m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cfc44132e614b0892cd4893bbfe0ee4 : SUCCESS in 57m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d46b4743c99791962eafd60f0ed3b622a2b6174f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 20:08:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86131dfbe6a046378b8f477cfffee211\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c7d2e200c97848949f30955a02fc1c6b : SUCCESS in 2m 13s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/68af3ae3d81f4da2aeec63bc2371b74c : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a076cb5eba16475f90df57b0ae730244 : SUCCESS in 8m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eff3947dd89a4772a01cb263097ce92c : SUCCESS in 2m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63559957488e4104bc8514c1ce41af09 : SUCCESS in 5m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/e29f16ae3b504257b0ed0a77df6fb255 : SUCCESS in 50m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bc66004dfffe4db287bd01c7ab84001b : SUCCESS in 44m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ed677a3bf2b4e519dd05b4aa989429c : SUCCESS in 1h 00m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b996c7f47df4acf884e910d09b59034 : SUCCESS in 18m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05061a88b120488dabd4d31308b99c06 : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a953d9382494c59bf84a60da4d200c8 : SUCCESS in 23m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f48dd8d10dbd407dac6fcbefffe614f3 : SUCCESS in 9m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ee31995a13d74128a45219dcc4cb2454 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/95f4e6420802440faabb15648a508803 : SUCCESS in 16m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e48ac1c2a7ba41b48d872b82ae6754e3 : SUCCESS in 11m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/442be615e92c45608190e3db84b8d6e8 : SUCCESS in 18m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/824ac52876494f8fb0c18f691120dd57 : SUCCESS in 16m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/72c331b90a3d4c369abbfd3f9843f6b1 : SUCCESS in 15m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/56ccb250c16b4be5a1ef74d1a08b9c56 : SUCCESS in 20m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/670454148a0d467bbcb05892705c4d7b : SUCCESS in 20m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9eadfeb53a8945c783a7a679ca76d806 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f13016082fd14f6ab19217f103edb9e8 : SUCCESS in 16m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0661a9fc972c41a1b999832f91bde3c4 : SUCCESS in 14m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0867f71c2899411786cb35551713b825 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/58c0dfe02d64432cbdb186e1d7320806 : SUCCESS in 15m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3e46345893ea4d9894e8b6502b4d2816 : SUCCESS in 6m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cba86b603fd8415d8fbadbd57189e3ca : SUCCESS in 47m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/728380289def45b493d99131fadb4150 : SUCCESS in 55m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5e2c4c804794c8da3703e70f669bb19 : SUCCESS in 39m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad48f70bda074664859f281cc3ec69ec : SUCCESS in 37m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e928d929dfa04805aa348322cb729027 : SUCCESS in 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11ff51457e3341eb981b70aac1abc2b5 : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd67ec1195754801a4e4e88115c8e68c : SUCCESS in 16m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd9a781b3fc24ee4bc888f36f8455da7 : SUCCESS in 51m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a915fcbcc26de9065b726740f87f34f781c95dc9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-26 20:21:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36887f1a4a2413324f04f72a4501f9b20ac65d31","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 21:07:30.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd0325146eca4c408d030473b9cf2f4c\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/388d7b3ff4544ee89cfed95fa3f174ac : SUCCESS in 26m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/97a7e7d65d024090a0d751fa12a1ef0d : SUCCESS in 23m 02s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/9b1918126712476fa48553e947fb8fa4 : RETRY_LIMIT in 3m 11s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/85a9525ca99a4affa18a3cd67f326d61 : RETRY_LIMIT in 3m 06s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/4ddc7439feb84bb1985e54709f99e7b3 : RETRY_LIMIT in 2m 44s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/7bd59edf34e04ce0a5053b13748a39b4 : RETRY_LIMIT in 2m 52s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/48e1b181f5d24f0ba1658797827928c8 : SUCCESS in 14m 57s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/16832cdc91254e3498dab509067018c4 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/627afa516b31410ab7b2ff3e8f9a2b00 : SUCCESS in 13m 37s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fe6e455c7f34e8d84ab785878e1f13d : SUCCESS in 15m 26s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/65d40a9e33c44c2086f9477fad8c4d21 : SUCCESS in 17m 06s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/fca3e1e9dd7d44758c4defcece074568 : SUCCESS in 21m 14s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/f4579dbcf6cb43c8938089be7132d937 : SUCCESS in 19m 11s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/13a2b60fd8454826ba865cd8a033d061 : SUCCESS in 20m 16s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/20c3792121844d6fa6520a82cb098ece : SUCCESS in 19m 48s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/ede37379bdde448881e2d39c2f94ac0d : SUCCESS in 19m 37s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/991fe939c2574404bb385bc7150505cf : SUCCESS in 18m 40s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/9e096e15daec424a80d86132dff64d6c : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":1},{"id":"279f81bc0ac4703753fb6fcd6e365e2068fec9a6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 18:11:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc2247f982c09a19d56509a60cc139af9a2aaca2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 19:10:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae9aa7974807470ea4907a559c626f86\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe472cf0928547d3991d12bc70e5426c : SUCCESS in 17m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b981c4e3671a49aba29e8b44b7e3b2ad : SUCCESS in 20m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/479df942aa1e4c5ea44a57967105ecf2 : SUCCESS in 57m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8be716d59f14ce8ac30d5cd4c395deb : SUCCESS in 20m 56s","accounts_in_message":[],"_revision_number":1},{"id":"a75933412178082aebf82f93bef809337fd9d4df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 19:17:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9cef7d81af947fc819765f898c47009\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/63fef2f92e794092977ebf67aa8cf70e : SUCCESS in 1m 36s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7ac4aa3d699345ad9a225359e0dea8fa : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce1f4f95bc1846fea0f8d419ca608d99 : SUCCESS in 7m 15s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a14c48af7d940a0920ff6bce705e355 : SUCCESS in 2m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/157bc3a9ec7742cfbab4531b9643f206 : SUCCESS in 6m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/ddacd53755e1471f9faf08f59605fbbe : SUCCESS in 53m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ec3327b396604161a2edf956e906b580 : SUCCESS in 53m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83d7382049d64e5c9a919db5f544f594 : FAILURE in 1h 01m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b65f09b9d224777a5751ba46fbd9bf2 : SUCCESS in 15m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ae3409d8e7e45e6b6ab21d804f2292c : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1eb2d028c9b4f9eb32f50dee856a565 : SUCCESS in 18m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/beb3456981c6495bb135c1af81c1c459 : SUCCESS in 10m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6b4c2fef05bd4dab8bdfa4ae638c2972 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/07d1655f752548c09ce6e53c8b234d86 : SUCCESS in 21m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f2fe52312e2e493b9e67e132f0f52341 : SUCCESS in 15m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ba167e84fa784a66a70721692cc2f7d6 : SUCCESS in 18m 12s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1f353f1831b5408abc7cf566817c2171 : SUCCESS in 17m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/73f6aee34fad4f3baaa664ed14b734dd : SUCCESS in 22m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a21c4bbc7f2e43e9ab70554b0bce46fd : SUCCESS in 18m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4fd07aefa8e14528ba48343474c1ec3d : SUCCESS in 18m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3a2b15b59ef5491da4b142029befffaa : SUCCESS in 17m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/423b25ddb6dd4315a7c3ecb7379d9eab : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dfac03e8c5714f8d99ce0007d0e559d5 : SUCCESS in 19m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/df6cf7e7082c40adb3e14f7d47486103 : SUCCESS in 17m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2e606c0bc0914bbd91f89fc2393116ed : SUCCESS in 5m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a1738f4cc96441d4ade2a5cef5fea7d9 : FAILURE in 4m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f0f231fdd04e451ab78ecd4333edb7bc : SUCCESS in 6m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/765dd428377d4d63bcf4ca611bf2e4ec : SUCCESS in 1h 03m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0779f8dd0b8241fd9d731e36289a15ab : SUCCESS in 53m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bfcdee98ac37450299eb742f6f9fd7ff : SUCCESS in 50m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bb8c2b5ef984bc4ab3502c288334900 : SUCCESS in 14m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/269f9d615ae948488016bc5d863660b7 : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/898dd9fb8b224715bb4f080848cea18c : SUCCESS in 16m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ef041f5d1934e289f665762632cc2ba : SUCCESS in 38m 20s","accounts_in_message":[],"_revision_number":1},{"id":"080338eba3e341848618b1cf8e163a95c81ed0c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 20:37:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cd4de8200d42e9a8db0136f81cb0635496d2765d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 21:43:11.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9f5cbb4a4e24c9ab488a80470dcc3bb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/600b59c5bb62496a979cbc0222c14bc6 : SUCCESS in 19m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92e35fd2b99642f0bcd31db5116d1367 : SUCCESS in 1h 02m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/69ec91dee2c541f6ad16db6847d66299 : SUCCESS in 20m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/00a70bf78fb643b8877401462cc13ddc : FAILURE in 51m 52s","accounts_in_message":[],"_revision_number":2},{"id":"a5439b04eaaa06b398f079649dbe5db40fadbe6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 21:46:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c9e9c60b1124cae85b0de0acdba5f15\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bc34ceeed9074b2e975c7bfed1a3708c : FAILURE in 2m 05s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8011e92dc45345caa2cc1e13493ea741 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd9f27854214bde914b5506b60d7efd : SUCCESS in 8m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1dff850e60a94609966a7b05de547bce : SUCCESS in 2m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/420d017276084f83b420a5c1351e1963 : SUCCESS in 5m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/308278c145bf48d7a22b1ccd6a65c45c : SUCCESS in 51m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6c4b2be559f3456a9f9c417909429ea2 : SUCCESS in 49m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6728d2e020564c19beb107b739e417d9 : SUCCESS in 59m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc615c70fc4945fab3bb77d4d7416610 : SUCCESS in 22m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3840791376ad41138e4d32207015b7e0 : SUCCESS in 17m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a0ccef92440430eacea8f8628a9b98e : SUCCESS in 18m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4514323c98fe4a86a9cda760dee1568d : SUCCESS in 10m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/91f531ecacdc4d639f183b74287446df : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4e2476db93bf4c889f7999b53ceafa58 : SUCCESS in 17m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f056b832f3b4470fb7f25aabbd2edca7 : SUCCESS in 18m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dc7b526821004eff832b42a71ef4555a : SUCCESS in 16m 15s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7e8545db2baf46d791d85e785698289e : SUCCESS in 17m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/69b1b6771de0459f8f6c464320316edb : SUCCESS in 23m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0d28f751ca854985aabfb21f64149700 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/86f56c27b2a24693bf76a3e105470de1 : SUCCESS in 17m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9a9ff06881a5437ba129d187e89c0457 : SUCCESS in 17m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/acc46630cdc1425ab9922064f67cb082 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fda404d726f245ffa7b600a46c89998b : SUCCESS in 15m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8da5f2eb42ff43299823df6bfd1ee32a : SUCCESS in 19m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e0db5b8313224a158204f81a1553943b : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/51b8bc5ee19646beaf61e5606793aa8c : FAILURE in 6m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dd6b6364549b446a8f10abe54fa6c3b7 : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e64a136c4be4e819d4ca4a008f81f74 : SUCCESS in 1h 00m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55dbd5e80508422abcbab67fd976bffb : SUCCESS in 55m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4d6d5e8d72b04eba9c0adcc0000c0ad6 : SUCCESS in 46m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24643c5005b74b21915946933789d8b5 : SUCCESS in 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5444d535ee844aa0aed3e8d3174a3c0d : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8ca82cac95e64778983a5e7f2a8806ad : SUCCESS in 14m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8984df47a94c47d6bb53593ddd0189a6 : SUCCESS in 45m 40s","accounts_in_message":[],"_revision_number":2},{"id":"f26852d816c3045869b396a94a830ad0fb4210b8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 23:07:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"86210e73b1b00b0bf5dfba7ffc471373f91041eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 00:17:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e6896cb6d454e11884a47b70209187f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/52b2d3bc1ab84fe4abe4f290dac8dc98 : SUCCESS in 1m 37s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/91eaf82a687a4e4998bfdf4629a3a9dc : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c59b7e479d5c4df8bf4183c28fbf2038 : SUCCESS in 8m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/14b541abf3ff420dad4a06119a57aafa : SUCCESS in 2m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fff6888204ec4e97a93599856fcc35fc : SUCCESS in 5m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/821d597d3b7f449481457195d96a3645 : SUCCESS in 1h 04m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/20e287302e40486d844ddf211bd9ac72 : SUCCESS in 42m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0314f5d77d14da7afda62751f990df5 : SUCCESS in 53m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf4a0b36de164c529704cf31b8d028c9 : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34db9c4e0cf7492fb310b50b0b089d3f : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d976e4695cac47faa8a9ac99ebf0f359 : SUCCESS in 21m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aec32894ca0f40bb85583dd8d3d3edfe : SUCCESS in 9m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c59c0afffd334a099c5a2ea924af3316 : SUCCESS in 6m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3739be8b6e5f4683a6a89c48255f4ea3 : SUCCESS in 19m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a9406d695a354d9083c1f0b2e051eb68 : SUCCESS in 18m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/38ce146a90444613b8881fbc5cab27b9 : SUCCESS in 17m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0072bb6a45714bfdb78521e589149faa : SUCCESS in 13m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b77d93b9b6a84f91a5b74815fc1eafe1 : SUCCESS in 24m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f2ffc0ba5f1143438964f63130321c59 : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9ac7850050fa4df0a177e4c4a2c40a81 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f9efd7bc13294d6e8c6ebf19626aea00 : SUCCESS in 16m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d73714eec13b4f4e9f1abc2751c62d09 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8f767e0b56b942a3933e85ee2c34bb55 : SUCCESS in 14m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2066e09d65874aef9db761a280d9be1f : SUCCESS in 17m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e8730320b09493faba9c8c8e9cebfbb : SUCCESS in 5m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2a59064e891448fdb6572c9d89a0a059 : FAILURE in 5m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/10d8ff31f52a47eba2dac61ac27fcdbb : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/43758c7c0abd4ed39caf30f55f65011e : SUCCESS in 46m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/58f8279b97104ef0a481cae4f92537a1 : SUCCESS in 49m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b93ebfebe17a49e689391d0eaff4071e : SUCCESS in 53m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69340817bc5143c38f4a837104a2f5b5 : SUCCESS in 14m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdff517347904fd7879d7e88a1596aa9 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/16333e7de2ff4e629011f0d685622e74 : SUCCESS in 17m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f468428260354f739bece999ed254d52 : SUCCESS in 57m 28s","accounts_in_message":[],"_revision_number":2},{"id":"22221673da78b8b0a3700184590b98e94d1a4f65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 00:17:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef6fb71ebdfe42bc9725141795a1b39d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/623abcabbb2d4090909d14fb48083d41 : SUCCESS in 17m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/03dbcbdc39b44a3bbc6ad9e7c870f695 : SUCCESS in 56m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a36abd06f1b74bbab9822bd1a4f56c39 : SUCCESS in 24m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f2e37d64aad44c9ad843da73f731810 : SUCCESS in 1h 08m 02s","accounts_in_message":[],"_revision_number":2},{"id":"8c4af4f7dd9b90b7b90437cc96b9e079b005e1d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 00:17:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"65309d01f927378d8286215edeb54524e15878ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 01:32:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b6129d5524f44e7bede0e6047d5f536\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/40814e40496845d88dd879e86f504815 : SUCCESS in 1m 44s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bb5f95a0c4aa448cbc0d5eb8f381079e : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb7ba4b9434d461e88756a801fed4a93 : SUCCESS in 8m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/014f40e919784e84bdfdf624cae30500 : SUCCESS in 2m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1b0b697fe6a457bb0bb164a95f18de2 : SUCCESS in 5m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/ea29620de7ac437fb7f49120e2a0b75f : SUCCESS in 1h 00m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b90a6c856961485dbecc51b7b36e5f15 : SUCCESS in 48m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9e70f7140d5440fbb0621b9c4a695a1 : SUCCESS in 1h 09m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4f07bc9558da46cfb96c74c4b50973ea : SUCCESS in 10m 14s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f42f3e6e24dc4bd49f15423ad0874d5e : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8ec1be4973394a45a40bbe3888245e66 : SUCCESS in 18m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/abc80c56465a4f6db493d36f09025637 : SUCCESS in 23m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fced4f81adca4c66bc81bf217df956a3 : SUCCESS in 13m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6aaffae0c9774edea47bc5ecc26925f5 : SUCCESS in 13m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d07e13b111d0448ea4ad0857b9f3c23d : SUCCESS in 21m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fb4ea0ab8dce43d68f467d2f678222eb : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b7dfff521176406aab9744caf6f0614b : SUCCESS in 17m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/21d497ac949f45019dd4116c58eed9a1 : SUCCESS in 16m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4802cc364ef041f2ad3be8fbf6816018 : SUCCESS in 13m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/27b3abe233ff4fbf9532855469460197 : SUCCESS in 14m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a0b2364776ce46719e4e0c8825df617f : SUCCESS in 17m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/50d62d989c274d50a75afc3f0dc8c19b : SUCCESS in 6m 16s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2b20953afee0426ab6e85e419eefb690 : SUCCESS in 7m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/997221c69db44f4493cb51be2b2baf28 : SUCCESS in 49m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b86e6885be2f4d95b3c4864318fa3da1 : SUCCESS in 40m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/57bf81614ea243959eafc16017541af7 : SUCCESS in 40m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec4a1d900e59481aa98ac9435adb079a : SUCCESS in 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ebaca6e97494db3a4cdea501fcad7a8 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6846abd2b4ec4c329afb562c1d827d7c : SUCCESS in 17m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ccd2ba970c64efa8384a35b91a9c3d6 : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":2},{"id":"56de572d00bd80646c29b96da407f7704ddc3345","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 01:32:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"99aa4d8333e7dd94a43d82f07800f101f84bd485","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 01:34:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11ab9de1e2724793bae2bfad5b73104c\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/95edb37c5bcd4c4a916d40230bec6193 : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7c4adcab935438891966fc1cff6d71d : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92af908f46aa4256bfa230ccaa6e0352 : SUCCESS in 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/861e75992eb74b7dbc8048bc3a69494d : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8dad423cc26a4a19b582202ad0498fa2 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6220a473f3ad2b5238717de832f508b557c6586c","revisions":{"917b83697bdf25d13b59d72e3a2b22ee86e47399":{"kind":"REWORK","_number":1,"created":"2023-04-26 19:00:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/881642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/881642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/881642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/881642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/881642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/881642/1"}}},"commit":{"parents":[{"commit":"0518c60235abc2405f532f80a9e1b998955957a9","subject":"CI: Use [tox]requires to pin virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0518c60235abc2405f532f80a9e1b998955957a9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-26 18:59:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-26 18:59:53.000000000","tz":-420},"subject":"Support tox4","message":"Support tox4\n\nChange-Id: If15c3fee706d295de678f812839a5d0759089ccf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/917b83697bdf25d13b59d72e3a2b22ee86e47399"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/917b83697bdf25d13b59d72e3a2b22ee86e47399"}]},"branch":"refs/heads/master"},"6220a473f3ad2b5238717de832f508b557c6586c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-25 20:37:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/881642/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/881642/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/881642/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/881642/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/881642/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/881642/2"}}},"commit":{"parents":[{"commit":"fcbe00a5cdfaa5aaa316f991b0b5aece0a23abde","subject":"Merge \"tests: Check proxy-logging stats on the wire\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fcbe00a5cdfaa5aaa316f991b0b5aece0a23abde"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-26 18:59:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-25 20:36:42.000000000","tz":-420},"subject":"Support tox4","message":"Support tox4\n\nChange-Id: If15c3fee706d295de678f812839a5d0759089ccf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6220a473f3ad2b5238717de832f508b557c6586c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6220a473f3ad2b5238717de832f508b557c6586c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
