)]}'
{"id":"openstack%2Fswift~881648","triplet_id":"openstack%2Fswift~master~Ic0813e6a9844da1130091d27e3dbe272ea871d11","project":"openstack/swift","branch":"master","topic":"errorMsgChange","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-04-28 10:16:31.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2023-04-28 11:37:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic0813e6a9844da1130091d27e3dbe272ea871d11","subject":"Error logs changed for ChunkWriteTimeout","status":"MERGED","created":"2023-04-26 21:35:04.000000000","updated":"2023-04-28 11:38:20.000000000","submitted":"2023-04-28 11:37:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"881648-errorMsgChange","meta_rev_id":"8253556ec7aef3333b7548cf1b1864f03ba63680","_number":881648,"virtual_id_number":881648,"owner":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-28 11:37:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-04-28 10:16:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2023-04-28 10:16:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-26 21:40:29.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2023-04-26 22:52:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2e83f797a92c9b24813500609df20bbd83df7ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2023-04-26 21:35:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27f552f42e2de98d632a603457d5f9cd96ffc10e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-26 21:40:29.000000000","message":"Patch Set 1: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cfb8247a87f34326830f53e59001031489571124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 22:52:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e7c0cffd8d94de690aad9acd3c62124\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae44fce6c9754a599150787f3b8cacf4 : SUCCESS in 7m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/2300b04483dd4a6d9f2659f3feb793c6 : SUCCESS in 44m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d1e324b40aed4537bab074555c297514 : SUCCESS in 39m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12cb50813a54474899a40173a579854e : SUCCESS in 53m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3640969587ba495c8eef15484ea1ccf7 : SUCCESS in 23m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1998fbc339514d779c6b3b836d8ff1a0 : SUCCESS in 25m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2806b1a5c2ec4118bf55ffa9d469614a : SUCCESS in 17m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/92e9913817dd425ea8e8fc38e65221b9 : SUCCESS in 11m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/eaf79e49487d46fc8b5b3edb478a8995 : SUCCESS in 7m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ee6e2d46099e4157a96b2f10dbaf98bf : SUCCESS in 17m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f1ebf9ce9c648a2bc47ff2532901eb2 : SUCCESS in 16m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5c9353bbdf494516a549887f5ed87356 : SUCCESS in 15m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9deceff08c8c4d768ab3ac379a64cab8 : SUCCESS in 17m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/adfdaf7c54be4a27adb2d867aabc4031 : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/908f0b35b71442b080956d5ffc53589c : SUCCESS in 17m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/53b58917529d4e609eaeaaa74a439dca : SUCCESS in 17m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6d7120aaf93a4fd783f19c7855aebb9f : SUCCESS in 13m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/37e5362ed35d459fb62fc82b21a36f20 : SUCCESS in 13m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/69e002f985734756be7b6d32d5af678e : SUCCESS in 14m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f46bcc6086de4adcace5c44466afdcda : SUCCESS in 7m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c6b3bf866f3d4ba49e5575b3f2f3cb20 : SUCCESS in 15m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c1a9910b1e6e4243b88ce4112e74c7b1 : SUCCESS in 7m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/325063f537f24821b0cbefb02e5eb42b : SUCCESS in 59m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/92ba55a8ad0d4dcd8acb71a399ea8ddd : FAILURE in 54m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/368aab9d29964aa5a0fb7f84207f92b5 : SUCCESS in 53m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f92eb0d357fa4aaf9ddf62a1ef337d6e : SUCCESS in 57m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c24e70c08c9443b846e4d9753b82867 : SUCCESS in 22m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6bf8754f57a4e7ebe096858dd135424 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb82d6531496460380e3122f54ccab00 : SUCCESS in 22m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f7352f10ce9484581d43358a9d8081e : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":1},{"id":"41430b8465236b6422e66a022c70aad58edbc38a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 23:08:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/662fcc10071b45f39401882422bf68eb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4506a7e517ee42329893612737771e6c : SUCCESS in 55m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3de2fc1e862429caed6c6e538c29c91 : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a7db5fae1f7498d8620c713767d2694 : TIMED_OUT in 1h 31m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d77901fa3e974fbc807e5e20f5b0b4a2 : SUCCESS in 1h 14m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e32eaf1f477da15851197aa405c19eb6a9762b2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2023-04-27 23:20:06.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f52bf1a4dd178f6ed123664cb551533044223f12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2023-04-27 23:21:04.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c4cbfd7fd4da6f58ab08542d79a538f94da6bc0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 00:28:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f314c9aa5af43dc8ae9bcf75437d3b0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b58869ddeac94986adc14a6b1c8e3754 : SUCCESS in 1h 06m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ffef5a83795e490789d4a10823da6698 : SUCCESS in 1h 03m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d2c14cb272240fcaaeb9db0ad81f672 : SUCCESS in 20m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a7315e657ce4ae1b4c5752bdd8990d5 : SUCCESS in 18m 15s","accounts_in_message":[],"_revision_number":3},{"id":"e97ccc5c0ece0bb6a319f87bcaa37a66e98b9c0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 00:43:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6e7b7f491614b2196611353efbc74e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c30b3edfa7cc4bdcae52c99afed1cc31 : SUCCESS in 7m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/0c6506f10e5a4d98aeaf9098094a794c : SUCCESS in 1h 11m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/82e4ee3199db4247be5cc68012b40ae7 : SUCCESS in 49m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c94029539a34a9686156607748ad1e6 : SUCCESS in 1h 03m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5aca87039e8e4b59b04645b5c107d024 : SUCCESS in 15m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/edd2c57a06494d8d92dc83f922237fd9 : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b0e96b4fc984d568c328c4684c85455 : SUCCESS in 16m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/29cc6d1465df41349532f40064099da0 : SUCCESS in 9m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/70bdffde556e428e9a618d575669b217 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/db328230110b4d6ea42f693d785cdc17 : SUCCESS in 13m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a1408546b25a40df9747feb5fde352cf : SUCCESS in 15m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/27af365aa0ae4619bf7ef7af32a5d205 : SUCCESS in 15m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1721ea52148c4d1e917ade082d664e78 : SUCCESS in 15m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8c155c311f0144dfb20b435ed72b20ae : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d5aa5edc1bdc4dee85ad193556495700 : SUCCESS in 19m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7606d856046f40b8a9eef1fb9b5ffa25 : SUCCESS in 17m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/32115cebdc204decb7a64c14a59c32cc : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a72865ff6a66484c9bb826fa11af5a46 : SUCCESS in 15m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/10bb696f3cac4210ba38756aa71dca37 : SUCCESS in 15m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b8dc9364a168406daa136fb0f21829bc : SUCCESS in 7m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9b60b605f115425285b554253b4d11b2 : SUCCESS in 15m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/edc8eb5763d34608b355973560c169c6 : SUCCESS in 7m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9753c8ce0fe4490aa06dd96fd2906f4b : SUCCESS in 51m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/312cf27ed0474277a488c4696eef9dd5 : SUCCESS in 1h 01m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5c40b3b48eb24286bdadb5a92b35e929 : SUCCESS in 47m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58698b637423421981aa652c32da1622 : SUCCESS in 52m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df91cffbed6049f4bb4d610ca11ce02f : SUCCESS in 18m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87238d0b261c49a783ff7b20e28c3cc8 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/629d828d77c3463f830f614627505da8 : SUCCESS in 19m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67578af5894442d3bd3a93082f8a901b : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":3},{"id":"4b63a13c27f01b49c76dd8b8a05567a5ef51c5bb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-28 10:16:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d86c2e582107bd010313d20a4145985efb118285","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 10:17:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"61ce740f7650923d16727b209c368296f936fcbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 11:37:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c02a7ec72c374aab93fe70ac32fd2de5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99d8ae43e57c497ea305009070fd627b : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/c72ff2a13a1d4e2db00651322011272e : SUCCESS in 1h 12m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/84fb967958ad4d4e868e9764af158dc6 : SUCCESS in 47m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cea525eff324af78794ea470814fec8 : SUCCESS in 56m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f109865e35634ac8aa9467ec78191648 : SUCCESS in 8m 58s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/cdf779fa89b740e19c3b00b0198d21e9 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5c887fda9d764b57bf35c39fc77c0035 : SUCCESS in 19m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b057d252a002461a93c2cf3159fe49ce : SUCCESS in 16m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/92c3aabb46004e96babdfaa2d7809422 : SUCCESS in 14m 45s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1490f3fa84b44030b20c86a19db387fe : SUCCESS in 15m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ee75b691c2714f28926dafee83579dda : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7c1321f9cfc94a9a827603a95634f709 : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/26dd84278e634add92d72f2f1297181a : SUCCESS in 16m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2fb8b7c7c3e94ab1855d69285c1bcc81 : SUCCESS in 12m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/31db92cea16b4f768eee5c5dec6be920 : SUCCESS in 13m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1f916232de7a47aebae0e9a3b25e6e2c : SUCCESS in 14m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a2b3e750e21b461aa4a7f47796501c0f : SUCCESS in 7m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8a56062750ea4ba59dc5f195c55e7755 : SUCCESS in 7m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e83c2f05a64f4314b92a8b725b5948ce : SUCCESS in 1h 00m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/20b6dadd2691461b80105ce1fa97f234 : SUCCESS in 1h 12m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/25046bd29c0a4842ab71f3225151cfcc : SUCCESS in 36m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/676dc5c4bd544def9755d775c3f0c852 : SUCCESS in 48m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52954ff3630d464b852e1094256a882c : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2492573d4bef4e50a00d573009936ce3 : SUCCESS in 4m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0515495f3e194082bd63c9fa023e0aa7 : SUCCESS in 59m 34s","accounts_in_message":[],"_revision_number":3},{"id":"b684e206b71ab273a989458fedfb2a75f37b07ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 11:37:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8253556ec7aef3333b7548cf1b1864f03ba63680","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 11:38:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c74f634a84d144479841369fdb9aec21\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdda5e6846c24364bde4a75b28baef6f : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d9c58c02d2524f3a8e5c9dba57fc3d8a : SUCCESS in 55s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/dfddd0492d0840169f792e3312f22824 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9dfbac7c4d600a90fa0eafcc55faf72a8daa41bf","revisions":{"b974ba945b47752c9e70f7d77a0c04e67f43dadf":{"kind":"REWORK","_number":1,"created":"2023-04-26 21:35:04.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/48/881648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/881648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/881648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/881648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/881648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/881648/1"}}},"commit":{"parents":[{"commit":"9f7527973997baa5e685ea57d1563232234bc872","subject":"Merge \"Assert ChunkWriteTimeout errors are logged\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f7527973997baa5e685ea57d1563232234bc872"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2023-04-26 21:29:19.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2023-04-26 21:29:19.000000000","tz":-420},"subject":"Error logs changed for ChunkWriteTimeout","message":"Error logs changed for ChunkWriteTimeout\n\n---All unit tests passed-----\nThe log message phrase \u0027ChunkWriteTimeout fetching fragments\u0027\nimplies that the timeout has occurred\nwhile getting a fragment (from the backend object server)\nwhen in fact the timeout has occurred\nwaiting to yield the fragment to the app iter.\n\nChange-Id: Ic0813e6a9844da1130091d27e3dbe272ea871d11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b974ba945b47752c9e70f7d77a0c04e67f43dadf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b974ba945b47752c9e70f7d77a0c04e67f43dadf"}]},"branch":"refs/heads/master"},"0847bd24699fb1a9481268dd1b8097f017c48cbb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-04-27 23:20:06.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/48/881648/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/881648/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/881648/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/881648/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/881648/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/881648/2"}}},"commit":{"parents":[{"commit":"9f7527973997baa5e685ea57d1563232234bc872","subject":"Merge \"Assert ChunkWriteTimeout errors are logged\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f7527973997baa5e685ea57d1563232234bc872"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2023-04-26 21:29:19.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2023-04-27 23:20:06.000000000","tz":0},"subject":"Error logs changed for ChunkWriteTimeout","message":"Error logs changed for ChunkWriteTimeout\n\nThe log message phrase \u0027ChunkWriteTimeout fetching fragments\u0027\nimplies that the timeout has occurred\nwhile getting a fragment (from the backend object server)\nwhen in fact the timeout has occurred\nwaiting to yield the fragment to the app iter.\nHence, changing the message to \u0027feeding\u0027\n\nChange-Id: Ic0813e6a9844da1130091d27e3dbe272ea871d11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0847bd24699fb1a9481268dd1b8097f017c48cbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0847bd24699fb1a9481268dd1b8097f017c48cbb"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9dfbac7c4d600a90fa0eafcc55faf72a8daa41bf":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-04-27 23:21:04.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/48/881648/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/881648/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/881648/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/881648/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/881648/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/881648/3"}}},"commit":{"parents":[{"commit":"9f7527973997baa5e685ea57d1563232234bc872","subject":"Merge \"Assert ChunkWriteTimeout errors are logged\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f7527973997baa5e685ea57d1563232234bc872"}]}],"author":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2023-04-26 21:29:19.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2023-04-27 23:21:04.000000000","tz":0},"subject":"Error logs changed for ChunkWriteTimeout","message":"Error logs changed for ChunkWriteTimeout\n\nThe log message phrase \u0027ChunkWriteTimeout fetching fragments\u0027\nimplies that the timeout has occurred\nwhile getting a fragment (from the backend object server)\nwhen in fact the timeout has occurred\nwaiting to yield the fragment to the app iter.\nHence, changing message to \u0027ChunkWriteTimeout feeding fragments\u0027\n\nChange-Id: Ic0813e6a9844da1130091d27e3dbe272ea871d11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9dfbac7c4d600a90fa0eafcc55faf72a8daa41bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9dfbac7c4d600a90fa0eafcc55faf72a8daa41bf"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
