)]}'
{"id":"openstack%2Fswift~881671","triplet_id":"openstack%2Fswift~master~Idf263a7f7a984a1307bd74040ac8f8bb1651bc79","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-04-28 15:21:34.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-04-28 17:36:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idf263a7f7a984a1307bd74040ac8f8bb1651bc79","subject":"ring: Centralize device normalization","status":"MERGED","created":"2023-04-26 22:36:18.000000000","updated":"2023-04-28 17:37:54.000000000","submitted":"2023-04-28 17:36:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"881671","meta_rev_id":"b970649926a71f8d146b503eb5087da9dc9345cc","_number":881671,"virtual_id_number":881671,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-28 17:36:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"date":"2023-04-28 15:31:30.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2023-04-28 15:21:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2023-04-28 15:21:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-26 23:42:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-26 23:44:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-27 04:44:19.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-04-27 15:19:15.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"f4b97a3c73242e730b3e6ab21a1e0730e2b21be4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-26 22:36:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdf03e408dad786146777a20d91e0cc1609cbf05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 23:42:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b77f4fe92057471f9ca2f0cbf16575ae\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a93f371809b4e9999fcc513a10af106 : SUCCESS in 15m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df097ba78b2c403094289d50a7d7fa92 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b9921c8f86c7464f90ed17957075da53 : SUCCESS in 1h 04m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/60e706d13a274331b9beebb8c5a145d4 : SUCCESS in 18m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9dd1d34927b9aada532297f4cd1307064663b76e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 23:44:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78e0099dd3a84b73a55f73c732b87894\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f580fdfd7d85472db015c21542679830 : SUCCESS in 6m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/0dc2c8d5a9d4462cae3919c6005000e2 : SUCCESS in 57m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bf89616864e042c7892fbd4b2e5462ce : SUCCESS in 29m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85ab4636682c4bae87bb252cbe061912 : SUCCESS in 1h 00m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1dfde0c98fac4a43a4a5a1db99e06473 : SUCCESS in 13m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89c3e2ffbc3f470aba55cb38c7858a4f : SUCCESS in 15m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc751a9b80d542518d542d2049b76b12 : SUCCESS in 14m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ed1c8d828d87400cba046b094a6cb142 : SUCCESS in 9m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/007c30de13fb4f71944a59861f38da7b : SUCCESS in 4m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1533f06363a24b8f937fcb534d4aeba5 : SUCCESS in 17m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/71b3d378e7224f658f33de4015a288fd : SUCCESS in 12m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c6c5b07d155143bf82ff2667c77df3d0 : SUCCESS in 16m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/169474bd4ed14b99ac16ec368d99283b : SUCCESS in 14m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/466f32d89f20492faa8eeec4bbb59cab : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5a4cd4db80b64b79b41a0788b1869ec7 : SUCCESS in 18m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ffa8d820636044c7b94e1f88f140f6ed : SUCCESS in 16m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/68f82cb4607844d2bbd1405817b82da2 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b1553ef9a977409eb6d2bf0deccca34a : SUCCESS in 15m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5ba8ac47047949fe9aa5c4027e82fffc : SUCCESS in 11m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/90cb9235677b4f8393591cded4492f3c : SUCCESS in 5m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/36f4931691644c7ea38ef0d085d0e8e7 : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a68b2b94666d42c1bff4e932a6f3ff2f : SUCCESS in 6m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cf4c1d792ef44701905cba1757d5e212 : SUCCESS in 53m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e6a2c0f52cc486685f8f06f0eff5a75 : SUCCESS in 1h 03m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9be5460a16974f04961c4eeac4bde707 : SUCCESS in 47m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c165de4e9424e8283831d0dcf8dab3e : SUCCESS in 42m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b143124f58cc439081a9f9f292ff0f1b : SUCCESS in 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75c287b2282042f085c83a0b5407ffeb : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e1ef95f83feb4a1893b752afa05c9d65 : SUCCESS in 19m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6dc2dc59c4d04e00b1d427e9a5eb800e : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":1},{"id":"63eb442722a0541af888a744f379801d7a476abe","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-27 04:44:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"61d23e9bf3f2d7646f924f50045a6fd7d7caf233","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-27 15:19:15.000000000","message":"Patch Set 1: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cf6771dde65215e302086692daa5618b4dfdff38","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-27 16:36:37.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"74ef46828b43315411c6909a1d1df2417e5556ed","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-28 15:21:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1c196eea1b98eab4a6c20160f18a047fd383bc09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 15:22:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84dbcf7dd52bd866f3ccd7454fd5cfd2fa35c942","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-28 15:31:30.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"3ad5662483797ae9306e39c989c7aef2612a2952","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 17:36:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8cf3f0630bc4e579b1cfc1287fcca6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1363a625b8748cc9a236fa0ee8967db : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/b57108aaaabc46c8a8ea65445f1381aa : SUCCESS in 50m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/611198a17a184849a87816e2ce1977c9 : SUCCESS in 42m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f9b8c8f10c54c0ea2cd6e27b0457e91 : SUCCESS in 1h 08m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7264b0b89f1b45998d6e1319ac40f39c : SUCCESS in 10m 03s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/285042e74e8f429c8cb8da82bcaabd8e : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f4910a5abdbb49d38692c6a65ea3d9cc : SUCCESS in 40m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5f24d7fdb9f54db08e2a6e3b38b70d57 : SUCCESS in 17m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/34ccb05a0d6b434f969105f80fedc0c3 : SUCCESS in 14m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5470766cd1bb40bd9364c3e168cd335d : SUCCESS in 16m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bc8128891caa4e43a43d1d148f944250 : SUCCESS in 19m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/78ec6a04e870439eb7da9a6ee394a0a4 : SUCCESS in 17m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a7086054e36d4beaa8b3e33af60bc708 : SUCCESS in 17m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/32ed78027a414be5ac5c3809829cbaf2 : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/de52dc6d032442ff856eb397fb921e5a : SUCCESS in 31m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8a6966f30a064562bba3dea638e4ca9d : SUCCESS in 16m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b392318aaa094be4b91a657a194f397c : SUCCESS in 6m 16s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1c90394541a244fdb48fefca5122ce36 : SUCCESS in 7m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d0e8a57815db456db03526f5dc36ee3f : SUCCESS in 51m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/33aad2481f5642089acd61ee8d0c39b6 : SUCCESS in 1h 33m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/747824f6b0a745a3ac605005ace026bb : SUCCESS in 57m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a8c1751bb59a4b4b8e6f67af27e63a9e : SUCCESS in 42m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3536de452ef4132bf524cfb310f42d0 : SUCCESS in 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81923aa098644557a644dfc81a379f45 : SUCCESS in 5m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84e302c6890d4cbb88b0b235f8f9a16d : SUCCESS in 52m 59s","accounts_in_message":[],"_revision_number":1},{"id":"951ac14729f4048ddcbffb3ee3491be8c387e6b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 17:36:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b970649926a71f8d146b503eb5087da9dc9345cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 17:37:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b17b69e2544641d8826d94c79b441d5b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af8a291758a456dade9d3fe2259f881 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/bc1cd3d413d24239b9559e9c0bd749e1 : SUCCESS in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2eb32110528a409fba76a04170fb5706 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8f0a0ed5cf9dc79ed29073c15c9996cb55e4efb","revisions":{"b8f0a0ed5cf9dc79ed29073c15c9996cb55e4efb":{"kind":"REWORK","_number":1,"created":"2023-04-26 22:36:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/881671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/881671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/881671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/881671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/881671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/881671/1"}}},"commit":{"parents":[{"commit":"9f7527973997baa5e685ea57d1563232234bc872","subject":"Merge \"Assert ChunkWriteTimeout errors are logged\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f7527973997baa5e685ea57d1563232234bc872"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-26 22:34:04.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-26 22:35:58.000000000","tz":-420},"subject":"ring: Centralize device normalization","message":"ring: Centralize device normalization\n\nThis puts replication info in antique rings loaded with metadata_only\u003dTrue.\n\nCloses-Bug: #1696837\nChange-Id: Idf263a7f7a984a1307bd74040ac8f8bb1651bc79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8f0a0ed5cf9dc79ed29073c15c9996cb55e4efb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8f0a0ed5cf9dc79ed29073c15c9996cb55e4efb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
