)]}'
{"id":"openstack%2Fswift~881673","triplet_id":"openstack%2Fswift~stable%2F2023.1~Ia2b3cc4f719347705dec15f1fc97d982e72bfb34","project":"openstack/swift","branch":"stable/2023.1","hashtags":[],"change_id":"Ia2b3cc4f719347705dec15f1fc97d982e72bfb34","subject":"CI: fix lower-constraints job","status":"MERGED","created":"2023-04-26 22:45:59.000000000","updated":"2023-04-27 17:21:43.000000000","submitted":"2023-04-27 17:19:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881673","meta_rev_id":"503d4a56072d3e7e60818132b2f5a68eba21af0a","_number":881673,"virtual_id_number":881673,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-27 17:19:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-27 16:06:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-27 16:06:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-26 23:44:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e34eae6c29aae060bb8d098f9befa00443a0f5da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-26 22:45:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71a026a1a572188b0b8d802216f06ed26c38eb51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 23:44:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fb2c27574b4483e8e3339569eb06a5a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/100a7270136f4f67996d2241bad98e64 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e9561bb3d704dafa17983fe8c63d962 : SUCCESS in 5m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/8a64a73593df4d0e880a54c5a6fd738a : SUCCESS in 42m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0f22dc4a764f4a1cab010f61decbd869 : SUCCESS in 33m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00c3bdcea93940838a1c7e1ab01c38f5 : SUCCESS in 45m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f55aa7fa29fb4b89806c0ebcfa7f30b6 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b70eed81a3904fba881c182cf6184dd2 : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91002a28cab54b9190c677b9fab1b6ea : SUCCESS in 20m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f81a0f1951694233880feb506ab4005d : SUCCESS in 8m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9b6236cedfe0447f8adb358e770fd604 : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f51656f43f8b46f2acf2b7dde26b0c23 : SUCCESS in 17m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d290ce3cfa3a4c4ea3ce8ea488bbd09f : SUCCESS in 15m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/99ccfb0cc0cd4bb5908db578cfaa4cdb : SUCCESS in 15m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/480f52e582da46f39b42cc5d4c79a4bb : SUCCESS in 15m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b76f6b3d976f4593bbee16a3ba61fb41 : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5ecf762e3fe54f5ca4c608e339432d2f : SUCCESS in 17m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c92843dcb58c40908f7a4851bd72d34a : SUCCESS in 20m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7c43ab1c343947c1a01efaff17117cf6 : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e81510fda4e146a6835d640362d51847 : SUCCESS in 12m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7c10a24df6bd49fb82bb15cf22e798a0 : SUCCESS in 17m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a9dadcd1d3804db48c907ff1fedcb604 : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7a1ff87fd67849ef8be9922e8e4cac78 : SUCCESS in 13m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1268182ae3164e1ba306cbd09a790b24 : SUCCESS in 6m 25s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bb0d87e7e00546989e1f1508451188dd : SUCCESS in 49m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b662b2449584c6e8b0486e7170f130d : SUCCESS in 53m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e28f1d5911d846bfb51620e3913653d2 : SUCCESS in 41m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/286210a5bda6449e946c3778cab02325 : SUCCESS in 41m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/170378e5172d4937baff71c38b73636d : SUCCESS in 19m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e664bf4490d3495f9d04f1a71b8e265e : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6130b5b05b004b18b545428d899b36ba : SUCCESS in 19m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6618d437e4524061b45e8e3abe8003b3 : SUCCESS in 53m 31s","accounts_in_message":[],"_revision_number":1},{"id":"17b6a40a0449ff09361d341198076dd6ad53fd43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 23:48:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea547522dd894252b7dbd6d7eb1fec06\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f076afe169f483398a946e858891dda : SUCCESS in 15m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b659fa7be4014e008e0a7a33a416615c : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b348ea55685740878f8a78851ae2ede1 : SUCCESS in 1h 00m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b097981e4b0046ef9b45bc7ec664be18 : SUCCESS in 19m 00s","accounts_in_message":[],"_revision_number":1},{"id":"70cfa2bed61d03d6c5f22fe0e55f0211880b8839","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-27 16:06:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"780c9646a2b4951c4e464a4d33d154a7f6bc6e97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 16:07:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"747c7222aa1dd753d4592d0cc0ccdc192c3d1a6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 17:19:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4368f3baaf2c4070bd6aba3c0c948828\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28064197c09f4e379028d1ad7580b415 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fad29608589444db4ea747c19f517d9 : SUCCESS in 4m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/a71da1ac5dce43d690981b6bce949535 : SUCCESS in 59m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/65ad80f0ccea4dd888d9cc0bdf8fb5fa : SUCCESS in 40m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e808c31e6cbc44fd9332e9869d1e7513 : SUCCESS in 46m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/da81307655bd4656b5920e0e098cac40 : SUCCESS in 8m 33s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/78222a4b873e412b8035aff4792e8cb5 : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/af6ecb3588a340dd8f0826df2ef99ee8 : SUCCESS in 18m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f966810679484574823c50200de148b5 : SUCCESS in 16m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9da4319855564108b53f7193abaa79a4 : SUCCESS in 17m 45s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bdfc3b7094d84b34a83b8c498c5df37c : SUCCESS in 14m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2af8497446304368b5a4c8c168c1cb15 : SUCCESS in 18m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3d5c79bf07194f398f4d3c29306ef4e4 : SUCCESS in 17m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a82c132fb44f47c9a5b8641e90af9621 : SUCCESS in 20m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/90dfbac3707f485a85c591a94f369e4c : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fa258bec646143b795d0e7efdb68ebdf : SUCCESS in 15m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7396af747c124d318f3b091bae18e274 : SUCCESS in 15m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/93211ca4cb434d32baa0345781750e95 : SUCCESS in 7m 00s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6f79b4fde4304cecb9dea419fb50591c : SUCCESS in 7m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9d3d4bc16b814fdfaf1763fa90f9f92c : SUCCESS in 52m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/842ab060eb764bfb936a27e33e46208f : SUCCESS in 1h 00m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/39d04c0934504382973ab1fd319df953 : SUCCESS in 52m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/474813c7ab7c42269bc3b2df20878c2a : SUCCESS in 50m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76ac4840b1a74b038d52be4fd7abe475 : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/844cf1d47b334da98b5c930cdc014586 : SUCCESS in 4m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ac1f2040d4f44299d29732144be236a : SUCCESS in 57m 15s","accounts_in_message":[],"_revision_number":1},{"id":"12b07a7daf921ee311459f9a01febae01b5bda55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 17:19:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"503d4a56072d3e7e60818132b2f5a68eba21af0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 17:21:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eb352fcb23c4d57ae862dfdbddfaa53\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2744078283b14c0c867168276a0d687f : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcb42c1d92044cd492c1708942bfd43a : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/03fc512d87df45a8a8d15f389d69b972 : SUCCESS in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0517e353902549b48b19cd597f685d0f : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44aee248073728d56bc7f41a815198520ad0c605","revisions":{"44aee248073728d56bc7f41a815198520ad0c605":{"kind":"REWORK","_number":1,"created":"2023-04-26 22:45:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/73/881673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/73/881673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/881673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/881673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/73/881673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/73/881673/1"}}},"commit":{"parents":[{"commit":"16df8acc91c5707667c247d9f9d149cbf97c6cc0","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/2023.1\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16df8acc91c5707667c247d9f9d149cbf97c6cc0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-20 17:27:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-26 22:43:23.000000000","tz":-420},"subject":"CI: fix lower-constraints job","message":"CI: fix lower-constraints job\n\nThis is a combination of 2 commits.\n\n----------------------------------------\n\nSwitch from focal back to bionic so tox gets installed with py36. This\nensures that we get virtualenv\u003c20.18.0 so we can still create py27\nenvironments (support for which was dropped in virtualenv\u003d\u003d20.22.0).\n\nNote that this also means we can drop our tox\u003c4 pin, as newer tox does\nnot support py36, either.\n\n----------------------------------------\n\nCI: Use [tox]requires to pin virtualenv\n\nUsing an antiquated distro to keep us on an EOL\u0027ed version of python\nthat *just so happens* to have had support dropped for it around the\nsame time virtualenv removed the ability to create py27 environments\ndoesn\u0027t seem like the right way to address the problem.\n\nMake sure we also include the tox\u003c4 pin that we\u0027re using project-wide\nso we don\u0027t get some nonsense about \"Multiple top-level packages\ndiscovered in a flat-layout\".\n\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/881035\nRelated-Change: I32a161e555179ca34d306ac37e4097611853e36b\n\n----------------------------------------\n\nChange-Id: Ia2b3cc4f719347705dec15f1fc97d982e72bfb34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44aee248073728d56bc7f41a815198520ad0c605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44aee248073728d56bc7f41a815198520ad0c605"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
