)]}'
{"id":"openstack%2Fswift~881709","triplet_id":"openstack%2Fswift~master~I856b690b7319100fb8ea80d08d75b7de48c70e4a","project":"openstack/swift","branch":"master","topic":"timeout-detail","hashtags":[],"change_id":"I856b690b7319100fb8ea80d08d75b7de48c70e4a","subject":"Unit test for logging Timeout elapsed time","status":"ABANDONED","created":"2023-04-27 10:02:13.000000000","updated":"2023-04-27 15:09:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1282a4da6ca059f76a80f648fc2d1e5b33a38331","_number":881709,"virtual_id_number":881709,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-27 11:11:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-27 10:26:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-27 11:11:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee81c5e95636c61153e0010ec57cfd7b1610849d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-04-27 10:02:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f378ad850cf39a10880d47e42a4b1ee502753569","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 10:26:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/deef422c2aa943309f8c0d389ff60c9e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab2710a437484ae99ada67038f34febc : SUCCESS in 19m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6bbaaf86248247cca5f7ba53888a88a1 : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ddd15fdd5aa04e97bb776cd981fe186e : SUCCESS in 23m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/079337720d104c73b8d45be581d45613 : SUCCESS in 23m 27s","accounts_in_message":[],"_revision_number":1},{"id":"04134e1c3e9a34c07c908f66982af47a1bbe05f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 11:11:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8fadd1de81c4e09b63f579f61310443\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/746f67fe71a84c9fae1272c2d8b0a8ea : SUCCESS in 8m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b1067f503a447559cc9fa85f6314e6e : SUCCESS in 27m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a49f5bb6d5c3449f8b937f8d4266aba8 : SUCCESS in 17m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3fc2032ef9d0440f98f7484cac3c82ce : SUCCESS in 19m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3d132fc576a24c5aac5affe78261f915 : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/18fcc32b64844ca0b09756c724ded61e : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/13603d890b4b4f239da6ffbd7e8c12be : SUCCESS in 14m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9938c7bab1bb4c7d9ec54b318d0ce0d4 : SUCCESS in 16m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f763edce0c01455dbe853ed8b8dc74ab : SUCCESS in 14m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2afd4d3d494049bea243502e6418f745 : SUCCESS in 26m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/052196e22d5943a48d54729187cba7ba : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fa4e6378b3d540fe80ed10369008f73f : SUCCESS in 19m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/36914d9a5c7a4e5a83ad190746e66e8b : SUCCESS in 18m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/46b9f89a5aaa4860a1f03b87c398248b : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1b78af804928406e9732577c652d392a : SUCCESS in 17m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5e7e5e9d85d04a54be9cbc4a32c5da01 : SUCCESS in 17m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c761edfa9714975a2e2014240c6f83b : SUCCESS in 52m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/78fbc17f093445cabbe9a863efb293b1 : SUCCESS in 56m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f8eaf0cb5c846279e934ae6b2350c91 : SUCCESS in 20m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01d9752817994cb1b1beb9796de57500 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8689da323832427a98097d258300c350 : SUCCESS in 24m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1282a4da6ca059f76a80f648fc2d1e5b33a38331","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-27 15:09:03.000000000","message":"Abandoned\n\nsquashed https://review.opendev.org/c/openstack/swift/+/881520","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46443fe2e0133b7cb6375fc252958907f870a782","revisions":{"46443fe2e0133b7cb6375fc252958907f870a782":{"kind":"REWORK","_number":1,"created":"2023-04-27 10:02:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/09/881709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/881709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/881709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/881709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/881709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/881709/1"}}},"commit":{"parents":[{"commit":"a596cb162c43d91269c8c733aae0aa91eddc8a72","subject":"Log (Watchdog\u0027s) Timeouts with duration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a596cb162c43d91269c8c733aae0aa91eddc8a72"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-27 09:56:52.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-04-27 10:01:23.000000000","tz":60},"subject":"Unit test for logging Timeout elapsed time","message":"Unit test for logging Timeout elapsed time\n\nChange-Id: I856b690b7319100fb8ea80d08d75b7de48c70e4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46443fe2e0133b7cb6375fc252958907f870a782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46443fe2e0133b7cb6375fc252958907f870a782"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
