)]}'
{"id":"openstack%2Fswift~881752","triplet_id":"openstack%2Fswift~master~Id8cb4d3583e6e3ca8184c6ef38027634dbe843a2","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-04-27 22:23:42.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id8cb4d3583e6e3ca8184c6ef38027634dbe843a2","subject":"DNM: turn on debug logging for reconstructor in probe tests","status":"ABANDONED","created":"2023-04-27 17:24:07.000000000","updated":"2023-04-27 22:23:42.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"80110bd9d83b7a0c95dfba5ba1fb7d59ca42be9f","_number":881752,"virtual_id_number":881752,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-04-27 20:55:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-27 17:47:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-27 18:42:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-27 20:30:09.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"31ae7a190f564c7b753261273c24b28fd72c5c3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-27 17:24:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17324b2af39a7a7fe46f85c17cb00b47aa3953c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 17:47:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cff58cd8c97848efab61860d7719aee5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df5ea814418e4ac1b3bbf33f9afd6d10 : SUCCESS in 17m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b221a5297f6498a9db08f9977c26f15 : SUCCESS in 19m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8265ef911c0049a9b785ea931c320adc : SUCCESS in 22m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4e0b3705c064335a89bff2717ec8c81 : SUCCESS in 21m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1e72c499cfd6751fb9717a0b277538a029657ecc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 18:42:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82b02bc0412a4803bbb39367ff1f9620\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb4be50583c744cebeb9b95af0f7ef10 : SUCCESS in 9m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8a5588386d1142b8a23b965b68326b0c : SUCCESS in 7m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9a9487d99b804800bf2cb27da83352a9 : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a3e6135033334e41a8575e1df52f1beb : SUCCESS in 8m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e0fedaee09554577be2335f2bd60f4f7 : FAILURE in 56m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a90a852a07d4c85b2b21deb956079f7 : SUCCESS in 1h 08m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb501552a8f74227a78c30884fe64afc : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"56247779a722a7e0e12bc4a2430d74e3938c5fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-27 20:21:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4a1c3232fa49bf14e6bd43078852ae73f42b0768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-27 20:29:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6a9c6d1a72fc7bdd3afc446ead35c99301b172dc","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-04-27 20:30:09.000000000","message":"Patch Set 3:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@master\", line 678, column 31","accounts_in_message":[],"_revision_number":3},{"id":"16c6f76375d6fce34bed846f19e059827beab285","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 20:30:15.000000000","message":"Patch Set 3:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@master\", line 678, column 31","accounts_in_message":[],"_revision_number":3},{"id":"283e6cc3cc5465c9d3b999a2eb22377ba69e7df2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 20:30:21.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/swift/.zuul.yaml@master\", line 678, column 31","accounts_in_message":[],"_revision_number":3},{"id":"41c4fb0d89b22536198a6ac66974f648132d8802","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-27 20:31:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"552d4359d5f6079173bba4ce20ee73b2c9300e9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 20:43:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bab8dd846e94795b66e586be0ac3f37\n\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/458adb1704f241cbae3306226f6b5038 : SUCCESS in 5m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a37af43909341d2ac23c99d151746b3 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":4},{"id":"94a1b9ab982a75bdf7f9f840ed90584f9622c095","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-27 20:44:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d454dad84682daa640559078662946a5aa687d1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 20:55:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/44480dcd8fb84eb8aaf42ec2d538f2fc\n\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/562bceadce144b05972d030fbfb5eb6e : SUCCESS in 5m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ceb37fa23b554b169543f44cf7de345b : FAILURE in 5m 54s","accounts_in_message":[],"_revision_number":4},{"id":"80110bd9d83b7a0c95dfba5ba1fb7d59ca42be9f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-27 22:23:42.000000000","message":"Abandoned\n\nThink I found the root cause -- see https://review.opendev.org/c/openstack/swift/+/881672","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9d87a40d083afe2b27bfe66311f7417a9062846a","revisions":{"9f9469659a297fcd0ec04f6801e7d55299948879":{"kind":"REWORK","_number":1,"created":"2023-04-27 17:24:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/52/881752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/881752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/881752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/881752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/881752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/881752/1"}}},"commit":{"parents":[{"commit":"b5a2453ed75e4f31d63f719a9c7af9b59bcd81f7","subject":"Merge \"ECFragGetter: simplify iter_bytes_from_response_part\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5a2453ed75e4f31d63f719a9c7af9b59bcd81f7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-27 17:23:16.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-27 17:24:01.000000000","tz":-420},"subject":"DNM: turn on debug logging for reconstructor in probe tests","message":"DNM: turn on debug logging for reconstructor in probe tests\n\nRelated-Bug: #2017021\nChange-Id: Id8cb4d3583e6e3ca8184c6ef38027634dbe843a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f9469659a297fcd0ec04f6801e7d55299948879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f9469659a297fcd0ec04f6801e7d55299948879"}]},"branch":"refs/heads/master"},"b1df8b0334f102a2dae1a3cef03e28d81a7610fa":{"kind":"REWORK","_number":2,"created":"2023-04-27 20:21:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/52/881752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/881752/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/881752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/881752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/881752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/881752/2"}}},"commit":{"parents":[{"commit":"b5a2453ed75e4f31d63f719a9c7af9b59bcd81f7","subject":"Merge \"ECFragGetter: simplify iter_bytes_from_response_part\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5a2453ed75e4f31d63f719a9c7af9b59bcd81f7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-27 17:23:16.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-27 20:20:32.000000000","tz":-420},"subject":"DNM: turn on debug logging for reconstructor in probe tests","message":"DNM: turn on debug logging for reconstructor in probe tests\n\nRelated-Bug: #2017021\nChange-Id: Id8cb4d3583e6e3ca8184c6ef38027634dbe843a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1df8b0334f102a2dae1a3cef03e28d81a7610fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1df8b0334f102a2dae1a3cef03e28d81a7610fa"}]},"branch":"refs/heads/master"},"3ac894903df2ac0495bd7a8994e44ae21246c92c":{"kind":"REWORK","_number":3,"created":"2023-04-27 20:29:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/52/881752/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/881752/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/881752/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/881752/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/881752/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/881752/3"}}},"commit":{"parents":[{"commit":"b5a2453ed75e4f31d63f719a9c7af9b59bcd81f7","subject":"Merge \"ECFragGetter: simplify iter_bytes_from_response_part\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5a2453ed75e4f31d63f719a9c7af9b59bcd81f7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-27 17:23:16.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-27 20:27:12.000000000","tz":-420},"subject":"DNM: turn on debug logging for reconstructor in probe tests","message":"DNM: turn on debug logging for reconstructor in probe tests\n\nRelated-Bug: #2017021\nChange-Id: Id8cb4d3583e6e3ca8184c6ef38027634dbe843a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ac894903df2ac0495bd7a8994e44ae21246c92c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ac894903df2ac0495bd7a8994e44ae21246c92c"}]},"branch":"refs/heads/master"},"9d87a40d083afe2b27bfe66311f7417a9062846a":{"kind":"REWORK","_number":4,"created":"2023-04-27 20:31:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/52/881752/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/881752/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/881752/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/881752/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/881752/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/881752/4"}}},"commit":{"parents":[{"commit":"b5a2453ed75e4f31d63f719a9c7af9b59bcd81f7","subject":"Merge \"ECFragGetter: simplify iter_bytes_from_response_part\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5a2453ed75e4f31d63f719a9c7af9b59bcd81f7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-27 17:23:16.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-27 20:31:18.000000000","tz":-420},"subject":"DNM: turn on debug logging for reconstructor in probe tests","message":"DNM: turn on debug logging for reconstructor in probe tests\n\nRelated-Bug: #2017021\nChange-Id: Id8cb4d3583e6e3ca8184c6ef38027634dbe843a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d87a40d083afe2b27bfe66311f7417a9062846a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d87a40d083afe2b27bfe66311f7417a9062846a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
