)]}'
{"id":"openstack%2Fswift~881753","triplet_id":"openstack%2Fswift~stable%2Fzed~Ia2b3cc4f719347705dec15f1fc97d982e72bfb34","project":"openstack/swift","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-05-02 20:38:26.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ia2b3cc4f719347705dec15f1fc97d982e72bfb34","subject":"CI: fix lower-constraints job","status":"MERGED","created":"2023-04-27 17:36:32.000000000","updated":"2023-05-03 04:40:44.000000000","submitted":"2023-05-03 04:38:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881753","meta_rev_id":"a9806d6ac7b93dcaf94571bccc83bc587643757e","_number":881753,"virtual_id_number":881753,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-03 04:38:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-02 20:38:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-02 20:38:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-27 18:00:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-27 19:01:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c31decc07d835930e6e41c28588a37c0ad1d3273","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-27 17:36:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e011f92508bb2bf5ed1b7fcbcae80158091df2e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 18:00:25.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c35225eee27482c98ddc8a4e1b5b4da\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/54a962b35e9049cd8923527a57640a48 : SUCCESS in 16m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/13463400f90541d3bf9fb6ba0e8a9d5e : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8e5be4a48d946d59ef436fb9a35d851 : NODE_FAILURE Node request 300-0021080410 failed in 0s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1de55dbf20e42de8fdc182096534a9d : SUCCESS in 22m 00s","accounts_in_message":[],"_revision_number":1},{"id":"958afabda2bb3f039d28a62236103bf595d12f70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 19:01:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9464ebfb5d95401c97b2c58448fbaa14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4ad9c8955174f0e9a0b6be8b9b9fd3f : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8aef5f1e8265486ab89307f969a9a5b2 : SUCCESS in 4m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/174f017e71c749aabd973c029c4ebcfb : SUCCESS in 1h 14m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f0989aee80b04bb19c84b76b494636b0 : SUCCESS in 42m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea9d082312eb45dba7c74dea91b29545 : SUCCESS in 1h 02m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/278bb11431d845bd8c747a1a99dc3ec7 : RETRY_LIMIT in 4m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bee901291934432d9a3e9fbe23ca6650 : RETRY_LIMIT in 5m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/170d73cd98244fee9b830bb51666fde4 : RETRY_LIMIT in 5m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/955082d742564688ae97b40810e6e4a9 : SUCCESS in 8m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/776d378197b44f2aa183dd33e47cb892 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/af2c5ca14048444589bea2ceb4cd5d76 : SUCCESS in 18m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a27d39eaa94a42f3b83ecd7fcc07334d : SUCCESS in 15m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3218bd3c4153473ba4471c1ea0b0cdc9 : SUCCESS in 16m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dc2b113bd38b42e988666feb2da71f2d : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/43bb860d236e45ec913971cae28e1982 : SUCCESS in 16m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b84fc62c38484d9ea2fa49b539ac635a : SUCCESS in 15m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/640876745a844fb4a14d2f08923349c2 : SUCCESS in 14m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c64b5c31ea2045989572ce49b5957dcc : SUCCESS in 13m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f6dacd08ee084fb8a4897c9d636b7f27 : SUCCESS in 17m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3b7ce7cba2ee4758ad4a054379e51c57 : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9f944d1aae0341ec9a3ecae53126ab9a : SUCCESS in 15m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/80df1473021344acb2876038b8047814 : SUCCESS in 6m 25s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/32d124081b3546f497dc64d0269b49d8 : SUCCESS in 51m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec83788c1fb8498caac5daeb1e95fb84 : SUCCESS in 56m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2f8d0e25d685435c851f0cc59a8164ee : SUCCESS in 48m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/25c6b4a255ab47ed9580eafda512b4f1 : SUCCESS in 42m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29ee4e1515004852a0e70bd4ae3be94c : RETRY_LIMIT in 1m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bba933d0ba3b44959dc2139770b3238b : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dc396a8ab5f04fb8b4ef616008638c7d : SUCCESS in 22m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3528940d5bf04f8596fd1da32807c230 : SUCCESS in 51m 07s","accounts_in_message":[],"_revision_number":1},{"id":"db708595dcfc7be7668b56eeb3e59613f5b6fc0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-28 21:03:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"15c853c2d3d801df5280a1feba48d7519199f3e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 22:16:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/632550d793dd43ec877bda79cee72aae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59d7cb1a73e4601b5ac5cb75bcc52ff : SUCCESS in 8m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0496b27214574e6280e95ac0d4b2d800 : SUCCESS in 4m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/08d4b99d5d0e47019503304df3d88ecb : SUCCESS in 1h 02m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/691325fe4e714e6b94400acd5282a218 : SUCCESS in 51m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54969710cdc24bb483473efe40c46bf2 : SUCCESS in 48m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1bb2cb02654466ba98b36c86272a498 : RETRY_LIMIT in 6m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05657e84c2044b818e32a07159e43a10 : RETRY_LIMIT in 5m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/501c10cd76664bd78743eeb0d1559ca8 : RETRY_LIMIT in 6m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fcda892210074839b7b38d413006ec0c : SUCCESS in 9m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fc6f39cab35c4ea5acf45c12d4845e2f : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f5ade919bbfe4ebb964a3e93fc890f62 : SUCCESS in 15m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/65f480413776429996aba14522494a11 : SUCCESS in 15m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/de2a2a2e4f18461985270e8865265388 : SUCCESS in 29m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1787ad9875c84351a00832c0bc31ade1 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ce18bac9a88148049f58e67e73f82117 : SUCCESS in 29m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7e69ae46cfc54c86b2e0a3168137b541 : SUCCESS in 13m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2461fd8f7b4043f99b722368ce92d04c : SUCCESS in 33m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6f37bc13711745c5aad95449ed49f756 : SUCCESS in 16m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8ed3ec3d82894b7ebdb7b03208547bf7 : SUCCESS in 14m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d92b666763fe409b85c0c06d5bda8755 : SUCCESS in 7m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/403a5cadcbd14f3d8f1dc36b4612488f : SUCCESS in 16m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ceb3b1778c7642c3941c3918b6894174 : SUCCESS in 8m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d9f8ef469ec040b49286cc1936ef709b : SUCCESS in 51m 14s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d3ae7c1f1d374909bb4bce9f73d8d68e : SUCCESS in 1h 10m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/070b9e1d235646f39eee415ae049f294 : SUCCESS in 32m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7f869d4382724cb0b5dcb2fc12ce8b41 : SUCCESS in 39m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b57aa2a3a7d6486c9dc8f12cd5ba8f6c : SUCCESS in 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51ccfa6e8a464f1fa64acbec3299d952 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b1bb263bfdff410eafa15b12692ed6d0 : SUCCESS in 20m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e59e3a6ff1f4b518407e7a1c83b25c1 : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":2},{"id":"b2ffdcb3b20353ddd4c1b1d8bf894280d4dd0a98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 22:27:19.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2987d08b1ed46baa7994581eaac7434\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae781d2dde0e4139a6f7c3fb8f13dcf2 : SUCCESS in 1h 04m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/072d1129cc9c40f199c3d231296926f6 : SUCCESS in 1h 16m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab411252da1d47e5842ab548263e6ca3 : SUCCESS in 24m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/613df00f10fb4726b6b2da2cfffd6a95 : FAILURE in 1h 21m 13s","accounts_in_message":[],"_revision_number":2},{"id":"0617e700be6e8c9dde969e37186426aa5d1ab1e9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-02 20:38:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"36381dbc38f81bf9951019350f8bf0ae2e3b30c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 20:39:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e948f1639f1312ac88d57460c3f1ff140f325f23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 04:38:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ecb0df0324c48848485a43acfa1a68d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35ad4b285288425f87e7237715840219 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47a540dcc12c4e3e876ca0fad7bd6d1c : SUCCESS in 4m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/911aa44e11cd42a6badc439a66c30cc8 : SUCCESS in 1h 03m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a7672310fc6e4c9ca5d66db3eb9fa687 : SUCCESS in 44m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6f895ace1064a3caea3a4d559f1f005 : SUCCESS in 53m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8bbf507a7d9f4a588852ec098b3bdc86 : SUCCESS in 10m 25s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/9d4c571a16c641deaeeeebe70ce90cc9 : SUCCESS in 7m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ef045085ef604258ae8d63bdd1d8a90e : SUCCESS in 14m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8326582f99924df488b5782d3bf5e3f5 : SUCCESS in 14m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b560898d3b164d17b46956f755ab1dbd : SUCCESS in 35m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c84adebb5f234f4b8dec8a51b2ab88c3 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0141043976e420c9d1aa533db2075e9 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d3f914fd41504a9891ffd5278bcc6ebd : SUCCESS in 16m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9abdba08dcd44a5e8b97c8c28dc54447 : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/26b3e9383b9b4d1f8d8cb97f7b769bac : SUCCESS in 21m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2801f986788347c09348154a4f588d20 : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c9bfb59956664f73865d9102f3de82a7 : SUCCESS in 9m 37s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/abbe16b46d8749deb39fdaf18c15de61 : SUCCESS in 6m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/af3f5069ee00479fb0cd9768f6eabb45 : SUCCESS in 54m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd25e6d19c8a4fdf88c44affc65c01f0 : SUCCESS in 1h 17m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/05da8a10d0d941f0850b9ebf631454e6 : SUCCESS in 56m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cbeaa7e91005445e82d675fac0559a6b : SUCCESS in 50m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea795d85fb054992a87649449bc42b88 : SUCCESS in 17m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5905e4478c4013a09a246a68fca8e1 : SUCCESS in 6m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4610e8adbcef4e34bf9dee9d19a139ef : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":2},{"id":"4c5c97d6d36f0961c34b47df4f36dbf9527316f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 04:38:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a9806d6ac7b93dcaf94571bccc83bc587643757e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 04:40:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c24d944ffb864e579360d2186a201b05\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b9b4a6ce2bf4a32b18b614e538862e1 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b8c1d8a1f244ba5965ebd1e8537fc06 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/20fcd10475614499ae98258dce101d0d : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d40a092adfce46c9afdb2456e13295e6 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4d4fd0ce147d5602ab64d7483bdb1d816be294bd","revisions":{"88a4ebd5b8fa4b6274f28be31bcc2467fc1b8b45":{"kind":"REWORK","_number":1,"created":"2023-04-27 17:36:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/53/881753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/881753/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/881753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/881753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/881753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/881753/1"}}},"commit":{"parents":[{"commit":"708df56d5a88a3204c2e989f6a0f48ca383d5362","subject":"Drop bandit check B309","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/708df56d5a88a3204c2e989f6a0f48ca383d5362"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-20 17:27:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-27 16:08:06.000000000","tz":-420},"subject":"CI: fix lower-constraints job","message":"CI: fix lower-constraints job\n\nThis is a combination of 2 commits.\n\n----------------------------------------\n\nSwitch from focal back to bionic so tox gets installed with py36. This\nensures that we get virtualenv\u003c20.18.0 so we can still create py27\nenvironments (support for which was dropped in virtualenv\u003d\u003d20.22.0).\n\nNote that this also means we can drop our tox\u003c4 pin, as newer tox does\nnot support py36, either.\n\n----------------------------------------\n\nCI: Use [tox]requires to pin virtualenv\n\nUsing an antiquated distro to keep us on an EOL\u0027ed version of python\nthat *just so happens* to have had support dropped for it around the\nsame time virtualenv removed the ability to create py27 environments\ndoesn\u0027t seem like the right way to address the problem.\n\nMake sure we also include the tox\u003c4 pin that we\u0027re using project-wide\nso we don\u0027t get some nonsense about \"Multiple top-level packages\ndiscovered in a flat-layout\".\n\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/881035\nRelated-Change: I32a161e555179ca34d306ac37e4097611853e36b\n\n----------------------------------------\n\nChange-Id: Ia2b3cc4f719347705dec15f1fc97d982e72bfb34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88a4ebd5b8fa4b6274f28be31bcc2467fc1b8b45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88a4ebd5b8fa4b6274f28be31bcc2467fc1b8b45"}]},"branch":"refs/heads/stable/zed"},"4d4fd0ce147d5602ab64d7483bdb1d816be294bd":{"kind":"REWORK","_number":2,"created":"2023-04-28 21:03:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/53/881753/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/53/881753/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/881753/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/881753/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/53/881753/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/53/881753/2"}}},"commit":{"parents":[{"commit":"708df56d5a88a3204c2e989f6a0f48ca383d5362","subject":"Drop bandit check B309","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/708df56d5a88a3204c2e989f6a0f48ca383d5362"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-20 17:27:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-28 21:03:13.000000000","tz":-420},"subject":"CI: fix lower-constraints job","message":"CI: fix lower-constraints job\n\nThis is a combination of 2 commits.\n\n----------------------------------------\n\nSwitch from focal back to bionic so tox gets installed with py36. This\nensures that we get virtualenv\u003c20.18.0 so we can still create py27\nenvironments (support for which was dropped in virtualenv\u003d\u003d20.22.0).\n\nNote that this also means we can drop our tox\u003c4 pin, as newer tox does\nnot support py36, either.\n\n----------------------------------------\n\nCI: Use [tox]requires to pin virtualenv\n\nUsing an antiquated distro to keep us on an EOL\u0027ed version of python\nthat *just so happens* to have had support dropped for it around the\nsame time virtualenv removed the ability to create py27 environments\ndoesn\u0027t seem like the right way to address the problem.\n\nMake sure we also include the tox\u003c4 pin that we\u0027re using project-wide\nso we don\u0027t get some nonsense about \"Multiple top-level packages\ndiscovered in a flat-layout\".\n\nRelated-Change: https://review.opendev.org/c/openstack/swift/+/881035\nRelated-Change: I32a161e555179ca34d306ac37e4097611853e36b\n\n----------------------------------------\n\nChange-Id: Ia2b3cc4f719347705dec15f1fc97d982e72bfb34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d4fd0ce147d5602ab64d7483bdb1d816be294bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d4fd0ce147d5602ab64d7483bdb1d816be294bd"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
