)]}'
{"id":"openstack%2Fswift~881875","triplet_id":"openstack%2Fswift~master~I0715a849fc93f90035b9b9e61df6e1ae85e340d9","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-05-02 20:53:35.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-07-23 18:35:21.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-05-02 23:32:51.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I0715a849fc93f90035b9b9e61df6e1ae85e340d9","subject":"Prohibit number\u003d0 in Servers/Managers","status":"ABANDONED","created":"2023-04-28 17:05:26.000000000","updated":"2024-07-23 18:35:21.000000000","total_comment_count":11,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"558bdd210b96534bfaa4bcee533e603fcbcabed1","_number":881875,"virtual_id_number":881875,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-02 20:50:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2023-05-02 20:53:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2023-05-02 21:24:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-05-02 23:32:51.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-28 17:25:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-28 17:42:03.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-04-28 19:06:17.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2023-04-28 19:33:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-02 21:24:43.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"4b641927b4a0d8d82e8f1def402f21e632375ec7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-28 17:05:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73d68f36029d70b433d4c3934d30cb1b4d57befb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 17:25:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cc9d7a37749448c97ecf57d8d8fb18e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a20ce93107e47278ade4c1b74112769 : SUCCESS in 16m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe0f59f9c4094da49d0388eb2132931c : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/65d7e05b92cb40339ca3373f64c36c31 : SUCCESS in 18m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b91662ff4a234cd8827060297bd11b93 : SUCCESS in 18m 58s","accounts_in_message":[],"_revision_number":1},{"id":"75860712cdbd270271e9c19b03e2527ad11a7cc7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-28 17:42:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"da5e547b8b08a09d10dd69a0f7bd051000c3f930","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-28 18:50:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ae2149077ff67a58114d0f37363f7722e8fa65f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-28 19:06:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7175e49db46f05cab4710d745737a1812ff823f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-28 19:08:25.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f7703691019baac17a91b900022df83f6f11290d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 19:33:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa4e8254a843484683636ae2275e8e6a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f57aeab8b524e8f800563513687e127 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/ce571519d3b54a2ca2b957399397adb2 : SUCCESS in 52m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bfac331c8d664edfa125074bac29a7d7 : SUCCESS in 44m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/923b9cece0234fd2a4b2576c1a765734 : SUCCESS in 57m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a02ce1b33aa4341b4ea196cd8b5c2a4 : SUCCESS in 16m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7aa0b7fce07a4a2a9a2cd244dab4a6a7 : SUCCESS in 17m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/538dcd6256ac49ada6436013dbc326bf : SUCCESS in 26m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6b6c889ef6eb4491b45d62899b5e88aa : SUCCESS in 8m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0950d0d506f047cd9c774b225e2e2c3d : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c111abaab5144bcfb72328ba0de8404b : SUCCESS in 16m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/409ca979b01b4911b0ed88061d948417 : SUCCESS in 17m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d73a34d0e7cd448a8e3e4665d61fbb45 : SUCCESS in 16m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/655cd29daf884f6595161a27f03e40ad : SUCCESS in 23m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4b0b77b43def4a9ebfeea098aa6cdf77 : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a0bce9b27f034f6ba85819016d939729 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b93b459dafd549a7ba6de60008840d79 : SUCCESS in 19m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2f9f4494be7e413da9bf3ddec6aa883b : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/11cf4c76dfa24f22a5fcd5115f31498a : SUCCESS in 23m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6e2b8b8dca97420ea440f068049db45e : SUCCESS in 13m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a2453f7eae3349759e65e51a4de0aa38 : FAILURE in 6m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4e0d7bfe186a418fa189b56b6bb0a659 : RETRY_LIMIT in 5m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1ec92fa562514532b4e009d5cba9186c : RETRY_LIMIT in 7m 09s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/81099363e00645a7ad459c69d16ccd75 : SUCCESS in 57m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c2d21940e87045dbaa170c8ca6590348 : SUCCESS in 57m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2cd0b503c2304d2dabd40fb7a38e1c19 : SUCCESS in 47m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3b9a60a0487749c9bc4812b887109d51 : SUCCESS in 48m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d98ed36566264f25af56cf197a6d1397 : SUCCESS in 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ad40184a4de4b8c954aa78d004138d7 : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05f8bbb5fecf485dbdf50427dd0f7095 : TIMED_OUT in 1h 30m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76829d7e3f7b4f6cb5a2419fccc01b77 : SUCCESS in 57m 50s","accounts_in_message":[],"_revision_number":1},{"id":"673e226a0205ea5d221d2b3b2cf7f6b281884f92","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-04-28 20:06:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6ea94825596829e8a32b86e17e61ad675b25a042","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-02 18:30:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"23428582c02cac0dea9f57493854f0537a52b42d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 19:49:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4aeb4ad865bc48378d8228ae332456ec\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa2cea6b78c643929e5842e9fb2832b2 : SUCCESS in 1h 08m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40db19cfd0574d348b4d4e3b51fce90c : SUCCESS in 30m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/59c71de6152045f3b1660b73208dc1b1 : SUCCESS in 1h 15m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fb6753a2de64486bd84f190e9af74ea : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":2},{"id":"77d5c919069bee21b70712645100979cfe8e852e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 20:50:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/52559511d575471bb8817af1364d3450\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1e88bfc21944ed18f696236be5dea33 : SUCCESS in 7m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/0021cf7e6a094318bfa54d8e329d30c2 : SUCCESS in 50m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d933906cb93348078d949fae12defb8c : SUCCESS in 39m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0024a12a1a8340d490525ba7054263d2 : SUCCESS in 57m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/825e85d9e11a47b8ae006c394aa44363 : SUCCESS in 41m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cef7789fcc394e229523eb47c13b556c : SUCCESS in 18m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d7ee88abfe04651b5a917e7b69a151e : SUCCESS in 19m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c446e8db4078446aa2c4ee6e1b25f34d : SUCCESS in 8m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1334194849b74d829d909452b495b4a9 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/40859b1baee04e3bb7ef34cd1e8c0dd7 : SUCCESS in 17m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d9d7d6368e4c4ff6a8cf527d1e19d0ea : TIMED_OUT in 1h 30m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a7bf7b39781e478ab160a82385f7e8bb : SUCCESS in 15m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b9e73bb029f34722bb3235ec8990aa6f : SUCCESS in 16m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ce7b5b9331a249d0b9c60ae77822e819 : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b0a05ed020cc435a87d7f42344839da8 : SUCCESS in 36m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f6e269f8e5bc4189a38000213e839c08 : SUCCESS in 18m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e51d932391d74dcfbe830f5b912a5134 : SUCCESS in 12m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/78ea9e6eee8e481d89fc8d7f7c9a78f6 : SUCCESS in 14m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2904778625fe4b2787a138364300e870 : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8db15fc49dec4a6d99c33bce9aae35a9 : SUCCESS in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/77fe1741eb3644fe917915c89dd38060 : SUCCESS in 17m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/64dd0eb17b974314843e7a1a4a89d433 : SUCCESS in 7m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bfd54e1e4c3d4e6ca551519064aa87ac : SUCCESS in 1h 42m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4877e47d964e42dd96d2e62ccfdc4136 : SUCCESS in 1h 13m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c50235ea3fa449e4835483cd8ec834ae : SUCCESS in 47m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aefdacaf84a9489195d334bae38346a9 : SUCCESS in 47m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c02f23eeddb94e98bdb5c17bb5a78b3e : SUCCESS in 17m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/817437ba9d4845dfa0e045f4dfb14fdd : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/189cb3b7017e46c7a8b45ab2c5194133 : SUCCESS in 18m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f003a1335264aa58a513b390a16b0f6 : SUCCESS in 40m 25s","accounts_in_message":[],"_revision_number":2},{"id":"8f69819ad106179269d2066c277396799d8b57ef","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-05-02 20:53:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"248174d67b165443a15429e37219add40d424589","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-02 21:24:43.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"32bfef7b9788145513dbb2e5736f24c2a6233964","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-02 23:32:51.000000000","message":"Patch Set 2: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"558bdd210b96534bfaa4bcee533e603fcbcabed1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-23 18:35:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"75144cabab02304feda1d1769387d3858e55b201","revisions":{"af63e1d5dad0dbf212eb30317bb5271b66efc93f":{"kind":"REWORK","_number":1,"created":"2023-04-28 17:05:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/881875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/881875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/881875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/881875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/881875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/881875/1"}}},"commit":{"parents":[{"commit":"b61602b70c74a4fca67910cb1cc98639e6f52451","subject":"Merge \"docs: Fix broken paste/pastedeploy links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b61602b70c74a4fca67910cb1cc98639e6f52451"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-28 17:05:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-28 17:05:19.000000000","tz":-420},"subject":"Prohibit number\u003d0 in Servers/Managers","message":"Prohibit number\u003d0 in Servers/Managers\n\nChange-Id: I0715a849fc93f90035b9b9e61df6e1ae85e340d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af63e1d5dad0dbf212eb30317bb5271b66efc93f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/af63e1d5dad0dbf212eb30317bb5271b66efc93f"}]},"branch":"refs/heads/master"},"75144cabab02304feda1d1769387d3858e55b201":{"kind":"REWORK","_number":2,"created":"2023-05-02 18:30:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/881875/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/881875/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/881875/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/881875/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/881875/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/881875/2"}}},"commit":{"parents":[{"commit":"b61602b70c74a4fca67910cb1cc98639e6f52451","subject":"Merge \"docs: Fix broken paste/pastedeploy links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b61602b70c74a4fca67910cb1cc98639e6f52451"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-28 17:05:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-02 18:30:14.000000000","tz":-420},"subject":"Prohibit number\u003d0 in Servers/Managers","message":"Prohibit number\u003d0 in Servers/Managers\n\nChange-Id: I0715a849fc93f90035b9b9e61df6e1ae85e340d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75144cabab02304feda1d1769387d3858e55b201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75144cabab02304feda1d1769387d3858e55b201"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
