)]}'
{"id":"openstack%2Fswift~881936","triplet_id":"openstack%2Fswift~master~I5c3d260d677c61213c42662c6641207d9b7f026a","project":"openstack/swift","branch":"master","topic":"sys-meta-pretty","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-07-03 23:45:34.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-06-29 09:00:24.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2023-07-04 01:01:56.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-06-28 21:34:38.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I5c3d260d677c61213c42662c6641207d9b7f026a","subject":"Properly format meta and sysmeta in db info cli tools","status":"MERGED","created":"2023-05-01 23:42:11.000000000","updated":"2023-08-17 16:14:40.000000000","submitted":"2023-07-04 01:01:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881936-sys-meta-pretty","meta_rev_id":"9368195f020685a1a45a436047db1d13c9104434","_number":881936,"virtual_id_number":881936,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-04 01:01:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2023-07-03 23:45:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2023-07-03 18:47:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2023-07-03 23:45:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-01 23:44:12.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2023-05-01 23:44:12.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2023-05-02 00:40:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-02 00:47:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-23 22:16:03.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-06-28 01:41:45.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2023-07-03 18:47:01.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"d3bc4b74747670e18d7c8f9c3963c5d8b4c736d4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-05-01 23:42:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e4d6708c26ed70346f8ae405213747073dd33d3","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-05-01 23:42:17.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"3b1e907e5db787e1e6644583e6365398138fa342","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 00:40:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63ca4c972bba404e84a35044805dd65d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b5c4332525749ceb2d2a1a665f5a496 : FAILURE in 56m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b3012339490c4d3cb2c83e5e09b47677 : FAILURE in 15m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ad7a4ad97814947bc8d522abcc6a749 : SUCCESS in 16m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/49242032e9a34ded988c0ccaf67dc35c : SUCCESS in 24m 55s","accounts_in_message":[],"_revision_number":1},{"id":"187c5aee9311e0435c6513c8cdc97f36d8134d09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 00:47:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa5f5d9f1dde48fab01b138230b0f98a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edf800a7cf034642bc49db2f40deae70 : SUCCESS in 7m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/31ec6f42269142d9880d460ae26482d6 : SUCCESS in 40m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0e15fd37ee2e46ed83669edd5984a98a : SUCCESS in 35m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9099c3f1fb7473db1dcea8695236154 : SUCCESS in 43m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27f008a1e2eb425798cf6a4dce241014 : RETRY_LIMIT in 1m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bdd8198b4f142f4b4d6c29b84c730cb : RETRY_LIMIT in 2m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1303cf01bee4403b763ad64e48fa8d6 : RETRY_LIMIT in 1m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e2dfcd1b54ae4a7ca5c69ab5215330e0 : SUCCESS in 7m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/08d1774b8bae453f8045b7e03b903cfc : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/623fea8f1d514758b95e366032d3f36c : FAILURE in 19m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c1eae2d17fca4bf48e2a74ab6d6b31c9 : FAILURE in 12m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1804f7034f26468abf7850088937ea7a : FAILURE in 10m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/101974882e7d4c219f0d2ef1e57091c9 : FAILURE in 16m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/425edad5659f4ea0a121f9457771fdfa : SUCCESS in 18m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/675bcbade70c4ce6935f8998a577e39f : SUCCESS in 17m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3c80e6e77dd54176a559f06d6b7e30fb : SUCCESS in 14m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/baa75525c94c4bb4b7862835ef28a23f : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/12c1d765008046a0854e1aec1badf2f6 : SUCCESS in 15m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7c0ea87c057245d7b9cfd0ee15dbdf5f : SUCCESS in 15m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b3b6302944e419fbf07df537dacc4b1 : SUCCESS in 5m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fc92ec406cec4864b3f5b7cb6d020ff4 : SUCCESS in 15m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/29246b9b0b38487aa6ad5e558f340a1c : SUCCESS in 7m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f0582c63e936433489dd069575e47bb5 : SUCCESS in 56m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/83d9a48011c94f11afbae1c840dd986f : SUCCESS in 1h 04m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ae6282432b345df9ab3a9e5acaf488e : SUCCESS in 38m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d322fa9692a6495ba750e6d86697e1b8 : SUCCESS in 56m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebebdbeac49f4d5390057834447a71ea : FAILURE in 19m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b36738449c3484baf1d690a402980fe : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6eb7fab9c94941c5b567abb95ef3611c : SUCCESS in 22m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3dd8f38ccb6409fa82fbb1191bc83b5 : SUCCESS in 54m 46s","accounts_in_message":[],"_revision_number":1},{"id":"41aaed5c7ae4341b79d6999e55a17eb5480f2a0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-06-22 16:20:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"04fcb2e8e809f0da2532a933b92b61ea8b54dbd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 16:43:45.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19c2975b684441d88b63f1ca2f09d9e4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a023ec0f27cf4330b8cb9bf1594c6678 : FAILURE in 16m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15f827bfbbe64f24b202d121237bb35d : FAILURE in 18m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3fcf6aaaff54134843ae7c86e941769 : SUCCESS in 19m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5158301a7a3c4fd4b12658432cb4ea6e : SUCCESS in 21m 42s","accounts_in_message":[],"_revision_number":2},{"id":"bb92dae95ddd652417fc22cb9965827faf4cccce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 19:01:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cfb97d117e4746f7888fcde734b10a4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3034b6bb13cb4272af776b567b887987 : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/e3d96fd0d8f046aabf57f35f7e50d484 : SUCCESS in 1h 10m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b9113766c054ba597af64b88bea7f09 : SUCCESS in 1h 01m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f36d7a2a5a93452cb85ae3366d8d030a : SUCCESS in 1h 02m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b08370c356d64a338df012dda26dd83a : POST_FAILURE in 27m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5543f99401734abbabb4bf0f73949b8b : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49aa4c368a0f4f6d975f093d7b7a9708 : SUCCESS in 22m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/252fae5d6ce24487ba6bf785ddc42b88 : POST_FAILURE in 38m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ab520a55ab3f4684977a134a2deaee11 : POST_FAILURE in 34m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/350ddc43f4034e86835a52d5ca810410 : FAILURE in 15m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/19de90c6f2164621ab825d084a269e5a : FAILURE in 14m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0db9bf29013245f181952b3b5070422b : FAILURE in 18m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ad10a79209604afdb02e0fe2d1f6631f : FAILURE in 19m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d324563333694f8198cf673264281052 : SUCCESS in 20m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ee1dc263d48748c3bc58e6e1adfb7b33 : SUCCESS in 16m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9afe213700f34025a57fd41990580eb4 : SUCCESS in 19m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6f3f797d986442cf9742a295482f913d : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b2ef518705104e5882f4b0edc3d74f69 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5e0cf213fa8a4777bac04204691ed0ae : SUCCESS in 14m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/17f3aa41099c42be9c12039233b1ee59 : SUCCESS in 6m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d1b1329eb70241689314da0d02eec607 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2ba916f2994548fb886cde613381b6e7 : SUCCESS in 8m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/df3accb3067b4cc9a00f0e293c2b6234 : SUCCESS in 1h 07m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7933374dca0d42568bc0d08c598a9584 : SUCCESS in 1h 10m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/613b2c990fa7451b97df825f99017d49 : SUCCESS in 58m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7d335587855f4b3e884bb1df8f43cf90 : SUCCESS in 1h 17m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e75068ee8e44dc1a2bca6429ab31946 : FAILURE in 36m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26a94160bca046d99fc4ecd5ad86e76d : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e81254ef357449c986011a8471609bd8 : SUCCESS in 25m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f2786fc5c824907a68cb1e89622101e : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":2},{"id":"abd4a0fc6e1e7287dd245da00774641b87cad72e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-23 22:16:03.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3c029d7997206e63f684067f4485740fab542754","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-06-28 01:38:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e2871f2ae73bb3bb2bbf35c38130e0752f5d9f7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-06-28 01:41:21.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"40b401e41ac4b31c3d5181db56f87ecb563c10d3","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-06-28 01:41:45.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fd138dbdbf32622ed39b0f72be6820f29edd47f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 02:54:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/687e5a21554f4a3aaf08c40203fa03f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0773851e4c7740b3913cf9214d7181e3 : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/c981943dc58f443a917dffbd160cd3ca : SUCCESS in 46m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d3dd66d2a614647ba744f70e883da2d : SUCCESS in 1h 05m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16d61ad8bed041a5896d8ea116dad229 : SUCCESS in 49m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d2b06fe4251462cb1186aec7dd5ce04 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2389b3043c484a5796573f1f42573835 : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d00e41638694a55b905cb0ca346994c : SUCCESS in 20m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/692752db532141abb1cdb6fae16503a4 : SUCCESS in 6m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/03c193dff578437ab79b5ff9632c1283 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f0be7e6cc69f4dc0ab49c9878a2de4e2 : SUCCESS in 17m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9d3ce420247a4f31afaf3b3abf98a166 : SUCCESS in 16m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cb0c70edca984ca193aa19fa45f14c61 : SUCCESS in 20m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/069fa8d9a0b44d4292036dbbe6a8910e : SUCCESS in 16m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2b63e15456bf4933afc95364c5962bc0 : SUCCESS in 20m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/24c60ec2fc5e480db790c59bfb975523 : SUCCESS in 18m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eb7e6cab9c4547fcb26c908a3523dfba : SUCCESS in 21m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7bc5db8acdde4ae39d3bd52f2ef5c512 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a7476003bd9e45fca4faa5afc9097bbb : SUCCESS in 16m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/15c5c352ab84459baeee03f7864b7165 : SUCCESS in 15m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a837c7ade71748579ab4fe8fa4358ecd : SUCCESS in 8m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e0529e07726f4dc99e607cc42d97ad9b : SUCCESS in 16m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/af5a3688d83544e4a35a49b86abb47e0 : SUCCESS in 6m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/057a05073f8d4cc4aece2f2c10ea46b2 : SUCCESS in 1h 08m 10s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f572486ba86a4553833877071d9386ba : SUCCESS in 1h 11m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3edcf98364e447fbb020fa1645284499 : SUCCESS in 50m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8eb205f67e3e4e5494106db7b2956b87 : SUCCESS in 53m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ecbb2af9de34dacb14932e0b78b67b9 : FAILURE in 21m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afb27baf8d0a4e01b4aaefef2c9fec83 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f3a48d84a5d84e26b710198e042d3516 : SUCCESS in 25m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/322e87bd2dc84435a1682c6def03c340 : SUCCESS in 48m 48s","accounts_in_message":[],"_revision_number":4},{"id":"761e945ea9d56d5142267080c238c5290f130589","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 03:03:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84e76467c0d04114842aaff73cc34c04\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b919a3a7aa604a18995337304ab3fc7b : SUCCESS in 1h 05m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2a97701fbbf430187780039e0e12546 : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc7651f84c08425ba7d25f77e93aa6de : SUCCESS in 1h 19m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8be52340cff41a291d0f7ca53ae62cc : SUCCESS in 1h 15m 19s","accounts_in_message":[],"_revision_number":4},{"id":"6d851f546c5c87b85d81f872ed2b56e79b5bc733","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-28 11:47:13.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a17eb39576ec3ddab6a963cc6d70bb6d48c20d0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-06-28 17:45:32.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a7320300b96736df0a2e956e907f2fbf3f53a505","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 18:09:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccab347dea924a8194fd9478417aa94e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff27157f299c45d4ad2f1210ed3d2450 : SUCCESS in 15m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84990b95641c461399eb85a5cac3ca1c : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7538f28ce4747ffa6d2072c9f3ba69b : SUCCESS in 22m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3bec0fa4bf3c4519a95d005f0f02e262 : SUCCESS in 18m 49s","accounts_in_message":[],"_revision_number":5},{"id":"082a4a67d23e349cfaba311132d116d4207db814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 19:06:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e339d0ae578b4cb4b4d3074b1e7c4f27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a0479ca8c50455b82d56df739eeab52 : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/6e1c8d0fa7054276b1196cbb85bf5bbc : SUCCESS in 1h 11m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e7bf1d15870d4cd49fd2a3d6c297dd78 : SUCCESS in 52m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/824056dd28464ef6a44317b2cfe3b6ca : SUCCESS in 46m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba7d35564c7345f6b323dd85503cef5c : SUCCESS in 16m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/50040add0fd54a1a87172aac362ad8df : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a7f22bc97f84e12b1be2dbda5d7e13a : SUCCESS in 17m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b8b69451144148a2b4b50d6b9379964e : SUCCESS in 9m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a896c3ebe017430397c4342c49fb057b : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/046aac4052eb44819534d06b3c4dbc11 : SUCCESS in 23m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a17a012c743a4ace817c836ed00126e1 : SUCCESS in 22m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a310849a8dcc43d0ad3c1e9be4219d5d : SUCCESS in 14m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4be83386253f4db9a7f571e23330928a : SUCCESS in 15m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bed4a9b8b24646d0819fb0c64d63ebf4 : SUCCESS in 20m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4fa94cb0ada84a00b6c54cbb9747ad91 : SUCCESS in 19m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1a9aedfd1d5341948a16b4e5816945aa : SUCCESS in 22m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/845f485f68b845349d8fedef1f1ccc38 : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/736cb4d1e4e1445787b11e0fbd53c352 : SUCCESS in 18m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/df160b1b24434b168e0a81e928d7aa85 : SUCCESS in 18m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2f6d6433dce94565a3339c378bf011c8 : SUCCESS in 7m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2f7de0957ded4e1d924a15c6c703a604 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cf605a92bf9f43d3ba10d0b8d1a3f5b8 : SUCCESS in 6m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4415fded5837432c96b6494727253b94 : SUCCESS in 59m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d0fb4df4a544854829a365aaa2caff6 : SUCCESS in 1h 13m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5a9f2f7f3c25460396438b57edb41615 : SUCCESS in 1h 02m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a1f3206acaf44bedba71160d92983ceb : SUCCESS in 40m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb6a6e524b8b4a0ebaf255d8f3d5ca64 : FAILURE in 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/314175e1929f4277b348d7fc4fa4ed19 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e17c08921d664d9d9926e575f04f50ea : SUCCESS in 26m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10ea38a9bdd542d9accbefc4d36660fb : SUCCESS in 48m 04s","accounts_in_message":[],"_revision_number":5},{"id":"8ce04753040a7397cb96821f5d8c90c1f38616ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-06-28 21:31:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a575279deadb0c2f8fbed5dedf76385936a27930","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-28 21:34:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"25a3528f84401a0eb093c0ebba224dd5412e903e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 22:49:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77af0ec0d70f466da862f3ae6c124294\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc04c3401e9476888b4b368870ff90d : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/77739c0aaed5431da35c9dc4cbfb524e : SUCCESS in 59m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6611362ea5b143e5b81fd5afd1ccd560 : SUCCESS in 1h 04m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46f992d6b40c4e119f5c8788a3fc86d8 : SUCCESS in 1h 16m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f06cd73524047aaa99f1032c47012bb : SUCCESS in 17m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55bf8b92069b42569ead620a36ab49e5 : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9503b1581f040b0931039332d5e8085 : SUCCESS in 19m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/44c96412d6214ffa9ac290c3dec3bee1 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/021e19c7ca1b4cd2b6ce435c071819de : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/45cbd8de84bd45e6a5f10617e859cdb0 : SUCCESS in 24m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5e21d83be512426a9ecfb569677533ae : FAILURE in 12m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ce6ab87c14e64898b3f2f21c1ad22d49 : FAILURE in 16m 49s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6ce90a03122c45339c0afcc773d6e6a7 : FAILURE in 16m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6969a3b3680c4509a25b8b008fdff5bd : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/42c778937ab14abb9644aabacdfe5524 : SUCCESS in 14m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eedfe097f01549e8a379737a451ce962 : SUCCESS in 17m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/103a707c81a54c2f8d062e513c116d62 : SUCCESS in 10m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b1b7e7533be14e72b9ee11304313c150 : SUCCESS in 15m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e9b34745d4e540099aaffbca668d42db : SUCCESS in 13m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7e42d945bd1443d5a9cdc0aa9004865e : SUCCESS in 8m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fea74b12be8b4179ac385410f6cc145b : SUCCESS in 31m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6f2b548f62ef42a3ba6a2017caa8f63f : SUCCESS in 7m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/db151d85d447404a9ef9989258b58df6 : SUCCESS in 45m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ccf426e5fb1341ce95ea1dbd7c7a3c57 : SUCCESS in 58m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/957c6d169ac74ae199926b1d4735a754 : SUCCESS in 1h 00m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/61ec49c3fb244d95824151de84a6b399 : SUCCESS in 44m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25cf2db5b7dc44538f1b0a847070eb5b : FAILURE in 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72eb7c070dae489bb0cb2eab5d013c79 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7152807a0844404d9fb06c922afe7d2e : SUCCESS in 20m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a512d0f2c5bc4f4a8dd5978527c1f8e8 : SUCCESS in 1h 05m 59s","accounts_in_message":[],"_revision_number":6},{"id":"2f53632cdded5164474b30af7ef0d7397a2db56d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 22:52:27.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a789a089674f4fab826dd8c7190e0737\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e6fabd6959a481ab6fc86d81e53a0a6 : NODE_FAILURE Node request 300-0021569430 failed in 0s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/30d06515b811408bb03334c9358b6da6 : FAILURE in 27m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/64c89afc428b4c22b4a63943727e7953 : SUCCESS in 1h 18m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a24619c048164e938fdd0fdbb9cf2f25 : SUCCESS in 33m 21s","accounts_in_message":[],"_revision_number":6},{"id":"2dfd6b9705e6163ebf426e9544357110168c198e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-29 09:00:24.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0e61dbe62072fd28c8987c4eeaa7522581745387","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-06-30 16:53:34.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"b98e84feab896046a2d2b47d74978fce255b380f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 18:07:19.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff590db7275a4a6c98fcf34010df809d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac6d087c5aa6463bb891c1879cf080f7 : SUCCESS in 14m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b568cb3a77a422b844a7dae9973c133 : SUCCESS in 1h 11m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f797e638dfa6486d80809bb60e28fea4 : SUCCESS in 17m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/19b37a0afddd4fc3925968a2f67f77e0 : SUCCESS in 21m 04s","accounts_in_message":[],"_revision_number":7},{"id":"5741e340b15ee477f64a9746fc094b41850dff02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 18:09:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/708af3f4897e42dabd48602feb1525e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc94abcc1d774664bab2e380c621daaa : SUCCESS in 10m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/f018ec909a954dfc9505cc904e31f257 : SUCCESS in 41m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a0ba452f506949c08d09bc2044ba82e6 : SUCCESS in 59m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/887f0f835acf40138c4d22e4a61f7e36 : SUCCESS in 58m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/039c0d411937482ebf570be5be98a1d2 : SUCCESS in 17m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca6143a518014aba8aebd527277f35bf : SUCCESS in 16m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e8ae3fd9fa341b785dd70123308d968 : SUCCESS in 19m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d3324d55cb104bc2881a4656a1e11183 : SUCCESS in 10m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b8658f2e6143447ebef69ff768ba815d : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dcc5a4354605441e92405534336e18ac : SUCCESS in 18m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b504a6f25a48450c8c1334334b3a8bcf : SUCCESS in 18m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0578fa7b7a8a4c3aab386153afc8b919 : SUCCESS in 15m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/903ca945a0874326915fc98e53ffe6c5 : SUCCESS in 38m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c480390f5f384889995cde794081a64d : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8300c8a60e5b47c8b6ed6bd118496f8c : SUCCESS in 19m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/20c5619be5104bbca351eefbacba2191 : SUCCESS in 18m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e54e4acd9f2d41d2acc429d04edd9b3f : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ffd3b708c4db467296b917c76898945f : SUCCESS in 14m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/73b86e905f66449eac8138de7e54420d : SUCCESS in 16m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f59e425fef3748d69731a3ba6e33a3d4 : SUCCESS in 7m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/116b65e9c6c9419893726e134b0a7c92 : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1c188f5108514a2788455fb9ee39ccf7 : SUCCESS in 9m 23s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2a4e19bee28d440ea805d4dc5a440b69 : SUCCESS in 48m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e179d6ec6c7e4c9581ebba2a305e8835 : SUCCESS in 1h 07m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/db20fa78e6064700a9dcbb90e13f05b5 : SUCCESS in 44m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c22a4c5c75b74f4eb3d048149c97f2d7 : SUCCESS in 46m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfa348d25d134f78984f8df926058691 : SUCCESS in 21m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05d0e5a0d1734b23b2d4278540160161 : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6058c0f529da4c29b6eb8afb2b3952cd : SUCCESS in 24m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab48d4569077488c92cce7fc79f78306 : SUCCESS in 51m 37s","accounts_in_message":[],"_revision_number":7},{"id":"10c4b394ceac90f6029ff73c2871f3a6ad14a778","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-03 07:03:38.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)\n\nNice! Looks awesome! Only thing that needs fixing in my opinion in the commit message, as there isn\u0027t enough detail. Think what\u0027s the current situation, what does this patch do and how does it improve things.\n\nAlso the scope of this patch is off. It\u0027s more then just effecting the container-info cli tool, but actually changes the account-info tool too, which is great.\n\n  $ swift-account-info \u003cpath to account\u003e\n  ...\n  System Metadata:\n    X-Account-Sysmeta-Coffee: yes\n    X-Account-Sysmeta-Something-Else: 0\n    X-Account-Sysmeta-Chicken: salt\n    X-Account-Sysmeta-Potatoe: 99.9\n    X-Account-Sysmeta-Something-Really-Really-Really-Really-Long:   something_else_long\n  User Metadata:\n    X-Account-Meta-Coffee: yes\n    X-Account-Meta-Something-Else: 0\n    X-Account-Meta-Chicken: salt\n    X-Account-Meta-Potatoe: 99.9\n    X-Account-Meta-Something-Really-Really-Really-Really-Long: something_else_long\n\nI\u0027ve given you an example commit message as a comment in your commit message, feel free to use it, or whatever.","accounts_in_message":[],"_revision_number":7},{"id":"d397e42abffc5dea1ca9a18927fbd3c19ac7d5dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-07-03 17:31:50.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"df7aa2801b54baa67558e6ad722f5c496e1bf524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-07-03 17:31:56.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"7d7de4c3d7accb11505481a8eb8fe7920fdf8d83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-07-03 17:49:44.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"627943f585a4a0ac96a1f60c87247bce2950dc4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 18:17:22.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d8810e610a54ced9bcf7f0c3edec64b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/357cca8f532944f2887f358706e471db : SUCCESS in 16m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b64a16d06acc46989024abf62fa0e5f8 : SUCCESS in 18m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/916a053d396e4fd4be0ea41e4f4dd9a8 : SUCCESS in 24m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/afd0000f9a98400193aad82d51ac6a67 : SUCCESS in 18m 03s","accounts_in_message":[],"_revision_number":10},{"id":"2c1460f2bd1f396746dc07c4d9ef73f0bfda1cd6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-07-03 18:47:01.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"40abac16359ab18c8d9d34c27444d27514ce8829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 18:55:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2e7159e4d1d4f15b305e3fc0348c372\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54840ed3b6b244f8849dc3579e810052 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/3f90f552ea774db4aa52af2693995653 : SUCCESS in 54m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b83f2a7eb712425688bd78884e2069ed : SUCCESS in 40m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37cba5c128a24b48af7957b5c332da69 : SUCCESS in 50m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/649843fbed6944f48ad325c681bc4f15 : SUCCESS in 16m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7302e566029b4ebebaeb4a54e1be9b5f : SUCCESS in 18m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57c342461c00406299f26eacf1bafae1 : SUCCESS in 20m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/38980ff9466e44f69f34532c0e31fce7 : SUCCESS in 8m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e1b69c6f3b264e66841d75ee3f8ef612 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f48ee3a032684d1b9f3c6e242050efcf : SUCCESS in 17m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/85e988cf034548baac8987bede95bc59 : SUCCESS in 17m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/790b0fd4e7f4498d821e26e4e611a1fc : SUCCESS in 15m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5b1cff3c17ea4669a1a589b36ae6c8f8 : SUCCESS in 24m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/249b38e10881466480949d9a0b39e584 : SUCCESS in 23m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/237d5e5d76c14e9f87d61229cd6497f3 : SUCCESS in 25m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7766df68bb33489d853cb2122ebfae42 : SUCCESS in 17m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8e2786672f5f445f9c9be366f56814d3 : SUCCESS in 12m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4c705a5b365c49c0824579d4aaa867fa : SUCCESS in 14m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7054be55a4dd4168863cc6715c897409 : SUCCESS in 15m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4176be15c49b48128e27ea2d30ab4546 : SUCCESS in 6m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b0bc3f19bb384082a828a60bf2514e9a : SUCCESS in 22m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e766ecd8062442149a4f2b2c3b70a3f5 : SUCCESS in 7m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b490fec0b8ff4d88ba599526d9ca59f3 : SUCCESS in 1h 04m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2b6662ec0414b75b07c6980250dcc18 : SUCCESS in 1h 02m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b584a5e569354377b6ff2bae7b0283bc : SUCCESS in 48m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/33951d6144e04f32b723182568333e44 : SUCCESS in 55m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66607182c0bd423bb14016741b51bd6e : SUCCESS in 13m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65e99ccae0ac4275bc25a368baa31dac : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f20c1019d113487ba81aa2cf81303524 : SUCCESS in 23m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7c0a264900a428d8ebec2d9ef67149b : SUCCESS in 40m 09s","accounts_in_message":[],"_revision_number":10},{"id":"8f577f958e2a16d2522b56785f474493193148e0","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-03 23:45:34.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"49aa5a65f136a9c54bfa3ac475736af7f56a0ff0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 23:46:05.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"b0ac1622155aa0b373b7008701b5a5bddf3f7943","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 01:01:55.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c84c08c68d734a27826c5386e0d39974\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/415a05fbfdfc439184bedc44aa5df573 : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/7b9ee46edbeb4d0cad85544ab35a191a : SUCCESS in 1h 09m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4ecac2f886184c14b1fbd279820f9e89 : SUCCESS in 1h 02m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dce26192fde44426b396c2f7f4edeb20 : SUCCESS in 42m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/922933bf1d2f45afbb86284ec39c8160 : SUCCESS in 7m 58s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/4f6166a953bd4c7081ba2907d14d685b : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7a6d04e124dc40e7b97d2c7199381043 : SUCCESS in 20m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c434b888e33348c6a078fb08f89362d5 : SUCCESS in 17m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d2e910ed3a9c44f48707a6eb88009eb2 : SUCCESS in 15m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/319a70dae7e6434f818d49b367cbad21 : SUCCESS in 16m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/323ad96c651b4767a8a57db222639bb3 : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/28b25bbfcb614fa98c558595a817c656 : SUCCESS in 19m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/413b12add1fc4d4bbdf3202b6cd6551a : SUCCESS in 19m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1909e658a45c4c9a99f0e24cd5e2234e : SUCCESS in 10m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cba572315ed14d42aa2a83a2fe9b8ba3 : SUCCESS in 14m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4a18624c0969404fbe00d60fb84eb47a : SUCCESS in 15m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0771aebc3a684a9f8a95d2700853eeb0 : SUCCESS in 7m 55s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8144ace35bd1430383faf98e8dd16118 : SUCCESS in 10m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5424db8440834d00b8388b25dfeb182f : SUCCESS in 55m 43s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6fa81170d43b494a8ba092ba402a9f18 : SUCCESS in 1h 01m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/829672199f7d40c49c38ec70bb147f16 : SUCCESS in 55m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d4bf4a46c4c14e2dbbafddd396b57e54 : SUCCESS in 51m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6ff25af9cbc481198fd19ee0395387d : SUCCESS in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/314cbd6b05d644448879a93777609d22 : SUCCESS in 4m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2dc995cd6494284aa02cea5ffff1b3d : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":10},{"id":"6c640b163790c9c1e8318ba963946a59015b854a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 01:01:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"5d60152bd826710d08db37b08a28c0b542070123","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 01:02:59.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5d235548c924afb8549cf0b760a9b4a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e30cf3223e5543838f3f4d6a8bda2d39 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3a8a817cbfa2437aaf5b03dea5930105 : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0ee6b80c664f4704a38e6c0d0eeee193 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9368195f020685a1a45a436047db1d13c9104434","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 16:14:40.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e82536c64da655a5e35b967cb617092df8585c75","revisions":{"12a96095818b6b1555c882d68d6709dc5ba6769e":{"kind":"REWORK","_number":1,"created":"2023-05-01 23:42:11.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/36/881936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/881936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/881936/1"}}},"commit":{"parents":[{"commit":"f99a6e5762896c7789d168bc49d8cdcb47903264","subject":"Merge \"Log (Watchdog\u0027s) Timeouts with duration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f99a6e5762896c7789d168bc49d8cdcb47903264"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-05-01 23:41:49.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-05-01 23:41:49.000000000","tz":-420},"subject":"format sysmeta in swift-container-info","message":"format sysmeta in swift-container-info\n\nChange-Id: I5c3d260d677c61213c42662c6641207d9b7f026a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12a96095818b6b1555c882d68d6709dc5ba6769e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12a96095818b6b1555c882d68d6709dc5ba6769e"}]},"branch":"refs/heads/master"},"2f7192595e704794fb357922c67f7c0b6e6dee28":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-22 16:20:45.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/36/881936/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/881936/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/881936/2"}}},"commit":{"parents":[{"commit":"2343521f751388ba2e4158194fb8c2aabc28d08c","subject":"Merge \"proxy: print logs when write new shard range cache into memcache.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2343521f751388ba2e4158194fb8c2aabc28d08c"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-05-01 23:41:49.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-06-22 16:20:45.000000000","tz":0},"subject":"format sysmeta in swift-container-info","message":"format sysmeta in swift-container-info\n\nChange-Id: I5c3d260d677c61213c42662c6641207d9b7f026a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f7192595e704794fb357922c67f7c0b6e6dee28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f7192595e704794fb357922c67f7c0b6e6dee28"}]},"branch":"refs/heads/master","description":"Rebase"},"180fc7d38a3dcae561fed2f47e48b4598c7fbc6d":{"kind":"REWORK","_number":3,"created":"2023-06-28 01:38:41.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/36/881936/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/881936/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/881936/3"}}},"commit":{"parents":[{"commit":"a9491eb7dada1afea882451335c56544936ce902","subject":"Merge \"memcached: handle failed incr+add+incr better\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9491eb7dada1afea882451335c56544936ce902"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-05-01 23:41:49.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-06-28 01:38:26.000000000","tz":-420},"subject":"format sysmeta,usermeta in swift-container-info","message":"format sysmeta,usermeta in swift-container-info\n\nChange-Id: I5c3d260d677c61213c42662c6641207d9b7f026a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/180fc7d38a3dcae561fed2f47e48b4598c7fbc6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/180fc7d38a3dcae561fed2f47e48b4598c7fbc6d"}]},"branch":"refs/heads/master"},"ca0cffc005d6735bdb60ee41c97130e62a133029":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-06-28 01:41:21.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/36/881936/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/881936/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/881936/4"}}},"commit":{"parents":[{"commit":"39127fee5be6eb7733938d31e82bb6553ea9cbeb","subject":"Merge \"sharder: Fix completed metric emission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39127fee5be6eb7733938d31e82bb6553ea9cbeb"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-05-01 23:41:49.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-06-28 01:41:21.000000000","tz":0},"subject":"format sysmeta,usermeta in swift-container-info","message":"format sysmeta,usermeta in swift-container-info\n\nChange-Id: I5c3d260d677c61213c42662c6641207d9b7f026a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca0cffc005d6735bdb60ee41c97130e62a133029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca0cffc005d6735bdb60ee41c97130e62a133029"}]},"branch":"refs/heads/master","description":"Rebase"},"0f743aeac7b04e6090601f1db6526f022816469f":{"kind":"REWORK","_number":5,"created":"2023-06-28 17:45:32.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/36/881936/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/881936/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/881936/5"}}},"commit":{"parents":[{"commit":"39127fee5be6eb7733938d31e82bb6553ea9cbeb","subject":"Merge \"sharder: Fix completed metric emission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39127fee5be6eb7733938d31e82bb6553ea9cbeb"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-05-01 23:41:49.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-06-28 17:45:22.000000000","tz":-420},"subject":"format sysmeta,usermeta in swift-container-info","message":"format sysmeta,usermeta in swift-container-info\n\nChange-Id: I5c3d260d677c61213c42662c6641207d9b7f026a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f743aeac7b04e6090601f1db6526f022816469f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f743aeac7b04e6090601f1db6526f022816469f"}]},"branch":"refs/heads/master"},"73f364a0a1bf275485a5499f44399edc6bd1a0c2":{"kind":"REWORK","_number":6,"created":"2023-06-28 21:31:20.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/36/881936/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/881936/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/881936/6"}}},"commit":{"parents":[{"commit":"39127fee5be6eb7733938d31e82bb6553ea9cbeb","subject":"Merge \"sharder: Fix completed metric emission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39127fee5be6eb7733938d31e82bb6553ea9cbeb"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-05-01 23:41:49.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-06-28 21:31:09.000000000","tz":-420},"subject":"format sysmeta,usermeta in swift-container-info","message":"format sysmeta,usermeta in swift-container-info\n\nChange-Id: I5c3d260d677c61213c42662c6641207d9b7f026a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73f364a0a1bf275485a5499f44399edc6bd1a0c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73f364a0a1bf275485a5499f44399edc6bd1a0c2"}]},"branch":"refs/heads/master"},"8dc9a112eb809faa5eca031fb00399e0d8b90f7c":{"kind":"REWORK","_number":7,"created":"2023-06-30 16:53:34.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/36/881936/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/881936/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/881936/7"}}},"commit":{"parents":[{"commit":"39127fee5be6eb7733938d31e82bb6553ea9cbeb","subject":"Merge \"sharder: Fix completed metric emission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39127fee5be6eb7733938d31e82bb6553ea9cbeb"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-05-01 23:41:49.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-06-30 16:53:23.000000000","tz":-420},"subject":"format sysmeta,usermeta in swift-container-info","message":"format sysmeta,usermeta in swift-container-info\n\nChange-Id: I5c3d260d677c61213c42662c6641207d9b7f026a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8dc9a112eb809faa5eca031fb00399e0d8b90f7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8dc9a112eb809faa5eca031fb00399e0d8b90f7c"}]},"branch":"refs/heads/master"},"bae1589a1751d74e91c5601b19c38543e3e96589":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2023-07-03 17:31:50.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/36/881936/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/881936/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/881936/8"}}},"commit":{"parents":[{"commit":"39127fee5be6eb7733938d31e82bb6553ea9cbeb","subject":"Merge \"sharder: Fix completed metric emission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39127fee5be6eb7733938d31e82bb6553ea9cbeb"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-05-01 23:41:49.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-07-03 17:31:50.000000000","tz":0},"subject":"Propery format meta and sysmeta in db info cli tools","message":"Propery format meta and sysmeta in db info cli tools\n\nCurrently we make no effort in formatting the meta and sysmeta\ninside a db when we use the info (swift-{container,account}-info\ntools. This patch properly formats them so they are easier to parse by human eyes:\n\nChange-Id: I5c3d260d677c61213c42662c6641207d9b7f026a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bae1589a1751d74e91c5601b19c38543e3e96589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bae1589a1751d74e91c5601b19c38543e3e96589"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a9f2534ca6c1a33cdda7541409e0c12dc682ffed":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-07-03 17:31:56.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/36/881936/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/881936/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/881936/9"}}},"commit":{"parents":[{"commit":"a82d9c58b6035dc3a69d4274919084d6ec6950c7","subject":"Rename TestSource which is not a TestCase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a82d9c58b6035dc3a69d4274919084d6ec6950c7"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-05-01 23:41:49.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-07-03 17:31:56.000000000","tz":0},"subject":"Propery format meta and sysmeta in db info cli tools","message":"Propery format meta and sysmeta in db info cli tools\n\nCurrently we make no effort in formatting the meta and sysmeta\ninside a db when we use the info (swift-{container,account}-info\ntools. This patch properly formats them so they are easier to parse by human eyes:\n\nChange-Id: I5c3d260d677c61213c42662c6641207d9b7f026a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9f2534ca6c1a33cdda7541409e0c12dc682ffed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9f2534ca6c1a33cdda7541409e0c12dc682ffed"}]},"branch":"refs/heads/master","description":"Rebase"},"e82536c64da655a5e35b967cb617092df8585c75":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2023-07-03 17:49:44.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/36/881936/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/881936/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/881936/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/881936/10"}}},"commit":{"parents":[{"commit":"a82d9c58b6035dc3a69d4274919084d6ec6950c7","subject":"Rename TestSource which is not a TestCase","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a82d9c58b6035dc3a69d4274919084d6ec6950c7"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2023-05-01 23:41:49.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-07-03 17:49:44.000000000","tz":0},"subject":"Properly format meta and sysmeta in db info cli tools","message":"Properly format meta and sysmeta in db info cli tools\n\nCurrently we make no effort in formatting the meta and sysmeta\ninside a db when we use the info (swift-{container,account}-info\ntools. This patch properly formats them so they are easier to parse by human eyes.\n\nChange-Id: I5c3d260d677c61213c42662c6641207d9b7f026a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e82536c64da655a5e35b967cb617092df8585c75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e82536c64da655a5e35b967cb617092df8585c75"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
