)]}'
{"id":"openstack%2Fswift~882103","triplet_id":"openstack%2Fswift~master~Ib3c9b274bbd2e643f3febbdf54a8a43f4775944b","project":"openstack/swift","branch":"master","topic":"py2-metadata-on-py3","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-05-15 08:23:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib3c9b274bbd2e643f3febbdf54a8a43f4775944b","subject":"testing xattr metadata with py3.8","status":"MERGED","created":"2023-05-02 22:12:28.000000000","updated":"2023-05-15 08:26:31.000000000","submitted":"2023-05-15 08:23:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"882103-py2-metadata-on-py3","meta_rev_id":"8bf1a513f40712eb87aa4f9a5032272c66402043","_number":882103,"virtual_id_number":882103,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-15 08:23:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-04 16:55:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-04 16:55:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-02 22:54:07.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-05-02 23:08:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-02 23:14:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-04 16:55:08.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"16f91d8d75104fec27a9ba90f840980de440dc26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-02 22:12:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec0b76cca74098a721eb2835cca20dd69f219493","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-02 22:13:26.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"88241b0d6cdd8f06ab0cc29221d494e79f204df4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-02 22:54:07.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3af143ade20d6ddbd11633fbf2b7f5dba0068ca8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 23:08:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/438b4541063447a1b9836e9c4fba6dc9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e14d97688984797926f5aeb96ac2303 : SUCCESS in 15m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5409aabad70e4862ac417770eff266d9 : SUCCESS in 54m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7e23bc9b10a482eb513498aa2416001 : SUCCESS in 17m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cba99c2380084762a4786f8948c86196 : SUCCESS in 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"eec7bf44cbdb8d699e828414db1cbdfbee1c7eda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 23:14:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3505f6ce18be40aeb79e02de160d3228\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8cf71626d434434b8bb3ed42b21c456 : SUCCESS in 8m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/127269298a4041d8bd5e52abe5f26e4f : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94f759f465a44eb4a624675ffdeeba74 : SUCCESS in 14m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7dc7a5c27624fb493561e4a06391f31 : SUCCESS in 20m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b9a5ac69c3a9497ea007fc96af019e80 : SUCCESS in 8m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b2b6f77d57cf4434bd71c416da8c500e : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/40083abd6258411d8324d306eabbfcfe : SUCCESS in 15m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a32c7adb9a5416e9ee342ba1335cc6a : SUCCESS in 38m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b1102729e0cc41319bfde674acac5fc7 : SUCCESS in 14m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6bd45ab4a7e54620a02fae0706bb7606 : SUCCESS in 15m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4188c22046754772a54fda2b8fa4b71a : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4948c349ab324e11ad7bafb6a19cec60 : SUCCESS in 16m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/347a0b27b1fd46da821e88ffd0ddc504 : SUCCESS in 16m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/894e11aa945d42cc87e6db7d5d163720 : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/75f1f3ab0224405a828f874fd79f49f4 : SUCCESS in 15m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/af0cd990c0e14932a6a3235c5e1a3e8b : SUCCESS in 15m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b864ede26522437b917975f1a4fef325 : SUCCESS in 53m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2a048590e5a1492ea0df0eaae3cf1a9d : SUCCESS in 54m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/372f047c908e4ab681e04e880420b81d : SUCCESS in 17m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2720cefb2374d8aabbd7acdcacb9cb0 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d6911d970c4b4a0c93623eea9c85bd05 : SUCCESS in 19m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ed8159afce9d7a7d05e9617e04ab1e9d4f2a6f87","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-04 14:44:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"002cb1f59af55d2a0ff4cf4d46d6c412fa01a093","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-04 16:55:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1bc83a79fb9db0c3b4ad3a933345c8861b0fe0ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 07:02:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f75d7bc37731f4e8773f6d8ebf90f7ce9385818","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 08:23:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/268f2785fe1b427c8bf1348ca21e7437\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94da05d4228c4170a2c111772b1cc719 : SUCCESS in 8m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cbba93cab25d4c9988a741d2fa71d13f : SUCCESS in 10m 03s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/415d4e62c2f0475a8778128e19c9394e : SUCCESS in 6m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/94313ea52065403d9676f2f09e8160ce : SUCCESS in 16m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6ff86da8e3494f04a6d2a870be8d3e67 : SUCCESS in 16m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b8d0981de85c43d6a1fca24101b57a29 : SUCCESS in 12m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/0f27e2aada8a4c6a8ab772d9e41efe21 : SUCCESS in 17m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1ff347943ee746fcbd1292a28361faba : SUCCESS in 14m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d7c177cab1164b49a4774f20c0313b26 : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/420d8e2ab6324ab29e11dc1d6d863a1d : SUCCESS in 15m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bc1157709b6744b3aaf5ff0b4e8063ce : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d95b9d9e709f48868354f4bdc1a5a323 : SUCCESS in 12m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/af6c23034967449a9d9b863b41bc96e4 : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/74f2b285c41442f7888b62001c105569 : SUCCESS in 6m 45s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b937a935c38941c9ba1b1ba0d4bf70a1 : SUCCESS in 7m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8287e19147c94e98a09deb3775b99018 : SUCCESS in 1h 00m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/efef8485616e473d87b9b73e75356e93 : SUCCESS in 51m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a3b004a86844e2cb17910536260bd53 : SUCCESS in 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39a730aa32a242d88640a01a34cf1773 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6956c22e872f031fba4b30eb0ce7b83ceb30b251","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 08:23:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8bf1a513f40712eb87aa4f9a5032272c66402043","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-15 08:26:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a67d98f9e6504067a383571d15190638\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eb884fea54d48bcaf9a2869865adfbb : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a24224870c864093ae169c01b6083c4e : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/bcfc0d2615054bf09ff90a78495cce1d : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d393fa482abf4bb550d5a25a045c61f7a41b2671","revisions":{"d393fa482abf4bb550d5a25a045c61f7a41b2671":{"kind":"REWORK","_number":1,"created":"2023-05-02 22:12:28.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/03/882103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/882103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/882103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/882103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/882103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/882103/1"}}},"commit":{"parents":[{"commit":"780754096267851545f2aa97afb250064a2292e3","subject":"Properly read py2 object metadata on py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/780754096267851545f2aa97afb250064a2292e3"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-05-02 22:12:18.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-05-02 22:12:18.000000000","tz":-300},"subject":"testing xattr metadata with py3.8","message":"testing xattr metadata with py3.8\n\nChange-Id: Ib3c9b274bbd2e643f3febbdf54a8a43f4775944b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d393fa482abf4bb550d5a25a045c61f7a41b2671"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d393fa482abf4bb550d5a25a045c61f7a41b2671"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
