)]}'
{"id":"openstack%2Fswift~882105","triplet_id":"openstack%2Fswift~master~If14e4d2c1af2efcbc99e9b6fe10973a7eb94d589","project":"openstack/swift","branch":"master","topic":"mix-policy-mpu","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-05-12 19:57:03.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-05-12 17:37:28.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"If14e4d2c1af2efcbc99e9b6fe10973a7eb94d589","subject":"disable requests_mock pytest plugin","status":"MERGED","created":"2023-05-02 22:30:10.000000000","updated":"2023-05-12 19:58:10.000000000","submitted":"2023-05-12 19:57:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"882105-mix-policy-mpu","meta_rev_id":"ed48cc48fdab611e6a4c50b175d3b21100775dc9","_number":882105,"virtual_id_number":882105,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-12 19:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-11 03:45:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-11 03:45:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-02 22:37:43.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-05-02 23:20:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-02 23:43:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-03 21:17:47.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"c21362a4aeb3228534ebe1bf1f9925a3aed5cfa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-02 22:30:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62e4c5d28c12084ee92ddd7ee488fadf9a92c4ff","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-02 22:37:43.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ccc72ccfd5a0352f884cce9eadc865caad69a0d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 23:20:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/121c6e7668a740729339cdc417679a8b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1cb793f983784413b62d7f940f138a87 : SUCCESS in 44m 37s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29fa8e1860cb45b784a83c69f1a66dc3 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/577905dfbe444625978fd48bc7bf1fda : SUCCESS in 21m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/241b6751932847dd9aa304e4d192df00 : SUCCESS in 17m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8dd77f8fac372ffefadd2f0f0ab9b7acdd0070ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 23:43:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87de7598e6fc4a1d8c22c8796ee6592a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/093c14b69da146ff95c147648e58f0e7 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c08fe996af194c8a8c505e9aa484243d : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de436862ae524f8ba547fd9b28c837b1 : SUCCESS in 4m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/b75f31dc44da4ec9ba092e50cd9ffd5c : SUCCESS in 46m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8e8588929af44d72b31ea887ef51a406 : SUCCESS in 45m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f7d163367b240eaa83f147828d2c573 : SUCCESS in 53m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab690897e5bb4789a479fcea673f92f0 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40d34c10e73146aca02cab7edda36bca : SUCCESS in 18m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dec4fc93a47546ffb887825ccdc517b9 : SUCCESS in 15m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6e3ff9d3c7b64d4f97c72c26dfe1abfc : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1dd1680d157f490ea4de56e45c046a77 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9eb3d55e5a064e289e460ed4b7c94e38 : SUCCESS in 16m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bd67882b0d1a4cd1a1cf09d660c79ca2 : SUCCESS in 12m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/60e6d5e3dd1540b3a25cd83858b58f41 : SUCCESS in 38m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/352d0612df9a444caaf13ec06cd89f2c : SUCCESS in 15m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d7da1b072a26479086f6c89d1b7c8845 : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2e9e687725984f63b1a31aea6de80dca : SUCCESS in 16m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/96dd48b8caa641a2a8bd04aa700d0bdc : SUCCESS in 13m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/44edc7f637364fb992702787fe86522e : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/aa5d7e75f72f495a810c689abcb31b0f : SUCCESS in 12m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ffc658a9318d4344baf87d9e57702d28 : SUCCESS in 14m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c9cc12b5e77142bfbe53133a9b1af4f7 : SUCCESS in 7m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1f7ebd3ef18449ff9c875c0f4cb558c1 : SUCCESS in 15m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6de17ffbde294012a6fd5ae8a76c77ec : SUCCESS in 7m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/714d598254154496a802fcf328b9508b : SUCCESS in 1h 04m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c824a26e913d4c53b2d6b3b50b7d7d02 : SUCCESS in 1h 08m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d1749e8200094510ae5e0041d6a2c0a8 : SUCCESS in 46m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7bf92f2e0e114a59a5bc5493f52c3a98 : SUCCESS in 43m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd11a49fc5d34aa9b38ed658030d6174 : SUCCESS in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a67fb522c34de889e822cfa4e2ff65 : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/53c2d82b806e431582418c8f07a8403d : SUCCESS in 16m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ff4e7be624c4916987f02a0cd3c13ab : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e78440b10386008974070ddb5c132ebfcd2ee180","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-03 21:17:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ae3a13433b146e466fdf4012e0f73dbac25db1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-10 21:48:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"eff68616d30e899b7b1e572a0e7513431c4f426d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 22:41:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2d6048d65cf4e42802498db07f37c4f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a02bc91b8926460aac5f5de787e741e4 : SUCCESS in 50m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33b6e057741d42b7916ca421481c0b30 : SUCCESS in 13m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/11566fe586824bf4bd9ebf311e2b3fc0 : SUCCESS in 47m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e52104ed50654bc998e853995409dc2a : SUCCESS in 17m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c4c0a6e337525486242d796d0e71377c44c2dd8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 22:57:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfe610d9faa04a53b71fcc6053772fe4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3acadbee6f7a415a98fa9f00a415717c : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfa0f6c5199641c9aabafea2aa29bb15 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d84174b91e04922980935707813990a : SUCCESS in 4m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/0b4f6a87ca9646889a23b96eb805fed1 : SUCCESS in 1h 01m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/198e602ea68647a7997ccede0a0299e6 : SUCCESS in 49m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5bdb9edc45e4f8eb91d80f4a389c1cc : SUCCESS in 1h 03m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4440b63d8414004b7772e715dfba8f6 : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/446debba5d3b43ebb77d18482da5d2c5 : SUCCESS in 17m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/099735881df045aabb1bfbf3ca4b7d04 : SUCCESS in 18m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d8c250bc7bd94b119cbb658829c3715c : SUCCESS in 8m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6db9f42c546348718a9cca1229a55ec0 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2f732bbf28b64057bec3325203f49146 : SUCCESS in 15m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dc51b666dd4b4efd98acf124e30e6cfd : SUCCESS in 22m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7294148edb2748a6b6d52ab503bdad85 : SUCCESS in 22m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5ca539a4b16e4ac4b8c4c8e4dea95655 : SUCCESS in 16m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e07ea62169e940a2b783cfffbc93f873 : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5661b4e4817044bab8e6b11f0eac1a94 : SUCCESS in 23m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a9e2c7c291ea47ab8ac5fcdfea4bb6fb : SUCCESS in 13m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/91b4cbb12053463a9f1db820ee6b403c : SUCCESS in 22m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6d7a749d4ddb47a98570fe67b73ce103 : SUCCESS in 12m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7b242adcee4e40af8e8901e93ece3ce6 : SUCCESS in 22m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1838c9a8f45847568d42efc414e7d4e9 : SUCCESS in 6m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a11063d986dd4262a2f1bdcf9375fe46 : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cf7336fad3ac47c7bcbb7390e069f153 : SUCCESS in 7m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d7db0958e5924466b12f03b9ae77a51b : SUCCESS in 1h 01m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d78f89e7d3448588bcf2ca736f297d7 : FAILURE in 50m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/632ea8ddaa064a55a33ec08135cbe7e8 : SUCCESS in 44m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/408829e66e3d4b3dbd9bcf2a72d31b83 : SUCCESS in 46m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0516a0b8d974498c96e1d400b957d80f : SUCCESS in 26m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d12adb4dbcee454f82e0c2c11f6f5628 : SUCCESS in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d59758905dc64963b6fa0a17752f25c0 : SUCCESS in 21m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9799c246fbad443ba9cd0a9fb56f670f : SUCCESS in 43m 10s","accounts_in_message":[],"_revision_number":2},{"id":"b026e8c1ca0c39283fd1827ac3d2959de799380c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-11 03:45:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"098de349be78325aa3b8a30861b4f9a59abdce44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 05:09:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e5cb06e1a6747878a45cbe084fc7782\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8bc6b5e84f0f46dc81899da3c405da22 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c0be6207e7c4204b2883b4499b40989 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/043bc7813d35473fbc40ec9efc55054c : SUCCESS in 5m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/054e8997817546708b32ef8339f79655 : SUCCESS in 52m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2fadf4eca053478fa313e3e6f04d7bff : SUCCESS in 52m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79cc780ae3f4449f9c0a5edf30a5e2b3 : SUCCESS in 57m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de5abc304a90415baf774b6231d9fc72 : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/456785ddfc0b4fd8b6a61dad385e1355 : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e58c96e38df46cb8dd34dc2d019c709 : SUCCESS in 23m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9eb13d1e99f141aab93ce3da04043806 : SUCCESS in 7m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bfee5463cf674216aa57c55d45636eb3 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a33de96193eb404bac1e69dbade0c788 : SUCCESS in 25m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8adc55bbfbfd4e9099bb129313daed80 : SUCCESS in 24m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ff6efbda5b724b8c807a5c845cb777a3 : SUCCESS in 13m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1622491da59345b6857301fb4ece6e6b : SUCCESS in 13m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d35cb935894747378ef9892a83e1a106 : SUCCESS in 25m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4433e7892e284e5c96fb1cc93752e6e1 : SUCCESS in 15m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/260353e68d92487c8a548db55341b411 : SUCCESS in 36m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/100a061c3a334b2c93d818772e3fa386 : SUCCESS in 12m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f285b856ec0f42ba8d0f3db17e171d66 : SUCCESS in 17m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1eebbf81bd974d5887ddb218895f35c0 : SUCCESS in 19m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4ac1b0941f3845d292d1ec0930243ead : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/46e986385f5e4935839d97e333fa44d7 : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a4b9237c03f402f9d142c87be1684ed : SUCCESS in 7m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d5495b4819974f9ea1d83fda3721c179 : SUCCESS in 53m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f37f32a4184e456e9486c71d85ab201c : SUCCESS in 1h 05m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e72b8446bb7842cab378e4429deb3833 : SUCCESS in 42m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0c3b9066eef14dd7bc9eefcb3ed29205 : SUCCESS in 47m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88845f5264894f06a93f1c5d3aadbe40 : SUCCESS in 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a14ff67e4054c6498b97f175d2a8c7c : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b8d699f261a0480c87f294af3ef20ab0 : SUCCESS in 24m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3013a0a3ae1b48b581195e96ad5c81a3 : SUCCESS in 46m 03s","accounts_in_message":[],"_revision_number":2},{"id":"d0e909b951a49f0605c73e039d22d34dc404a017","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 05:10:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4253455f04f9d69486f3c2e2beee396a8bde7fca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 05:26:12.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdfc52a433d8476bb9751ca01cfff077\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3113f0ba84a48e48bf90f713b28a578 : SUCCESS in 1h 01m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a31fa9c933ff4a57a6e697fc9c490a65 : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c1f091590d74183a69c6426df3a9aa2 : TIMED_OUT in 1h 30m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0effe94d8d3b426fa5252c910211ad0d : TIMED_OUT in 1h 30m 59s","accounts_in_message":[],"_revision_number":2},{"id":"c10eaaec3ad8369e0a457e566ad9e5ac5773e532","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 06:49:43.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76f996aee5c449958aac3ac9453fda36\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/80949c27fcfa426d9686e2d42a1b9d2d : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8a1d1f77e0d4864b7270352920ea2f7 : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1c22149a7a84faeb54d5caec341097c : SUCCESS in 5m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/ff658157b8ac459abf3b4ff74d08269b : SUCCESS in 50m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a49e639dc1b7493bbf68cb3af5c0e11f : SUCCESS in 53m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d64a09a57f449f3ae271afdd336c7b3 : SUCCESS in 59m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ded3cf6dca7f4df6ad037d40bd2587bd : SUCCESS in 9m 45s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/0e4ef69812fa4336b3acd2953f0fc0b9 : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bba3031af5ad4635a227d8be0bf27e48 : SUCCESS in 16m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/18b0cfd3613d4acf9fcf931d974a2a55 : SUCCESS in 16m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/60f315ef772046faada216904763fe09 : SUCCESS in 14m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/42a9d4d802a34a75a84d26c09c1ee706 : SUCCESS in 17m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b9f55902c49f4a97aac448deffb3d307 : SUCCESS in 26m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9b297046f57a46f4bba45564f43ee25d : SUCCESS in 25m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8aa58653a63e447291621e253be02d70 : SUCCESS in 27m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7204d97a373e45a09a690e07bb938c19 : SUCCESS in 20m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3d1aa0e033c247909ee4bdc602d892e9 : SUCCESS in 21m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/493534cc53094a81b5612bd5ca7dbcf9 : SUCCESS in 22m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e80928123f544b288e49d59f26b7bf9 : SUCCESS in 6m 24s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/29fd7808f0834e20b35be57b18ac2232 : SUCCESS in 8m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/57abd24b8ea745469d6b2a80ff3ca60c : FAILURE in 1h 03m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6cced71d8ce648b3889f24be974185e1 : FAILURE in 1h 07m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/72b57413f127468681ee801c6b1b3967 : SUCCESS in 47m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/63fab58c4ec942a5a1cdeeebcec5f0c3 : SUCCESS in 48m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e3bdb0dc58c4d54820fd48dfc635143 : SUCCESS in 17m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a67aa68f1f5047969e0751211e5590f1 : SUCCESS in 5m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd88fff252f04d319fe2c4eca2bc816e : SUCCESS in 45m 32s","accounts_in_message":[],"_revision_number":2},{"id":"9b01de51636eb0723d527c84c641a5f119f40c2c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-12 17:37:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b4244e5f1f2fb43fa7772b21c20ff39e3a632fa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 18:47:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d3e9c3fea61473d9e3da2f0debee659\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/176cb56123814435b95330bf77398f55 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/922ba1225f64431ea9de53251417b544 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc3b2f3c6ed347d7a043d5fdb1adfb1e : SUCCESS in 4m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/541457d0f69649a2bece6372c788a921 : SUCCESS in 1h 02m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a17d25194af64ee0937e98ae638e3a62 : SUCCESS in 32m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2ffd516548d4623b3e34b0057d31add : SUCCESS in 43m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f521d9c9a12f4e9aae7bc92edef417a8 : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8970e9b33e284ec39970b71a04093d48 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/366e4329cc134007bc12bdd4a2ff8030 : SUCCESS in 17m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/978f830f604543bbba7d8c0552568dad : SUCCESS in 7m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f9022a86c9814eadb1e3eceec8e92448 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6f913ef1d7704c62a6d66ab25296d7e6 : SUCCESS in 16m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e1516b245c474894ad1b69e174a53d1c : SUCCESS in 16m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6d8ac3b2b4ff46b5b3ec5d7b8b79111c : SUCCESS in 16m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e3a99cd027ee4c2cb48e684ff9b0ba37 : SUCCESS in 16m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/713c5a7ecc4e43d99f83c1066d4281ba : SUCCESS in 18m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8c075aacea3046118306fc4fe492a667 : SUCCESS in 17m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/25a3395f04154493968a0a572219cd69 : SUCCESS in 18m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/871d542846bc438a89643893df03b221 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/66a694bfbcd94f00b1456153cc7f70d9 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3ce32d8f357f42da86debbdc12e38887 : SUCCESS in 14m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3331782787e84f83910a7283b40e9f34 : SUCCESS in 6m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cd14b2dc49964fb0a0e847dc58f8e7f0 : SUCCESS in 12m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9dd68049124e4d1faf155452caef04a9 : SUCCESS in 8m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6933b081f420473ca366da44b4105e72 : SUCCESS in 51m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5912be1101ef4d128e66b9297a28863b : SUCCESS in 1h 02m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d25d2caf37b14502b561cb8187bcc57e : SUCCESS in 48m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/570715d4f19c42368a56460aed314d6d : SUCCESS in 48m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f62fa2eb6b8949fc9dc9f8a5d6839b1e : SUCCESS in 16m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29be7e48f65644e387850c3be1d4c763 : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9bcf8178b0a544f6b073d0dc2a605186 : SUCCESS in 18m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d01bb45e39194f35a202f442a5bb08e3 : SUCCESS in 53m 31s","accounts_in_message":[],"_revision_number":2},{"id":"5241b2175b7b32e7ad6f90e77ae3912b49310741","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 18:48:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"966664251a459261480c8633b6add4c6382a5d62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 19:08:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85fc517c54df44a6b16b1b2dfcb9c7c0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41dbaf03c3d5490faba09fa7569172ca : SUCCESS in 15m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f61d7f169581439c9c4bb2ddb7063c0f : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2cce384988f6437996631aa95e67e9de : SUCCESS in 1h 28m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a45aac01411646a3a0b0d747f6874c8e : SUCCESS in 20m 08s","accounts_in_message":[],"_revision_number":2},{"id":"0dd2133b24e15627918c3fa42e5260ef6c854d8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 19:57:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/249587c8a65f418abcde80e1785af3fc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/414472847ac14d46ac24cfba8cb1e258 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8e8937a0c8047809c5dec584ca9e7e4 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d707fcfed5c47609aae38488ac529a4 : SUCCESS in 5m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/840c44db47d340e88de58b1b03776901 : SUCCESS in 1h 01m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c56405dbe4d14b92875a6527423d84c8 : SUCCESS in 56m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8eee0df9d5f1435789e2706f45635b40 : SUCCESS in 1h 03m 59s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/38556ffb099a45bfad301a9876c22606 : SUCCESS in 8m 54s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/5e28d6fff7c14344b6e08de15beafcb9 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/075cf3a16c234054a8e545c39b26ff2a : SUCCESS in 17m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b39f9d5c0d9f4ef8b881d9f251f18f70 : SUCCESS in 21m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/45d1e7ae75eb4c3d83ff4cc5217fb3a5 : SUCCESS in 16m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f3b1ce95c26043698aa705277f8ffcac : SUCCESS in 16m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b662677e309f4ce8aa0982248a84c56d : SUCCESS in 21m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a104a90645c542bdb8120275cb148512 : SUCCESS in 13m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/13f2b1dd65c444b59c8ce9958012c6ed : SUCCESS in 18m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/880b8625dbc5407291d95e6db0b86eb5 : SUCCESS in 14m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2cac3b6cc5fd486cad199a44e3a4afbb : SUCCESS in 15m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ce78c522484a41a0ba44aa27ee755e73 : SUCCESS in 20m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3c48b9457d804d17b63882375682bd72 : SUCCESS in 9m 38s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c1ff45c5ca554ddc9d3ae023a7239509 : SUCCESS in 9m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/285089d1544e493aa6793be45d43394c : SUCCESS in 42m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7aba06093a574650ab798278620e6921 : SUCCESS in 48m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8887f292bb0748ad97d6866f8181047d : SUCCESS in 42m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0bb1bb7565274237b900cba27e5928a2 : SUCCESS in 44m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e33a4ee8fe5c425bbf494cbf185824b1 : SUCCESS in 15m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f2a16bdf454164afa7ba8a1fb1a3d9 : SUCCESS in 3m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bc62cfac0ca4002a412928b34652eea : SUCCESS in 39m 22s","accounts_in_message":[],"_revision_number":2},{"id":"1c8a739ee221c382f8bbb948dd80bd5085f30f8e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 19:57:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ed48cc48fdab611e6a4c50b175d3b21100775dc9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 19:58:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2851c852d894c12b1b9fc3ec01d5e2f\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/ca74d2fb431a4ceebaa9e858bea289b9 : SUCCESS in 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d809bfa7bd4cc9972fd39744714532 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/309c99e3792c4b7d8d963cd2bebdef4d : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8a3c8645a185461686b04e576dd76de1 : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/06e56be69f8343c990531db93e73b8c0 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9adc3ce425bf2a05836247e032355ae54b62a271","revisions":{"340017dec906dcf70d84940c269aa9f3d13c8109":{"kind":"REWORK","_number":1,"created":"2023-05-02 22:30:10.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/05/882105/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/882105/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/882105/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/882105/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/882105/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/882105/1"}}},"commit":{"parents":[{"commit":"ba5b898f5ab73ed24a1f113a947f88e9b917a26c","subject":"Passing probe with bad behavior","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba5b898f5ab73ed24a1f113a947f88e9b917a26c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-05-02 22:28:21.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-05-02 22:28:21.000000000","tz":-300},"subject":"disable requests_mock pytest plugin","message":"disable requests_mock pytest plugin\n\nChange-Id: If14e4d2c1af2efcbc99e9b6fe10973a7eb94d589\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/340017dec906dcf70d84940c269aa9f3d13c8109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/340017dec906dcf70d84940c269aa9f3d13c8109"}]},"branch":"refs/heads/master"},"9adc3ce425bf2a05836247e032355ae54b62a271":{"kind":"REWORK","_number":2,"created":"2023-05-10 21:48:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/882105/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/882105/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/882105/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/882105/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/882105/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/882105/2"}}},"commit":{"parents":[{"commit":"f99a6e5762896c7789d168bc49d8cdcb47903264","subject":"Merge \"Log (Watchdog\u0027s) Timeouts with duration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f99a6e5762896c7789d168bc49d8cdcb47903264"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-05-02 22:28:21.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-10 21:45:33.000000000","tz":-420},"subject":"disable requests_mock pytest plugin","message":"disable requests_mock pytest plugin\n\nWhile we use requests-mock, we don\u0027t need the pytest plugin.\n\nChange-Id: If14e4d2c1af2efcbc99e9b6fe10973a7eb94d589\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9adc3ce425bf2a05836247e032355ae54b62a271"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9adc3ce425bf2a05836247e032355ae54b62a271"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
