)]}'
{"id":"openstack%2Fswift~882184","triplet_id":"openstack%2Fswift~master~I5d00015ae72407bb70b98ee0bc4fbbd9d9689f49","project":"openstack/swift","branch":"master","topic":"ring-v2","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-05-11 04:19:34.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-05-11 04:19:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5d00015ae72407bb70b98ee0bc4fbbd9d9689f49","subject":"ring-v2: Make RingReader/RingWriter operate on file-likes","status":"ABANDONED","created":"2023-05-03 22:55:39.000000000","updated":"2023-05-11 04:19:34.000000000","total_comment_count":17,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"f02ccb107fbbc0dc8d35fa13bf82ee9d6f5d3e0b","_number":882184,"virtual_id_number":882184,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-05 00:52:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2023-05-08 06:48:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-04 00:05:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-04 16:05:09.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-05-08 06:48:57.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"25a4f20a1c207102fabaa3bb961cdf26b101416b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-03 22:55:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"896eff89666184d87c117128297089d9f4f73673","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-03 22:57:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a4b2ccdea2efe17935834269eaef2f9b37e826b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-03 23:09:36.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"156e3ec5faf87913563e79fc0874ce9000bd99f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 00:05:26.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5aa452f956ad4cb98fc77bae40a41ac8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46947e3bd6604e84b6fd55ebee6793ab : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/0cbf208df93744beb4ffbf1ba4e65795 : SUCCESS in 41m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a1a40afefaa84f57b2b2fb1bdfd35814 : SUCCESS in 45m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48405aa30267447cbbc76bd912115da7 : SUCCESS in 1h 02m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ea64e58fcb3443397cd7d807e1d604a : SUCCESS in 28m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2930d3d3b1c54122b4da9c62fef786ca : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a09194da7e04445a928ffc3922770db6 : SUCCESS in 18m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f95fedf5c54645baaa1e752ff08946fe : SUCCESS in 10m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c9d20377713d4108a1d47875de3286fd : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/63258a5437fd4321906b134d684acb9e : SUCCESS in 16m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fd1544ac2ca74fa2a089172b64dee75f : SUCCESS in 15m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/16c62db858b542c49cc1861f6b2bc050 : SUCCESS in 29m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a0a42ccb4eb8486cb683c87c98930951 : SUCCESS in 14m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c44e54083e414a3b8b5bbcfe0fd60448 : SUCCESS in 28m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c768da9633c748bdb2d1c39f3d9d0212 : SUCCESS in 17m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3d0096797dec4a4ca269dfffe452016c : SUCCESS in 29m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9945f57be46b4760b733bf8c36b1a73d : SUCCESS in 27m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8ed28302739442f189d746c335ca8781 : SUCCESS in 29m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/818ad6b200cf4773b73a1ad83b44c90e : SUCCESS in 26m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6cca1b4658a248d6b340579759bd3baa : SUCCESS in 7m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/76c681c3c35646ae9b44e67bff1006a4 : SUCCESS in 14m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6b7ed94d22394d548e06c3ac92a3f21d : SUCCESS in 7m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b164bf9ebfcf435f8a0ef765b8be57ab : SUCCESS in 56m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/00a274319d8f4ba8b6959f312c9ab066 : SUCCESS in 58m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/150edff9d5fd4bd399d089f2229ab858 : SUCCESS in 57m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7a1b11b5f2fb421ebfed485bb8ba2edc : SUCCESS in 51m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54403ed3b1a745edaec9c10b5092b3a6 : SUCCESS in 17m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28d08d70727444ab479b4cd28c217fe : FAILURE in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b62df5c977a24c0689bb21cc0d02dd41 : SUCCESS in 20m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab30798cd46448dc842a21c9f84db256 : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2004b99cfcb3c9d9275d7c292e5cef4db7c9a13c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 00:17:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f195c08f2fe74ce4a8e3c4585c47400e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a25347348b4e4917a1eae9767e6ec777 : SUCCESS in 14m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f1a5eedf0484c6189b6b2e12d29a6db : SUCCESS in 1h 19m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8147a80db31e4395ada41bf02b1cba49 : SUCCESS in 17m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/73a989e9be814143817170c2977092c9 : SUCCESS in 19m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ef43adddd907c1b8d943807cd033f0d0089d943a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-04 16:03:20.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9c630e530477f2e833e77839c97f37ba00328773","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-04 16:05:09.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f77eaf922db47bb942d8637ee786b2e44ab0aaa8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-04 17:12:34.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4c3c8f3f3d8d340a2b132b798b4ea90131615f42","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-04 20:11:11.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bd124e672410757f805f7f2519825d996a1813e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-04 23:31:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"720fb5efa4fb615fdec14d689353c65ee6ecfdc1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-04 23:31:30.000000000","message":"Topic set to ring-v2","accounts_in_message":[],"_revision_number":2},{"id":"c4319a8d2789b36f8184b904cafddf32f1eca986","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 23:53:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c272c9a534bc40f39be782a4c9e1ea76\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f50513fa527a41b6b646ebbb3872c2fb : SUCCESS in 13m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd46423ccbba45c28400081272cddc81 : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a2f3c20345c49f193a94665e2c0fc49 : SUCCESS in 18m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/65110e2b03ad40e397f91146d5a99819 : SUCCESS in 20m 45s","accounts_in_message":[],"_revision_number":2},{"id":"3f3775ac1befb4fd429fa4e8bb648f4cbeeca9e6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-04 23:54:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5222498e7ebce049c10f559bda77e233ef8f91c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 00:52:39.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db305ec257b9479c9c11d93578f7662c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae839363592a464f969b30234e3fb536 : SUCCESS in 8m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/c77020be2a924fc1b70bca93c246dfcb : SUCCESS in 1h 00m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/24d2a3a2e24843ab8319f8ba78044b42 : SUCCESS in 50m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/094e4eeb988e4ec3abcd29d8234a62fc : SUCCESS in 1h 02m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9993577757a45998f5bd1f550f5a109 : SUCCESS in 21m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c569061f46cd45eabadf8546e244cb65 : SUCCESS in 22m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/538aeb92a9294e55b278196ce9d2f25d : SUCCESS in 19m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7ba36184e2064b3a91729257a024947b : SUCCESS in 8m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/52505e231cbc4c26a94e1fc3321d638f : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/78c352d5156d4b73a534bab7415b21a3 : SUCCESS in 17m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ddcc4e6321d44fc392a2325441ee11cf : SUCCESS in 16m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e7e0419343444647a3c4721a4d6eb49f : SUCCESS in 13m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7838b51fea6d4f97b1e9408b2116647d : SUCCESS in 21m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c264528c127f42b480b2ae5bcfd79c98 : SUCCESS in 19m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2ce3a1219d0f401b9e6fbd8a14393152 : SUCCESS in 19m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ff7dfc850ca6440db2ab0899be1d3e73 : SUCCESS in 15m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/709355b6d87f483cbbbeec9fa23c8619 : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/260fb6a8333a43039407ed81cc99671f : SUCCESS in 14m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c17a1e2ce8e044198e255ca298096890 : SUCCESS in 21m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/db02c7c5409b43fbbc858415df508efa : SUCCESS in 7m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/35dd6ed8232242b39a20f1e1cb366a9e : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/76aeb6e9aa164d1088e1a1c0c712c802 : SUCCESS in 7m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5dabd44199e54ab0992a92c056a2a335 : SUCCESS in 1h 10m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5775e68450424ac2be1330949409ffc6 : SUCCESS in 1h 07m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cfb2acda2c3443659d6e4f5185028587 : SUCCESS in 44m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/319270e7f36646028b435ef3122b7fdc : SUCCESS in 39m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc612f3d4fdb4c8b89ac43353fbd1abb : SUCCESS in 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71cd8ba1b3c1437aa8604e763e1b40be : FAILURE in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/703bb5c354a544359ae0c9a640438a4c : SUCCESS in 21m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc7cfd87d2e7484b848582d3c37e6d8a : SUCCESS in 56m 29s","accounts_in_message":[],"_revision_number":2},{"id":"566c31dc1a6aa79add83cb50b30da80d0d2b972c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-08 06:48:57.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f02ccb107fbbc0dc8d35fa13bf82ee9d6f5d3e0b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-11 04:19:34.000000000","message":"Abandoned\n\nSquashed into parent","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b7beb542a3518ce4266e670944b4b5387f28bbb","revisions":{"ec83e703bd6231cb9e1a21f8d7398ed308664662":{"kind":"REWORK","_number":1,"created":"2023-05-03 22:55:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/882184/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/882184/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/882184/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/882184/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/882184/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/882184/1"}}},"commit":{"parents":[{"commit":"c94f5a8bf78e661168c3f6de25df684edea6e7be","subject":"ring: Introduce a v2 ring format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c94f5a8bf78e661168c3f6de25df684edea6e7be"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-03 22:53:45.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-03 22:55:05.000000000","tz":-420},"subject":"ring-v2: Make RingReader a subclass of ZlibReader","message":"ring-v2: Make RingReader a subclass of ZlibReader\n\nChange-Id: I5d00015ae72407bb70b98ee0bc4fbbd9d9689f49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec83e703bd6231cb9e1a21f8d7398ed308664662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec83e703bd6231cb9e1a21f8d7398ed308664662"}]},"branch":"refs/heads/master"},"7b7beb542a3518ce4266e670944b4b5387f28bbb":{"kind":"REWORK","_number":2,"created":"2023-05-04 23:31:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/882184/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/882184/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/882184/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/882184/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/882184/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/882184/2"}}},"commit":{"parents":[{"commit":"c94f5a8bf78e661168c3f6de25df684edea6e7be","subject":"ring: Introduce a v2 ring format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c94f5a8bf78e661168c3f6de25df684edea6e7be"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-03 22:53:45.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-04 23:29:57.000000000","tz":-420},"subject":"ring-v2: Make RingReader/RingWriter operate on file-likes","message":"ring-v2: Make RingReader/RingWriter operate on file-likes\n\nAs part of that, make RingReader a subclass of ZlibReader and separate\nconcerns a bit in RingWriter by having it subclass a new GzipWriter.\n\nChange-Id: I5d00015ae72407bb70b98ee0bc4fbbd9d9689f49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b7beb542a3518ce4266e670944b4b5387f28bbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b7beb542a3518ce4266e670944b4b5387f28bbb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
