)]}'
{"id":"openstack%2Fswift~882671","triplet_id":"openstack%2Fswift~master~I2e0352bf8c0235438f92e202f403840b4aee337a","project":"openstack/swift","branch":"master","topic":"listing_sr_cache","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-06-19 04:53:44.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-06-27 06:55:43.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2e0352bf8c0235438f92e202f403840b4aee337a","subject":"WIP: Split the inter_shard_range interface into 2","status":"ABANDONED","created":"2023-05-09 07:02:36.000000000","updated":"2023-06-27 06:55:43.000000000","total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"c4db5c3302fed42f75cc928354d67258dc064232","_number":882671,"virtual_id_number":882671,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-09 08:14:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":-1,"date":"2023-05-09 07:08:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-09 07:48:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-09 08:14:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-08 17:59:15.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"b92a12614fdd605bc26ae762712730d985a662d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-09 07:02:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f2a5cbc86f079f6aea2b50310c22c6f9a04605f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-09 07:08:45.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3619a0bdb0c02d892d3da97f3ef520329bb827f2","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-09 07:08:57.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"90dcbaa6e61364b7364f6f9423ffcfaf7e8ce669","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 07:48:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abaf848fb5b94a4f8c42837330eb9a70\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c923356a6c5430a8a6bf3270dd68fd7 : FAILURE in 18m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/090277b387f44a56950dbe82cd7088a9 : FAILURE in 16m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e71a101a0b854a3fbf6a652f2fe991a0 : SUCCESS in 43m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e5fccb152c14d058e0829dc7388cc3e : SUCCESS in 19m 35s","accounts_in_message":[],"_revision_number":1},{"id":"496dd7afb6e273f801c72bed61d65b2e278fcff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 08:14:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfaded4574aa498ca7cd93c2cabd5917\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e99a292c224013917b2c71080b6a4d : FAILURE in 5m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/6e8cc44acd6c448b875bc7fd7a282506 : SUCCESS in 43m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/16a5bfd9f47f4606824d7092270eafe4 : SUCCESS in 55m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6302737b3214b8889f9d3c70518bf56 : SUCCESS in 1h 08m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b836a4ce3e434d28a6c9d15e3db45dcc : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96970b4c4c724b338c9f660bf3bde8a1 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ebb54cb3ed194cafb6fd01e69fa6649d : SUCCESS in 20m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/16a2beb011c04112a26213046d66bf27 : SUCCESS in 7m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/155376d9c2d94b9f9d662a099b749be4 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b6e298bfaf2448108e7f137a7434fadf : FAILURE in 18m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4aef27e490c84130aceeff293987deae : FAILURE in 16m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d4a2c3eccfb9409ab0faa839cf506f06 : FAILURE in 16m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f7c9ba7fe3d5416eb3c61e139137c122 : FAILURE in 16m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7c9589b59d014ed3b65bf9795d720980 : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5937d18f0dfb455a87159c58589e32c1 : SUCCESS in 18m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/228214c608af45b3b65f2fbbfa051af5 : SUCCESS in 19m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/405166a8191644fa87fe01b6d5953d49 : SUCCESS in 18m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/16941626bae741da8e50da6015807ab2 : SUCCESS in 13m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8d2bd9ff4c16413ab0f6b1a831a45a05 : SUCCESS in 15m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3b19c3b9beaf4b939338f4ec25187f87 : SUCCESS in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/30155c0fa9f94bc0adec6506f5ae5a61 : SUCCESS in 15m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/79a161afcf204978bbace33bc40ddfd6 : SUCCESS in 6m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d9c90299ba0e49119d7de3b11bbed8a9 : FAILURE in 1h 02m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a00d545626241c181d6356b9b5a5e0f : FAILURE in 1h 05m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/91b080ea0e9b48cf825ff5fb29ec0c22 : SUCCESS in 53m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/02b3fe2ec3a64f2390843a3c435f7e6f : SUCCESS in 54m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78ac5359443e4ae18ce7052b2a705a60 : FAILURE in 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19b265efbaa1492fabc18943cd3e5787 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ac16021d14c14eb9aa0c63b6e2d0547c : SUCCESS in 22m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74dafd4e39ea42d1aa5f13a53799bb32 : SUCCESS in 43m 14s","accounts_in_message":[],"_revision_number":1},{"id":"37acf46a491a188c62604b7bd270e9f4e7057741","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-08 17:59:15.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"18fdf0f2ba506b61b48c45c565ada7edb3f4b7f3","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-06-19 04:53:44.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c4db5c3302fed42f75cc928354d67258dc064232","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-06-27 06:55:43.000000000","message":"Abandoned\n\nsuqashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c68b23e6234f6b71be01b2641c541de5f5fedc6","revisions":{"4c68b23e6234f6b71be01b2641c541de5f5fedc6":{"kind":"REWORK","_number":1,"created":"2023-05-09 07:02:36.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/71/882671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/882671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/882671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/882671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/882671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/882671/1"}}},"commit":{"parents":[{"commit":"46ffe6c73fc0b1e9f1001f1288823a0365ce13f1","subject":"internal_client: Add iter_shard_ranges interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46ffe6c73fc0b1e9f1001f1288823a0365ce13f1"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-05-09 07:01:40.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-05-09 07:01:40.000000000","tz":600},"subject":"WIP: Split the inter_shard_range interface into 2","message":"WIP: Split the inter_shard_range interface into 2\n\nChange-Id: I2e0352bf8c0235438f92e202f403840b4aee337a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c68b23e6234f6b71be01b2641c541de5f5fedc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c68b23e6234f6b71be01b2641c541de5f5fedc6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
