)]}'
{"id":"openstack%2Fswift~882740","triplet_id":"openstack%2Fswift~master~I77c4f1f4015e43905146111a035de5afc67f14fb","project":"openstack/swift","branch":"master","topic":"p-memcache-incr","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-05-11 10:53:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I77c4f1f4015e43905146111a035de5afc67f14fb","subject":"memcached: log method and time_spent for exceptions","status":"ABANDONED","created":"2023-05-09 18:33:19.000000000","updated":"2023-05-11 10:53:34.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"4eccade3a9c5c86440e0130c8017c59dabf49304","_number":882740,"virtual_id_number":882740,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-09 19:33:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2023-05-09 23:01:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-09 18:54:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-09 19:33:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-09 23:01:36.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"fffc5aa23043e9653efb4461893f12bdab74d85a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-09 18:33:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e93abbff2cd6303222d1cc673db77644546c160","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 18:54:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2318e1b1630145d4b89ae52ea89d4423\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fda9f52cfb3c41c8842d1848f3a880d8 : SUCCESS in 14m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e724ae5e71e749f1a381f53db1d6160f : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f66578e58a1e42abb7729713ba144f4a : SUCCESS in 20m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4dcac2d64a4b40ffbd88ce20463a9d39 : SUCCESS in 18m 05s","accounts_in_message":[],"_revision_number":1},{"id":"86f8f201b5f73c54acdd6ae4e98cf6f56d954f9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 19:33:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2218fbabae94cd584128313181d6fbc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f5cbcb27519495b9e6539afa94d3857 : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/69633aed1e1c404ba0ae18fb5fd19c6e : SUCCESS in 50m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bc471706355d4c06ad266d80151846ec : SUCCESS in 44m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4981c8f09e624fb48bb140ebfd018cb8 : SUCCESS in 49m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a92ccbe3cced4687886b5725aad2916a : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0935cd70677c4a2dbe834709bbaf0de5 : SUCCESS in 19m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ca27d91a7fb4c6b8e3c813981920383 : SUCCESS in 20m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/93d0ce5c487644b1b448c1f2bdbf6a6d : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f86efe0ff44d4660891bb203843898dd : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/af6e9fe75dfc4f8586108b94c695cb5a : SUCCESS in 19m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/009c009219684bbc8b3e929c6e1e7ca1 : SUCCESS in 16m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/620d91b1eb614b4f9c9758add3e06d76 : SUCCESS in 14m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1317a84e129f41e5abc8d4dc609b33de : SUCCESS in 16m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/be82b21f23764cec871c67912b04c92f : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4d93a5cbf3474c00bd7459ff9104cb4d : SUCCESS in 16m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da0622acbbcf45e4a8f4a422be060dcd : SUCCESS in 18m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dfe5dfbe34f4422aa91eb0d6cd9479a1 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d0a8d69d5f80452184209deab0fefd03 : SUCCESS in 14m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/72172a35e849412aa5c220c15d81c085 : SUCCESS in 17m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c4b328e7377e4011a9bab84901f23b2f : SUCCESS in 6m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5d18ef2d3a0545f4801c45a74bfd8c14 : SUCCESS in 16m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/406afd3d09a943289f993f4459ac0fd5 : SUCCESS in 7m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b2413452f0214818a667a3cb58d2c21f : SUCCESS in 50m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/278525081b8245c9a275d3a7546d8676 : SUCCESS in 56m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce2d6ab213674f4f93f932e2a0bb196c : SUCCESS in 39m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/29d405969b78464f9c28fea67af95439 : SUCCESS in 45m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21a6d865afa148c1958d06f4fc83694a : SUCCESS in 18m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c37cb4d5c9432186f79e9f0dd2a382 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf37ed06b70045e6af6ac26d21c34173 : SUCCESS in 22m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8701d8a2e1154f3eb0a41e7c440b2158 : SUCCESS in 35m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b521b1ce7b6e1e88725f96212acd47b17b85167c","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-09 23:01:36.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eeb8a59afd4c99ceba5d68a6b35da337097c4866","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-10 20:23:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4eccade3a9c5c86440e0130c8017c59dabf49304","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-11 10:53:34.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6dc58f3882f20e31cb6c91ebafc66b887d9590c5","revisions":{"6dc58f3882f20e31cb6c91ebafc66b887d9590c5":{"kind":"REWORK","_number":1,"created":"2023-05-09 18:33:19.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/40/882740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/882740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/882740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/882740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/882740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/882740/1"}}},"commit":{"parents":[{"commit":"4d19482b9c25ee3f4a8399210273740f884b2ce6","subject":"Memcached: emit memcache timing metrics when exceptions raised","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d19482b9c25ee3f4a8399210273740f884b2ce6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-09 18:07:55.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-09 18:09:56.000000000","tz":60},"subject":"memcached: log method and time_spent for exceptions","message":"memcached: log method and time_spent for exceptions\n\nThe MemcacheRing method and time spent in the function is already\nlogged when a Timeout is raised. This patch does the same when other\nExceptions are raised.\n\nChange-Id: I77c4f1f4015e43905146111a035de5afc67f14fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6dc58f3882f20e31cb6c91ebafc66b887d9590c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6dc58f3882f20e31cb6c91ebafc66b887d9590c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
