)]}'
{"id":"openstack%2Fswift~882761","triplet_id":"openstack%2Fswift~master~Iae9f10ee6825fa7c67401edf2ae90a8db0158212","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Iae9f10ee6825fa7c67401edf2ae90a8db0158212","subject":"auditors: Track deleted DBs","status":"NEW","created":"2023-05-09 22:42:29.000000000","updated":"2023-05-10 00:01:56.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"eef7ace8b75700647f6205eb4e99e47001598920","_number":882761,"virtual_id_number":882761,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-10 00:01:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-05-09 22:46:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-09 23:46:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-10 00:01:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae30c8325d621700f73a96df5251ce51f37dee3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-09 22:42:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e2322b19d0c62a335591ec3c2a4384ba4540d04","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-09 22:46:21.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c688c98ec3d329816bac273f8f15df9172f5657c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 23:46:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24b82b50d376471b9f15b432efccf66c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b20f0c5b454472b98f6ac9202ee6863 : SUCCESS in 15m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0aecb3200aaa40b1a99c2a1973b4439a : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/55bf127f26c34adf81f50e1ffdfa632b : SUCCESS in 23m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c797c6eb6dc048d4973ada89cbe562e4 : SUCCESS in 1h 02m 53s","accounts_in_message":[],"_revision_number":1},{"id":"eef7ace8b75700647f6205eb4e99e47001598920","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 00:01:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62221484deb547e9925951f842faf216\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43a51a77fa6046ec8dc05256f5f36f51 : SUCCESS in 7m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/d1db06a640874ee4a0291a252dec8f0c : SUCCESS in 51m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8eb79ecbbbda4982b6fa7dba1e4e981a : SUCCESS in 30m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd9578fac1344f3da4cfe4846687a1af : SUCCESS in 38m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/120a2f78ea13484ca68dcc533dffb216 : SUCCESS in 14m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a6fe229cb044484803cc4b30eb476af : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ac3b0e9aa014256b10f69089b8390dc : SUCCESS in 14m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a67d0095a5ad4339b75b546eabd3a109 : SUCCESS in 6m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3e206c99f9254b80861926ea72dc3b42 : SUCCESS in 4m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e8adf860e134cba93de463de81e639d : SUCCESS in 15m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c51956f7a6d54e079464b8f1ce7d9219 : SUCCESS in 14m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ed80d31358ec41e7b06540c097b07128 : SUCCESS in 15m 01s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9ff437e90fc24a739fefc34cf4e2658b : SUCCESS in 15m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0e765ed7c9fd40a0b4f3771175c65ff5 : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/99577ab1ba1840628bf0f457f1899153 : SUCCESS in 18m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bf4f8d9c5f9246e48d7c3afdc34a1ff9 : SUCCESS in 25m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/27a6ab7efa42475bb70228ec81e4ee37 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c4a44ef284d54ef59480cccf0e258b4b : SUCCESS in 14m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b7a9d87136d449bebffab758e348600f : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3e5c8d7acc074b2b946df5feb2118f08 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3c58d809964d4e0c813393dfeaa606b1 : SUCCESS in 16m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b9715767fb584f7492e9749723ded475 : SUCCESS in 6m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/35ee13a526934ed7a793c35b2a1d409c : SUCCESS in 1h 10m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6686311007744da9cbf1e4aeb1f5e2e : SUCCESS in 56m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/34e6ef2d1b59494095fe2d5614f7b214 : SUCCESS in 48m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/538377b7d65244a289632d98fecde31e : SUCCESS in 33m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9894bf543a5644bdafd8f4a506de6777 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa896927ec8543b3aee5224aa48adaac : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/52ae671f6ec3439ab6fa420ccc6cd5d0 : SUCCESS in 17m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d69784f18e8043388fef78c32b5f49d9 : SUCCESS in 44m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"337cd50a1190917691d579bb434cbdc02f7744e3","revisions":{"337cd50a1190917691d579bb434cbdc02f7744e3":{"kind":"REWORK","_number":1,"created":"2023-05-09 22:42:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/61/882761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/882761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/882761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/882761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/882761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/882761/1"}}},"commit":{"parents":[{"commit":"f99a6e5762896c7789d168bc49d8cdcb47903264","subject":"Merge \"Log (Watchdog\u0027s) Timeouts with duration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f99a6e5762896c7789d168bc49d8cdcb47903264"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-09 22:08:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-09 22:37:51.000000000","tz":-420},"subject":"auditors: Track deleted DBs","message":"auditors: Track deleted DBs\n\n... in logs as \"deleted\", in stats as passes.deleted, and in recon as\n\"{account,container}_audits_passed_deleted\".\n\nChange-Id: Iae9f10ee6825fa7c67401edf2ae90a8db0158212\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/337cd50a1190917691d579bb434cbdc02f7744e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/337cd50a1190917691d579bb434cbdc02f7744e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
