)]}'
{"id":"openstack%2Fswift~883084","triplet_id":"openstack%2Fswift~master~I1f26483dbc5935052ec8df0c874ee38070b84d5b","project":"openstack/swift","branch":"master","topic":"pipelineTimeout","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-06-23 23:41:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1f26483dbc5935052ec8df0c874ee38070b84d5b","subject":"move test_GET_pipeline to BaseTestObjectController","status":"MERGED","created":"2023-05-12 17:25:32.000000000","updated":"2023-06-23 23:42:15.000000000","submitted":"2023-06-23 23:41:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"883084-pipelineTimeout","meta_rev_id":"f5abc0e3005ddd9d6292943b25834ca0d204cb59","_number":883084,"virtual_id_number":883084,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-23 23:41:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-18 21:53:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"value":2,"date":"2023-06-23 22:37:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"value":1,"date":"2023-06-23 22:37:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-12 18:24:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-18 21:53:48.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"REVIEWER"},{"updated":"2023-06-21 20:27:03.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2023-06-23 22:37:42.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"4e0c5bab6707cb4b229a113b719975f7545c61f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-12 17:25:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ed4d96b77e857f06ae7934a28af05819b767c9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 18:24:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa0e70749d0446a1bf7c1004bac08e0e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efdfe7676ca14a98af9073fe5275b77f : SUCCESS in 8m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d016b343238b4a679d631d547e4668ee : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f13d553fad9341c9b791ce607844ddf7 : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07e420496c3d45109463e8be20f35361 : SUCCESS in 20m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/42e1ee0d9cf642e9a62b6f70dd5f9103 : SUCCESS in 9m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2affe037e11640cc96df6e9254b06b62 : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d5e44dbf95f8431893103bae33599a0a : SUCCESS in 18m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e26082523f654e2c99ce99ebfaba26e4 : SUCCESS in 16m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/890e8186b2e141b2adc7fa005db7bf94 : SUCCESS in 17m 05s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/82ac64d210a544cc80ffa1d8cecb8a58 : SUCCESS in 17m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6ba3278897fe455188267f3b7147b04b : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0cfb1e91d174473494cd9e1e82b7967f : SUCCESS in 17m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/768077a580124175b7709f271f73832a : SUCCESS in 17m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cc3ac3d76ef34e4895e4b0460d23b8b1 : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8d0a4c5d6c384191bff3dc3df8745786 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/34592c1d03714949b173709e83dfc625 : SUCCESS in 17m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c8fa1de324244ce8e9f53a9176117c6 : SUCCESS in 39m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c40d05ba0e4f494abe736dd55d57eed4 : SUCCESS in 51m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0b13d63918e405ebeb9a4b1967a708e : SUCCESS in 19m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94cec31c56bb4cafb8ca981ef2f067a3 : SUCCESS in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b5180fbc93ab40b9b02c0acc5c6ae2c1 : SUCCESS in 22m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e4efafa18c6101163f82b17f878404cff6394e72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 18:56:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94d8d54806ab4b1cb041695ff30bf359\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/948b68bcc8e845b5b6ad9b69f47b7a42 : TIMED_OUT in 1h 30m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b1440d9c96fb4e6e93dc359cb6f71e14 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/18fb064eba9b4505923a70e602293397 : SUCCESS in 21m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6691a23ef4024decaff8a9cd42423e3f : SUCCESS in 20m 48s","accounts_in_message":[],"_revision_number":1},{"id":"dafa2790c63084ad1508e02a2d7be886a6ae7c18","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2023-05-18 21:53:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7615362c0abdb5b99d66c41f2d4bce0a0bf69850","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2023-05-18 22:02:57.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"704acd938f110e69a12149a26fe5bf1007a462d1","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-06-21 20:27:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"575e89c2be2119165459c7ca704f96e9c5cad585","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-23 22:37:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1e70dd17981fa1e38ce437017e09fbc13f9afd04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 22:38:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58ac09db7691f8a9196c9d832dea5436527b5f16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 23:41:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66f7c522f577441bbd1fc6699b4b0462\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09dfae324c124468951e0b00316cef0f : SUCCESS in 7m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/37bece58456f417f85db9fe0fc6d90c4 : SUCCESS in 7m 53s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f0055a8d62844b3f935bac16004b5a5a : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7128f074107e4673bfa3fb58a15adbc4 : SUCCESS in 36m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/de0158d4cdb64081ad7e9e8f00647489 : SUCCESS in 29m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/72bd3a87f1e64216a778944fa15c29ed : SUCCESS in 32m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/aa42f7719ad043d3b0bace56dc07ce7b : SUCCESS in 28m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/530c3e8293d14f5c8d12c6888dcb5892 : SUCCESS in 31m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d7900b33d9f348329c0820194df994b0 : SUCCESS in 31m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ddbfc1d112f44d65acc953a95477ff09 : SUCCESS in 32m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4f47f011ab1a498c8d8bdd6e044949a8 : SUCCESS in 29m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8c8f470ab02e4742bebc9017b62924cf : SUCCESS in 32m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/06a6eda79fad425cb9d1022123010f04 : SUCCESS in 32m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d6b694cdb8e94884a8b65772be295ef4 : SUCCESS in 6m 04s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/49e069d19c8744d0a0e6b414e8292061 : SUCCESS in 7m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a30ffa8182c84b69b05ea1bfdedb7712 : SUCCESS in 52m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7df5b0dbae344d669174daa39ba3b639 : SUCCESS in 56m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7be52a4b88b3451a90e0db7bf8693239 : SUCCESS in 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53522cb4f6d048f290412930e68c00d3 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"5592e27e423142b92fd610d0aee7aeb1632f0f20","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 23:41:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f5abc0e3005ddd9d6292943b25834ca0d204cb59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 23:42:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7947ccb988a8402ea3ea9c9454a11c59\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20ca1ba3bd3d4676a43556542106e112 : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/41a0f1185aeb46e9843ec39e20276b54 : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/9968f3c932354355a3a3a940612b548d : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4c9d5a40b112d6cb1c344a00a8a7c4c1b529588","revisions":{"c4c9d5a40b112d6cb1c344a00a8a7c4c1b529588":{"kind":"REWORK","_number":1,"created":"2023-05-12 17:25:32.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/84/883084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/883084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/883084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/883084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/883084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/883084/1"}}},"commit":{"parents":[{"commit":"647ee83906df8fae336b479c0f169be2ced1772c","subject":"Unit test for keepalive timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/647ee83906df8fae336b479c0f169be2ced1772c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-05-12 17:24:11.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-05-12 17:24:42.000000000","tz":-300},"subject":"move test_GET_pipeline to BaseTestObjectController","message":"move test_GET_pipeline to BaseTestObjectController\n\nChange-Id: I1f26483dbc5935052ec8df0c874ee38070b84d5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4c9d5a40b112d6cb1c344a00a8a7c4c1b529588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4c9d5a40b112d6cb1c344a00a8a7c4c1b529588"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
