)]}'
{"id":"openstack%2Fswift~883091","triplet_id":"openstack%2Fswift~master~I0e13e0148a6e19fd6d0a9438604aa791f948502a","project":"openstack/swift","branch":"master","topic":"sr_cache_log","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-06-14 12:22:29.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-06-14 11:08:44.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I0e13e0148a6e19fd6d0a9438604aa791f948502a","subject":"proxy: print logs when write new shard range cache into memcache.","status":"MERGED","created":"2023-05-12 22:37:58.000000000","updated":"2023-06-14 12:23:39.000000000","submitted":"2023-06-14 12:22:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883091-sr_cache_log","meta_rev_id":"c0055ba5b0fcbca5f06ba9f9e7c553cd03156551","_number":883091,"virtual_id_number":883091,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-14 12:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-06-14 11:08:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2023-06-14 11:08:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-12 22:38:50.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2023-05-12 23:00:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-12 23:50:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0bf495aad7986b66e113741cada118a50d85ccba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-12 22:37:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b86e9185e52867b7e317988f778f247f5d09ecec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 23:00:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36265713a379483eaa7284615a07d562\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3578cbc86a2f4f46b61fde4b75e8a877 : SUCCESS in 15m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1db02df8e11b4d129f40455f91db874c : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/018f335c944a4a56b1471afd390a70a6 : SUCCESS in 19m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/11e153f3ef6940c2bdd9a2a1432a2e3f : SUCCESS in 21m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ca82b07c0be52ef085800992291c693e9af77567","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-12 23:50:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d20051eab7834da6b5c5a8ecb3caee10\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaff095e64bc4ad19b5cbb06e0a785ec : SUCCESS in 7m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/cf73f80aded24492b8545689e101e970 : SUCCESS in 36m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d767170f160e420f83fc0e47ef7ee10b : SUCCESS in 47m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37dc3d0a8b554131bcdb5159a8b504fb : SUCCESS in 1h 07m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13b39f13c7644267bb48c961609f4c60 : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31850c1e5e914821832148dd37b790c3 : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3de7870b8fac44d2bb3246efa47e84a6 : SUCCESS in 18m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/020cee33e78a494383165840e05fab11 : SUCCESS in 7m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8e64bff2120147ee83c6bba365cb5beb : SUCCESS in 4m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dfa05182c3a94fd487391b34746e9d99 : SUCCESS in 15m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1eae71a1e7254c30a9a8ed298df4775e : SUCCESS in 16m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9318264103b7407ebda3b8170e0bbc9f : SUCCESS in 12m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8b53fe45dfbf4466abb3268de8628fad : SUCCESS in 16m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ac3dac3d5d204bf08495523af6bdada8 : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c6c8e6b2fcf74a419aaca6b9c068739d : SUCCESS in 17m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d9da4f3922954bc8bae937a62a8d30fa : SUCCESS in 13m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/acc8ef8dce0248e49783f40f8eb1bafb : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1e2abcef55f9470a8ef39b2b26423230 : SUCCESS in 11m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ce64814eb3a1428ebb61ae03c7b3c648 : SUCCESS in 16m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3e7593c1ad014982be6cdaf260e9cb1e : SUCCESS in 5m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/789e459e6ef0424baef3e38cdc10bb45 : SUCCESS in 15m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/af3d1ef4c0f34fa0adeebeef731c03c9 : SUCCESS in 6m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/15e616c30b484922baadea5063813fed : SUCCESS in 55m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd1953ca2ed14b28b0128fe6d471246d : SUCCESS in 58m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/39ba55c6a12f447c848498b57c894e55 : SUCCESS in 53m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2cea45872d644167a2cab6e4d163845a : SUCCESS in 39m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01d8955f050c4037b96698e55d9f3c7c : SUCCESS in 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89b8b1cd4bbe4102b8d05149082d09de : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ce8a664cdaa49688bbd17beb49679d9 : SUCCESS in 19m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44b91b5cc47548ce9e64e1725a180480 : SUCCESS in 50m 31s","accounts_in_message":[],"_revision_number":1},{"id":"aafab5664ab60d605619f94248d65d6667da4428","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-19 08:39:00.000000000","message":"Patch Set 1: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f677030efa34f1f51bf6389940ecc974b905e502","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-06-13 04:59:57.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e16b5fa8507f11577087696a0fcbfa4153376df2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-06-13 05:02:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"526db3f7567d1b3f24399c76ac5899a5ce104dd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 06:15:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c9139ec24554fc28df1513aa44e1b1d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35da157a9667467cbfbe96dd4d52e425 : SUCCESS in 1h 10m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cdccf008cb03495d8b4b17e28d8229ee : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a38a5537e284e45b98ac6ac51dbb2bf : SUCCESS in 1h 07m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e83e57186718480684f0a5d5fe197b2b : SUCCESS in 19m 05s","accounts_in_message":[],"_revision_number":2},{"id":"a45d6375c37ab682320a0b16caf895861f873a0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 06:24:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d3a918dd20a47c5bf1ad4e643bfd8e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b7fb7992a4a454ab24728a5a7284ec9 : SUCCESS in 10m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/62b7aed759cf488ca77893a4a5147442 : SUCCESS in 38m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6ea5396f651b43499e6406ce037f12ed : SUCCESS in 33m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f614c62cd6ea436a88d4d189cddfc0ca : SUCCESS in 45m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8f57164617a4fb9b19cfc5c11a0c07f : SUCCESS in 16m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/912fbe8b90ab43a38aaa21294f7e3ace : SUCCESS in 16m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6bdea750f31f468c959c7f038de6b4a6 : SUCCESS in 15m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/46a42d5029e84ce0a5ee3cd31cecc49f : SUCCESS in 7m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/56fab62189ca413fab2acc77773956f6 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/29d46174935b4e649cd61b40a7497080 : SUCCESS in 15m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a1bbeefb3f3d41d787e39c6465ad2f25 : SUCCESS in 13m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/396d43907c234297bba5e8e816861ce7 : SUCCESS in 14m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d173c21e4d094b4281adb6fcdc56b374 : SUCCESS in 14m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/60854776ed4547b2a201d79403b4000c : SUCCESS in 14m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d3a986a0d95b4f9e9e68e86d86044e1f : SUCCESS in 17m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b4696986863349c28ed9465f561502be : SUCCESS in 15m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5c47b70ce6ed48ff9e0f88957fe5f0d2 : SUCCESS in 13m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bc4dcbae3c204461b59e0281ded73ef2 : SUCCESS in 13m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a401d4786f2b48b5868854b076bbbb5a : SUCCESS in 16m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5229b2edf69d42079d4d59307c049709 : SUCCESS in 13m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dfe8eb14d6de4b6eb42bb500d9c46c26 : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1bc4f0bf52fd4d6fa594c2ae77e7233c : SUCCESS in 9m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d7f2d6a10a504586a755ace02b89b4e1 : SUCCESS in 53m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ddda53193bc496aa5220d3561c81525 : SUCCESS in 58m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a0e75cf8d856445ca410c6f6110dffe2 : SUCCESS in 51m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6abcef1eb25a47f38ca7d0bfa387979e : SUCCESS in 46m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/721eb9a76e4c45aebf094bb2ccfc8471 : SUCCESS in 18m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70a80df6e9fa44398fc7ec4351573c50 : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/edec00dfafe24fde998301407637adaf : SUCCESS in 21m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5776185ae826426881206ad462d6229e : SUCCESS in 40m 05s","accounts_in_message":[],"_revision_number":2},{"id":"705824b7911118ee88c50a62f215b7981a6b9b4a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-13 09:11:24.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b5b84360cde7b24980565aa812893b2a37b41ad8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-06-13 18:21:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4e097ea8d1cb2b54c212add47bc50e5ccf636a66","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-06-13 18:21:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f0706721e3ca190f01c0406685e6a2ba423d452","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 18:41:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50eafce0b11a4ebb9c65afb3cf151c62\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d9ee28870a349e4bf15b393b81e2f7b : SUCCESS in 15m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/17b6c8a287154ea88713d72a2a6a5a50 : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9610c02e4244d928853bd948245c4c7 : SUCCESS in 17m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bcd41df1b9204985a20656ecbe62496b : SUCCESS in 19m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b610d1e9137dd08ec6110bdc9e5175d3afbcafa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 19:41:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d1a8046e1214bc08f6368c785b644fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fabdcf4178514cf0a714dd9138b6675f : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/24248fe365ab4073a95fc683748ce7a9 : SUCCESS in 42m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f05baed00065423ab05f1a4d88d8cd02 : SUCCESS in 55m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12164ae614d74092a794c5f533db89a4 : SUCCESS in 1h 05m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/064c0c0286d64b9c967fc1498c2c6f9a : SUCCESS in 18m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f309ab159b6c4326a362a309acd41607 : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34d4921a98ba4b2a8142275965dba71b : SUCCESS in 21m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d4095e9850254d338b06901683667086 : SUCCESS in 8m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/024005f89f1847b6b826d2556a54e395 : SUCCESS in 4m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fca164f5fd5c49b88d61e269c2dfa5ca : SUCCESS in 17m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0cf88b86e37b4da0985ab3c7e93be677 : SUCCESS in 18m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2c1bf566abbc4ee4bd55fc7b8706c47e : SUCCESS in 19m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/657063b4acea4aa3a06614df9764a648 : SUCCESS in 19m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/43db55d661a141539f5e344ab42f24d7 : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c63bf34af4bb4067a0af0a50a7cddb6d : SUCCESS in 20m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9cd59201afbd46e2a76f0606319a7328 : SUCCESS in 20m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0b9a4128e0fb4e378327a6f7e8416642 : SUCCESS in 18m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/35f50717ca69479cbc383b16532f1b2f : SUCCESS in 17m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fcc3d5235b014b49b8fe036ac4b27356 : SUCCESS in 13m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e5756e2c74ce436f95bd7f1096bcce98 : SUCCESS in 7m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0d7ed2e37fa14d18a754ceecedf68161 : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/93c8c9bb29c04f15aaff32b87ef1812e : SUCCESS in 6m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f26a4595baa04ed4ad71d22d7fc3ad4a : SUCCESS in 54m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c2a13776a6347d5adc835b95578ec38 : SUCCESS in 55m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/98a387459fec4653904b3ef0f20dd6e1 : SUCCESS in 45m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cc78329e3c2843f9adef28935c00a125 : SUCCESS in 37m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32a863b7983746eb92b6013d385b97bd : SUCCESS in 18m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1a1c6cc0cfd40c68560fbe6804de2c6 : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/67994d5d96a84feea688aa8c8d76b2a9 : SUCCESS in 17m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a16cb3538e254057a846c5f541833bc9 : SUCCESS in 52m 31s","accounts_in_message":[],"_revision_number":3},{"id":"e1fe92386f4716d3f2aef302c2d2e59f1a76604d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-14 11:08:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3e92b1b3fc15105148c2f6ef88fbe5d74c3cd5b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 11:09:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"51b641117124514022641983e855ef465e2f4a59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 12:22:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/093714a6175149f2a87f0ecc79e98d05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7df945c2566749a3a0f314fe369a6f3d : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/d9ca83c43bc74354ad3d464b143ba986 : SUCCESS in 1h 01m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e17965f561654330aa7fd9a555863065 : SUCCESS in 37m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff9112ff50514a1ebe3191a159e09e20 : SUCCESS in 49m 11s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e4045bb9fddb4dc8a03e746638ffc2e7 : SUCCESS in 9m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/5ce4c8aa49e144a2b1f11f370550985f : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7975951570dc48259fe2d8459d07b347 : SUCCESS in 17m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/50616221d4774ad8b6326481f5e0b29e : SUCCESS in 16m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d773dd8fd004417dabda6482de3b4624 : SUCCESS in 11m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/faf515afa5f742489b86715fabb896bd : SUCCESS in 18m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a38cea01fe9c4b3cbfa6bf9fcddc8aff : SUCCESS in 17m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b9124650ed6849758aeb6fcce2b215a4 : SUCCESS in 14m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8769d55490d24e968a62ae78f0977257 : SUCCESS in 18m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ceddc5024be74baea1cf17aa7ad22533 : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e1f70e7e4d3c4b5192a463dba298a15b : SUCCESS in 11m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/91e50fa33730485b865caacd4bfda9bb : SUCCESS in 11m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c54c50e0d55b4b80a438d4db40db407f : SUCCESS in 7m 01s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/059ea6a3d0e9407ea83505162d1beb61 : SUCCESS in 7m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/93bf46add550482395dfcef704ebcee7 : SUCCESS in 54m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4392f18f8e8405fb447018a8b47bebd : SUCCESS in 1h 07m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f9eb5041f60f4521ae9ad92a243bd503 : SUCCESS in 1h 03m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba1f9a89936a4f0389f159ba017bec5a : SUCCESS in 44m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96fbb5e8616245fca3140e6d0df4b2f8 : SUCCESS in 18m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df6fd1bca68b4f29bdc070765a01db0a : SUCCESS in 4m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8f3c9b462ea4d48a1ae155c944beb77 : SUCCESS in 1h 00m 28s","accounts_in_message":[],"_revision_number":3},{"id":"27cc09a88c296fc750a4a3513289231e00c37dc5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 12:22:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c0055ba5b0fcbca5f06ba9f9e7c553cd03156551","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 12:23:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/228f4af54b05497e8c366c5647e256ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aa6a19cc827496aa205e0d09242d408 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f7799bfa58b1443395e5f9adec19102b : SUCCESS in 39s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8c0f607764bd4114a6bc82719a79864f : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f150a5035ced99f40cad0888c2301bd1469ec279","revisions":{"23f24bdffd8875aa3cba7b4dccca1e3b1298d03d":{"kind":"REWORK","_number":1,"created":"2023-05-12 22:37:58.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/91/883091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/883091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/883091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/883091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/883091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/883091/1"}}},"commit":{"parents":[{"commit":"8f84aef7005d1a797408dbcd66355c641f7b2a9d","subject":"Merge \"disable requests_mock pytest plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f84aef7005d1a797408dbcd66355c641f7b2a9d"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-05-12 22:37:20.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-05-12 22:37:20.000000000","tz":-420},"subject":"proxy: print logs when write new shard range cache into memcache.","message":"proxy: print logs when write new shard range cache into memcache.\n\nChange-Id: I0e13e0148a6e19fd6d0a9438604aa791f948502a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23f24bdffd8875aa3cba7b4dccca1e3b1298d03d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23f24bdffd8875aa3cba7b4dccca1e3b1298d03d"}]},"branch":"refs/heads/master"},"725bb2d1eb5bbae62c25c073912edd69ab95731a":{"kind":"REWORK","_number":2,"created":"2023-06-13 05:02:33.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/91/883091/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/883091/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/883091/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/883091/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/883091/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/883091/2"}}},"commit":{"parents":[{"commit":"cc26229eefa24b33d54bb1420895be685705d4d5","subject":"Merge \"container: Stop pulling alternate stats when getting policy index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc26229eefa24b33d54bb1420895be685705d4d5"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-05-12 22:37:20.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-06-13 05:02:19.000000000","tz":-420},"subject":"proxy: print logs when write new shard range cache into memcache.","message":"proxy: print logs when write new shard range cache into memcache.\n\nChange-Id: I0e13e0148a6e19fd6d0a9438604aa791f948502a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/725bb2d1eb5bbae62c25c073912edd69ab95731a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/725bb2d1eb5bbae62c25c073912edd69ab95731a"}]},"branch":"refs/heads/master"},"f150a5035ced99f40cad0888c2301bd1469ec279":{"kind":"REWORK","_number":3,"created":"2023-06-13 18:21:17.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/91/883091/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/883091/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/883091/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/883091/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/883091/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/883091/3"}}},"commit":{"parents":[{"commit":"cc26229eefa24b33d54bb1420895be685705d4d5","subject":"Merge \"container: Stop pulling alternate stats when getting policy index\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc26229eefa24b33d54bb1420895be685705d4d5"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-05-12 22:37:20.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-06-13 18:18:18.000000000","tz":-420},"subject":"proxy: print logs when write new shard range cache into memcache.","message":"proxy: print logs when write new shard range cache into memcache.\n\nChange-Id: I0e13e0148a6e19fd6d0a9438604aa791f948502a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f150a5035ced99f40cad0888c2301bd1469ec279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f150a5035ced99f40cad0888c2301bd1469ec279"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
