)]}'
{"id":"openstack%2Fswift~883314","triplet_id":"openstack%2Fswift~master~Ibed9dc0afbdb922d06f7798bdac01db7c55b19f1","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ibed9dc0afbdb922d06f7798bdac01db7c55b19f1","subject":"docs: Update versions in Getting Started doc","status":"MERGED","created":"2023-05-16 22:36:06.000000000","updated":"2023-05-17 00:18:54.000000000","submitted":"2023-05-17 00:17:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883314","meta_rev_id":"0bcabd500bde334acab6eafd17cdac0bb070c151","_number":883314,"virtual_id_number":883314,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-17 00:17:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-17 00:04:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-17 00:04:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-16 22:46:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"099df1ec1c9748b1b347a9c070bb6fa652b54154","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-16 22:36:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80dc84b874c2e45a71826e0a7443448d700cecfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 22:46:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/458931443034459ca7ace57b58823764\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b63f64f98c9344398b21f95b20556002 : SUCCESS in 8m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ab62269f36a5182ce50beec705b94e521ecc632a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-17 00:04:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"535d44b5870366f694404dfed1c0ced829044c83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 00:04:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac7208d392677bb70234d02009fcfdd487612a7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 00:17:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6be4adbc07f4f6cb91a87595c7b1264\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fafe9dd2dd0546e786e85a476355e9ac : SUCCESS in 9m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8dcde3a107984525be0bfffb0592e740 : SUCCESS in 7m 40s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/08cbead1855c4b73bc2df57f35d2093b : SUCCESS in 8m 11s","accounts_in_message":[],"_revision_number":1},{"id":"2598039db1576d44e3ac97e3dcd7c4c0c29bafcd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 00:17:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0bcabd500bde334acab6eafd17cdac0bb070c151","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 00:18:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/909e37726e7e4847bd37ca643ea1d9b8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21c05596456943f6b810494cf2bedd57 : SUCCESS in 41s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9edfe3c9e97d499192ce3c979d3423fc : FAILURE in 28s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/5bec178093ea45a4b82fa5726799aead : FAILURE in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cff092b6822385e93d8d42b579a67f10291f795f","revisions":{"cff092b6822385e93d8d42b579a67f10291f795f":{"kind":"REWORK","_number":1,"created":"2023-05-16 22:36:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/883314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/883314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/883314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/883314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/883314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/883314/1"}}},"commit":{"parents":[{"commit":"e6792bccef1a18336e8f266a5264bd2222388e8c","subject":"Merge \"testing xattr metadata with py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6792bccef1a18336e8f266a5264bd2222388e8c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-16 22:36:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-16 22:36:00.000000000","tz":-420},"subject":"docs: Update versions in Getting Started doc","message":"docs: Update versions in Getting Started doc\n\nChange-Id: Ibed9dc0afbdb922d06f7798bdac01db7c55b19f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cff092b6822385e93d8d42b579a67f10291f795f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cff092b6822385e93d8d42b579a67f10291f795f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
