)]}'
{"id":"openstack%2Fswift~883315","triplet_id":"openstack%2Fswift~master~Iad3011e8603cd2771add97942d46122434d048cd","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-05-25 09:05:09.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-05-25 10:56:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad3011e8603cd2771add97942d46122434d048cd","subject":"Drop more translations","status":"MERGED","created":"2023-05-16 23:17:16.000000000","updated":"2023-05-25 10:57:29.000000000","submitted":"2023-05-25 10:56:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883315","meta_rev_id":"d1b6e401039918e96a8a7431c4ca246162c70459","_number":883315,"virtual_id_number":883315,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-25 10:56:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-05-25 09:05:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2023-05-25 09:05:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-17 00:30:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-17 00:30:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-17 09:04:42.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"84a8327bf911693fb820111761f8d7327bad7300","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-16 23:17:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"813abfcc74f3b3b90597e721112962d31f776109","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 00:30:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a86a966d8f04ffd868b34d4e2af500f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ed4803c7af241b1a3310a9bdc0b9167 : FAILURE in 14m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dfb07799867c4111a9d11a9721d265df : FAILURE in 13m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/861e6800e3fa44709c1c5287f432b06c : SUCCESS in 1h 11m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d500f7432ab9424fac2162361673af7a : SUCCESS in 18m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a4e13777e9ba2002967f53f3aa95f001ba70fd43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 00:30:21.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95163a5cd1bc4189be033341b840b793\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abbd8750a2054d7fa71c1c364bff3e4d : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/16b3498331f342bf9b5a1ffbe1208a03 : SUCCESS in 1h 04m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/37ffbe559a224cd88f9d422010190868 : SUCCESS in 51m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/982312a8decb4f5cbb807e1421b36015 : SUCCESS in 1h 04m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d8971cbe50714a21bc44d63f9cffd9d2 : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee03e6b0b8e14f2889608fc5f52ece0e : SUCCESS in 17m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a053085c3c04f0aaf7ff5cb406a70a2 : SUCCESS in 17m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a3ec9c881f114a8cac6a839049402c4d : SUCCESS in 7m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6b4a38e826d148c9b2277063a622a062 : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/679e88147f9b4630a54236e6d4838b35 : FAILURE in 18m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a8884a07d9244bc8b008c1e8a6353189 : FAILURE in 15m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1755882ba87742708377f676dfa9ccc3 : FAILURE in 18m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a29a9ce08f8f4ef8bf5a3caa46128759 : FAILURE in 19m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/05c57296327f45509971a9182155fe38 : SUCCESS in 18m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dd918d13b79a4827bd875382201acfb2 : SUCCESS in 18m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5d6cf8330895426fae59c5fab5509319 : SUCCESS in 19m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/68f31ada2bd74bd0b2d8acbd9ec332e6 : SUCCESS in 12m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1c860c34bd8e4b21b7ce361217bf6f2a : SUCCESS in 14m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a9e47855a51f482381ece2d88d2bdfad : SUCCESS in 16m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9af127b19604338a112cb3926fb0c8c : SUCCESS in 7m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/498633989a0d427895f9c4aaa8bb3a12 : SUCCESS in 16m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/77ee9a6a9e3d4f04bb207e950c987448 : SUCCESS in 6m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c5b2ac6cefe641a9bbc90c6d0d9bb7e6 : SUCCESS in 54m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c06272f7f9ef41c698f9fcc525f5792f : SUCCESS in 1h 06m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0e252f2b203a4afc87f69d0dccd6e825 : SUCCESS in 52m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/653810d40c924c6b84fe897f09950c3f : SUCCESS in 33m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/984f5812cebf427d802560c1923c905a : FAILURE in 17m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3262abc650441f3b607a38461ae261a : FAILURE in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/74c4a307a8aa48539fec9f1e79d33ccd : SUCCESS in 18m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7745fcdc8f3342d6a2aa56043438a82a : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4e9a48400176a10b3a22ad70f606deab34502737","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-17 09:04:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f1e45a24dd3046d49ea8a9775333c558c902a54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-24 20:58:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"01ba47ec6b2fff907b08ab8f0a0aa68fa006f7db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 21:24:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75fe0815d6e940ffbe49530b218ebed4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbec4df25eb8447caadd210dc5e20d47 : SUCCESS in 15m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b77b444cc75b45349a403c1065e72f5c : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/eac10de0007a444992bd11b0b93012cd : SUCCESS in 17m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c0c44c99a57462099bbaa58dcd95e00 : SUCCESS in 21m 54s","accounts_in_message":[],"_revision_number":2},{"id":"814da324c59d66ce290fe2b8981fe5cc81b5a615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 22:15:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c445e4dcb8694fdd8efd363d63c99739\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa6d580174fe4573bc7773f4b0702728 : SUCCESS in 10m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/487653a8f3404acb979e65e591347412 : SUCCESS in 48m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c5705f0bca7840118855c0bb5c70699b : SUCCESS in 48m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2752c062c804168a415b647a5346ded : SUCCESS in 47m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8458946f88ab43f6be2201a609190d64 : SUCCESS in 14m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b337d057422f48d7a04cbcbc38db9a9b : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa9e4e1ba28a4397b09908e4bbb7e861 : SUCCESS in 29m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9d060a2e08904ca4a8d2b3cf0ce65ea9 : SUCCESS in 8m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9c2ef11bc76f497d84013dc8e78990b7 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/35efa87965994e21a47059d8804addd8 : SUCCESS in 16m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aeb47343f53c46749502efb85986af2f : SUCCESS in 16m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e8782cceddaf4d6e968c8ca0c5166dbc : SUCCESS in 25m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3d38f4fd0f1c4a249d65f053a3e0cb4d : SUCCESS in 23m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7dc9101d3b9e4c76a08ef544945a0e7d : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aba70c76e179481082a55bc62470b800 : SUCCESS in 16m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/80d0e1796d7942b2b1fe41cc24b5fb54 : SUCCESS in 17m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bd627c78215e458193a0918b65b2eec7 : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6837121d4c1f4bcc96a48519febe3a20 : SUCCESS in 13m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e419e7a7ce8241d4963d8db3ac0b27f8 : SUCCESS in 16m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9ae59dc79413464cb09df36613ac9449 : SUCCESS in 6m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9a30a7a6a08d43298c25fde135a6ff88 : SUCCESS in 22m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee8733ba56bc427c9d06a425289302cd : SUCCESS in 6m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ad47e5248bc549ed9be9383b5791ede0 : SUCCESS in 54m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f0f9b285edd4650ac5d6972a8b7c5f1 : SUCCESS in 56m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/044d19970e8549d9937fd8d197e76ccc : SUCCESS in 51m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e70fba91cffb491fa070c7a661a54a28 : SUCCESS in 50m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c34548adca3e48768f42fb06ec2a5254 : SUCCESS in 17m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a52c58b7dcb402cbc4fc7ca4538165a : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1f36304579aa4d0ea74c9161ff5a99b5 : SUCCESS in 19m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d91510fc78a49ba89b5eb3f9d6805d0 : SUCCESS in 1h 10m 58s","accounts_in_message":[],"_revision_number":2},{"id":"db04a2e7756010c980d8a2acb64d3a6a648bfb54","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-25 09:05:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"596c8c1d99374b0d2c32d4c3478b1af9d0d5d4ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 09:05:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"222a1186ab655aebb47743d1d6ec2ce3faa3c004","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 10:56:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea425b0b14c445c4b5bd145553bd489d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d52a1038a84cda8564169f5db8fead : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/41bb423e8a8e4f05b579729e088d5948 : SUCCESS in 59m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b0676aef216e4d4b9539d8f2a3c2906f : SUCCESS in 59m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45a4f4feb12241ae80ee8ed1189d214e : SUCCESS in 1h 05m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f750fe76a1b6480b86973eff2da2de5d : SUCCESS in 10m 39s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/9eaec750c02b46f2a06b0842964a3fca : SUCCESS in 6m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0edfaf6c224b4b3eb7cf0fdc0eeb03df : SUCCESS in 16m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4d74c90f30e34efba0731522ca3eb167 : SUCCESS in 17m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bef061b6d5b94df48a78b23afc63b21a : SUCCESS in 29m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cdb44973c5ba4d30a9dc383bd56af4d1 : SUCCESS in 30m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/47d64d91a2a24fb6907a306e4ad1b70d : SUCCESS in 27m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6806c468c71e43fdba53afb72b239daf : SUCCESS in 29m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bd4648d3394745f1a4f62e90b06df0c5 : SUCCESS in 28m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ebd88dd2768249e2b7aecf3db23cc8d1 : SUCCESS in 29m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5152fb5550dc479aa9377a189523b9eb : SUCCESS in 30m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d806212bd60d4f869c6c8ec3366af975 : SUCCESS in 30m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5a96b77b83654bec89825b2b42066d82 : SUCCESS in 8m 33s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a44a95f9c17b4daf96b61a18fefeca83 : SUCCESS in 8m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f3bc4234fd50482681f1e5acbb966a7c : SUCCESS in 1h 05m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae6ce68a501c45e0af10a6a69b41b2b7 : SUCCESS in 1h 09m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8af7bf1b5174637b0af6767928d98fa : SUCCESS in 51m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7f58a075835d40a6956591b70b5a8940 : SUCCESS in 54m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7f0a24a61c5400a8290e1a4774b4760 : SUCCESS in 30m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06eccdc1da744355b27643a575b78b59 : SUCCESS in 4m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21a9853e23eb4f28b307bad604a4229c : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":2},{"id":"9f73ad7fd23d6589337b51647007b00561148bdb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 10:56:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d1b6e401039918e96a8a7431c4ca246162c70459","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 10:57:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/116e1794f41e4c65ac1e4ab25cef927d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d448b474e5742ff9933398f2f1cc404 : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/44155233f4964e5091122c8789913f3a : SUCCESS in 37s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/dda9bdae189341c082a1ead480e8560a : SUCCESS in 30s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d3d503f905e9afb47b7c90e438d35342533939bd","revisions":{"56fccb84d63fd0efea694e6f84254bd550cc1670":{"kind":"REWORK","_number":1,"created":"2023-05-16 23:17:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/883315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/883315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/883315/1"}}},"commit":{"parents":[{"commit":"e6792bccef1a18336e8f266a5264bd2222388e8c","subject":"Merge \"testing xattr metadata with py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6792bccef1a18336e8f266a5264bd2222388e8c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-16 23:16:21.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-16 23:16:25.000000000","tz":-420},"subject":"Drop more translations","message":"Drop more translations\n\nPartial-Bug: #1674543\nChange-Id: Iad3011e8603cd2771add97942d46122434d048cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56fccb84d63fd0efea694e6f84254bd550cc1670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56fccb84d63fd0efea694e6f84254bd550cc1670"}]},"branch":"refs/heads/master"},"d3d503f905e9afb47b7c90e438d35342533939bd":{"kind":"REWORK","_number":2,"created":"2023-05-24 20:58:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/883315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/883315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/883315/2"}}},"commit":{"parents":[{"commit":"e6792bccef1a18336e8f266a5264bd2222388e8c","subject":"Merge \"testing xattr metadata with py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6792bccef1a18336e8f266a5264bd2222388e8c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-16 23:16:21.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-24 20:58:24.000000000","tz":-420},"subject":"Drop more translations","message":"Drop more translations\n\nPartial-Bug: #1674543\nChange-Id: Iad3011e8603cd2771add97942d46122434d048cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3d503f905e9afb47b7c90e438d35342533939bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3d503f905e9afb47b7c90e438d35342533939bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
