)]}'
{"id":"openstack%2Fswift~883367","triplet_id":"openstack%2Fswift~master~Idd6b767f1a746c72c106199f5d1fada3615b1e97","project":"openstack/swift","branch":"master","topic":"p-ec-put-sleep","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-05-18 19:41:40.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-05-19 23:14:32.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-05-19 23:14:32.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-05-19 22:12:07.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Idd6b767f1a746c72c106199f5d1fada3615b1e97","subject":"proxy: add periodic zero-time sleep during object PUT","status":"MERGED","created":"2023-05-17 12:31:42.000000000","updated":"2023-05-19 23:15:31.000000000","submitted":"2023-05-19 23:14:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":29,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"883367-p-ec-put-sleep","meta_rev_id":"e1ed8d3f87d16bd849057a103acfc692767b0e03","_number":883367,"virtual_id_number":883367,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-19 23:14:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2023-05-18 19:41:40.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2023-05-19 09:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-18 19:34:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-19 22:12:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-17 13:36:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-17 14:26:02.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2023-05-17 16:55:25.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-05-17 21:00:03.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-05-18 18:34:33.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-05-18 19:41:40.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"2a082bca05e438f64efe7fcb2243b03aef4b8fa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-17 12:31:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c46b732649b6a714854e67a711607ad1971f7a4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-17 12:37:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca84310a77e9660822ede4db654526fd6119f923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 13:36:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6502fd0214c04ac5b7ae6d52070559bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a314d5014fd45ef8f57d05ce80b271a : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/a085aa95718d4adcb389b2c755d442ee : SUCCESS in 45m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cd58294e5aa544ac9ebc93bc0e57c850 : SUCCESS in 46m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/809f83c8ab7a41fe81f487d31d796d6a : SUCCESS in 48m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3fc1fe85afde4e81b1e6812cd0f81a80 : SUCCESS in 23m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b59c406a16094a6aa9bd57c072118b7f : SUCCESS in 17m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58a6f2316358456dbfe616a388c19fb5 : SUCCESS in 26m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/690d8a51bd544ec482163619b9f25867 : SUCCESS in 8m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/61a9cf5b75d5444c8a23e5d70447544c : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5e32c475f1d846bd8ef51d3e3d817734 : SUCCESS in 14m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0d75734098b940c6a0625d817a37caa0 : SUCCESS in 16m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7bede11f636a498d8d1191a09ee38d6c : SUCCESS in 15m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/23bd7be586644db89856ca461ed892a2 : SUCCESS in 16m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/919abbf4aa4b474eb0c39bd8d44da12f : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/38b1638e909f4c56b517d6ec1f740e7e : SUCCESS in 15m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/54a3cb6afb954ef8aaca122e25c7a14b : SUCCESS in 17m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6d694771b79c4ac4ad07769d3ed777fa : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cac0206e107a40f19b2891a185a099cc : SUCCESS in 15m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9c152d9713464784a89c58d4bfe45a65 : SUCCESS in 14m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/20521321090e44bcb81331bd00caa013 : SUCCESS in 9m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/000c500d00a946b79723dbd78b5cf48a : SUCCESS in 15m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/36213bc4669d40eaa3b21933f3519163 : SUCCESS in 9m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c8df255a302942a49b21ab5ca60ced7b : SUCCESS in 52m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a124dc87872d49418964d0f52a30d138 : SUCCESS in 57m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8d3ea79a1f774d3f88e07bad205c48ae : SUCCESS in 48m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bfba295f4f3d46c8ab78ad6357175eef : SUCCESS in 50m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c9617911a38486abb7bbaaa822bd530 : SUCCESS in 26m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c801e904465646008287141cd7b0c4ec : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4979026d7ec84b4bbd5a71a889800a52 : SUCCESS in 18m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/365f45524ece419eb592d9c53dea503d : SUCCESS in 39m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c14d426ddf7ee85ed342aa7f7ac2839aa1a6a33a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 14:08:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09e81132b121496a9cf3db4d8c7d10eb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/239bfa37dd4d4bd3bfa74c7e0ba3719b : TIMED_OUT in 1h 30m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d8cafa170dd43d0b4dbd2e38fd034d7 : SUCCESS in 18m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7ab48d6e5854025905d31087c70c5c8 : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3aeea7d0a0ee4422b01fc2a19a0ce999 : SUCCESS in 22m 20s","accounts_in_message":[],"_revision_number":1},{"id":"98b568b71612afd3bd9a06db8285f3c0d58276fc","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-05-17 14:26:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"805b19e9541d50db9ba015a8bd06d27ee23e6392","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-17 16:19:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b9f0ad790ad55bad41b69e1ae8103aacd43a2724","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-17 16:55:25.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8e15262fa9ce0903ff49e4984d3130ee2b431af1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-17 21:00:03.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"46e29ac095f77787152dac83811e35107a31ca04","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-18 15:44:45.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e32433ac79efc201b58960d93318163670d91fe7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-05-18 17:04:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7886bc2e2e00fef587b4a2277fc8a85424b563e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-18 18:02:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"be7a0a8662502fa55787d85e16269e4df184fd39","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-18 18:11:41.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"43f0c48df53370036fc1e2f322a25669f58cd7a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-18 18:18:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"54d61832474c0914196a6d222d00e25ee6513ec8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-18 18:34:33.000000000","message":"Patch Set 3: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"616b638a3eebbf68198f6e32b8d9f765f95b7ff3","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-18 18:52:51.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4e2fb32dcd3671d7e8dc24dbc78e7cf2f7da1190","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 19:17:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d02271b2db48440d88fea1234f5dfab2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31738bac4a4743fc8ae929a8ba70137c : SUCCESS in 16m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9e97870b6e24780a8f7849eb48f26bf : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e14f58d377d432eb9596d88a5d136be : SUCCESS in 16m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb990d133c614462bd77f9bff0d84bb6 : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":3},{"id":"5b974db9a72a8e305e5bc96fc540b05963a3c2e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-18 19:31:16.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"06b11523de9a7f55e4e900f6a2a15b4bf75dc498","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-18 19:34:40.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e8ac70be7fe511cb81d672cbf995a20b857eb76e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-05-18 19:41:40.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"285cd2a69e4449b42b3138df161078bee633f22a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 20:32:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce23b63ae1694578bceb03c3510c24b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/348b97cd2bee40b4be28f9009390ea32 : SUCCESS in 7m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/c024c29bf5f04aa9a28eaa7404b2afbc : SUCCESS in 55m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7e004338fd75438eb27f6608f3897fc0 : SUCCESS in 34m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ccc6f33358784b699058e24a42f60475 : SUCCESS in 44m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07deeb7dd114445489687cbe2f3906c2 : SUCCESS in 13m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cdb48947f9684e239cb9c49a34036c1c : SUCCESS in 23m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e840ca6b2ed54ff29ec80020aa2fb9c4 : SUCCESS in 17m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/53f0038319114baf841f9ce8e208db40 : SUCCESS in 9m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6fc9391e18d34d32aae75242e33b0581 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/147bf0f7700f4977a339e8bfa8736081 : SUCCESS in 13m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3d8316bbbca245c8add4a152fbaf1814 : SUCCESS in 14m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b7daccf352eb4146b4d8db63cf010138 : SUCCESS in 11m 46s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a0f013c3ea2945b2897f7d9e485aa521 : SUCCESS in 13m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b81528cc3db44fca978be6083a62ee07 : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1961db72aa764e588d1267bdcd423613 : SUCCESS in 13m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d6e775f5fe01448b9d57bde683f16249 : SUCCESS in 17m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9f13d79682644cb58c6b254ebca94e21 : SUCCESS in 12m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/123eb400bc7e4328b59ac47bd678f9d1 : SUCCESS in 14m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/10798fafbb9c4fa2a847a529faba3561 : SUCCESS in 15m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/18be31b8643a41d8aeb0188cc65c58ba : SUCCESS in 7m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9c5b4a48c1d842bfafee59d83edc37c9 : SUCCESS in 15m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5348dd8eb2b242a39d7c68bb7db7fd2f : SUCCESS in 6m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e9805849ec58411da774255b48bfdf64 : SUCCESS in 52m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5bcc2557b8a443959c9f150082785be7 : SUCCESS in 55m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/184ace5302ab4ffca28f23dd2973523e : SUCCESS in 34m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0ee42b0a4f6646f5878a8fdb2c917803 : SUCCESS in 51m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22a9b3c5a336412795911ac7f59b2c43 : SUCCESS in 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcf1854f16c049f9a1136668e36785a3 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b2c95168d9144caa76660ab7326a004 : SUCCESS in 16m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/336a2aa335224cbba49f5d1f46ef8b1d : SUCCESS in 56m 08s","accounts_in_message":[],"_revision_number":4},{"id":"80842b4952a3adb794f64da7e17c3cbc130c18f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 20:43:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99b3f0eeb3fd4ba3b335884790206d1d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d333cba716684df7b1375ff2e0bb5794 : SUCCESS in 16m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/73a31f3714654023b961e4f532512c67 : SUCCESS in 1h 09m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e58a958a8c44f348402dbb0dc52fbb1 : SUCCESS in 1h 09m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b101ea2e8564e0890de0286e8b097b2 : SUCCESS in 1h 07m 43s","accounts_in_message":[],"_revision_number":4},{"id":"6fa369346c39f38e0e4f970eef3be9e2c2fd67e0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-19 09:02:43.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2ed681ca09ec4cbf54c0ca9e074565434f8d4ded","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-19 22:12:07.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"648539885650a8c85c65500644cfb8b41069b1f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 22:12:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dbc494c5fe795e699773e640d01c80abc588f14f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 23:14:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ca2d35fa5054d1ababe7e3cc9ae0518\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e95a3c931bc4299875aa9974a20bff9 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/f70bd41ee04641f58bd2d08c110404f3 : SUCCESS in 55m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/78ae427d0d8142a4bfab693590284174 : SUCCESS in 35m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf22ad7ca34a45b1b63bb382681353b4 : SUCCESS in 44m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f195ab8120d6442da0ae572f56935608 : SUCCESS in 7m 38s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/95c127c42f514dad8fb7d1660d2a2c8e : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/79070b4bb246439795ef1674e1c4d050 : SUCCESS in 20m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/99ee6c27e6c54b65b87cc955b2946595 : SUCCESS in 10m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3372baf711d5498d8546e152fd3e85da : SUCCESS in 10m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e09cd6cf7f8249e78867cc36d3846479 : SUCCESS in 21m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/98a2201438524a72a09785230ab03b22 : SUCCESS in 20m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b0b46a52ec994befbd48b0c46ada168b : SUCCESS in 18m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4e65ad4f74874d47b1e146fe742b51f1 : SUCCESS in 18m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9bc088b0fa3b4e43bc47311e920a1c3e : SUCCESS in 14m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c25b49d690a44b1392afab45d8a6cb4c : SUCCESS in 14m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a78309fc901c47ae874288ba5444b98f : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b55e140379384426a0e803f69f397e2b : SUCCESS in 6m 16s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4d626d63df5b4fe393fc088bfbc60831 : SUCCESS in 7m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7f0f1dac9939417783377c1c335f54b6 : SUCCESS in 41m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/eaa927accf3b4a19b523519090a83c7a : SUCCESS in 1h 00m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f1e5f44c532c4a4f8586b52ded688e22 : SUCCESS in 49m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c7ea4da19d8c4757bea5c8b54cf9721f : SUCCESS in 1h 01m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8aba416e2d9c4e66aa38303e672aa248 : SUCCESS in 17m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27bc1307c6db4a7582f689fc8d66b607 : SUCCESS in 4m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8faffd60682d41cf9326c7de5647d943 : SUCCESS in 55m 47s","accounts_in_message":[],"_revision_number":4},{"id":"ab5b0ddf4b6c088371fd4227450b58f04df32f98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 23:14:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e1ed8d3f87d16bd849057a103acfc692767b0e03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 23:15:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5554bbc8b87495eaf54b0979aeddb77\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e42b1e9b645f4f36a0b4fc05d4d618c1 : SUCCESS in 39s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b47d4fbd4ae247e1a3d768ea96288554 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8e8387b991b447e0847c3d68e6a00b2d : SUCCESS in 30s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"aa96cb3dc675415f48d8bb2f8b61a4f394759f53","revisions":{"7ab509811c93558220791588822ae0b345918265":{"kind":"REWORK","_number":1,"created":"2023-05-17 12:31:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/67/883367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/883367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/883367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/883367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/883367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/883367/1"}}},"commit":{"parents":[{"commit":"a0c5ac29a1e4f5dd5ee66ceb3d319774017cadfc","subject":"Merge \"docs: Remove references to out-dated install guides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0c5ac29a1e4f5dd5ee66ceb3d319774017cadfc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-17 11:54:45.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-17 12:30:31.000000000","tz":60},"subject":"proxy: add periodic zero-time sleep during object PUT","message":"proxy: add periodic zero-time sleep during object PUT\n\nPreviously it was possible for an entire object PUT data transfer to\nexecute without the greenthread sleeping and allowing other\ngreenthreads to run. This was more likely with an EC PUT because the\ncomputation of EC fragments might be slower than the rate at which\nthey are drained out of IO send buffers, so IO never blocks. In\nextreme cases this could cause timeouts in other greenthreads to pop.\n\nThis patch adds a periodic zero-time sleep in the object PUT data\ntransfer loop. An existing pattern in the GET path is re-used, and\nextracted to a new PeriodicSleeper helper class.\n\nChange-Id: Idd6b767f1a746c72c106199f5d1fada3615b1e97\nCloses-Bug: #2019955\nRelated-Change: Iae27109f5a3d109ad21ec9a972e39f22150f6dbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ab509811c93558220791588822ae0b345918265"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ab509811c93558220791588822ae0b345918265"}]},"branch":"refs/heads/master"},"3a7a8bd2173673176406d6c04c2503fa67b4c970":{"kind":"REWORK","_number":2,"created":"2023-05-18 18:02:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/67/883367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/883367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/883367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/883367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/883367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/883367/2"}}},"commit":{"parents":[{"commit":"8909604540436e227f21bfcc03dfc2bf2a149551","subject":"Merge \"Memcached: emit memcache timing metrics when exceptions raised\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8909604540436e227f21bfcc03dfc2bf2a149551"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-17 11:54:45.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-18 17:55:32.000000000","tz":60},"subject":"proxy: add periodic zero-time sleep during object PUT","message":"proxy: add periodic zero-time sleep during object PUT\n\nPreviously it was possible for an entire object PUT data transfer to\nexecute without the greenthread sleeping and allowing other\ngreenthreads to run. This was more likely with an EC PUT because the\ncomputation of EC fragments might be slower than the rate at which\nthey are drained out of IO send buffers, so IO never blocks. In\nextreme cases this could cause timeouts in other greenthreads to pop.\n\nThis patch adds a periodic zero-time sleep in the object PUT data\ntransfer loop. An existing pattern in the GET path is re-used, and\nextracted to a new CooperativeIterator helper class.\n\nChange-Id: Idd6b767f1a746c72c106199f5d1fada3615b1e97\nCloses-Bug: #2019955\nRelated-Change: Iae27109f5a3d109ad21ec9a972e39f22150f6dbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a7a8bd2173673176406d6c04c2503fa67b4c970"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a7a8bd2173673176406d6c04c2503fa67b4c970"}]},"branch":"refs/heads/master"},"975bf0e6677300485b33aca5471dcff75b0c2c71":{"kind":"REWORK","_number":3,"created":"2023-05-18 18:18:25.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/67/883367/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/883367/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/883367/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/883367/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/883367/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/883367/3"}}},"commit":{"parents":[{"commit":"8909604540436e227f21bfcc03dfc2bf2a149551","subject":"Merge \"Memcached: emit memcache timing metrics when exceptions raised\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8909604540436e227f21bfcc03dfc2bf2a149551"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-17 11:54:45.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-18 18:18:16.000000000","tz":60},"subject":"proxy: add periodic zero-time sleep during object PUT","message":"proxy: add periodic zero-time sleep during object PUT\n\nPreviously it was possible for an entire object PUT data transfer to\nexecute without the greenthread sleeping and allowing other\ngreenthreads to run. This was more likely with an EC PUT because the\ncomputation of EC fragments might be slower than the rate at which\nthey are drained out of IO send buffers, so IO never blocks. In\nextreme cases this could cause timeouts in other greenthreads to pop.\n\nThis patch adds a periodic zero-time sleep in the object PUT data\ntransfer loop. An existing pattern in the GET path is re-used, and\nextracted to a new CooperativeIterator helper class.\n\nChange-Id: Idd6b767f1a746c72c106199f5d1fada3615b1e97\nCloses-Bug: #2019955\nRelated-Change: Iae27109f5a3d109ad21ec9a972e39f22150f6dbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/975bf0e6677300485b33aca5471dcff75b0c2c71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/975bf0e6677300485b33aca5471dcff75b0c2c71"}]},"branch":"refs/heads/master"},"aa96cb3dc675415f48d8bb2f8b61a4f394759f53":{"kind":"REWORK","_number":4,"created":"2023-05-18 19:31:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/67/883367/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/883367/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/883367/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/883367/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/883367/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/883367/4"}}},"commit":{"parents":[{"commit":"8909604540436e227f21bfcc03dfc2bf2a149551","subject":"Merge \"Memcached: emit memcache timing metrics when exceptions raised\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8909604540436e227f21bfcc03dfc2bf2a149551"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-17 11:54:45.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-18 19:30:58.000000000","tz":-420},"subject":"proxy: add periodic zero-time sleep during object PUT","message":"proxy: add periodic zero-time sleep during object PUT\n\nPreviously it was possible for an entire object PUT data transfer to\nexecute without the greenthread sleeping and allowing other\ngreenthreads to run. This was more likely with an EC PUT because the\ncomputation of EC fragments might be slower than the rate at which\nthey are drained out of IO send buffers, so IO never blocks. In\nextreme cases this could cause timeouts in other greenthreads to pop.\n\nThis patch adds a periodic zero-time sleep in the object PUT data\ntransfer loop. An existing pattern in the GET path is re-used, and\nextracted to a new CooperativeIterator helper class.\n\nChange-Id: Idd6b767f1a746c72c106199f5d1fada3615b1e97\nCloses-Bug: #2019955\nRelated-Change: Iae27109f5a3d109ad21ec9a972e39f22150f6dbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa96cb3dc675415f48d8bb2f8b61a4f394759f53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa96cb3dc675415f48d8bb2f8b61a4f394759f53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
