)]}'
{"id":"openstack%2Fswift~883415","triplet_id":"openstack%2Fswift~master~Ieb35e6ecf87506d8fda101b78f5e841fa9f53fd1","project":"openstack/swift","branch":"master","topic":"s3api-429","hashtags":[],"change_id":"Ieb35e6ecf87506d8fda101b78f5e841fa9f53fd1","subject":"s3api: add option to log \u0027503 Slow Down\u0027 as 429","status":"NEW","created":"2023-05-17 16:04:02.000000000","updated":"2024-10-18 17:08:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"816b635e8b05c4a01eb4db923d5feaf378ff1f8f","_number":883415,"virtual_id_number":883415,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-10-18 17:08:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-05-17 17:08:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-17 17:36:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77b0d8a30f4cb1bf875d5b9710fd63668041515a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-17 16:04:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"008f22081deb2e1d5397659be6d10ab64e58f8a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 17:08:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc81e3075611462b9c87f993061a9c26\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d37e051d887a4879b6d1939977f2b72d : SUCCESS in 15m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0b4cba5087e42b48f656caaadf3ebe3 : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/86ab6463996d4e2c87c51e1d64a94f31 : SUCCESS in 18m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/584e49c8ba884755b8f7cc0c12c2ab4c : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":1},{"id":"eb69a58e0dc52aac861adf69ed37ff061530d27e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 17:36:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9aded2e00c1f4c96abadd90ad20384ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/309ff31b400445b8802020d8a2d98094 : SUCCESS in 11m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/683bb5da4d1f455393a6ceaaeb903469 : SUCCESS in 1h 05m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bb54b41a01f84f76b977c9c92e049cb5 : SUCCESS in 41m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bd344df07e54890ade3a001e57921a3 : SUCCESS in 53m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c52660bd01046efb074cc3c4364e221 : SUCCESS in 18m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/938ca2fa81fe4cd0a62c72dcb8ef656b : SUCCESS in 19m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de23b7941b794ceba1d9f466446c5da8 : SUCCESS in 19m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e9111a69ff414c28be81d326b84c36e6 : SUCCESS in 10m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fcd42888e7ee4fa59a6076b5a000fd8c : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d1174b8c00754d7aa0c57ffc0cf4bee2 : SUCCESS in 17m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fcbb46045d134b9e9bc23ac481f0d197 : SUCCESS in 15m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bf50577560fa41de96a5177b1c0bd351 : SUCCESS in 15m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5bb6e4086dd84d13a0fd5ff907a7b6e2 : SUCCESS in 19m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/625bf814aa5c47db8e5325d0d02cf591 : SUCCESS in 19m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dadc8645baa04c9087e480305a750f7b : SUCCESS in 16m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b895b12328f84573a0cdfb818bdd0b1d : SUCCESS in 16m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/452f89d77c6841fcab8fa8cdc2132de7 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ba23e4e7dcba413a8d235cffd4133b49 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fab8cb40fa614bf59074f4854037ce33 : SUCCESS in 14m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f3578b6ebd174061804a529f7cf5bfaa : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e703443831ec43e4a420a0119c764097 : SUCCESS in 24m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6441e9667ff841a8bdefbe9eaed965e0 : SUCCESS in 8m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0e02055f58184945867670ebe19e40e2 : SUCCESS in 1h 05m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/55aeeafafbaa40fb9bb30b477b354551 : SUCCESS in 1h 09m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e5a3efdbfbb4e399ce1485ad88294f9 : SUCCESS in 57m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d17b80d6426645939bd24d0ef4d34aa7 : SUCCESS in 1h 04m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b518d9a7bad48f8ae51254787f458ec : SUCCESS in 18m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a0b14cb01e144cbb1c861b0baefcc0d : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5ef633822cdf4e2e9ff4a78a090f0210 : SUCCESS in 18m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef0630803f61452697af282c563e3486 : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":1},{"id":"8a034952105fe28b157282e36592d8457fab3773","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-10-30 16:29:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5080bf4b37984291d384d342218e6b6bc0b2cb62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 17:40:53.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0195809bca74b10a606941f0d367d08\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc44a1f8c0044a21abc33604215cdf12 : SUCCESS in 1h 02m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af7ef872546c4558b8b9df9b672e360b : SUCCESS in 1h 09m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/62bb159eb7da41a0b676ba3ffcefaff0 : SUCCESS in 23m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a134aadf38144c19daf9bf109469e2e : FAILURE in 1h 08m 15s","accounts_in_message":[],"_revision_number":2},{"id":"f434e920a4db2935e9152c9d468c31c5dc4f96ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 18:20:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1db9a50cc96a4426a2f1cbb21a2e5176\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63d29cc01c10408aa013969d114b8642 : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/d03715b7e8a14f7eb2064b2c46dd71a9 : SUCCESS in 46m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b308fcd2f9954467ad165f37623e7d5f : SUCCESS in 57m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/42908cdf977647a99d514dcc3a8d490b : SUCCESS in 34m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf47c3dbd00842bfa75a93686c19df84 : SUCCESS in 59m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/924b4bae865a4a9f90545f891c732304 : SUCCESS in 17m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab8e2bf9d7674eb69aabfa1e4028b1d3 : SUCCESS in 17m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71dd62ae41ae4d4cb8f41f8fc1d4313e : SUCCESS in 25m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6c203e36f4e4714bc2c6259ceec9dae : SUCCESS in 8m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b091713420d943aa82d1064198db2aa6 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f1aaaf53d6624ceb83d0fe35c80d0af4 : SUCCESS in 17m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d86067efdfc247919a160aa9a2f10561 : SUCCESS in 15m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0b1e8b7e272848bca15490107f578f22 : SUCCESS in 14m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3d5d566548d045498ebf556d8dddc749 : SUCCESS in 16m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9f7866efc6d0401a946472417a3063f3 : SUCCESS in 15m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/192bfa11c69040f4a06776e12b4ddaf9 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8904cbcb59ff4686990824921f5b57b4 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a6445a37d78345f2814bb873e9fac798 : SUCCESS in 17m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fc299d5dc70e4f36b8330478f424448f : SUCCESS in 23m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8aa859903a5b402494f14048e3067feb : SUCCESS in 16m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0f0394bc7f1c4f9b99d58a6016769e00 : SUCCESS in 17m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1d315fbc73934536afe6c8f08d390482 : SUCCESS in 7m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4e90b8b046a64a13ab0ef8c720543fc6 : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/10e08a9773dc4dc8818c6295133b478c : SUCCESS in 7m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/10826821e23844f39d25b580818ea9a9 : SUCCESS in 48m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/01f67b06a58a4f89a8a2faa3425e6fa1 : SUCCESS in 1h 04m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1ae38c5dda6342ae9e540f57163783ec : SUCCESS in 52m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/50f09ba6c1554b4581237a9b27af7c32 : SUCCESS in 51m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d9527eacffa4653a3365b7c9449ba88 : SUCCESS in 17m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c68bb1805b96424e9b92f1d88710fb34 : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a718a6e4801841c1bec91bc62aa27ded : SUCCESS in 26m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe6c24bfc6e043f49b8919d76e3eb9dd : SUCCESS in 47m 03s","accounts_in_message":[],"_revision_number":2},{"id":"cb917681b75d5e099a05d52285da7e5a3a08e3de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-11-06 17:45:07.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5852e412a57bb4b1c13b9a1e660e334f90b2c3aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 18:56:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da8cd979b04a4d048995d4c8296c1a84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d29b7bcf9e04ddda9341abb2c90fa4d : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/37a2b47024bb49089cec608288badc3f : SUCCESS in 47m 32s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bb2ace429602499ca327735f94e3a97f : SUCCESS in 40m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fe68092c7a7f43208247bdda3ee04438 : SUCCESS in 40m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28ebd9b0b8b94f4585ab79e387590871 : SUCCESS in 1h 05m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e7d6e03f5df4138aff73d64084d505c : SUCCESS in 22m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c1f0b5a698c47f4a3aab64d447c6325 : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83aa2c1cd9b443b3a4f69087cabb9976 : SUCCESS in 16m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f593c931c0514cd499eed3c19eec4d79 : SUCCESS in 9m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3fd966aaa0ec49178803822a18ff1e12 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a4a1031f09654f39ba84d678e73c36d7 : SUCCESS in 17m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/46564c74a3ad45859778e62c1962a0ad : SUCCESS in 16m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ccf18449c3b54603838fa69627a9ef32 : SUCCESS in 17m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9ce1fee26a2943e292c3ba2e6b74ad08 : SUCCESS in 16m 50s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e3cacfb08b4c432e8013804446a2b44e : SUCCESS in 15m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9d3fdca14a6c444d856e334ca9249666 : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/35ff097a8f2d492babe014773bc21b72 : SUCCESS in 22m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4bd81ce668344997ac0cef2d10f41ac0 : SUCCESS in 17m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d9e668f24fdc4544b975851d2cf88547 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f22289ae71e84942b6505919f977171a : SUCCESS in 14m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e103abd657dc49949a0703e107a8974e : SUCCESS in 17m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9458566cd1e4b06869d55a2eb25f626 : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/daa95e77f8ab4a7b9000b22509061e47 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/016537a7454440339dd3a2756b574931 : SUCCESS in 8m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/37d4bed8c3924ed6aec3db0cccb61e19 : SUCCESS in 55m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e7f3bf2ee8c433183f681dd04f33c3b : SUCCESS in 1h 01m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/efe60cdf34b14e06bdcdb35a49a94a92 : SUCCESS in 43m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/044f53cc140540fe8874f738401d7240 : SUCCESS in 49m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc523251daa6425ebdc1677c3442de4c : SUCCESS in 17m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c599dcabfe214330bb33f33863cdd477 : SUCCESS in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d0b7d148509b4948a0dcb188433b8e6d : SUCCESS in 24m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a426a735e9f64b7cbd575628594327c6 : SUCCESS in 36m 35s","accounts_in_message":[],"_revision_number":3},{"id":"36ac10f0ec7761e2998fc0f1f693e88f9d86c3a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 19:02:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ba35966c5904ac3a70993b7662d59e5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c425ada48b234c58a2e0814df4e321e6 : SUCCESS in 59m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e87d0ee4cf14dfeb6f373d9dc4d91bb : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b99395a52a243e8aabcdb166d054561 : SUCCESS in 1h 14m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2cce2abde71c46b5b4e7f4d4ee10f5d8 : SUCCESS in 18m 00s","accounts_in_message":[],"_revision_number":3},{"id":"80db6f52b035a2b7d691f0582498bb18536f497b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-17 23:20:15.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nThe following files contain Git conflicts:\n* swift/common/middleware/s3api/utils.py\n* test/unit/common/middleware/s3api/__init__.py\n* test/unit/common/middleware/s3api/test_obj.py\n* test/unit/common/middleware/s3api/test_utils.py\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b7af2563e6e8ab2325befbd5fd98e48a3d3636bb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-17 23:28:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"19e8ae990983c366d36ec388dbeb6596aa695ae8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 23:38:36.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66fbad9ca06e475c82eb334e169f2796\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4bdd67da580e4a77be10bbef3423673a : FAILURE in 8m 25s","accounts_in_message":[],"_revision_number":5},{"id":"e94cea925c62d950cfeb6320d1b3b4c3f1701005","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 00:32:22.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4b314edc9bb4c6ebdb6647927daf3ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d73bc0833534b3cb06c697323efb39e : SUCCESS in 7m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/4c8c5c5c133a4587af7ee303305d707e : SUCCESS in 37m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0734cedc356947b2a8f54fe3d55746e1 : SUCCESS in 37m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d2fb16ca09a547d7aa8059f74acc7aa5 : SUCCESS in 57m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ea1374634a448929099c960d393e63d : SUCCESS in 39m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc6b8e3436a94062b0f04925ac8e7475 : SUCCESS in 16m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45d28dfc55dd4d29a7c3b48dfb74f4c2 : SUCCESS in 11m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53434816cb804bd5adb61f069a406a44 : SUCCESS in 19m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a33f2d1fd83d4f60b09a34bc315401a6 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/48aae0d539bf4b5e880f46902e7750f1 : FAILURE in 3m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9dd1a7b164cb46f581c3bfaeb419e752 : FAILURE in 4m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f9f1b224bbac4741a0a4e237d89cd01c : FAILURE in 4m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2d02dfd1655f44dfae4d58f6b891a43c : FAILURE in 3m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6a7dbd7d8c8a451fa50be0b9076d8013 : FAILURE in 8m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d445d47712b142d5b04849b09d912de2 : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b290401c98284317919de6705388d578 : SUCCESS in 14m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6ca061fb75ec4a38a1df9e2688f0585a : SUCCESS in 15m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bba92df80a28410daeea97c39a72d3ba : SUCCESS in 12m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ed60741f93e847afa22f8c13d391eb0f : SUCCESS in 14m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/12d5299ef3224eb6a6288ad8967a75fd : SUCCESS in 13m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/30da408a2eee427c9bccfca02d8432c2 : SUCCESS in 5m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/625aa701162e45f58b6db9dc9ef809e4 : SUCCESS in 13m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f40a638a3cf547c2bda64a8442dc7825 : SUCCESS in 5m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/217ce142a50c480dbaad89fb14c981c6 : SUCCESS in 39m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1ec1cd4ec7ae4518afc34138745f0c2d : SUCCESS in 32m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c54e6bc333e2477cab978d389bdc8463 : SUCCESS in 34m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50e53a686d974a2aaf344c69445c1f33 : FAILURE in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6963717b0ed4d05b318bfb9d0cfa637 : FAILURE in 3m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7a1c8a4b4d9410e995daed79983b140 : SUCCESS in 15m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/220666409b4347dbbbf28c32a342301c : SUCCESS in 37m 37s","accounts_in_message":[],"_revision_number":5},{"id":"9cc11d4756360b548ac9d8cc1e5c6a20a64d072f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-18 15:45:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3ac7d0ea344ad2a39580f2697c89a2e19294aa04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 16:00:47.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79a55a405ec146e18d111cf339ade050\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9c384eb95d3e448f93b8e8fce515c977 : FAILURE in 13m 23s","accounts_in_message":[],"_revision_number":6},{"id":"816b635e8b05c4a01eb4db923d5feaf378ff1f8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 17:08:24.000000000","message":"Patch Set 6: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81f4e4fbea3943f2b0158dd563cc71d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46a5f0db76d749caae36df03771eb70b : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/b91057326d9742298c2802c43d6277ca : SUCCESS in 1h 15m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a449f780cc64bd783343d96d2df03e7 : SUCCESS in 1h 05m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7347ebed23464791864f7883c2d031c2 : SUCCESS in 50m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb2c8acf08c449fa94997a6f8e386c4e : SUCCESS in 42m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f1292cbcbcb4758bdb9d4a5faaba03c : SUCCESS in 11m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5983eb4b58514de395dcc81c12057baf : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c580142a74854b0bab5dd35f780b9381 : SUCCESS in 18m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8bb6d4fbc3df4a6284f7b4a40bc3e5bb : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/02a78eb6d2be46f0a0ad375521f4f169 : FAILURE in 6m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f55e5c89f406444f901e84e2ab6bf7ea : FAILURE in 4m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3437903f3a524309968ec22984670c06 : FAILURE in 5m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5f15dc3c124545a28dd919a875b05e99 : FAILURE in 4m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9c9af3660bcf41988dc73db12e38aaf8 : FAILURE in 7m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f798f297840448ce9b281fca960e91d1 : SUCCESS in 10m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4d87d9dd85ab41c0a61f701c0f7ce103 : SUCCESS in 10m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6973df22df9446ad8378369bca4f49e9 : SUCCESS in 16m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6db2aaf43f05483caf8f30a976aaee70 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c2d276ecde434b1698dc15b64d6dd5b9 : SUCCESS in 14m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/121c0523cd6645da8d1fe077346ba67b : SUCCESS in 13m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e06a81416b464542a49b5d08570185bb : SUCCESS in 7m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e7905cae3374426a88eb9048caa435ad : SUCCESS in 9m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d3ee7034369a47b7961c2ae2c123e3b2 : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/423cfdc968e24360ae72873909a573f3 : SUCCESS in 49m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed5e1fe10bf847a0a82a375143035a44 : SUCCESS in 50m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/02805cc2d2d54e7782a2dc86159b69bb : SUCCESS in 58m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96750aca6c0142e4ba6a63bb32129f30 : FAILURE in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f71b75f4b6254a15bb938b28a6040417 : FAILURE in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/128094f9d8d141d582c18a11ee61b8e5 : SUCCESS in 12m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cbd1c1a84b946489ccf4703d42f908f : SUCCESS in 45m 08s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"bda9e95781a1a6ef394e4698c6e09268e60af60f","revisions":{"09b7b42b5e71350bd02a4689412fdd7584492dda":{"kind":"REWORK","_number":1,"created":"2023-05-17 16:04:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/15/883415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/883415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/883415/1"}}},"commit":{"parents":[{"commit":"7b46fe05d8e42a5c8ad477c809ba3e488a368ff9","subject":"Support swift.proxy_logging_status in request env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b46fe05d8e42a5c8ad477c809ba3e488a368ff9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-17 15:29:33.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-17 15:56:11.000000000","tz":60},"subject":"s3api: add option to log \u0027503 Slow Down\u0027 as 429","message":"s3api: add option to log \u0027503 Slow Down\u0027 as 429\n\nAdd \u0027log_ratelimit_as_client_error\u0027 option that will force\nproxy_logging middleware to log rate limited requests with a 429\nstatus int even when a 503 is returned to the client. This can help\ndifferentiate between rate limited requests and server errors when\nlogs only report the status int.\n\nThe new option is ignored if ratelimit_as_client_error is True, when\na 429 status int is already logged.\n\nRelated-Change: I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6\nChange-Id: Ieb35e6ecf87506d8fda101b78f5e841fa9f53fd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09b7b42b5e71350bd02a4689412fdd7584492dda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09b7b42b5e71350bd02a4689412fdd7584492dda"}]},"branch":"refs/heads/master"},"49344265679da21d88b68f839933608ea5c5a64c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-30 16:29:44.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/15/883415/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/883415/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/883415/2"}}},"commit":{"parents":[{"commit":"8a3c8a33037aa1deabbf064516bfe694872ec8c8","subject":"Support swift.proxy_logging_status in request env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a3c8a33037aa1deabbf064516bfe694872ec8c8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-17 15:29:33.000000000","tz":60},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-10-30 16:29:44.000000000","tz":0},"subject":"s3api: add option to log \u0027503 Slow Down\u0027 as 429","message":"s3api: add option to log \u0027503 Slow Down\u0027 as 429\n\nAdd \u0027log_ratelimit_as_client_error\u0027 option that will force\nproxy_logging middleware to log rate limited requests with a 429\nstatus int even when a 503 is returned to the client. This can help\ndifferentiate between rate limited requests and server errors when\nlogs only report the status int.\n\nThe new option is ignored if ratelimit_as_client_error is True, when\na 429 status int is already logged.\n\nRelated-Change: I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6\nChange-Id: Ieb35e6ecf87506d8fda101b78f5e841fa9f53fd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49344265679da21d88b68f839933608ea5c5a64c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49344265679da21d88b68f839933608ea5c5a64c"}]},"branch":"refs/heads/master","description":"Rebase"},"da762ab20f4d37627ae226b19c5978047925590d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-06 17:45:07.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/15/883415/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/883415/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/883415/3"}}},"commit":{"parents":[{"commit":"3268c5bb12c85955f5b1bd6b5d2e63d6c6dd2cd0","subject":"Support swift.proxy_logging_status in request env","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3268c5bb12c85955f5b1bd6b5d2e63d6c6dd2cd0"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-17 15:29:33.000000000","tz":60},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-11-06 17:45:07.000000000","tz":0},"subject":"s3api: add option to log \u0027503 Slow Down\u0027 as 429","message":"s3api: add option to log \u0027503 Slow Down\u0027 as 429\n\nAdd \u0027log_ratelimit_as_client_error\u0027 option that will force\nproxy_logging middleware to log rate limited requests with a 429\nstatus int even when a 503 is returned to the client. This can help\ndifferentiate between rate limited requests and server errors when\nlogs only report the status int.\n\nThe new option is ignored if ratelimit_as_client_error is True, when\na 429 status int is already logged.\n\nRelated-Change: I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6\nChange-Id: Ieb35e6ecf87506d8fda101b78f5e841fa9f53fd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da762ab20f4d37627ae226b19c5978047925590d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da762ab20f4d37627ae226b19c5978047925590d"}]},"branch":"refs/heads/master","description":"Rebase"},"fc7f7115defd59a54df31a451abf82ad90baca91":{"kind":"REWORK","_number":4,"created":"2024-10-17 23:20:15.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/15/883415/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/883415/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/883415/4"}}},"commit":{"parents":[{"commit":"8ded39bccd9192b570e24cf039a3a4c8e3ccc216","subject":"Merge \"logging: Add UPDATE to valid http methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ded39bccd9192b570e24cf039a3a4c8e3ccc216"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-17 15:29:33.000000000","tz":60},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2024-10-17 23:20:15.000000000","tz":0},"subject":"s3api: add option to log \u0027503 Slow Down\u0027 as 429","message":"s3api: add option to log \u0027503 Slow Down\u0027 as 429\n\nAdd \u0027log_ratelimit_as_client_error\u0027 option that will force\nproxy_logging middleware to log rate limited requests with a 429\nstatus int even when a 503 is returned to the client. This can help\ndifferentiate between rate limited requests and server errors when\nlogs only report the status int.\n\nThe new option is ignored if ratelimit_as_client_error is True, when\na 429 status int is already logged.\n\nRelated-Change: I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6\nChange-Id: Ieb35e6ecf87506d8fda101b78f5e841fa9f53fd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc7f7115defd59a54df31a451abf82ad90baca91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc7f7115defd59a54df31a451abf82ad90baca91"}]},"branch":"refs/heads/master","description":"Rebase"},"19f348e73d2b3c2e2be4c76d84bf13c86618e84d":{"kind":"REWORK","_number":5,"created":"2024-10-17 23:28:43.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/15/883415/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/883415/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/883415/5"}}},"commit":{"parents":[{"commit":"8ded39bccd9192b570e24cf039a3a4c8e3ccc216","subject":"Merge \"logging: Add UPDATE to valid http methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ded39bccd9192b570e24cf039a3a4c8e3ccc216"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-17 15:29:33.000000000","tz":60},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-17 23:28:36.000000000","tz":-420},"subject":"s3api: add option to log \u0027503 Slow Down\u0027 as 429","message":"s3api: add option to log \u0027503 Slow Down\u0027 as 429\n\nAdd \u0027log_ratelimit_as_client_error\u0027 option that will force\nproxy_logging middleware to log rate limited requests with a 429\nstatus int even when a 503 is returned to the client. This can help\ndifferentiate between rate limited requests and server errors when\nlogs only report the status int.\n\nThe new option is ignored if ratelimit_as_client_error is True, when\na 429 status int is already logged.\n\nRelated-Change: I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6\nChange-Id: Ieb35e6ecf87506d8fda101b78f5e841fa9f53fd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19f348e73d2b3c2e2be4c76d84bf13c86618e84d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/19f348e73d2b3c2e2be4c76d84bf13c86618e84d"}]},"branch":"refs/heads/master"},"bda9e95781a1a6ef394e4698c6e09268e60af60f":{"kind":"REWORK","_number":6,"created":"2024-10-18 15:45:20.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/15/883415/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/883415/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/883415/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/883415/6"}}},"commit":{"parents":[{"commit":"8ded39bccd9192b570e24cf039a3a4c8e3ccc216","subject":"Merge \"logging: Add UPDATE to valid http methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ded39bccd9192b570e24cf039a3a4c8e3ccc216"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-17 15:29:33.000000000","tz":60},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-18 15:44:54.000000000","tz":-420},"subject":"s3api: add option to log \u0027503 Slow Down\u0027 as 429","message":"s3api: add option to log \u0027503 Slow Down\u0027 as 429\n\nAdd \u0027log_ratelimit_as_client_error\u0027 option that will force\nproxy_logging middleware to log rate limited requests with a 429\nstatus int even when a 503 is returned to the client. This can help\ndifferentiate between rate limited requests and server errors when\nlogs only report the status int.\n\nThe new option is ignored if ratelimit_as_client_error is True, when\na 429 status int is already logged.\n\nRelated-Change: I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6\nChange-Id: Ieb35e6ecf87506d8fda101b78f5e841fa9f53fd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bda9e95781a1a6ef394e4698c6e09268e60af60f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bda9e95781a1a6ef394e4698c6e09268e60af60f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
