)]}'
{"id":"openstack%2Fswift~883455","triplet_id":"openstack%2Fswift~master~I6b60e1a6225c5b83f5108ef4e2a236d322861cc6","project":"openstack/swift","branch":"master","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-05-18 00:06:06.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6b60e1a6225c5b83f5108ef4e2a236d322861cc6","subject":"probe tests: Skip versioning tests if not enabled","status":"NEW","created":"2023-05-17 23:03:16.000000000","updated":"2023-05-18 20:24:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"563c18ce08bf9d9c461e7149f4e3d1c1a9b8df44","_number":883455,"virtual_id_number":883455,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-18 20:24:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-18 00:02:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-18 00:06:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7cea19049d7c3d14a5f808b971d81d162d148d8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-17 23:03:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dabf597ccc26e5b794ce30d60d1bfa0e7de0439b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 00:02:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd17a24e60c04a95b05be6bc63fd8bb3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa89da2c79684c2dbbf2d8fb41474a00 : SUCCESS in 56m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/13c44405dd4c43d9994a5186d48145bb : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a7e61898e7c46c4ae728c8be4aa491f : SUCCESS in 52m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8f7350d4ab542c4b4e0fb7682ab31c3 : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3728990de4c1013112def9ed1a4fb8bea16112e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 00:06:06.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/822a18116a1d4e138c5f517dca34b8d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8933ad4d88af4aa89446704da398b43e : SUCCESS in 9m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c637516afde34d48a4867a82512c0442 : FAILURE in 53m 45s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d152436485274c5a9a84a12f1074690a : FAILURE in 58m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/950440a2568d45809c95d5b4c318a7a1 : FAILURE in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e129aacbf01920b5f6118bcf56caa4289c3c5846","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-18 19:12:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0c0a39b811e3ec98fa5cb4755ba7e5eaa1427227","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 19:33:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/107bab7a78104dd5a554ce38ca95bc31\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2658545b06be40419dfe93781d876dd2 : SUCCESS in 14m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b3cbbe368e14654b71894f5ba2fb962 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/19de468dad2a44ca91fd8468bc83f4ca : SUCCESS in 18m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f42d4153ba4041228502512fdd77d49b : SUCCESS in 19m 19s","accounts_in_message":[],"_revision_number":2},{"id":"563c18ce08bf9d9c461e7149f4e3d1c1a9b8df44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 20:24:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91083b229a7249818b95b38c00e7e9dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e934012c9359449d878582d3af1ca32b : SUCCESS in 8m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/563a91bfb535445a831455e025173528 : SUCCESS in 54m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/77305aa7abab4e429642bd3b48d313fb : SUCCESS in 1h 04m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a1b545b064946529c4a5caea7e516b4 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1e4bd96e247fc65f2ab9042847ebef90fe401c5f","revisions":{"c22fa0696c58077cc83837b16f128fd7126cd554":{"kind":"REWORK","_number":1,"created":"2023-05-17 23:03:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/883455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/883455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/883455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/883455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/883455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/883455/1"}}},"commit":{"parents":[{"commit":"a0c5ac29a1e4f5dd5ee66ceb3d319774017cadfc","subject":"Merge \"docs: Remove references to out-dated install guides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0c5ac29a1e4f5dd5ee66ceb3d319774017cadfc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-17 23:02:42.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-17 23:02:42.000000000","tz":-420},"subject":"probe tests: Skip versioning tests if not enabled","message":"probe tests: Skip versioning tests if not enabled\n\nChange-Id: I6b60e1a6225c5b83f5108ef4e2a236d322861cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c22fa0696c58077cc83837b16f128fd7126cd554"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c22fa0696c58077cc83837b16f128fd7126cd554"}]},"branch":"refs/heads/master"},"1e4bd96e247fc65f2ab9042847ebef90fe401c5f":{"kind":"REWORK","_number":2,"created":"2023-05-18 19:12:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/55/883455/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/883455/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/883455/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/883455/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/883455/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/883455/2"}}},"commit":{"parents":[{"commit":"a0c5ac29a1e4f5dd5ee66ceb3d319774017cadfc","subject":"Merge \"docs: Remove references to out-dated install guides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0c5ac29a1e4f5dd5ee66ceb3d319774017cadfc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-17 23:02:42.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-18 19:12:18.000000000","tz":-420},"subject":"probe tests: Skip versioning tests if not enabled","message":"probe tests: Skip versioning tests if not enabled\n\nChange-Id: I6b60e1a6225c5b83f5108ef4e2a236d322861cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e4bd96e247fc65f2ab9042847ebef90fe401c5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e4bd96e247fc65f2ab9042847ebef90fe401c5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
