)]}'
{"id":"openstack%2Fswift~883636","triplet_id":"openstack%2Fswift~master~I6501284e8b749be4efd4e3257945b56933c619d2","project":"openstack/swift","branch":"master","topic":"sharder.sharding.completed_as_float","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-06-27 18:19:16.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6501284e8b749be4efd4e3257945b56933c619d2","subject":"Fix sharder.sharding.completed timing metric","status":"ABANDONED","created":"2023-05-19 05:53:04.000000000","updated":"2023-06-27 18:19:16.000000000","total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"e9679c259b3071d6959e7077c6b32e6053911159","_number":883636,"virtual_id_number":883636,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-19 07:04:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2023-06-27 13:23:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-19 06:50:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-19 07:04:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-19 08:01:57.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2023-05-19 08:43:55.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"c7242b7d026d81c52e11bc95634e0ed8e6720f71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-19 05:53:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4df924244fc133d6fca95874149de299eb88e8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 06:50:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0450d73d92a4c8c9a09e9b352cb694b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f325e27615664f21996b0c90af5d6142 : SUCCESS in 16m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57690d9dd57d414c972941391348a573 : SUCCESS in 55m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a9c9da125b740d5bb3dae54d8f0a3f9 : SUCCESS in 21m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0d890806cbd4509bdc76026e15e9e04 : SUCCESS in 52m 54s","accounts_in_message":[],"_revision_number":1},{"id":"40fd1939ad921c866f6f14f92f4665fc2ba0c307","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 07:04:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fdae7dcca5b402194e91a3ea3c4e18e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eeef36601c24b85b7318a7837a9ca00 : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/d69c51aac25c4b61bebd8e7579ccb6f5 : SUCCESS in 1h 03m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b2837352a15c422e948b9633f397442f : SUCCESS in 49m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cddeaa3cb84e4c67a1ae4698b4592c04 : SUCCESS in 57m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/459de56a9ed44e62bf8b43738816a5de : SUCCESS in 23m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3726ef9cbf04d61957e975906ae7fc8 : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9409df8553da4b7d90fe41d561f158ad : SUCCESS in 27m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e413260d76744f22a1a3a3c945172ecd : SUCCESS in 7m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/62f1273453e9427180ea9c59ba1baf20 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/192abbff4c6e412b80a96dc37e33f2c6 : SUCCESS in 17m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e50bd670cad34535a71276cc95135d18 : SUCCESS in 24m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/556a95f2a42b409da530dcccff575d89 : SUCCESS in 24m 36s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/91e09c3e3c7e48939ddd212333a6ff9d : SUCCESS in 25m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1843a8728d484291a410e14e4f237aac : SUCCESS in 22m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/31c495df5ff14f82ab4aa66e47afb85e : SUCCESS in 22m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6aa5b472023b44e991e4f8c623e1bb5a : SUCCESS in 19m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e1a71eb3ca7e4d97b663bf7a0c89fdd0 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/db25aca4fb4a4b5a84638700eafa7569 : SUCCESS in 20m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7ae5e860c8dc4776ad95632886f7c883 : SUCCESS in 15m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fcf61b0622084d89ab7eba6a85522fbc : SUCCESS in 6m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7d6d832432fb4701ae2677f225f6c0da : SUCCESS in 16m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1de9c989c7fd4b8a9c15403396d7ea81 : SUCCESS in 7m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/48fc3eed74dd4aac938660db56118edb : SUCCESS in 58m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e804eb566fab421aa0c6f8a8c3f3ea16 : SUCCESS in 1h 06m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/522085d2295e4e4da1628f04eafc376e : SUCCESS in 36m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/87cb5579bf7d4353922206479ee7ac3d : SUCCESS in 52m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70d729cdad4940dd9f907c0775e26c0b : SUCCESS in 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da17e3c9cd104001ab3094fe02419f68 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7337262b6b09408e99721a9440513fad : SUCCESS in 17m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8fc60c8b44984d5e82e62dea4046ca92 : SUCCESS in 41m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b18e5fdfc3f16504951915d5966b570f7b1ee0ec","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-19 08:01:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7bd9b73db318208d297a501275b9217393ad198b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-19 08:43:55.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ece8cedf2b44b944f1355e9afced18571186ad89","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-06-19 05:00:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f11c9f9f71349b70c93e788f0f1c533928f5879","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-27 10:42:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65720e23ddc32bea62943b7dc743f78c2f654281","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-27 13:23:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9679c259b3071d6959e7077c6b32e6053911159","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-27 18:19:16.000000000","message":"Abandoned\n\nI think this one\u0027s safe to just abandon -- if we really want, we can always bring it back.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"909052c881efac96c487ca7d3f11e126cf86c10c","revisions":{"909052c881efac96c487ca7d3f11e126cf86c10c":{"kind":"REWORK","_number":1,"created":"2023-05-19 05:53:04.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/36/883636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/883636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/883636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/883636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/883636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/883636/1"}}},"commit":{"parents":[{"commit":"eda3b33a65071bd5699107afe34a06e03052bf08","subject":"Merge \"tests: Make dark data probe tests pass with sync_method \u003d ssync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eda3b33a65071bd5699107afe34a06e03052bf08"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-05-19 05:35:27.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-05-19 05:52:08.000000000","tz":600},"subject":"Fix sharder.sharding.completed timing metric","message":"Fix sharder.sharding.completed timing metric\n\nWhen a sharder finishes sharding it using the epoch Timestamp to\ncalculate how long it took to complete sharding. But the\nlogger.timing_since method takes a float (time.time()) not a timestamp.\nThis causes tracebacks with:\n\n  return self.timing(metric, (time.time() - orig_time) * 1000,#012\n  TypeError: unsupported operand type(s) for -: \u0027float\u0027 and \u0027Timestamp\u0027\n\nIn a python shell we can see this:\n\n  from swift.common.utils import Timestamp\n  ts \u003d Timestamp.now()\n  ts\n  1684474287.80183_0000000000000000\n  time.time() - ts\n  Traceback (most recent call last):\n    File \"\u003cstring\u003e\", line 1, in \u003cmodule\u003e\n    TypeError: unsupported operand type(s) for -: \u0027float\u0027 and \u0027Timestamp\u0027\n  time.time() - float(ts)\n  19.668439626693726\n\nThis patch makes the timing_since method support support Timestamp\nobjects.\n\nChange-Id: I6501284e8b749be4efd4e3257945b56933c619d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/909052c881efac96c487ca7d3f11e126cf86c10c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/909052c881efac96c487ca7d3f11e126cf86c10c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
