)]}'
{"id":"openstack%2Fswift~883638","triplet_id":"openstack%2Fswift~master~I126d51dbde3bb20665b641f44cc59afbe46e798c","project":"openstack/swift","branch":"master","topic":"container_info_cache","hashtags":[],"change_id":"I126d51dbde3bb20665b641f44cc59afbe46e798c","subject":"Proxy: container HEADs to use cache before reaching backend.","status":"NEW","created":"2023-05-19 06:13:48.000000000","updated":"2023-05-19 08:11:59.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":2,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"b0c98153ab6340012905c12f6e5cfc0655aa69ec","_number":883638,"virtual_id_number":883638,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-19 08:11:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-05-19 07:46:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-19 08:11:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb408df68975cc656d7a626c12c8578a270a0ff5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-19 06:13:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"256941a0c5e758f8f2bbeb451fe20a2def87b740","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-19 06:14:11.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"23893411c34b9c0168a7447b373bc38b54af53f0","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-19 06:15:35.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"ee243a692a7af61d058538d42169dd019a3648cd","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-19 06:15:38.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"6138cf66e664bbb2f6e4b95324e10964807891a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 07:46:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e33488e91dd42ffbdadc54b0d686d7e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/699fa43711aa44f2b11e125194205cee : FAILURE in 14m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41939994fb044f69866514e0dd122245 : FAILURE in 1h 10m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d6e6ced846a4199b2e1114eb14c363f : TIMED_OUT in 1h 30m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a416a75326f2433f888b82e3bf09f713 : TIMED_OUT in 1h 30m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b0c98153ab6340012905c12f6e5cfc0655aa69ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 08:11:59.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5ee10920dca4459a84b091361809bd4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd256d9708374bceb43372e954dea11b : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/198e9bb7258546f6877bbec31ac6b49c : FAILURE in 39m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1f420b59b5d5490287f8f76313ba0647 : FAILURE in 40m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f5d9822cb3c40cb90b5e0190570d386 : FAILURE in 42m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/69ae054a42f84d7dbfee90db091af6e0 : TIMED_OUT in 1h 31m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d052848b3254f86b5a25618b2a5cb63 : TIMED_OUT in 1h 30m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f35a89e2df044bd9a23bb0bcc720a83 : TIMED_OUT in 1h 30m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5fa671f7b382461a9f3e5b8da26307f2 : SUCCESS in 9m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a874648809c94902b2004d19ae480419 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e6f574c1c8d4100aab9403248b98045 : FAILURE in 16m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a7ddcd128e254d89ae87dad05b2c629d : FAILURE in 17m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eb631a60fc324b2395e039326578c629 : FAILURE in 15m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5dff2274a6fb471f87dfb31f02c42231 : FAILURE in 15m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/77b946286065407694c88a0bafd7af67 : TIMED_OUT in 1h 30m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eee7ce4afe5e4fa89f3e2d154f1f7c0f : TIMED_OUT in 1h 30m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8d24a3c946e04de2ab80815bffc8a219 : TIMED_OUT in 1h 30m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0dddf17ba1fb4d478172345c723dabac : TIMED_OUT in 1h 30m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4ba21f7077b74fe6b5cdbfe37042f5f2 : TIMED_OUT in 1h 30m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8d691ed6c8714249a6182993fa4ff2f1 : TIMED_OUT in 1h 30m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ccc0fce969a84e7f98016046d93f3721 : SUCCESS in 6m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e92e1f55ab86444b91bfdc771ee1af46 : FAILURE in 16m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/928af8c0bd084dbdb344c50ec25f521a : FAILURE in 11m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ba141f665f6742028e201d276aaa7e27 : FAILURE in 55m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/21a8c7accfdb4b2d8da8c0dd16d22e70 : FAILURE in 1h 04m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/235fc274ee6c47c8a5fc36900c01339d : TIMED_OUT in 1h 31m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2a694977281d45bf86c83cfa0d95c635 : TIMED_OUT in 1h 32m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ec26b31ae4b480cbe29fa8e73fee503 : FAILURE in 16m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff006df429784fcd974500aeb233fec4 : FAILURE in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f71ddd7030c42db9ca8045b1eb0663f : SUCCESS in 15m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/780d54a136174c65832e986234a29b76 : FAILURE in 38m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83a2597ee97a9a3a5bd4ac6702aafb6a37cb6d21","revisions":{"83a2597ee97a9a3a5bd4ac6702aafb6a37cb6d21":{"kind":"REWORK","_number":1,"created":"2023-05-19 06:13:48.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/38/883638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/883638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/883638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/883638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/883638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/883638/1"}}},"commit":{"parents":[{"commit":"eda3b33a65071bd5699107afe34a06e03052bf08","subject":"Merge \"tests: Make dark data probe tests pass with sync_method \u003d ssync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eda3b33a65071bd5699107afe34a06e03052bf08"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-05-19 06:12:18.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-05-19 06:12:18.000000000","tz":-420},"subject":"Proxy: container HEADs to use cache before reaching backend.","message":"Proxy: container HEADs to use cache before reaching backend.\n\nChange-Id: I126d51dbde3bb20665b641f44cc59afbe46e798c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83a2597ee97a9a3a5bd4ac6702aafb6a37cb6d21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/83a2597ee97a9a3a5bd4ac6702aafb6a37cb6d21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
