)]}'
{"id":"openstack%2Fswift~883700","triplet_id":"openstack%2Fswift~master~I17bc7d7489e8947f53e77b1cd6eb9f792f750481","project":"openstack/swift","branch":"master","topic":"fast-shard-usage","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-06-13 23:09:24.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-06-13 23:09:24.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-06-13 21:13:52.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I17bc7d7489e8947f53e77b1cd6eb9f792f750481","subject":"Use the db for get_shard_usage","status":"MERGED","created":"2023-05-19 19:17:13.000000000","updated":"2023-06-13 23:10:27.000000000","submitted":"2023-06-13 23:09:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883700-fast-shard-usage","meta_rev_id":"be81403d3fa5e0fba043eaadeb49d18219c39031","_number":883700,"virtual_id_number":883700,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-13 23:09:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2023-05-22 18:42:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-19 20:46:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-13 21:13:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-19 19:26:24.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2023-05-19 20:23:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-19 20:34:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-22 18:42:18.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"a5669ca8de9ec723cd59d34d59bbb3c454f155cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-19 19:17:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c6aa78df814eed300a35b3743050f5f0884c0fe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-19 19:26:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11ea39200d6dc513725835439e89ab0355f35a71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-19 19:28:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2b41f8eecf8e3e044850ce266d5f117fb4b09f84","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-19 19:29:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff9d1d50489909ecff062b6a514fe25294962a03","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-19 19:44:37.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"87ddebf62e7b89f4434efafec4fc93b2d99005aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 20:23:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/930194b9430943ddac9c44488bcb8ac7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4136f34469a9445aa40d898a477fded7 : SUCCESS in 15m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b02f9e7dc4464e3cb2bb6bfb8e8066b7 : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/97063792acf84ff18e877927639903cd : SUCCESS in 52m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab226f48f4934bdd9baf0e93c0ed3763 : SUCCESS in 22m 08s","accounts_in_message":[],"_revision_number":2},{"id":"de431dff0833ab5e4df9029fae95bd212cd700b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 20:34:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22956ff8860844e0b85512aec7b2913a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7e1b26bd7214bc3b36304c21ba6e176 : SUCCESS in 7m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/d3ac756bda60450abe0f609f423e2606 : SUCCESS in 54m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0182d8a79b884d73910285a0cc939108 : SUCCESS in 34m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e7c50a85b954c4d866c39367c92e6dc : SUCCESS in 51m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d35de1840844413bcb68c20918d1679 : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2657c9d2ddc14e649872a1ad64509865 : SUCCESS in 21m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63ea2a1fac374d96b71e844687aa012b : SUCCESS in 21m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eb94a39e6c564a4181db3bffe88e5160 : SUCCESS in 7m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4722217ed82140dcb323b9b82353f8c2 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4151bdd0560b483aace056e2eab55010 : SUCCESS in 20m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b0d7aee2ef894e4580db10a69a2f5e18 : SUCCESS in 18m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f669cb36c7c4e2aadfffd8d23365876 : SUCCESS in 14m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/13fa9d68190a49f489c889f8bf65a098 : SUCCESS in 22m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de76e1b8b95940c8a0d86134503c6e8b : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eaa5970e7d444fe89246be7942cb7aab : SUCCESS in 17m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/589792991e5247dea042a841d282c5c8 : SUCCESS in 14m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c80d1fea87274e70803ccf77bba62ec0 : SUCCESS in 14m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/00955f9d4cd840c6906541bc81e05811 : SUCCESS in 14m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/57d9aea5a1e644228e60137dec41ba9c : SUCCESS in 14m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9b06d3f63c89438a95c2fefce7c79c36 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f378d1a1da904bdca2d0689f27a91f4b : SUCCESS in 17m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9633fd1c040e45a4a80c0dfc9346e310 : SUCCESS in 6m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/86c43a336eed45c4a2cc2601355f7fe9 : SUCCESS in 54m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/48ef70de113942ccaee2c5caa312da72 : SUCCESS in 52m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ec86248279ae44eebd407ba02b7aae3b : SUCCESS in 48m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/afff0d47945741d6ac39949b136fffda : SUCCESS in 51m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bce081fe72042dabbb964298503cabe : SUCCESS in 15m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98b08c17e235423c8d50cc5be65039d1 : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/17f09368576d47c589a6e79bcf138772 : SUCCESS in 20m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c34fbcc4ee714b479141fc17c56cc6ef : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":2},{"id":"a2916dc688c7d8bed4af98c94abe417bf31f3203","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-19 20:46:22.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8e560c15b0ec92cd6eeb2f9d952dd9d0f3ce941","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-22 18:42:18.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b0cc93c6fa770f5ab48eed98730e3e91260c4559","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-13 21:13:52.000000000","message":"Patch Set 2: Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"bef8b10e314ce39c930aed0ecde78baa5f1b05d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 21:14:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"57c77fabb943cf87a9dd43f973b33fc33373821b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 23:09:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e2f6141968e4ded9198f7a69f252c96\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5059039fddcd44d28c7ad9c5136620d9 : SUCCESS in 7m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/3ef0757456f044f5982d774c7153e075 : SUCCESS in 47m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a75833801ad84787b4149036b01edf2a : SUCCESS in 40m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b4ba079bcae4732a80e7afa4e5dfd9f : SUCCESS in 1h 00m 33s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f0fbc9e08a154a9fa78c99815ec81863 : SUCCESS in 7m 45s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/92b18e9a7b45491b96fd2a6f516741f9 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/993577fb6b8c4670bcb92316329a0a59 : SUCCESS in 18m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bbfe5a0ef52444418f6be5a37dbe474a : SUCCESS in 15m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a26892d2a9aa4e5e85bb0030b9dc0ce8 : SUCCESS in 15m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/576a8a95448542eeba69ff4b0a64d78f : SUCCESS in 15m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/702fb8119d0f4484bac45b0d7a8c3c04 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2cd16bf885164818a5021cd33d2ab6ee : SUCCESS in 21m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a772865333c94bffbe3a06afaaddbfd0 : SUCCESS in 16m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/81a8ba21decb4f8d84b1374137f34cbf : SUCCESS in 14m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8ec0db2d8fe3410b82bdaae7d7da6e78 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/717cfd99bc0343b3bb5f774d8935193b : SUCCESS in 15m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b9ab23fbfd4e4015bc751819f0bda4d0 : SUCCESS in 6m 55s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/23a5d62d96734148be0d2bbdf7832c12 : SUCCESS in 8m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eb0ade8ed5bb4ad7b33d7c04d1b25689 : SUCCESS in 54m 09s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/46b8bf6ee8bd46c58c41858f18fb5f8f : SUCCESS in 58m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ad1904b8233442ae8b3190cbaa106db7 : SUCCESS in 47m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7f34b78fb6964e79aa9a8dd58546a9b2 : SUCCESS in 46m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78abfa64d2c84be09bf24a62bec46d3c : SUCCESS in 16m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2802af107dbe4b069b0149b5abc9738b : SUCCESS in 4m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b2ebbce225e41039b618380d54c3e1a : SUCCESS in 49m 08s","accounts_in_message":[],"_revision_number":2},{"id":"99e528f6caa284aa873c9491f78c2d9f5810df4b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 23:09:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"be81403d3fa5e0fba043eaadeb49d18219c39031","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 23:10:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc0c7450eefd42a78b16d93a049e9a6f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e9de8bcd99748babd736fcec8b7a921 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/77ced21db78d429c8b3fd60b4fbcd372 : SUCCESS in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/569168de44b84f4eb79716bc78e57760 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"776a9d9a121884b2f400ad632c15c9517aab2162","revisions":{"1cfa92a4ba375dee8f6a40d010535d36ccee385a":{"kind":"REWORK","_number":1,"created":"2023-05-19 19:17:13.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/00/883700/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/883700/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/883700/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/883700/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/883700/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/883700/1"}}},"commit":{"parents":[{"commit":"e6792bccef1a18336e8f266a5264bd2222388e8c","subject":"Merge \"testing xattr metadata with py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6792bccef1a18336e8f266a5264bd2222388e8c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-05-19 19:16:11.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-05-19 19:16:11.000000000","tz":-300},"subject":"Use the db for get_shard_usage","message":"Use the db for get_shard_usage\n\nWe noticed on a large db getting the full list of shard ranges and doing\nthe sum in python is slow; move the query to sql to make it faster.\n\nChange-Id: I17bc7d7489e8947f53e77b1cd6eb9f792f750481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cfa92a4ba375dee8f6a40d010535d36ccee385a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cfa92a4ba375dee8f6a40d010535d36ccee385a"}]},"branch":"refs/heads/master"},"776a9d9a121884b2f400ad632c15c9517aab2162":{"kind":"REWORK","_number":2,"created":"2023-05-19 19:28:46.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/00/883700/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/883700/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/883700/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/883700/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/883700/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/883700/2"}}},"commit":{"parents":[{"commit":"e6792bccef1a18336e8f266a5264bd2222388e8c","subject":"Merge \"testing xattr metadata with py3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6792bccef1a18336e8f266a5264bd2222388e8c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-05-19 19:16:11.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-05-19 19:28:37.000000000","tz":-300},"subject":"Use the db for get_shard_usage","message":"Use the db for get_shard_usage\n\nWe noticed on a large db getting the full list of shard ranges and doing\nthe sum in python is slow; move the query to sql to make it faster.\n\nChange-Id: I17bc7d7489e8947f53e77b1cd6eb9f792f750481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/776a9d9a121884b2f400ad632c15c9517aab2162"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/776a9d9a121884b2f400ad632c15c9517aab2162"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
