)]}'
{"id":"openstack%2Fswift~883731","triplet_id":"openstack%2Fswift~master~Iea4142db99540063c7e816bb4f9d58dfe4a50b6b","project":"openstack/swift","branch":"master","topic":"sharder.sharding.completed_as_float","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-05-22 13:55:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iea4142db99540063c7e816bb4f9d58dfe4a50b6b","subject":"squash: fix the traceback in sharder.","status":"ABANDONED","created":"2023-05-19 22:59:46.000000000","updated":"2023-05-22 13:55:45.000000000","total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"0a90eb7d8f2d355617613bb236a8d9da925b986c","_number":883731,"virtual_id_number":883731,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-20 00:12:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-19 23:32:53.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-05-19 23:32:53.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2023-05-20 00:03:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-20 00:12:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e5eb3794ed42ec97240c979b674a52fcfb2e918","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-19 22:59:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a15146640c5bc685c7ebeb5bce543d104e408298","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-19 23:18:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebcb2e94077b0b47ca7f5eb77448cc17629f7db1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-19 23:32:53.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b2a7ad49f3568d7565abdd1ddc8716f3f799df83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-20 00:03:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8983a22a201c4a11a5b8ecf9b7d7d9d8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/419c9f61b6b34856bdf59ce8eeb7ec45 : FAILURE in 16m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0e14f8245e7b422cb6045e8b6bdbd1b7 : FAILURE in 16m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f8433d255fc46648bd6e3fdb38776de : SUCCESS in 20m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfea0656ae514f8aaec015d560ba6beb : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e4cb088014d5d5c93c61b6240fb47b2319a5a1c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-20 00:12:29.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/be238bceb66a4ddbbe1f01274d6bad9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1c52db9b50d43e5afc2507c50a72214 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/8fe7c3a49aa0461b886342465936d50a : FAILURE in 35m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bf21379e18814991bef87d346396a5b4 : SUCCESS in 36m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84a944fb06a3450999cc386cbd2c8a58 : SUCCESS in 42m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e39cdd425bc49ccac4dc40fe92168ee : SUCCESS in 21m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c12852e941f4063a525706839fe6b1a : SUCCESS in 22m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63b476b99ef2482e832da07823835eeb : SUCCESS in 19m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c5c8218e092b4a8e8229f6793d51ba1e : SUCCESS in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/351c9ba2679f42dca69c8d4d75f53854 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6ef3ecb42b7b41bfad376480244206aa : FAILURE in 17m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/82cf47cc4cb74939a76a36c40e3e6932 : FAILURE in 14m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/159c9fe3853e42de8a19225c5cf55eb1 : FAILURE in 19m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a363804c7ef74581aa8a3e9246a3b28e : FAILURE in 16m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/04e341e2705e473aa14bc71ab6615342 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/db7fa522dc434d88bc3f4944c18c1cca : SUCCESS in 22m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cff77e4f8dc34d1c98852c1e3ab87a0d : SUCCESS in 22m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2e0977248ee542dba09404d6adab0222 : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/04bb274e76944c76a90f88ac923912f5 : SUCCESS in 13m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7e804ea2b5bb4bb481219a56faae9685 : SUCCESS in 21m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/423d70ab5dae4ccebab7f10e44f4a63a : SUCCESS in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f838f3cd6871458a8a1dd098a4a3e38c : SUCCESS in 14m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eb4f62a201004d2292eae289e8540650 : SUCCESS in 7m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/50f6f2794143430580696b6ad716bf83 : SUCCESS in 51m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b62c4392f464e61a98637f803facd2b : SUCCESS in 1h 07m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aff89ca883ec4e31b6db3248fcaee18a : SUCCESS in 57m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a70ee133c79347c7b5e218dcfe171805 : SUCCESS in 41m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b6642c13a6f4f3f83794488dd9223f8 : FAILURE in 22m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bd4cf79cb454764a421fed6cc4cb6ea : FAILURE in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/efa6e6d8fe454638b1c3ed9dfacc395c : SUCCESS in 23m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c55f4ebbf6744ff78dcb18e851935ff8 : SUCCESS in 36m 24s","accounts_in_message":[],"_revision_number":1},{"id":"0a90eb7d8f2d355617613bb236a8d9da925b986c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-05-22 13:55:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81cd2f6017866107747cbf630b4cfa0d53260538","revisions":{"81cd2f6017866107747cbf630b4cfa0d53260538":{"kind":"REWORK","_number":1,"created":"2023-05-19 22:59:46.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/31/883731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/883731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/883731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/883731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/883731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/883731/1"}}},"commit":{"parents":[{"commit":"909052c881efac96c487ca7d3f11e126cf86c10c","subject":"Fix sharder.sharding.completed timing metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/909052c881efac96c487ca7d3f11e126cf86c10c"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-05-19 22:59:01.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2023-05-19 22:59:01.000000000","tz":-420},"subject":"squash: fix the traceback in sharder.","message":"squash: fix the traceback in sharder.\n\nChange-Id: Iea4142db99540063c7e816bb4f9d58dfe4a50b6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81cd2f6017866107747cbf630b4cfa0d53260538"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81cd2f6017866107747cbf630b4cfa0d53260538"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
