)]}'
{"id":"openstack%2Fswift~883936","triplet_id":"openstack%2Fswift~master~Ie8e0560f90207eede1992c8665e4d8571a505580","project":"openstack/swift","branch":"master","topic":"refac","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-05-30 10:58:48.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"Ie8e0560f90207eede1992c8665e4d8571a505580","subject":"Fix GetOrHeadHandler unit tests to use public interface","status":"MERGED","created":"2023-05-23 12:39:15.000000000","updated":"2023-05-30 12:56:45.000000000","submitted":"2023-05-30 12:55:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883936-refac","meta_rev_id":"b18012e7e47cbcea8e8bf19537526bf10c2aa731","_number":883936,"virtual_id_number":883936,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-30 12:55:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2023-05-30 10:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2023-05-30 10:58:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2023-05-30 10:58:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-23 13:30:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-23 14:05:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-23 16:29:57.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"2ebcb19acb43b622a7ab808aa83b5f7f256f9455","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-23 12:39:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24bfafc1e9a68531271bf573fe6f30cd52dfc71b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 13:30:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89d0126e1fc645dca25a32869cee45c1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d97b274a654149fd9e5d6c5ecc1c5c09 : SUCCESS in 44m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31eaa96d7e5c4624a9807129c57384e2 : SUCCESS in 49m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f9b6545eec046039d7e5a6266285e82 : SUCCESS in 18m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/96b982f83e85480896173fd84995cd6b : SUCCESS in 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a9f2a31d25f459ab7f4d40d6bf8a84b8b81fa4f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 14:05:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8428903b1dab4953acb90726a7c914c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e2c09e09c5047ce821fc4c25f3c7649 : SUCCESS in 9m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72c0317e316f4561b2f7922c90772dad : SUCCESS in 16m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc55d448ee3347159cba1604073540db : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7cb66e44b14146f2a4fca7c124b38240 : SUCCESS in 20m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dbbe8942c10b4970b23b8d91394103ff : SUCCESS in 8m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/18757b7214c44af5b1c51ee92fbf8d33 : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/942ed1e783be4c318d381c40b6151f2f : SUCCESS in 17m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e33c2c3b86624d0aadd51604c64de3e4 : SUCCESS in 36m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/533a93aa09de4eb28db1a8cbeb1590cd : SUCCESS in 33m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4d8031655a8f40f4b587dd2c3774c666 : SUCCESS in 15m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/44ad695693d54578ae56ea080f374f50 : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e1af297f288a4c08bd573bc8d04b816f : SUCCESS in 33m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c3cff678b07641cdae2a58a4633de573 : SUCCESS in 15m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/76948d2a38e94a55b5df66c81efbe32a : SUCCESS in 14m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/03e762ae13724219af3da32eed234013 : SUCCESS in 14m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/24adcdf45c754c2ea1cbf68732c3de79 : SUCCESS in 13m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8d73396db4344191a36de7c968c2e543 : SUCCESS in 1h 00m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e1d30851250c42afb52510dc7b2b9bd9 : SUCCESS in 48m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea99db17b7b4428788e40743357f074b : SUCCESS in 21m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16173726df904ba5b539dffbf531f3a7 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/987859b7ba0244999ebd88a3e0af36d3 : SUCCESS in 20m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"159155e025161ee24a056654ba9118db2700d8c8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-23 16:29:57.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce36b2adfebc5f06be56a6c6e752155d49509d87","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-30 10:56:42.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f594d5596d14ab55a3fc743230c85979f3bd2099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-30 10:58:02.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"29f3f489002aed617fc373c29e512fce87fbcae6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-30 10:58:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"44926be27c9cd126832b32c8e6b20127a0533d2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 11:03:13.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbec2eabfb3f4fa9a7f8a042d810d783\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5816c51d3fdf47cc81213e0b6db75690 : NODE_FAILURE Node request 300-0021345640 failed in 0s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32e4ab0173cd4335aad7b8a175d48414 : NODE_FAILURE Node request 300-0021345641 failed in 0s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/71c6c92f52644456b4484bfa7dba1073 : NODE_FAILURE Node request 300-0021345642 failed in 0s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/28743131cd78454fab54dce999692ba4 : NODE_FAILURE Node request 300-0021345643 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"ba181746799f7813ce3b68383ac97d46213e9c7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 11:55:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1755caf2f5c45a8ad59aa7f276a8bde\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3df816b6c6c4910b681f76489ada645 : SUCCESS in 8m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9afdcb0f8b3b4a3dacac4e112835757c : SUCCESS in 22m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c058c9719e3a45bfb766be3f883887cc : SUCCESS in 23m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b4bf4c86d2f4f9c849527e85befcab2 : SUCCESS in 17m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3593f1c14dd74c2ba448e812d40cf6fc : SUCCESS in 8m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/949dc1a51a074989a8f04b84931a2f78 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0422bb68a1494423af8c2c4ead59b8ba : SUCCESS in 19m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ed2ebe0be95848a8ae1bfcc74c97423f : SUCCESS in 18m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c7fd76f665414aa5940817024c379323 : SUCCESS in 17m 57s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c760689768c6402385fcf0b6a8003eed : SUCCESS in 15m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d6fd5c5bbae44493954a794b763dc11e : SUCCESS in 17m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c7ad022309394e168645eba492c2b56a : SUCCESS in 22m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/84918de5598041f088ffd3fe5a6ee38e : SUCCESS in 19m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ba1e2070f0184874839ceea1efe7021e : SUCCESS in 13m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/39c0fa3db6364bb1be6bdd14ede9849d : SUCCESS in 13m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2a8b5d27ce6147849bb97d05770dffe9 : SUCCESS in 15m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/65f323fdf02c4872a6f2edc7d18c6820 : SUCCESS in 41m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/101e256e373d4ba0960bd4ff10439dcf : SUCCESS in 52m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3230a8a533f4d0ea3ab530e3ff9f927 : SUCCESS in 17m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/777b2973dfa54d2e9f52f20677b14b4a : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5e5d1afa2ff041b6a18b0b05a6e2681d : SUCCESS in 17m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2003cae59420d47185dfcc1964ac8d3091ab32b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 11:56:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"280a7d8f59119d96f029d4e31941d727e1e283e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 12:55:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4dd2239342e41e48e4b8184389feb6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f740feabc742eea35295ecb4367266 : SUCCESS in 7m 56s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5bc7367672cc4d56a7e7b1bc49532968 : SUCCESS in 8m 16s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8f160171da874d86a4320643818e2f86 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28d7981e29d441d6bf840493228968ff : SUCCESS in 25m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29e8fb35797a412c9c4866f836c297b6 : SUCCESS in 13m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3e20dd74792740ce853f47fda1a58499 : SUCCESS in 14m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d2f30c8557084e3c94408557ff187c23 : SUCCESS in 15m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c959fd842a7543618fd6f9dd867b2483 : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5dbb4123387c446791ec98ca67c44391 : SUCCESS in 16m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d13dabd87aa54eb083dc6e08d9981b35 : SUCCESS in 14m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0cf6680dee7e4b2c8a8d746aa29ad6ef : SUCCESS in 12m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/eb127f5b3c694934a31e657d72e9c60a : SUCCESS in 15m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d870af9dc24d4fab97f82f172b29270d : SUCCESS in 13m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fa7708c4c70341d590302c16a909ee63 : SUCCESS in 8m 21s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eaa0a94a45484da58bd2fb15a7dc8119 : SUCCESS in 7m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3400fdf85c114bccb000a95b123676fe : SUCCESS in 53m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a41c42fd5c0b4a9d8d3d4289ace4deff : SUCCESS in 59m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa36552a200e4f17803680fbaa220893 : SUCCESS in 16m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7aef52bc46b4a508f6fee068ad7de3a : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":2},{"id":"800db5c1708fb8694dffd8c8ef25829e61b868d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 12:55:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b18012e7e47cbcea8e8bf19537526bf10c2aa731","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 12:56:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99cf17af36144bd6841ca35c310a0925\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2ab606deb9f4b62b5a64f8c23a7547c : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/86cf4454592e43eeb7fa5a2288605f5d : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/42657d77b30a4089936efae76aa6c90f : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"149b617c28514c36d1dec57232cce883375f1dbb","revisions":{"f4fe135d716c8c1cad0cbfc5fc632d402bf71a82":{"kind":"REWORK","_number":1,"created":"2023-05-23 12:39:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/883936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/883936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/883936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/883936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/883936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/883936/1"}}},"commit":{"parents":[{"commit":"4f463c72970b4aa0c1f8f04d2f917172c67bf5bb","subject":"Refactor for extract closure to method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f463c72970b4aa0c1f8f04d2f917172c67bf5bb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-23 12:36:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-23 12:36:41.000000000","tz":60},"subject":"Fix GetOrHeadHandler unit tests to use public interface","message":"Fix GetOrHeadHandler unit tests to use public interface\n\nModify some unit tests to call the get_working_response() method of\nGetOrHeadHandler rather than a private method.\n\nExtract and re-use a TestSource helper class.\n\nChange-Id: Ie8e0560f90207eede1992c8665e4d8571a505580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4fe135d716c8c1cad0cbfc5fc632d402bf71a82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4fe135d716c8c1cad0cbfc5fc632d402bf71a82"}]},"branch":"refs/heads/master"},"149b617c28514c36d1dec57232cce883375f1dbb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-05-30 10:58:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/883936/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/883936/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/883936/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/883936/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/883936/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/883936/2"}}},"commit":{"parents":[{"commit":"3e08fcd7b19d0a304b45562a94bdf06c5cbc9f03","subject":"Merge \"Refactor for extract closure to method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e08fcd7b19d0a304b45562a94bdf06c5cbc9f03"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-23 12:36:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-30 10:57:34.000000000","tz":60},"subject":"Fix GetOrHeadHandler unit tests to use public interface","message":"Fix GetOrHeadHandler unit tests to use public interface\n\nModify some unit tests to call the get_working_response() method of\nGetOrHeadHandler rather than a private method.\n\nExtract and re-use a TestSource helper class.\n\nChange-Id: Ie8e0560f90207eede1992c8665e4d8571a505580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149b617c28514c36d1dec57232cce883375f1dbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/149b617c28514c36d1dec57232cce883375f1dbb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
