)]}'
{"id":"openstack%2Fswift~884433","triplet_id":"openstack%2Fswift~master~I6bdff90f9d3d26a46b1f19c87db0f3c60260c5b7","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I6bdff90f9d3d26a46b1f19c87db0f3c60260c5b7","subject":"Use format_exc, not format_stack","status":"MERGED","created":"2023-05-25 18:30:16.000000000","updated":"2023-06-23 19:14:01.000000000","submitted":"2023-06-23 19:12:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884433","meta_rev_id":"0133cd2a915a198cb7fefd811b1a128560284461","_number":884433,"virtual_id_number":884433,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-23 19:12:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-23 17:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-23 17:34:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-25 19:33:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-25 19:43:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d8b547a9aeedf9b1b2d9fb20f07e2d3db5830b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-25 18:30:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fa4b70d7dcfbd714d04580ecc734ff85cf51839","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 19:33:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a642c37d10f14e4994be90660d190974\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a3248bb5b5645b38744e322e6ff57ac : SUCCESS in 41m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df20b381350d4fbfa1e65a110100dd08 : SUCCESS in 1h 01m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e5a3c98c26b4137a1e22e87e5865752 : SUCCESS in 58m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7949ef45f0145b58f37685c88454da0 : SUCCESS in 48m 18s","accounts_in_message":[],"_revision_number":1},{"id":"446f7968dd4a5af617c21db03906a78cabb72cc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 19:43:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3336c0110d1f4883b476b6b6015c7767\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f28c69060bca442084bb37eeb9d94f48 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/5f6ec96f992447c98e70ebf30b2f6a63 : SUCCESS in 1h 03m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/afd0c0acb94f4da6a62d4ce3ad16cead : SUCCESS in 1h 06m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f41b16bb540e449da749041004c15551 : SUCCESS in 1h 01m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b03bec5ed79c41daab91c5803fca3b69 : SUCCESS in 21m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70172fe767c640d78d6df59135c94fde : SUCCESS in 16m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ca424bd84aa4a03b8964616974de8ea : SUCCESS in 17m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77a938a3ed05479fb5297a38e3fb2138 : SUCCESS in 8m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0059e217437d49c883ddd334770a6c75 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1913031aa51f42999e278a0f155d2380 : SUCCESS in 18m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4a412f1eb5464336ac968cf65cdf6d3d : SUCCESS in 15m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2a6f92842f3f4a168dec09561f7fe364 : SUCCESS in 16m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a915859a111149c1a7585747d09c2059 : SUCCESS in 13m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/30b9dad4e3134be5aeed40825f9442c5 : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/606b60052be6455a92143443f46833bc : SUCCESS in 16m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ceeb1858c7db4f4181b12ca8fec943cd : SUCCESS in 17m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/818754bd71d14677adcb13e37749f5b0 : SUCCESS in 14m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/60ca90e3a6684eaf85893988905b3474 : SUCCESS in 15m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d6c42ac50f9a404cacf77392e9475889 : SUCCESS in 14m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d07ad38608514deb852b3573ff0243aa : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5bd90c57dead443193dfdfc39e8b117b : SUCCESS in 14m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/94f4bd8a706e43b987b5b6b9a278d522 : SUCCESS in 7m 06s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5e061eef0fda43fc9ff66850c2b07f0d : SUCCESS in 53m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/15e318f7629d4b74ab0d499067b64167 : SUCCESS in 56m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8d158e9610a9422296fa21e1fcae0030 : SUCCESS in 53m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6e3296bb49d240e799fa197408734136 : SUCCESS in 35m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/454173c6cc3a4a639eb1c47a89d434db : SUCCESS in 21m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3e1805ebf4e4280aafad46ff052a9c7 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/875bc68e44c246ef802378adf1f7881c : SUCCESS in 19m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/139cf6f67f1a4ba3bb7aa724ea1d0adc : SUCCESS in 58m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a70d79ab815072a7e6e97bfa3907e21795d4b09a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-23 17:34:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"33c06d3624bc5495e42af1228a7b5bc831585a73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 17:35:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"606c86ca295cc0e57e439639f5b1a13144b57baf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 19:12:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2b9a9374ad642ee81da65932b964a01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b85a45e0d92c4cd08048af291270c8eb : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/6727fe38cb0946a4a6f0dfe5f1c0900d : SUCCESS in 1h 04m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/741bd6242420409583ecbdc59d01372f : SUCCESS in 40m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d53ee66b93534d71a88ebc6a525c54a4 : SUCCESS in 49m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/549384078b81446294e0502d28b7c7fa : POST_FAILURE in 39m 04s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2ddc6f71119f4078b2d14d2c499cf6ae : POST_FAILURE in 36m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/598d210efbda40a188b68a861ae14664 : SUCCESS in 15m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b5d556267fe24b269896e5e40731eaf2 : SUCCESS in 15m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2154cfffa5c547c0af8469c14e48bbff : SUCCESS in 17m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3a358022ad6e45c9a99cfbd6e64a69a1 : SUCCESS in 13m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0ba991ebf7e1415dba5496ffd8d6c30c : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b0787769cf1e47cd8f4511796a8e9c18 : SUCCESS in 17m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f732a1746402407cb6c6bc969115739a : SUCCESS in 18m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/78704f11d1a7448fb92e922df21ade16 : SUCCESS in 12m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fbce1da30d304747a02afac105ed3f1f : SUCCESS in 14m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fb5f408df3a445adadb8cef956bd60b3 : SUCCESS in 13m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/39eae3c66eb6405ebb18351a494ef1c7 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/71c2b551c71e4bb4b23020bed3d57ba7 : SUCCESS in 8m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cbdb1f9c806f4e85918f86e9330c105d : SUCCESS in 44m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d74d3298ea404d66821883eb2f123f7c : SUCCESS in 51m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/129b9eaabeaa45e8b6858bcbd026acc9 : SUCCESS in 53m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/945b6b9efe204c98b67ee03a28d3740e : SUCCESS in 50m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b4ab557d64a4d0f8e41f3e641b3aa3a : SUCCESS in 18m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93124d4c064c4c20a655b4ece1ca0ce3 : SUCCESS in 4m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc3e8b90c2ae46d09bb346437070457c : SUCCESS in 1h 02m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e732dadaa3941679e5aa73eb12417a16ffa7675c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 19:12:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0133cd2a915a198cb7fefd811b1a128560284461","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 19:14:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b05b54ecca54d7e88dbd8faa46a4087\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d13d420dc76843e9897290d5fe390c9f : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/55f957f2dc094dc7ba3702aa3eb15ea2 : SUCCESS in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/36563d82a39c4844bf98fcd1af609359 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"effd4931d13f40d3b30a8cc9736215a0a81f3bc3","revisions":{"effd4931d13f40d3b30a8cc9736215a0a81f3bc3":{"kind":"REWORK","_number":1,"created":"2023-05-25 18:30:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/33/884433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/884433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/884433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/884433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/884433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/884433/1"}}},"commit":{"parents":[{"commit":"fabb5cc8a18b5c74edd6521401ffbd459b2bc6d0","subject":"Merge \"Drop more translations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fabb5cc8a18b5c74edd6521401ffbd459b2bc6d0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-25 18:29:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-25 18:29:51.000000000","tz":-420},"subject":"Use format_exc, not format_stack","message":"Use format_exc, not format_stack\n\nThe stack showing how we call self._logger.error isn\u0027t nearly as\ninteresting as the stack for the caught exception.\n\nChange-Id: I6bdff90f9d3d26a46b1f19c87db0f3c60260c5b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/effd4931d13f40d3b30a8cc9736215a0a81f3bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/effd4931d13f40d3b30a8cc9736215a0a81f3bc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
