)]}'
{"id":"openstack%2Fswift~884931","triplet_id":"openstack%2Fswift~master~I3f531f904340e4c8407185ed64b41d7d614a308a","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-08-02 11:39:05.000000000","reason":"Change was submitted"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2023-08-02 11:39:05.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-08-02 10:26:08.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-08-02 11:39:05.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-08-02 11:39:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f531f904340e4c8407185ed64b41d7d614a308a","subject":"proxy: Bring back logging/metrics for get_*_info requests","status":"MERGED","created":"2023-05-31 18:59:16.000000000","updated":"2023-08-02 11:40:03.000000000","submitted":"2023-08-02 11:39:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":78,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"884931","meta_rev_id":"adaa1e18689ae6313254646d576674d3d5a8fe29","_number":884931,"virtual_id_number":884931,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-02 11:39:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2023-08-02 10:26:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2023-08-02 10:26:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-31 20:19:22.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2023-05-31 20:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-31 20:58:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-05 14:14:29.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2023-06-08 18:24:21.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2023-06-12 19:51:46.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2023-07-13 06:53:54.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"de9680301ae73f0ffaabcdfb93d49f40971e9b48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-31 18:59:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6451e2bd9889daeafa81561e6ed5b1dfa89e1443","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-31 19:18:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2fd86f2a9541a8211c9e1589846bad1f28359e03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-31 19:21:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c74d8e8114e443afb4bdae82763ef434bd8f07e1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-05-31 20:19:22.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1e397fe62b28a766119b94aafc791ea669a5f100","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 20:50:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd3028072a784bdb9de276086ba7a07d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8e73c7f7d2f4effb844bec8df021089 : SUCCESS in 14m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd4e9d6525794dcfbcbe33fece4af25d : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b394979fa3c740b7a3aff166ccf35f60 : SUCCESS in 1h 26m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/763c9cebb9be48d1a334beb30479777f : SUCCESS in 1h 16m 50s","accounts_in_message":[],"_revision_number":2},{"id":"a26184abe7ae1a9e9f0e4aa1d6c2bdcf92a3b903","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 20:58:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/166a221e15204b5a83116efd2507735d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/636e6565f8c54daaa274e1f327204328 : SUCCESS in 9m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/2c007c61f6874254addd05624de72761 : SUCCESS in 1h 02m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/750d37da70cd4b6cab504ea3da7a6538 : SUCCESS in 55m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd29bb87d54e40339c3b15c56e7f4613 : SUCCESS in 48m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb97f41d175648daa4fb25c30fb66669 : SUCCESS in 16m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/002d96643f324f23a2f0b81f83d9add2 : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0071df9f2df482b9254cc58dd1c7860 : SUCCESS in 20m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bb45fd2396684c4aafb860e9bb0c3901 : SUCCESS in 8m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4b1137452c174c86b44f8b377971833b : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6c36fa7534db4374afa8725fdce3f2a5 : SUCCESS in 24m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b4478ed1d8ab4769ac28d54a077d16b8 : SUCCESS in 24m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/34667a2199f8440cb72327b454469aac : SUCCESS in 15m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4db61b6b8afd48f28ad146353ff5f719 : SUCCESS in 24m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8282a9ed839f4ccd949eea5b592c6766 : SUCCESS in 23m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a3d2ecd426b449feb58195ae3aff5351 : SUCCESS in 24m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6de854c17aa54baf8938b51b2731c21d : SUCCESS in 26m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ecc3324794b24471857d8406d3257b9e : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e78fc9448b95490087af4598a1f2691b : SUCCESS in 17m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f82c1153817846499975fceab53f7e80 : SUCCESS in 17m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ad39ff1e429641a88c94999d44c47a61 : SUCCESS in 9m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c3ada2f4905041508de81d2af918881f : SUCCESS in 17m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fec6ba3fb7494da2ab1e040384703701 : SUCCESS in 8m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c5cde2af3d174ade9c16aed279792d05 : SUCCESS in 1h 05m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5c43aa7ccda4c9097e0cf55de2ebe4b : SUCCESS in 59m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/17e989724267484ebace9bf4208a3948 : SUCCESS in 54m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7560235be885484ba9c860479db31daf : SUCCESS in 39m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fef6ad5fc8684ae285e1448c1379b7a9 : SUCCESS in 20m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50d7d78dc27b4299bf1bb93854552ddc : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb3cf57799f6440781fc67921e3315e6 : SUCCESS in 20m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a741026e2b3e452d9a60962723014df6 : SUCCESS in 1h 10m 28s","accounts_in_message":[],"_revision_number":2},{"id":"f91df175dae48f638b4e7bed2581ddacecad4ba4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-01 17:26:37.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dd587627cd7cdce1928fe0517eddbb1384a87fe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-02 16:40:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7c0d8f4f38a72f09418d9786f433a7919a022c02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-02 16:42:27.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"680334dceefd4939c8afbe89f30182f177b11e70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 17:14:18.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf92afdd6d484c4bac983723bb01d3ce\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9669e43e89044e7b8adeb8996c2f9308 : NODE_FAILURE Node request 300-0021384709 failed in 0s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78be20d5514641dda204ea7f21e59da4 : SUCCESS in 20m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/83d1c554e29f465eb4b2825ba1e6bc8e : SUCCESS in 27m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff52900afc904fb1bb6f50596cc15b46 : SUCCESS in 25m 02s","accounts_in_message":[],"_revision_number":4},{"id":"504c2bf2317df4a67fa2d7ff31ef60a85deb0b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 17:54:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02ed815a1bcf472cbc655bccf6d1b913\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4321dcc9fde34a74a41d66887d601f8e : SUCCESS in 11m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/9450e4378ce246aa8dce4564e51b8ce4 : SUCCESS in 50m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d021ffba72b3403abad0e41b5057dfd2 : SUCCESS in 45m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c14572bf66b48de9001114615298de6 : SUCCESS in 48m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ea43cdd65da4f5b9b30948c16dd9371 : SUCCESS in 25m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41ef27b7d9b545a3b6ebb2827a461e56 : SUCCESS in 15m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9bea9a7e2b7d4cc0b0b4ea0310cf8878 : SUCCESS in 16m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0a8030a005fd4a51b5ccb1e33bf2c757 : SUCCESS in 8m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c10e769730a44bddbc68843defcca193 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38c692d97d65453d9840a65f62dc1e79 : SUCCESS in 18m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/06aac0e902454a189f1b0db53d02d1a5 : SUCCESS in 15m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2063abc91b914258916edbd23c1dc070 : SUCCESS in 25m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/77591b5c33e0489195ca5f59f8447110 : SUCCESS in 24m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8bd5a98894434a4ea5ab9c668f2d68d4 : SUCCESS in 18m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e049a5bf9d4d49d59df6cc69ead88ca3 : SUCCESS in 20m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/49b6602608d84837a47a23001da1ba1c : SUCCESS in 23m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1f5c35cbc09e4a1a9c4c9b69cc0bf78c : SUCCESS in 22m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/23fde330bb8e4447811c58ad2c5789bd : SUCCESS in 18m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7f09b8b45fbd482b98903ed0e1ef9a7a : SUCCESS in 14m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/470496a43b864401939f111402ff980d : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f8bb82452eda473a885e06d69ae6c72f : SUCCESS in 21m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d789fb6c6c7442e58a34c285a70dc64b : SUCCESS in 7m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4d9fb9983b2346b3b7f68823fc40b3fe : SUCCESS in 59m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/505b760a3587445bb42ee2ccad9f899b : SUCCESS in 1h 00m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6a7d3e99592045ddbca206dcd293fcdd : SUCCESS in 59m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0442178a3f8741148fa98c08dd048221 : SUCCESS in 55m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2917ad3d2afe49ed9e44d8ad55921d54 : SUCCESS in 18m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aca629e6346e4b54a78a5a6dc25c0451 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bb799098ac1f4fd6a5fd865ff5ab50fe : SUCCESS in 24m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82ff0e34a05a421cb0aec274acf2a955 : SUCCESS in 1h 06m 38s","accounts_in_message":[],"_revision_number":4},{"id":"fa7b7b5611435e3bfe74a577b8eea38a4d6bc686","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-02 21:56:43.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"51bfbe8bc823811194d19b03dc0c53cfe2c4ed25","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-04 20:12:04.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"04fc4dd692792b9977e7b55243dc93ac37cad4fb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-05 14:14:29.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fdf0504a2bd41bdf8a4cf5df530e479c2ab1d296","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-05 22:50:52.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"96eb857a26377f4244d37bcbbfa18e749eba0d9e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-05 22:51:16.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5f07ba146c361a12b376f2fef3d5358036417ead","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 23:56:12.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f0833643df9436ea4737d17bf7b8789\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fdedfe42da4d478897a2a78cf0db5e6e : FAILURE in 18m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d41606689b8449da85e1a00201783b5 : FAILURE in 1h 00m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/979456a5ba2a427d8ed821fc0432b6c1 : SUCCESS in 21m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/453a94bd6a9542cba783e7c3e4b5c628 : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":5},{"id":"af6b767536c51251c8080a05f43d1f0c82fe4a4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 00:26:57.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/46e581d9cf214dae9fc9271b707b26ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a81b86e0e3874c518f083ec211af21cf : SUCCESS in 11m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/2421221ccf8b47038a96c85bc57053a7 : SUCCESS in 59m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/91f07f2e4772437792f9dc6d7a5b7924 : SUCCESS in 42m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7109a48e028c487a8d628320f6fd72b0 : SUCCESS in 54m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83d0497d7b734e4a9fb7242331819833 : SUCCESS in 23m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25a0a01717354607af97821d896c3d34 : SUCCESS in 18m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5bf59ce6983f441e9d6d043aea2517b1 : SUCCESS in 24m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/934a240127434257b1c8fca87ebc6d4c : SUCCESS in 10m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fc74eb4fe24141e28ace1f5f62c0b31b : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/494ae289976946eda79db9f070d2d369 : FAILURE in 17m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/296706900c4f49fabe2cdb0c97a3bb81 : FAILURE in 25m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6cba3cfc0a92472d913f8045bc4a1eab : FAILURE in 16m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c272a64bc86844feaa34e3855a014448 : FAILURE in 22m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/627858774d00411f81ee9efae67bee73 : SUCCESS in 21m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bb49ab4d2ae04a0ba8c3ff440b2e8301 : SUCCESS in 24m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c92ebb24763a4a708a7c6359ec3ac29e : SUCCESS in 19m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f80da3701f9a47d3bbe0f501d5c602d1 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9bcb852b33d54549816d860dd2966b50 : SUCCESS in 15m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/37684b42a42742b28305d98d0418bb68 : SUCCESS in 22m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e07599aa6ba6499c9dc12fdc88118ae1 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cb5888dee1154ad1844e158f26e0a805 : SUCCESS in 20m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cd721444178c43aaad0535631f28a786 : SUCCESS in 7m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4200cd6706844011b57da1ef0e27791b : SUCCESS in 56m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca08fdb639864bd89b2415c39220be9d : SUCCESS in 1h 00m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5c27c0f7e10747bb927c3a39d7529d86 : SUCCESS in 38m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7ef435ea8a7448cd81f0f09a1be46584 : SUCCESS in 47m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a09daedc0b584601bca474b699e3a25d : FAILURE in 17m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf4119d4f2d4f759c344281d57e542a : FAILURE in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9e4404eded704eec9514fc6af04d3c7f : SUCCESS in 26m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/608ecb2135d8433d88b7b0d266a12a4f : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":5},{"id":"209f5e8db37db6940bec42beaa29cbd6bfc870bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-06 04:12:52.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"cc76a5129950d8e814fdfb1eed6d3d0c44379d73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 04:39:50.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db661f28dd9b417d8d145569e01363a7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3cdc7d6473e441a598086d65dc0ac5d4 : SUCCESS in 18m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bdf666940c724e5e9750e48806df7c3d : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9a3031d3e874deeae711d96faf235ce : SUCCESS in 20m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a712edd806f340dbbf825bbe49f70f19 : POST_FAILURE in 21m 27s","accounts_in_message":[],"_revision_number":6},{"id":"4fc22128bc9510ec5a670ea25de91cca347f3d17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 05:28:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/330182a4b6fc40108b336c0d9075fbeb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ce822a574b4899b5a5689622354201 : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/def2462a58324e62a4b169fdb39ae58e : SUCCESS in 1h 08m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/09c57b296b5143bfa08c52b9fcdcacc0 : SUCCESS in 35m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f1aacaca4bb47a89d4659c9bae1ff26 : SUCCESS in 52m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7fa7cc8581fe4b6abe0ec9a1de960b46 : SUCCESS in 15m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c985f3f82d7f463694f034cff5528cb9 : SUCCESS in 22m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c281b3f28ed4c4dbe167ff09b13ff4d : SUCCESS in 19m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c81c5bda4314bf38d2578bc4be49a51 : SUCCESS in 8m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d06c754b642846c08b73645378c68870 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/520664e89a8c44aba8101be671ee51c2 : SUCCESS in 14m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/81781e2686fa4c85a1464a90b25fe0e2 : SUCCESS in 14m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ffeefb2b03a0413c93e21b9a47989aaf : SUCCESS in 13m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/40001b179c634e35adce44e510761bee : SUCCESS in 16m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0238d7894a154fa49022c9d451169da9 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/309371c3948f4894b2a169e6c6883106 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/de02ef8fa9ca48f888bb257aee66eaa6 : SUCCESS in 17m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/93c6af64599545e7b76b9cb64c82747a : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c58526aa4a5f4edd920904fa3d3d52e1 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/55cc92f0d22b4fc4a3b813632103f009 : SUCCESS in 15m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/07211470b5834a119fe3fcd9ab35fb74 : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dfedf5512c6e4216a6ce59d9fdee8843 : SUCCESS in 16m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0e9ee5bf54074dea84f38463aed3b0b6 : SUCCESS in 8m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cf2a3110508e491ba879912966370026 : SUCCESS in 58m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/236fb2c9801843b4b40e9bd301b3e3b9 : SUCCESS in 58m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7adcf59f97f1459da48b0fe3f50b1add : SUCCESS in 44m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb4326c3c66d445f9c75aa765cf4180e : SUCCESS in 55m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff935866750e44b588df70b7d56d9f09 : SUCCESS in 13m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebe4a0c067fc450c8d4e4a00e115e296 : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fd82527c843d4466befd5c033c2e02ba : SUCCESS in 17m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d27001ddcbc742a08db9d584ae451357 : SUCCESS in 47m 13s","accounts_in_message":[],"_revision_number":6},{"id":"cc5dadb6cac8d5d67db500b00781de46d388f805","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2023-06-08 18:24:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e31fcea35951a8e03b3196fc7eaf0659aefd4dd4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-08 21:19:15.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"416422d6f3810778a16b30429b7df47d00233d5f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-09 14:01:27.000000000","message":"Patch Set 6:\n\n(11 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7512289c7f76adee9d99d990091612ec21c1cffb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-12 17:22:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4c6f82d7f58ea943a09a876e76bc09a0190ede03","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-12 17:53:59.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b7f1d4c5e3edfe8723a877824582f25f67a8cd6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-12 18:35:54.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"539879aff0cac9df96c7d93d075e7b5b9f2fed4c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-12 18:37:02.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b744aa79b962add6be63dc6e07f2d0d8545260a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 19:48:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/097e578f17694775a3a75baad06424a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40284a192fe24a88a006ef028f8d0465 : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/bb81bba92ebf438db2a133887bc2912e : SUCCESS in 57m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/58aab8923a954e4b8aabb5ef50f8c34e : SUCCESS in 53m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf3136882e094e0a9abd5d360fbea2c7 : SUCCESS in 1h 00m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/322810ad14ec4891bd008d9f36c17e2a : SUCCESS in 17m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/435821d2c17547a58f0344bddce54aee : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d94965f31f44b3ea1ab7d7a82085f53 : SUCCESS in 16m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/53eac5d981054b5d8b833ae9d72b5784 : SUCCESS in 7m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/57de835427b64294bcc027111ff6da86 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4caf3ffd06c2401699344c7a7a366a9a : SUCCESS in 16m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/002439c1d2134690ac92028fa1ca1402 : SUCCESS in 17m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/67846984ab074ca4b5282faf394f5329 : SUCCESS in 15m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/60edc53d9c2b4d92ac24fc7375b8af65 : SUCCESS in 15m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cdc767b1fe7843159507bcc00e31b76c : SUCCESS in 30m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/736e1861ddfa42e982f94101fa7583f8 : SUCCESS in 16m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/df95481c599f4955b9fdda47346e327a : SUCCESS in 34m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/52516d9850a747e2b2e735d0793300af : SUCCESS in 27m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8a80a08aa328465b8159577f72730a9a : SUCCESS in 28m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c6ec3ff34112423d83729867e6289ce5 : SUCCESS in 33m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b33f494d62c348e0a4cfc0eb81a3469e : SUCCESS in 7m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d1eb3e61c8ba420c9ed56e9d391da44b : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9d487c82f0084620819c443262c118b7 : SUCCESS in 7m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c17e342bb0c24c0ea9661930c2f2c722 : SUCCESS in 1h 10m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a20244845e0471fa65460424adfe7f1 : SUCCESS in 57m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f50254e433914847b384fb93ac6f2958 : SUCCESS in 37m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8170e1c3ea9c4867bfd2cd2bd64181eb : SUCCESS in 40m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0cb6f6f7f9e431a97e1db5ef1a6e1ce : SUCCESS in 14m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373a353b53e14ffba48ee134d0535c50 : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2b1c6a0d8517497f9f0f6024957c62bc : SUCCESS in 19m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8deb997fdb034b2faddffd187212442e : SUCCESS in 38m 46s","accounts_in_message":[],"_revision_number":7},{"id":"bfd21f23a76c24bf9d69e2cafa3247088a38a2c9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-12 19:51:46.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cd20c5af9afc439dce399bf2f3c4ec1e050ba566","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-12 20:09:17.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/025303c07ff84ebb90b9dc7ec30efbf0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d8ddf5d776f47b7b2b5bb142b077123 : SUCCESS in 14m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f156d30bcab34e42887dceef5fe2fda5 : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca7e139a199e47b083d02f46027ef180 : SUCCESS in 21m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/697d20e467e34c8e8157fa0a4a11cb08 : TIMED_OUT in 1h 30m 54s","accounts_in_message":[],"_revision_number":7},{"id":"028196baa42e5cd6612fb3e057c34c1b9e5b8aba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-14 11:49:50.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"472ab9be9e1a1c5e716deb741756af6ef5c0dfad","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-14 11:52:29.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"d9b94a7efc8b3872d29881b5ccc7c887e9fbd977","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 12:13:01.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/493c83e0eef04099a7f58ec742bc690d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bec0ee58854e4805827cd9d77fb01914 : FAILURE in 19m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af6c0b48d7064bb0baaba896d3218152 : FAILURE in 17m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/667b0317288b4f5c9c5f1183385e8e99 : SUCCESS in 22m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ece6213843e447d5a390bcf9e946beb4 : SUCCESS in 19m 46s","accounts_in_message":[],"_revision_number":8},{"id":"9b52b3ada53070161c1fb4c3871f9d9669473c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 13:04:20.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a7c3e4179ed4c78a0732c941ecb36df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d3f2aab7f4442b6a8a6a6eafef2be6f : SUCCESS in 8m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/28fc36560e66419181d599d8112a1b6e : SUCCESS in 42m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dacc667b12c1401091dce39e8388fffb : SUCCESS in 52m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c595ff3a18304f68badf2b727f2f5d06 : SUCCESS in 57m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8244595472c54d81989f4e41f3e2447a : SUCCESS in 12m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/762a4aeba9014f72b3256feed98e1164 : SUCCESS in 38m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1123a6db38d54b2498cbc549dafff726 : SUCCESS in 35m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f2c37a67e6545c88d924c914b65c298 : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/38f07ded99b14c22a0b57e4dc8505089 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/04c8b96841b54e42a534c939f284b111 : FAILURE in 17m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1de1d8c4f855430ea34ac702f3ece4ed : FAILURE in 12m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6f36a938d2a54af492b2a548b20094b1 : FAILURE in 14m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2d6ecf99a2ae44a097117545c4b74bb5 : FAILURE in 16m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c9d646d7154642449a0af4efea667403 : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0a4df139d2894f68918c427e11eaae8a : SUCCESS in 18m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4a6734e043cc410b83ad4acfdfc24d54 : SUCCESS in 32m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/850d4be63bf142a9a851bf472452a178 : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/430665125f7c48a994899c27001f18c5 : SUCCESS in 11m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b3ba2b4e382e412fa26565df014b4f76 : SUCCESS in 28m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03baf996f6b445c08cab4d19d3357d8d : SUCCESS in 7m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f58581867f594a12813e94125a696017 : SUCCESS in 27m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3ee2247b7ddc40df827bce01e37bffbd : SUCCESS in 6m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0b17924b0ba842fd89c4f5103f0b3316 : SUCCESS in 1h 11m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d0e8557f4db45598c6e0ea79c5f7f6e : SUCCESS in 1h 13m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c3c5c7f63949411b8cb2d42a8d49745f : SUCCESS in 44m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ac23f77d7dd14d9c8e1b09a77071ef0b : SUCCESS in 55m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46c56ab2a8b1474296b20626b27934be : FAILURE in 31m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c8c3ad57bbd4726ad114f9ccb65b769 : FAILURE in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/233ebc5d32fa495497d88aebbf2b57b6 : SUCCESS in 24m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/338706ecf5f241eb9df119a725fd1d5d : SUCCESS in 57m 40s","accounts_in_message":[],"_revision_number":8},{"id":"14b3af0d4e7e7c0fe33750491180720b27dfd6d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-14 14:26:04.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"c9d61d9b7aa206494b7a8dc75e11b7910b0f04d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 15:43:42.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34187b2600d345f59eee8c84d5888e36\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edfd68da394d4ceda951a3ee1fc99e55 : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/1a98898dc67d435e96a49e190db6dd32 : SUCCESS in 46m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cb303fe8d9e943f0b9bc4e30dc0606d2 : SUCCESS in 36m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77d200ed66ee4f7184fd70bbc6d799cb : SUCCESS in 46m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e50738b450d5444c86be3a2d2ca2760f : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/988163e241304d88a90a712afdb54686 : SUCCESS in 21m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7822d2f5296d43d08906c96bf56ef588 : SUCCESS in 22m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7e83e7eb808b45d3bb66bbd1e60ae24d : SUCCESS in 8m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/075b796f9f8b41c9be482a02e1fe8878 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0ab076514dac40e1b18f25ba66e36668 : SUCCESS in 18m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1b6ff133c3ea4092bcfc5405b1296f54 : SUCCESS in 15m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2da551df912a4c48866bc3dd4bb7faad : SUCCESS in 15m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/45d55230e0d348d4a35dd3271a6fa62e : SUCCESS in 16m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ce20b90c1af047e1bed3bb8dccb4c784 : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef54b64af9f640e5b6adb6d0c3e29666 : SUCCESS in 16m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/667e59fcaa0441e391ba3f8e54ad300f : SUCCESS in 16m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/96cb03c8aea448f8b82ed51454cb5dfe : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d72e80667f794101991afcc8ac22d445 : SUCCESS in 14m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/444cc8e2747b4038b8c4b5561642200b : SUCCESS in 15m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c738c12128124a64b68e0417f7274703 : SUCCESS in 6m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fc1d61c7a8a742e5990873b812afcc39 : SUCCESS in 16m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f2e365fc5c994f738a001e6bf8e71157 : SUCCESS in 7m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b47b241311684091a0be17e67677c832 : SUCCESS in 48m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6f489b6680e4b84ba1223730695abbe : SUCCESS in 1h 04m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae26222f21f944448e907a2c3c1bc3cb : SUCCESS in 36m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ca6a762bceb246cbb43109f570b3aa7e : SUCCESS in 34m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ddbdd0ec47b4d37b990fbd4550f4f1c : SUCCESS in 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0c35a6cb7f641048e4d503a7caeb70a : FAILURE in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d5287b7d939649ed9144f452b4146d4e : SUCCESS in 23m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e986fc5c06c6425c8fa36060d3a5ecc2 : SUCCESS in 45m 27s","accounts_in_message":[],"_revision_number":9},{"id":"411c5570cf7f8ff97665f4700bfc65bb13627b56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 15:59:19.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f0aa93eb1294c82b9d3cfafee3daa9c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5fb36a4ccaf84869b4b606c5e824ab00 : TIMED_OUT in 1h 30m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67b8e47c75d748a698e7f70bda8373b8 : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/59f4b0a7ab194b70ae6091d8f636b239 : SUCCESS in 1h 23m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/108e5bd64d0443bc9dd52a50492b4ec8 : SUCCESS in 21m 28s","accounts_in_message":[],"_revision_number":9},{"id":"796546175fe952d79346e5f9e00ef15f784cf667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-14 16:23:35.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"fb3058ab1b468ee4be948b769eab72146af90241","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 17:28:23.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d0b73587c0e48e294c13bc940eb52f1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9cf9104f4c94bb1ae60fdfb6de71804 : SUCCESS in 1h 02m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/241a21bbaef14a1b8a698fb2a291db6f : SUCCESS in 57m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/034754e729184b7e9e5e292ca39d5459 : SUCCESS in 1h 02m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/62c7e1ae67d14de3bda9b909a768e963 : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":10},{"id":"27f2d47363cccb1a1e0147aafeace9577b7799e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 17:41:14.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e66147769984b68ae7f1815a6c738c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cea56432846e4ec884e817385a2b7b5b : SUCCESS in 7m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/2d7bb60b33ad49f686cdff3f9abb3331 : SUCCESS in 56m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0bba1414b6074f63ba71d6f0d968a7ee : SUCCESS in 48m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7c5c1611e8a4d7d9f94997a89d0a7ed : SUCCESS in 59m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d4eee18f0a24feca32b6d8c3210ed61 : SUCCESS in 32m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af60a351c19945e785b817ac205bd017 : SUCCESS in 15m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9847e139d9e4120b51a3fd991a2f367 : SUCCESS in 33m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7245e10d5c8940c591c454fb7f0de75c : TIMED_OUT in 31m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1ef9253ab97e4abdaafae288a516d750 : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae8519e6bd2a468e9a4f371e69ded2e9 : SUCCESS in 16m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8210566f976548b5830cabfb06a43a51 : SUCCESS in 31m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/835000c96a92422b94bd6f97d195f050 : FAILURE in 1h 12m 30s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3c3aaf48be394a979c0fb4f9ca8630d1 : SUCCESS in 32m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d118ed87ed004659be74dedecc36b6d4 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/42e5cfb7d4794878aad5d2b494fcb8d2 : SUCCESS in 17m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1b0c9d8f60b84f7eb24e7747968633a2 : SUCCESS in 19m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f46edcfe2b0e4ca99ef5cf3d3cd56aa0 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/079cbcef4a784363aa1e83e29d0d8a2d : SUCCESS in 16m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1dbb83b86be04ab1a37239e0fb97d439 : SUCCESS in 17m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e34f2fb82cf14df99942a7e293a23201 : SUCCESS in 7m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fe34ce5613084a308776b6d46e452598 : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/19b42f7438144406af68e598a9aa1993 : SUCCESS in 7m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6fbe5a5e750044e8907ea79164586d64 : SUCCESS in 54m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cfa42a1844c94b779c9273770a0da92c : SUCCESS in 1h 05m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/390b0ea0b4e34ecdaa1bf74ebb66633d : SUCCESS in 35m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c854fe8ba131494d8116748357475ca6 : SUCCESS in 55m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2422d8513c5e45ab9df0e1f0f6e28a70 : SUCCESS in 17m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e615465b7404c8bb9c1b6632128b93e : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a5ba5775bfd94e6093c0a8ca2e175f61 : SUCCESS in 23m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8df282c84bde4f6b907ce7d563cf10a4 : SUCCESS in 45m 25s","accounts_in_message":[],"_revision_number":10},{"id":"57fb031e78b0c9d4d37e7cff5ee99acbacf8508d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-14 18:42:00.000000000","message":"Patch Set 10: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"ae6ea10ff5b1cda2b67ae483cdf321238efaa1ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-21 17:11:20.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"b2e6e96463b1a050b2b9c829eaeba9104c26b98a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 17:46:00.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/785f504c626f48ce86eab8c728aa0893\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31fa87ebb8b9480384c46681db95d89a : SUCCESS in 24m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/092205a18f7b452d85ecf2d42cab6aab : SUCCESS in 18m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d651d725f37d40dcbc7b499fd477ce2c : SUCCESS in 31m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a4593549f074ecfb115178fc83f2c33 : SUCCESS in 26m 17s","accounts_in_message":[],"_revision_number":11},{"id":"440cb96efec2d9cb11a0f49ff930720cf5a879ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 20:04:55.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f623cf48fe946848cc66e23520125a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b77cbe982cda4fbb8584fafd35f1d15d : SUCCESS in 7m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/3432dc3fff78496293eb3fc37438742c : SUCCESS in 1h 15m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1b6e897df87b49e6acc3dccc444727cc : SUCCESS in 1h 19m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6764a4b77dd040ed897a80ba02e9dd74 : SUCCESS in 1h 13m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2e03cea44db41e6a42f788269acb0f7 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f66c9cbe1de4f58b3b17612abafb369 : SUCCESS in 21m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20e9074956f34117b7519cc66f470926 : SUCCESS in 18m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b4ee56dba2342b09ccf1f973dc7b05f : POST_FAILURE in 39m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/27ac904fe6d24e6f860943ae7f7facf1 : POST_FAILURE in 35m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2ec99249ceb14cffa9d1eee49c147e4e : SUCCESS in 17m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b5971c3a3f3d4f89ace669c7a6460d06 : SUCCESS in 15m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b12caf224a404a07983ce0e22cd7a22a : SUCCESS in 30m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/774e069396744d44bc8b507fbe86b757 : SUCCESS in 16m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3d38d25b3a804f9db85a79c14c5dee05 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3f28b95822e64242bae4742f59c30171 : SUCCESS in 18m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/63bd7a2b61104e54b34e3ca4b47dd573 : SUCCESS in 22m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/99397d7ce9ed464e88fd4fc8c8d340c9 : SUCCESS in 29m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0f879d60580a47c6a49b209360aed157 : SUCCESS in 17m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8f4964b2a4fa4848ab940d2ba3c8c1f5 : SUCCESS in 16m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/31da77880e844c28bb735066a176894d : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4c690f94233346baace4b2eddf9e941b : SUCCESS in 23m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/100009b477ba4e94ba7db67bb4a59b15 : SUCCESS in 7m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cdfa5e2e9e0f4891b509e375727c1bda : SUCCESS in 1h 03m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b445a69b1e742e5a76780c70967a616 : SUCCESS in 1h 04m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fed0d08fcce8471db6cb4b08391cab0a : SUCCESS in 48m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/15552efea445428ab5fda31d2fe49cbb : SUCCESS in 57m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68283e8ae17f4123ad8ee9e4b23e807d : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c69835ab1214d8782adb0ee0a8f8f1d : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/82c54100bb524bd5ab061607b3b1c308 : SUCCESS in 21m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68885988bcae4059869e5efdc21f9e41 : SUCCESS in 47m 23s","accounts_in_message":[],"_revision_number":11},{"id":"78e743eb25ebcb2451e975c6fe53a42482359d99","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-23 14:31:17.000000000","message":"Patch Set 11: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"154450367411f341fc1c991fba5e0ddb2aec4edb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-23 23:30:20.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"00319689c62f8606a56ec821bda3413f3e3209bd","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-13 06:53:54.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"c958d66129e0b4d253c10cd9f08d3ef07b85cd4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-07-13 18:23:17.000000000","message":"Patch Set 12: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"6b813aca6409075a9a9464457d76b580803a88f4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-07-13 18:35:26.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b6ab74e88bbc1a07d63ef097a4f251d87d5f94e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 18:42:18.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85758f820c984111bcfac17f6f299042\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bbfffd474814110a6f30ac62feae770 : SUCCESS in 14m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6fbc0351cc244f23a4b9b6bc3e23eff2 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/afd25f33ce2e48768d54062a4ed1379d : SUCCESS in 18m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4608515b09d4ad3b647875177f4c155 : SUCCESS in 17m 26s","accounts_in_message":[],"_revision_number":12},{"id":"ff217e662feea963e92421d958041442bedb3560","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-07-13 18:51:03.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"6fa28f1fcb07028f0e12832fb527dec579bd4ce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 19:36:21.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9096d0a132644a76beaf371db5ec5b47\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9636e2e4436f4a9c94b6757b9134ed7c : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/59fff9820b41484fb9092ab88ae87575 : SUCCESS in 58m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d0e665dfa84f49b999656c8480d44aaf : SUCCESS in 58m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b917bda5843405f863704f985e53188 : SUCCESS in 50m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3a11cf13c8f4b91bc8a832b4bce5211 : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b3318ef43f044e09f6277ee05bffa37 : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8415ea44a04b48eba7b06822c86f1db5 : SUCCESS in 16m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/954d04ef1df14eb9aa6234537d473100 : SUCCESS in 9m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8ebc4a0ff9244555bb373cf37502ced9 : SUCCESS in 7m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ad2d5f6ac6864bf3a92fe4f28fb65e5a : SUCCESS in 25m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e6bf46bdc16d4795adb67b363c9a7499 : SUCCESS in 20m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1977492faba24bbb93c62bf910ce8930 : SUCCESS in 14m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a622d68d4c994e52a878315372165884 : SUCCESS in 16m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3d8b85ec7d044d119b62a0648ea9fa85 : SUCCESS in 14m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/64dd6d287b344e76be4e5a003a4f51ef : SUCCESS in 19m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7105d5bf2d1f46dd9d20b6c6a6aa186c : SUCCESS in 18m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1d9b16a71fbc45a784ebeca2403f9bc8 : SUCCESS in 17m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9e911103f6b4455e881b90a932587929 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8974230ed7c644fba8cc51c5fb74f37a : SUCCESS in 21m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fa63101bd42847a7bdcce480a7ae369a : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/26ca14f518ac41dcbb03c6bbd20d3d61 : SUCCESS in 7m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/06e4b6cb86114f35a3db9992c3aaf93f : SUCCESS in 25m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee3e7c278ed94d669b5cc5ff0a55b615 : SUCCESS in 7m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/55a4977d41cd413db173ec9f2d1e32d2 : SUCCESS in 1h 08m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/141e92ddaec449e0a15f83507883e7e1 : SUCCESS in 1h 06m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3d762aae48044dc49f62f8e7f8a4a2f9 : SUCCESS in 52m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c37e9fb77f46490d9ebc70fdb87f63f4 : SUCCESS in 51m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ec918925b8249ff91e343e9afcc8341 : SUCCESS in 18m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07d1dd3a53fd4163ac29139c1612b61e : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d4ee348bf6a420eb025b02d28a1ef2e : SUCCESS in 26m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38998d056c644f4d9bb16de1cc8e2933 : SUCCESS in 53m 31s","accounts_in_message":[],"_revision_number":12},{"id":"e223ddce90c0e8d8cd50ea69f821c09d7dc6148f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-07-14 01:48:56.000000000","message":"Patch Set 12: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"55d9d1ee84cca954d5c9c8f745e09c22f12709a8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-14 09:00:25.000000000","message":"Patch Set 12: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":12},{"id":"207907fa71ee6d19a8f17ae3e8b1ff3f4661df79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-24 15:43:51.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"2d424f035aeefaf488f01107de5596265b7173fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-07-24 15:58:51.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"6cb2a5ad5d4b81264aba803bb64d7690e8b8c7be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 16:23:06.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaabbcf311ba4be6a47803be6ed4cb14\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/28689854a34b416fb0c1365af7757931 : SUCCESS in 17m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff5faa4092744915a63838423ec1808a : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1221b202e34e460ca6e5f021d7a248f6 : SUCCESS in 20m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0c575f28a7c4e8b9ef690acb4be0916 : SUCCESS in 23m 05s","accounts_in_message":[],"_revision_number":14},{"id":"231b075f7c3edec3db056b53e0c237c8a6da9c12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 17:12:16.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0892d899ef04f078b329541ec0d4a4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa5c66ae025a4a14b3fbb82becc754e6 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/dda1e50e263149b5977e8d0e6d533101 : SUCCESS in 1h 00m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/810b981649c548d1adb92ea86935807e : SUCCESS in 48m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a93abed58ae41cabc1d3d964591ddd1 : SUCCESS in 58m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9325ad4378e4416da79e422365aad28f : SUCCESS in 27m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/708e7eeea5fd4485adb2b0b72bceb103 : SUCCESS in 27m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2376379cd054b55a3941224435d7724 : SUCCESS in 20m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/734c9fcff2644c21b771dcac3455e640 : SUCCESS in 9m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a5f36fb56b014471866a6a3b8dcd2d55 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6bd12d276f934f63abd64b73442013d5 : SUCCESS in 26m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5b70ae1c0e6c44fcadfcbba6b56cf418 : SUCCESS in 18m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/64e5137363364d62a50841b7c98367a8 : SUCCESS in 26m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/075fa8eb684349e8bfaa73c491ea9ff6 : SUCCESS in 24m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fdb43bbcbe244c02aef0d48531c3a2b7 : SUCCESS in 16m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2cc2b10327b2413296f8e9a47bb8d9d9 : SUCCESS in 25m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7c2b9a05ec414d0e8d6b999d7e044dc2 : SUCCESS in 25m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d814ce4e330d499d8e0f45870781767e : SUCCESS in 26m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8245de33dbf041e280b1acd89a5e5144 : SUCCESS in 25m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e1cf7e7f1ef14bba919d9606ce1ffba4 : SUCCESS in 23m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/efe2855059284604b3e447bcdbe3a972 : SUCCESS in 16m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e2a3b11a866949628e71185041e7266a : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7dc78503132a4b8bad80ffef82eae954 : SUCCESS in 17m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cf8a94b5ebe64aecbf21f7ffa5f123bb : SUCCESS in 8m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3463620762fd4d5187dea1ca10a224a7 : SUCCESS in 1h 02m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6213d11a900745f3b7c9e9899588a090 : SUCCESS in 1h 06m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9daf3f5159f44049bc2f7aaa6400ee89 : SUCCESS in 55m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0b127051f6bb45c0ad0308a698670e4a : SUCCESS in 50m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99720b5f8cf6420a8825adb5959c8171 : SUCCESS in 17m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01dab83ed6c54f278cefca4224c592ca : SUCCESS in 5m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ab86558cdcf24617a1b08ac1224b089e : SUCCESS in 20m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b059bddb51464793a0fb913c862f9e82 : SUCCESS in 53m 39s","accounts_in_message":[],"_revision_number":14},{"id":"30b025d1e920bcd3e01816d1bd808c376eef7dfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-07-31 19:50:16.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"4398470b0f7d1cb7fc1671d70a3570068119a15b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-07-31 19:51:16.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"20cceff730450804fe0044be407bb71552ab0903","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 20:07:45.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9200ed594234b0d931df90c804e22cd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35f780f258d04ab79065f36a09821886 : SUCCESS in 15m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a376d2fd7f04d92a602566782231f14 : SUCCESS in 13m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e53772b516f24b79a8e9dc0aa82ec191 : SUCCESS in 16m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/27acc3c93b5048cbb6e24a0eefa61b7c : SUCCESS in 16m 10s","accounts_in_message":[],"_revision_number":15},{"id":"feffffe015e99e441fca87bd735a0e7e6b812679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 20:57:43.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35b6c8b537824a469275b797a2cbc1e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e572c4836f4bd5b1ed6255f02abbd6 : SUCCESS in 9m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/1d066ebd89f84329adbc20e139c38d6f : SUCCESS in 59m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b2135a390b754b81a0ae5247227d8949 : SUCCESS in 36m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6fa171bb3aa486b9b1b8e2783325a05 : SUCCESS in 47m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f37844337a5f459e91bf0687c473fc54 : SUCCESS in 18m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/164dc28dea8d45f4952ccc92956376f6 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cabca0b1d01b44a690fc14f2a5c6d4b0 : SUCCESS in 17m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1d70fdf9fddc4ba0a7532f968c2ad9fa : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ddaa8f92e3e14e1f9b2240e79cb9e573 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9a857bcf94a749d2b80cd188eb3f0cb8 : SUCCESS in 17m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e402f88facef4d8091f5037e1b13d77d : SUCCESS in 16m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e2823eff06874993898e2a0b768cd0c2 : SUCCESS in 17m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c8f75c4be17b432095d738aa4a68b3cd : SUCCESS in 18m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1670ff92c3b446e297a020b394bd4af6 : SUCCESS in 21m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a76dd43e0cdf48cb81851a141c06bb2b : SUCCESS in 18m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/290e35171d944b599eaa2b6efb278518 : SUCCESS in 18m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/36f1eac0c1d4496abca03d9c4c0a39d8 : SUCCESS in 18m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/754134ac81cb40118537c50e75deb3ba : SUCCESS in 12m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2ca120ccb3fe427d97106caf60570436 : SUCCESS in 17m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/22394d8be7304b55854537d801b1ad7c : SUCCESS in 14m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3cc87359b30d425f9fbda94b4efc4ba8 : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0a145546abb1484c9529146d968de804 : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/27f682645f354838bd50d6a3c92f6036 : SUCCESS in 7m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2481a3afc3664d0088609a93fe4a8550 : SUCCESS in 55m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/35347addb3c440cea98474ec9b8ac4c9 : SUCCESS in 50m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5611165e566246058bb63658d677f0bc : SUCCESS in 41m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/62a4fc797d774dcd8cd60f8c5ee8eadb : SUCCESS in 36m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fd12a71c3214a6a8aaaa51f8978a279 : SUCCESS in 20m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/711b54bbb4b548999929c27fbd2aa79e : SUCCESS in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d77df64e132341a9a5356e6bdf172c83 : SUCCESS in 21m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f06740e0473f470b9cb34f5a5529f444 : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":15},{"id":"add6c447d97ac79115239f6a095c53185efdddf3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-01 12:13:19.000000000","message":"Patch Set 15: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":15},{"id":"ffa0047ec440c8d559b60b1d84137011b74ae5ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-01 22:59:19.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"b92973cbe0ab9de211123710a635c9b9456807c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 00:05:31.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fe072ecd6694ccb920d3933b3416f43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef5594e35d7649fda846861beeb37c91 : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/5acb924c0c89493db1a4a3398a9a186c : SUCCESS in 43m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/591bbbe981a54dda8913120876b98a9e : SUCCESS in 53m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c21ad7dbac44b70b1c96f07fc4c28ae : SUCCESS in 43m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/caec144a39f3441f989fa6583579616a : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/342aa8ee4fde46ca866ff4f3070af5ce : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/384b1b89d6b443d1b56345dc777be011 : SUCCESS in 21m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9f5825036d7e4462b226b76693c57539 : SUCCESS in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/96955fb2f6ea426892a28b3ef6530bcb : SUCCESS in 4m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1b02de2f999c4280b9e3e3077d38ce49 : SUCCESS in 17m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f0d268a9d4944fcd8dccf5661eb80f37 : SUCCESS in 16m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dd471b3b824d462284ac947a7d377acb : SUCCESS in 16m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/09650480959e43a392618ae19cc7e6bd : SUCCESS in 15m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8d0b070398424ac097899d2e699890df : SUCCESS in 17m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f65efea2e52d4fabaa456d71e1541026 : SUCCESS in 12m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d00a0d05e93d4d6abb19d996bad67af1 : SUCCESS in 13m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e04a3ad911f94568949c203e4d2e2ec8 : SUCCESS in 15m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bc2c9b83071b453397130b355fdb7037 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/710cd7da003b4876af504332340ee5aa : SUCCESS in 14m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0c13d8ac4a6a4874a7a3361b6cbecb3c : SUCCESS in 15m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/39a3e0fa55dd4337afaec9a621dc1ad4 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/10e002aa970348fd80a44e73733d5b60 : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee31ed1a3ff5461ebeb763f1679d045b : SUCCESS in 7m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d4fe8b18b1dc4232973cb68396976f56 : SUCCESS in 1h 01m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4fad972ffac4e35831c159383917a99 : SUCCESS in 52m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a2f1455f5a984028aa465feb8219880d : SUCCESS in 41m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6f37488c0359460a9cf20ffce78f1c53 : SUCCESS in 52m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ced7bd1f1c3482598a0827df7704c13 : SUCCESS in 17m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/885ccee34343428fb346c25d5fc5e411 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/197a2ba8d725412b85fa078336094637 : SUCCESS in 23m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c9ad195b2a8469b91f42086f8e50c95 : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":16},{"id":"005e2745694e9fba156d0556d6c62494678858f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 00:32:33.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/028fe7bfc6ed44b5a77685666d199baf\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b53ef2e45574ed4a9199c02a5fdac4f : SUCCESS in 13m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a32b819b365c4d30bc9b939607a05499 : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7db1df5582614bcfbc61a29e5183197d : SUCCESS in 18m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/18dbfa0032fc462daf9732301fb698ee : TIMED_OUT in 1h 30m 51s","accounts_in_message":[],"_revision_number":16},{"id":"cebe69b9622eedc19a5a6e49d7f745438e7a1db5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-02 10:26:08.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"21ece7eaab01ddb64370c1b2ad459c6009a6e954","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 10:26:45.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"43f4b4ea563f39a3d3416ff8c2d0fed4bffcfd43","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-08-02 10:34:27.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"be8d40abaab431a0479f29158ca3574542c0efb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 11:39:05.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2bdbe7c3b594e24970bfe4adcf3962d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/654f3815d71a4724989d13c04d0f15a8 : SUCCESS in 9m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/09a197b8aa4447119cf837c06148386f : SUCCESS in 58m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/74e1245c0d294569ab5c7cd7bbf561af : SUCCESS in 51m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a3e72e27e52482a84e2650d11c41630 : SUCCESS in 57m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5b578f0957bc4f67aa923f59e5f7e9cb : SUCCESS in 8m 18s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1341098c23f04a4682e65eda8811d391 : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/caaa5ad8a6f148cb8ab26f5265ba2cbd : SUCCESS in 15m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a31e9b2a33c475896f38806db571255 : SUCCESS in 16m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/841e24c8a9e74337b0dc8f41b236ff8d : SUCCESS in 14m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/18194f9a9cd1421aa2fba151f5af6d25 : SUCCESS in 14m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a2352f9576c54f4bb5b06c4758d60480 : SUCCESS in 15m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7bfb93e97b254e18adf86ac1486a0ab1 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c61ef9f2017d41c1b07fc86db0333117 : SUCCESS in 15m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/92e91d63f06a4d2693e9c0e1ae566b3b : SUCCESS in 17m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a5d499ddc65c46aaacb7ac2d17b2498e : SUCCESS in 13m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0cc39cb55d0f4ac49e4ac43f41bb0ff5 : SUCCESS in 25m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ff4cc4fdc8ce4c04a32c090e99e98d18 : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03a18161034e4bca89dde4c009a9e629 : SUCCESS in 7m 14s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/970f29c92ae944a8b0aa700d9e5f5ff9 : SUCCESS in 8m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/409121762b804554b8e8c463c034986e : SUCCESS in 51m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/63df045ac38845ebbcf561251f88beed : SUCCESS in 1h 11m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/414e07f75fe0417da0053ef59f8facf8 : SUCCESS in 51m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2dfccf56de7a46a3b6932400c228ff28 : SUCCESS in 57m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2b329bc07954da0ae94d4674dea8d8c : SUCCESS in 16m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f6c280bfd8941468b5da6e61877e964 : SUCCESS in 5m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a598d99ceec84339a350d688407853f9 : SUCCESS in 50m 07s","accounts_in_message":[],"_revision_number":16},{"id":"6e54be854896321b44f24d36763c8883c770f6a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 11:39:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"adaa1e18689ae6313254646d576674d3d5a8fe29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 11:40:03.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11111f74a7054ecd8bc73dda6e00fc6b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/176375a67304469ebd058fc46718ac70 : SUCCESS in 38s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a848c35b442148889f191a8f82640732 : SUCCESS in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/9a1c1c0520234d67b55312bcd6c4eaa4 : SUCCESS in 27s (non-voting)","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"c51e81f640274db31bf45b5253d8e99d19a7e3d6","revisions":{"ecbf1b82c96bca2485276c5c7f54ee43af54ee04":{"kind":"REWORK","_number":1,"created":"2023-05-31 18:59:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/884931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/1"}}},"commit":{"parents":[{"commit":"e29e2c3ae5fd32f6b8c3d70e6ea4db786f980151","subject":"Move IP-address-related functions out to new module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e29e2c3ae5fd32f6b8c3d70e6ea4db786f980151"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:59:01.000000000","tz":-420},"subject":"proxy: Bring back logging for get_*_info requests","message":"proxy: Bring back logging for get_*_info requests\n\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecbf1b82c96bca2485276c5c7f54ee43af54ee04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecbf1b82c96bca2485276c5c7f54ee43af54ee04"}]},"branch":"refs/heads/master"},"cbd8c41b273d92806f2757163b05ec6f8446ce59":{"kind":"REWORK","_number":2,"created":"2023-05-31 19:21:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/884931/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/2"}}},"commit":{"parents":[{"commit":"e29e2c3ae5fd32f6b8c3d70e6ea4db786f980151","subject":"Move IP-address-related functions out to new module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e29e2c3ae5fd32f6b8c3d70e6ea4db786f980151"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 19:21:11.000000000","tz":-420},"subject":"proxy: Bring back logging/metrics for get_*_info requests","message":"proxy: Bring back logging/metrics for get_*_info requests\n\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbd8c41b273d92806f2757163b05ec6f8446ce59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbd8c41b273d92806f2757163b05ec6f8446ce59"}]},"branch":"refs/heads/master"},"79d48ca538ad21237340515576437e3eedd7ef37":{"kind":"REWORK","_number":3,"created":"2023-06-02 16:40:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/884931/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/3"}}},"commit":{"parents":[{"commit":"e29e2c3ae5fd32f6b8c3d70e6ea4db786f980151","subject":"Move IP-address-related functions out to new module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e29e2c3ae5fd32f6b8c3d70e6ea4db786f980151"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-02 15:44:32.000000000","tz":-420},"subject":"proxy: Bring back logging/metrics for get_*_info requests","message":"proxy: Bring back logging/metrics for get_*_info requests\n\nA while back, we changed get_account_info and get_container_info to call\nthe proxy-server app directly, rather than whatever was right of the current\nmiddleware. This both reduced backend request amplification on cache misses\nand removed the possibility that some 3rd party middleware could taint the\ninfo results.\n\nHowever, it *also* meant that we stopped emitting logs or metrics in the\nproxy for these backend requests. This was an unfortunate and unintended\nbreak from earlier behavior.\n\nNow, extend the middleware decorating we\u0027re doing in loadapp() to include\na \"logged app\", i.e., the app wrapped by it\u0027s right-most proxy-logging\nmiddleware. If there is not logging middleware (such as would happen for\nthe backend servers), the \"logged app\" will be the main app. Make account\nand container info requests through *that* app, so we get logging and\nmetrics again.\n\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79d48ca538ad21237340515576437e3eedd7ef37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/79d48ca538ad21237340515576437e3eedd7ef37"}]},"branch":"refs/heads/master"},"ba89a93a5e14e58fb6808fc976bcd249eaedbb2b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-06-02 16:42:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/884931/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/4"}}},"commit":{"parents":[{"commit":"e29e2c3ae5fd32f6b8c3d70e6ea4db786f980151","subject":"Move IP-address-related functions out to new module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e29e2c3ae5fd32f6b8c3d70e6ea4db786f980151"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-02 16:40:51.000000000","tz":-420},"subject":"proxy: Bring back logging/metrics for get_*_info requests","message":"proxy: Bring back logging/metrics for get_*_info requests\n\nA while back, we changed get_account_info and get_container_info to\ncall the proxy-server app directly, rather than whatever was right\nof the current middleware. This both reduced backend request\namplification on cache misses and removed the possibility that some\n3rd party middleware could taint the info results.\n\nHowever, it *also* meant that we stopped emitting logs or metrics in\nthe proxy for these backend requests. This was an unfortunate and\nunintended break from earlier behavior.\n\nNow, extend the middleware decorating we\u0027re doing in loadapp() to\ninclude a \"logged app\", i.e., the app wrapped by it\u0027s right-most\nproxy-logging middleware. If there is not logging middleware (such\nas would happen for the backend servers), the \"logged app\" will be\nthe main app. Make account and container info requests through\n*that* app, so we get logging and metrics again.\n\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba89a93a5e14e58fb6808fc976bcd249eaedbb2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba89a93a5e14e58fb6808fc976bcd249eaedbb2b"}]},"branch":"refs/heads/master"},"ff24428f417cbf0bc71ce54a6ec9528ff618485b":{"kind":"REWORK","_number":5,"created":"2023-06-05 22:50:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/884931/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/5"}}},"commit":{"parents":[{"commit":"eb3fe260d51d021674b2bccc4b9d1b4cc0757450","subject":"Merge \"memcached: add method and time_spent to exception logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb3fe260d51d021674b2bccc4b9d1b4cc0757450"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-05 22:49:14.000000000","tz":-420},"subject":"proxy: Bring back logging/metrics for get_*_info requests","message":"proxy: Bring back logging/metrics for get_*_info requests\n\nA while back, we changed get_account_info and get_container_info to\ncall the proxy-server app directly, rather than whatever was right\nof the current middleware. This both reduced backend request\namplification on cache misses and removed the possibility that some\n3rd party middleware could taint the info results.\n\nHowever, it *also* meant that we stopped emitting logs or metrics in\nthe proxy for these backend requests. This was an unfortunate and\nunintended break from earlier behavior.\n\nNow, extend the middleware decorating we\u0027re doing in loadapp() to\ninclude a \"logged app\", i.e., the app wrapped by it\u0027s right-most\nproxy-logging middleware. If there is not logging middleware (such\nas would happen for the backend servers), the \"logged app\" will be\nthe main app. Make account and container info requests through\n*that* app, so we get logging and metrics again.\n\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff24428f417cbf0bc71ce54a6ec9528ff618485b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff24428f417cbf0bc71ce54a6ec9528ff618485b"}]},"branch":"refs/heads/master"},"ff478f78f324691f8129f480f7df1e2a257ee7a0":{"kind":"REWORK","_number":6,"created":"2023-06-06 04:12:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/884931/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/6"}}},"commit":{"parents":[{"commit":"eb3fe260d51d021674b2bccc4b9d1b4cc0757450","subject":"Merge \"memcached: add method and time_spent to exception logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb3fe260d51d021674b2bccc4b9d1b4cc0757450"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-06 01:54:36.000000000","tz":-420},"subject":"proxy: Bring back logging/metrics for get_*_info requests","message":"proxy: Bring back logging/metrics for get_*_info requests\n\nA while back, we changed get_account_info and get_container_info to\ncall the proxy-server app directly, rather than whatever was right\nof the current middleware. This both reduced backend request\namplification on cache misses and removed the possibility that some\n3rd party middleware could taint the info results.\n\nHowever, it *also* meant that we stopped emitting logs or metrics in\nthe proxy for these backend requests. This was an unfortunate and\nunintended break from earlier behavior.\n\nNow, extend the middleware decorating we\u0027re doing in loadapp() to\ninclude a \"logged app\", i.e., the app wrapped by it\u0027s right-most\nproxy-logging middleware. If there is not logging middleware (such\nas would happen for the backend servers), the \"logged app\" will be\nthe main app. Make account and container info requests through\n*that* app, so we get logging and metrics again.\n\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff478f78f324691f8129f480f7df1e2a257ee7a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff478f78f324691f8129f480f7df1e2a257ee7a0"}]},"branch":"refs/heads/master"},"24f183c63aa4853ede08161b9b0d8a07f6fc8b00":{"kind":"REWORK","_number":7,"created":"2023-06-12 18:35:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/884931/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/7"}}},"commit":{"parents":[{"commit":"eb3fe260d51d021674b2bccc4b9d1b4cc0757450","subject":"Merge \"memcached: add method and time_spent to exception logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb3fe260d51d021674b2bccc4b9d1b4cc0757450"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-12 18:35:41.000000000","tz":-420},"subject":"proxy: Bring back logging/metrics for get_*_info requests","message":"proxy: Bring back logging/metrics for get_*_info requests\n\nA while back, we changed get_account_info and get_container_info to\ncall the proxy-server app directly, rather than whatever was right\nof the current middleware. This both reduced backend request\namplification on cache misses and removed the possibility that some\n3rd party middleware could taint the info results.\n\nHowever, it *also* meant that we stopped emitting logs or metrics in\nthe proxy for these backend requests. This was an unfortunate and\nunintended break from earlier behavior.\n\nNow, extend the middleware decorating we\u0027re doing in loadapp() to\ninclude a \"logged app\", i.e., the app wrapped by it\u0027s right-most\nproxy-logging middleware. If there is not logging middleware (such\nas would happen for the backend servers), the \"logged app\" will be\nthe main app. Make account and container info requests through\n*that* app, so we get logging and metrics again.\n\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24f183c63aa4853ede08161b9b0d8a07f6fc8b00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24f183c63aa4853ede08161b9b0d8a07f6fc8b00"}]},"branch":"refs/heads/master"},"6481cb4ce9e18be7bd279ab50eadfa76863283ac":{"kind":"REWORK","_number":8,"created":"2023-06-14 11:49:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/884931/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/8"}}},"commit":{"parents":[{"commit":"406bbb6abd684e3b168c21b2ac8e7766f0fcd73a","subject":"Merge \"proxy: Remove unused arg from set/clear_info_cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/406bbb6abd684e3b168c21b2ac8e7766f0fcd73a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-14 11:49:35.000000000","tz":60},"subject":"proxy: Bring back logging/metrics for get_*_info requests","message":"proxy: Bring back logging/metrics for get_*_info requests\n\nA while back, we changed get_account_info and get_container_info to\ncall the proxy-server app directly, rather than whatever was right\nof the current middleware. This both reduced backend request\namplification on cache misses and removed the possibility that some\n3rd party middleware could taint the info results.\n\nHowever, it *also* meant that we stopped emitting logs or metrics in\nthe proxy for these backend requests. This was an unfortunate and\nunintended break from earlier behavior.\n\nNow, extend the middleware decorating we\u0027re doing in loadapp() to\ninclude a \"logged app\", i.e., the app wrapped by it\u0027s right-most\nproxy-logging middleware. If there is not logging middleware (such\nas would happen for the backend servers), the \"logged app\" will be\nthe main app. Make account and container info requests through\n*that* app, so we get logging and metrics again.\n\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6481cb4ce9e18be7bd279ab50eadfa76863283ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6481cb4ce9e18be7bd279ab50eadfa76863283ac"}]},"branch":"refs/heads/master"},"3ca406a39b5fead8d6f7c48d82eda38c39ae071f":{"kind":"REWORK","_number":9,"created":"2023-06-14 14:26:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/884931/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/9"}}},"commit":{"parents":[{"commit":"406bbb6abd684e3b168c21b2ac8e7766f0fcd73a","subject":"Merge \"proxy: Remove unused arg from set/clear_info_cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/406bbb6abd684e3b168c21b2ac8e7766f0fcd73a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-14 14:25:55.000000000","tz":60},"subject":"proxy: Bring back logging/metrics for get_*_info requests","message":"proxy: Bring back logging/metrics for get_*_info requests\n\nA while back, we changed get_account_info and get_container_info to\ncall the proxy-server app directly, rather than whatever was right\nof the current middleware. This both reduced backend request\namplification on cache misses and removed the possibility that some\n3rd party middleware could taint the info results.\n\nHowever, it *also* meant that we stopped emitting logs or metrics in\nthe proxy for these backend requests. This was an unfortunate and\nunintended break from earlier behavior.\n\nNow, extend the middleware decorating we\u0027re doing in loadapp() to\ninclude a \"logged app\", i.e., the app wrapped by it\u0027s right-most\nproxy-logging middleware. If there is not logging middleware (such\nas would happen for the backend servers), the \"logged app\" will be\nthe main app. Make account and container info requests through\n*that* app, so we get logging and metrics again.\n\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ca406a39b5fead8d6f7c48d82eda38c39ae071f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ca406a39b5fead8d6f7c48d82eda38c39ae071f"}]},"branch":"refs/heads/master"},"167d3b5eaf8f92975f932af2ea544b434db247da":{"kind":"REWORK","_number":10,"created":"2023-06-14 16:23:35.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/884931/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/10"}}},"commit":{"parents":[{"commit":"406bbb6abd684e3b168c21b2ac8e7766f0fcd73a","subject":"Merge \"proxy: Remove unused arg from set/clear_info_cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/406bbb6abd684e3b168c21b2ac8e7766f0fcd73a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-14 16:23:26.000000000","tz":60},"subject":"proxy: Bring back logging/metrics for get_*_info requests","message":"proxy: Bring back logging/metrics for get_*_info requests\n\nA while back, we changed get_account_info and get_container_info to\ncall the proxy-server app directly, rather than whatever was right\nof the current middleware. This both reduced backend request\namplification on cache misses and removed the possibility that some\n3rd party middleware could taint the info results.\n\nHowever, it *also* meant that we stopped emitting logs or metrics in\nthe proxy for these backend requests. This was an unfortunate and\nunintended break from earlier behavior.\n\nNow, extend the middleware decorating we\u0027re doing in loadapp() to\ninclude a \"logged app\", i.e., the app wrapped by it\u0027s right-most\nproxy-logging middleware. If there is not logging middleware (such\nas would happen for the backend servers), the \"logged app\" will be\nthe main app. Make account and container info requests through\n*that* app, so we get logging and metrics again.\n\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/167d3b5eaf8f92975f932af2ea544b434db247da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/167d3b5eaf8f92975f932af2ea544b434db247da"}]},"branch":"refs/heads/master"},"1e9b477eadd87eebc6ef87e9f6fa86ee92821d1d":{"kind":"REWORK","_number":11,"created":"2023-06-21 17:11:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/884931/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/11"}}},"commit":{"parents":[{"commit":"406bbb6abd684e3b168c21b2ac8e7766f0fcd73a","subject":"Merge \"proxy: Remove unused arg from set/clear_info_cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/406bbb6abd684e3b168c21b2ac8e7766f0fcd73a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-21 17:08:25.000000000","tz":-420},"subject":"proxy: Bring back logging/metrics for get_*_info requests","message":"proxy: Bring back logging/metrics for get_*_info requests\n\nA while back, we changed get_account_info and get_container_info to\ncall the proxy-server app directly, rather than whatever was right\nof the current middleware. This reduced backend request amplification\non cache misses.\n\nHowever, it *also* meant that we stopped emitting logs or metrics in\nthe proxy for these backend requests. This was an unfortunate and\nunintended break from earlier behavior.\n\nNow, extend the middleware decorating we\u0027re doing in loadapp() to\ninclude a \"logged app\", i.e., the app wrapped by it\u0027s right-most\nproxy-logging middleware. If there is not logging middleware (such\nas would happen for the backend servers), the \"logged app\" will be\nthe main app. Make account and container info requests through\n*that* app, so we get logging and metrics again.\n\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e9b477eadd87eebc6ef87e9f6fa86ee92821d1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e9b477eadd87eebc6ef87e9f6fa86ee92821d1d"}]},"branch":"refs/heads/master"},"783c980c3e2e3431e7e6d3a4c2e5b280c43c92c7":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2023-07-13 18:23:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/884931/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/12"}}},"commit":{"parents":[{"commit":"406bbb6abd684e3b168c21b2ac8e7766f0fcd73a","subject":"Merge \"proxy: Remove unused arg from set/clear_info_cache\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/406bbb6abd684e3b168c21b2ac8e7766f0fcd73a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2023-07-13 18:23:16.000000000","tz":0},"subject":"proxy: Bring back logging/metrics for get_*_info requests","message":"proxy: Bring back logging/metrics for get_*_info requests\n\nA while back, we changed get_account_info and get_container_info to\ncall the proxy-server app directly, rather than whatever was right\nof the current middleware. This reduced backend request amplification\non cache misses.\n\nHowever, it *also* meant that we stopped emitting logs or metrics in\nthe proxy for these backend requests. This was an unfortunate and\nunintended break from earlier behavior.\n\nNow, extend the middleware decorating we\u0027re doing in loadapp() to\ninclude a \"logged app\", i.e., the app wrapped by it\u0027s right-most\nproxy-logging middleware. If there is not logging middleware (such\nas would happen for the backend servers), the \"logged app\" will be\nthe main app. Make account and container info requests through\n*that* app, so we get logging and metrics again.\n\nCloses-Bug: #2027726\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/783c980c3e2e3431e7e6d3a4c2e5b280c43c92c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/783c980c3e2e3431e7e6d3a4c2e5b280c43c92c7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"88c9c3bc0b4a4f7f3c9c500e4f5962bdebab8dc3":{"kind":"REWORK","_number":13,"created":"2023-07-24 15:43:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/884931/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/13"}}},"commit":{"parents":[{"commit":"740636b8e42bf30fb020e40d492c981162686aef","subject":"Merge \"Add more unit tests for ranged GETs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/740636b8e42bf30fb020e40d492c981162686aef"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-24 14:53:26.000000000","tz":60},"subject":"proxy: Bring back logging/metrics for get_*_info requests","message":"proxy: Bring back logging/metrics for get_*_info requests\n\nA while back, we changed get_account_info and get_container_info to\ncall the proxy-server app directly, rather than whatever was right\nof the current middleware. This both reduced backend request\namplification on cache misses and removed the possibility that some\n3rd party middleware could taint the info results.\n\nHowever, it *also* meant that we stopped emitting logs or metrics in\nthe proxy for these backend requests. This was an unfortunate and\nunintended break from earlier behavior.\n\nNow, extend the middleware decorating we\u0027re doing in loadapp() to\ninclude a \"logged app\", i.e., the app wrapped by it\u0027s right-most\nproxy-logging middleware. If there is not logging middleware (such\nas would happen for the backend servers), the \"logged app\" will be\nthe main app. Make account and container info requests through\n*that* app, so we get logging and metrics again.\n\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88c9c3bc0b4a4f7f3c9c500e4f5962bdebab8dc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/88c9c3bc0b4a4f7f3c9c500e4f5962bdebab8dc3"}]},"branch":"refs/heads/master"},"d2c9c5002169723266f4100b6d6d2aa03f89c555":{"kind":"REWORK","_number":14,"created":"2023-07-24 15:58:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/884931/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/14"}}},"commit":{"parents":[{"commit":"740636b8e42bf30fb020e40d492c981162686aef","subject":"Merge \"Add more unit tests for ranged GETs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/740636b8e42bf30fb020e40d492c981162686aef"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-07-24 15:52:59.000000000","tz":60},"subject":"proxy: Bring back logging/metrics for get_*_info requests","message":"proxy: Bring back logging/metrics for get_*_info requests\n\nA while back, we changed get_account_info and get_container_info to\ncall the proxy-server app directly, rather than whatever was right\nof the current middleware. This reduced backend request amplification\non cache misses.\n\nHowever, it *also* meant that we stopped emitting logs or metrics in\nthe proxy for these backend requests. This was an unfortunate and\nunintended break from earlier behavior.\n\nNow, extend the middleware decorating we\u0027re doing in loadapp() to\ninclude a \"logged app\", i.e., the app wrapped by it\u0027s right-most\nproxy-logging middleware. If there is not logging middleware (such\nas would happen for the backend servers), the \"logged app\" will be\nthe main app. Make account and container info requests through\n*that* app, so we get logging and metrics again.\n\nCloses-Bug: #2027726\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2c9c5002169723266f4100b6d6d2aa03f89c555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2c9c5002169723266f4100b6d6d2aa03f89c555"}]},"branch":"refs/heads/master"},"6d82511e3fc81577bf7ef3eb812ee0ace820b2d0":{"kind":"REWORK","_number":15,"created":"2023-07-31 19:50:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/884931/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/15"}}},"commit":{"parents":[{"commit":"740636b8e42bf30fb020e40d492c981162686aef","subject":"Merge \"Add more unit tests for ranged GETs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/740636b8e42bf30fb020e40d492c981162686aef"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-07-31 19:49:35.000000000","tz":-420},"subject":"proxy: Bring back logging/metrics for get_*_info requests","message":"proxy: Bring back logging/metrics for get_*_info requests\n\nA while back, we changed get_account_info and get_container_info to\ncall the proxy-server app directly, rather than whatever was right\nof the current middleware. This reduced backend request amplification\non cache misses.\n\nHowever, it *also* meant that we stopped emitting logs or metrics in\nthe proxy for these backend requests. This was an unfortunate and\nunintended break from earlier behavior.\n\nNow, extend the middleware decorating we\u0027re doing in loadapp() to\ninclude a \"logged app\", i.e., the app wrapped by it\u0027s right-most\nproxy-logging middleware. If there is not logging middleware (such\nas would happen for the backend servers), the \"logged app\" will be\nthe main app. Make account and container info requests through\n*that* app, so we get logging and metrics again.\n\nCloses-Bug: #2027726\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d82511e3fc81577bf7ef3eb812ee0ace820b2d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d82511e3fc81577bf7ef3eb812ee0ace820b2d0"}]},"branch":"refs/heads/master"},"c51e81f640274db31bf45b5253d8e99d19a7e3d6":{"kind":"REWORK","_number":16,"created":"2023-08-01 22:59:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/884931/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/884931/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/884931/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/884931/16"}}},"commit":{"parents":[{"commit":"8a5baf8e1316f39b169b3cb60cd243994f335a3b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a5baf8e1316f39b169b3cb60cd243994f335a3b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 18:49:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-01 22:58:58.000000000","tz":-420},"subject":"proxy: Bring back logging/metrics for get_*_info requests","message":"proxy: Bring back logging/metrics for get_*_info requests\n\nA while back, we changed get_account_info and get_container_info to\ncall the proxy-server app directly, rather than whatever was right\nof the current middleware. This reduced backend request amplification\non cache misses.\n\nHowever, it *also* meant that we stopped emitting logs or metrics in\nthe proxy for these backend requests. This was an unfortunate and\nunintended break from earlier behavior.\n\nNow, extend the middleware decorating we\u0027re doing in loadapp() to\ninclude a \"logged app\", i.e., the app wrapped by it\u0027s right-most\nproxy-logging middleware. If there is not logging middleware (such\nas would happen for the backend servers), the \"logged app\" will be\nthe main app. Make account and container info requests through\n*that* app, so we get logging and metrics again.\n\nCloses-Bug: #2027726\nRelated-Change: I49447c62abf9375541f396f984c91e128b8a05d5\nChange-Id: I3f531f904340e4c8407185ed64b41d7d614a308a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c51e81f640274db31bf45b5253d8e99d19a7e3d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c51e81f640274db31bf45b5253d8e99d19a7e3d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
