)]}'
{"id":"openstack%2Fswift~884954","triplet_id":"openstack%2Fswift~master~I61c7780a84a1f0cee6975da67d08417cf6aa4ea2","project":"openstack/swift","branch":"master","topic":"bug2020667","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2023-08-04 21:20:29.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-08-04 21:20:29.000000000","reason":"Change was submitted"},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2023-08-04 21:20:29.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-08-04 17:37:36.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I61c7780a84a1f0cee6975da67d08417cf6aa4ea2","subject":"Add non-ascii meta values to ssync probe test","status":"MERGED","created":"2023-05-31 23:07:48.000000000","updated":"2023-08-04 21:21:31.000000000","submitted":"2023-08-04 21:20:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"884954-bug2020667","meta_rev_id":"0d5ac608656fcd4a2f6aef26d1d35e112a036816","_number":884954,"virtual_id_number":884954,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-04 21:20:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-03 19:36:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-03 19:36:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-01 00:02:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-01 00:17:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-21 21:13:31.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2023-06-22 14:39:15.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2023-06-26 07:50:03.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"852e9478fee7fb71713212d1115b95c31d677045","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-31 23:07:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"396c9d8191cb1d1c1895d621523d248fbe11f737","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 00:02:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3672b4ae1c3f4490973dcc74b0b70bd1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/865e775064b145f7a60e6e1af7a7f4bd : SUCCESS in 17m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/089dc2d2e8e34987b9e95f8af48245ae : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5e68f21bce4479391d670dc319faaa3 : SUCCESS in 19m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc3ab76ec0a14c8ea90f553da4693975 : SUCCESS in 52m 18s","accounts_in_message":[],"_revision_number":1},{"id":"58d9dfec4435d9cfacba20afb6074d11390b3399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 00:17:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/83df23c315c448c8a71b1c33bc3b56dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0350f907b3514bbc98b77aef4b1a05b7 : SUCCESS in 7m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5e57eef37ea04ab89c26c92305e01a2e : FAILURE in 47m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/01f539ab8f674997a678c10fb6cc2763 : SUCCESS in 1h 03m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f53f3aa8046e415daf60941d4cc3db88 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"3b85a29537ec8ad3d82949f1ca3315fb64b459c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-01 16:19:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"73ff2c7212675447320cd67222a9cf68630dc869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 17:32:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef1de616c46b4f3fb9b5c2158d3c388b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a7752aeed9c4a4e8a5dac2905a45599 : SUCCESS in 10m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8e39fb2909ce47d5ab934d7787241455 : FAILURE in 59m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/45856b97213f42ad8c4ed3c47f5b52e9 : SUCCESS in 55m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68e7763a26644b7f8367024d0f5651f3 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":2},{"id":"20123f5e0d240a996399b3c616bcb6ab18bf77a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 17:42:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7993504063ba4256a00697ea1b35c267\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7bd75338672f415e839b04cc5eb1d357 : SUCCESS in 1h 20m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a65ba087ca24f51a6ff0fa271317c7a : SUCCESS in 14m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1eb452b059fd4233af8d9147e1badfcc : SUCCESS in 17m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ea027de426a40ba90b773e02d1be7f0 : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":2},{"id":"99393a1a88493a99404f35d85e0c0fb9e5ebf2f8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-01 19:47:34.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5fd5604ce94fadf71be6a389f626d3b0ab8d6a01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-01 20:38:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"10234dd0332ba5da5f543230c76116186cfa9f50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 21:43:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e9cd8e49b07404396d5e6d8b5516592\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed370cf403974050bd15df63c88f1655 : SUCCESS in 11m 09s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3aff2b0ca6a34d03b0ffe7c144d76514 : SUCCESS in 48m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f9bc64a5f2ae4a558e7266ed07a3fff6 : SUCCESS in 1h 03m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f7f435d4e94a0d92669d2fdd0a6f29 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":3},{"id":"001ec5f8d98912800f850465a4203d204e726835","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 22:12:33.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/432e472282f24ffca6452e81cc83a07d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f40f1cfdf334950a9aa9418b578c7e9 : SUCCESS in 14m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d04be855a90400bb5058dcdf5daec69 : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2bb427aac17043cf8017710c0612411a : SUCCESS in 19m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b4b455d477d478083ba1e968e9b2638 : SUCCESS in 17m 50s","accounts_in_message":[],"_revision_number":3},{"id":"824768046b8c8f885155bd28e42c7c583db176ad","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-06-21 21:13:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"644057df3a8c1a0d348115ad0374c86841ecc00b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-22 14:39:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b6e37d57a8a509fdf53b220cb67828d2f577f01c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-23 21:23:46.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"20857d64a88e46b645dac77744112fa04fcee25f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-23 21:53:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d888beae02c2bf07fbe9f6640eefe4cfd5d99c27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 22:33:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0e192a3013d4d5ab55ece2d771e1407\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9580c37b41e54543947f579e288374fc : SUCCESS in 8m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/94b0b741f30a46489a76ca5d4410074a : SUCCESS in 54m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/953159c4901945b19c010cade2c6367c : SUCCESS in 1h 09m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b28960046e334308ada88ad09f10f713 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":4},{"id":"8c562b6131cce77d7b03f4de782d134c4d0cec00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 22:48:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac74e3a0ee9a4052aad502b96095aa47\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca1aa254df4f4a468cd4de514a25ffac : SUCCESS in 14m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/018bfff6c0a340a1bda607918ed25cc8 : SUCCESS in 1h 23m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/da3ef67df20749e5b03b23e36f6528c1 : SUCCESS in 21m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1eff99549f0d4d33903df6b0f278d123 : SUCCESS in 17m 11s","accounts_in_message":[],"_revision_number":4},{"id":"49d1f67cdab30975e0058837f2eecfced1d385ab","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-06-26 07:50:03.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"08d5cc22db02728ec5ba7f1a9837d88f2ca90f6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-03 19:34:23.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"73b08f17f117d1d0683a31c0a4cfa27c51fd9b44","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-03 19:36:14.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"802f2f2101b89ea93b6e39f8f762e125c9256bf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 20:44:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07bbac7e1b0d4a229eed7f1ef53cd3fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b6a561036b34bda85b8226c2c28e809 : SUCCESS in 11m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1885ed4ddc934551b3569279a415f72a : SUCCESS in 53m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/161e23d855a747b3b39977cb368739b9 : SUCCESS in 1h 03m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb132f856ecf46a6b41ddac5623aa318 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":5},{"id":"d9c0320a7346ff29d7621de66eaf0e68ad0e224d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 20:45:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4469ee723425860638f0ca54111a47760cde92b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 21:07:23.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e30ce0b8a48e4be7a50987cb0cec370b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3bb7474037374887989b7e2f82bf9593 : SUCCESS in 15m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78e354c36e1649a0b9f572ba7ab16d1f : TIMED_OUT in 1h 30m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc69b45474734951a787ea5a0f378afb : SUCCESS in 19m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8dae658e45bb42a58b39d587bd9b37d4 : TIMED_OUT in 1h 30m 53s","accounts_in_message":[],"_revision_number":5},{"id":"539ddb5fcc947c810d73a3d133564e354f07eb38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 03:33:19.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb7c77613beb415d8b23e41802a077e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d38d845c223f414889333598b4bdf858 : SUCCESS in 9m 27s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c4dcccb03b2549509ac40c445c71a81a : SUCCESS in 18m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b1cf9fbfe5c6465aa023e627a87ddbae : SUCCESS in 38m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dd2334a1c69a4277b8fece3a16fbe87f : SUCCESS in 13m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ec865aa943c5447abe290029ccdb58e1 : SUCCESS in 18m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/40dca2b0e63c40d8b31b99113de22184 : SUCCESS in 16m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b05444d11eae43feb3ec0fc7778ff1f3 : SUCCESS in 21m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/41ae0b5e62494503899198b6ae68a031 : SUCCESS in 18m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/20d3d4b5202e401cb317dfe13e0af178 : SUCCESS in 18m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/56d8ec12d5e7409db69d29b8409d5ceb : SUCCESS in 18m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/148f99c54c534e94a5e981a5c47cf21e : SUCCESS in 21m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/96e39404f5234bde8f8b15bb52a8dfd1 : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a802fe3f301d497b80c93c9d5d4c000d : SUCCESS in 7m 25s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/639fb19d89f34ff781b0894756b970fe : FAILURE in 11m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2df357ff38ce43a0bff31db7792b8602 : SUCCESS in 54m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/566541d3faf6417db45154340514f26f : SUCCESS in 59m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73392e21ec1a4d0cb585f08a9c11f105 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":5},{"id":"31de625c11ad43189ccb7bfa47cb12ba0646cc94","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-04 17:37:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"927290e539399f42cc4a38f11521586fb156adf0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 17:59:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbc51deb047041f694628bd2fdaedc1c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a990e7785fec40b1856e608dabbee2f8 : SUCCESS in 15m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2cd40d6e5339499e99a2653b1f632080 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/37336e6b33fb458cbce3050d8812eb1e : SUCCESS in 19m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ccae728799e445e9cc68ee5acc1b21a : SUCCESS in 20m 26s","accounts_in_message":[],"_revision_number":5},{"id":"f33e4fb00dc584a192400d49ceaed14c3202a8bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 18:46:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9828c4dcff8c4108b9868e8b12075f8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d4956dcf0ae4a4dadbdd3afe2547f1e : SUCCESS in 6m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a19bcc66ea914019ac429ba79427a2d7 : SUCCESS in 1h 02m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d454a079b33b492b84c1c56e98d230a1 : SUCCESS in 1h 01m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aa45cba77524f2186b838813ecce728 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":5},{"id":"ea10f825aacae80fc95851b2876740c695c508b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 18:46:37.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ce35a4027dae0f79fac574921fc7bc7d000036c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 21:20:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0396defeee4e4382bf0103f719b294df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/311d0f2b79d44ad58d1cbd52891a5d9a : SUCCESS in 8m 22s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f626f3c4a01b4a12923aba3816565b2d : SUCCESS in 25m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d82474762d53407e9d81972af0cc0da2 : SUCCESS in 21m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3d3a404a475d4fb89c34a6ad5f51211e : SUCCESS in 20m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2045e00046cd4b7182e4a87b5748f8bf : SUCCESS in 20m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1a230d02766b46d1bfc4ff7a3b9237e2 : SUCCESS in 13m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f9396a34273c43e29344a30c574447b0 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/48dfa23f732f4a5583f35e4dc13ab8da : SUCCESS in 16m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/47cfadf883804a88aff630f0062ab866 : SUCCESS in 17m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/80210dfe651f4795b7250197a3f7072e : SUCCESS in 19m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5e856e53267147389549acb7796c41db : SUCCESS in 13m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/30f8bc2c7dcc4f688363eab3f218d976 : SUCCESS in 29m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e1f96e4aed0943b18b480d626af11ecc : SUCCESS in 7m 35s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/058fdb53cede4de49f78356ef0b0e6a7 : SUCCESS in 7m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2b413c41f1a9465e8f693169126eab66 : SUCCESS in 1h 03m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4599614005be40f8b7ada49e2af50148 : SUCCESS in 57m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2142c24744344644a0d801ad2d6e78bd : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":5},{"id":"7a85bf1ae16acc2c7f0eb6e9a1a61c15c9eb4b59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 21:20:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"0d5ac608656fcd4a2f6aef26d1d35e112a036816","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 21:21:31.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b77437b86ff04c2a826d6d59e36635c5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbacd91e7d054ae8abed327c1d56d628 : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8bad415a77a54b7598e1af90911d9787 : FAILURE in 26s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c9fccf2334ce4f9490934793f9955bec : FAILURE in 26s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"cb468840b9c14ec3de748d9721685e507db446e8","revisions":{"75fa5e5bea408e647ab2676dfab65308345b72f6":{"kind":"REWORK","_number":1,"created":"2023-05-31 23:07:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/884954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/884954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/884954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/884954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/884954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/884954/1"}}},"commit":{"parents":[{"commit":"1e167471ea4c4d7e28e6fd1378d7c7d3d058c96f","subject":"Encode header in latin-1 with wsgi_to_bytes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e167471ea4c4d7e28e6fd1378d7c7d3d058c96f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 23:04:06.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 23:06:55.000000000","tz":-420},"subject":"Add non-ascii meta to ssync probe test","message":"Add non-ascii meta to ssync probe test\n\nChange-Id: I61c7780a84a1f0cee6975da67d08417cf6aa4ea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75fa5e5bea408e647ab2676dfab65308345b72f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75fa5e5bea408e647ab2676dfab65308345b72f6"}]},"branch":"refs/heads/master"},"b389de3677624189710bda33632d13f38e0c18d7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-01 16:19:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/884954/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/884954/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/884954/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/884954/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/884954/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/884954/2"}}},"commit":{"parents":[{"commit":"e6179e98211a9488a5581f50427df9bdd647c783","subject":"Encode header in latin-1 with wsgi_to_bytes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6179e98211a9488a5581f50427df9bdd647c783"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 23:04:06.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-01 16:19:46.000000000","tz":-420},"subject":"Add non-ascii meta to ssync probe test","message":"Add non-ascii meta to ssync probe test\n\nChange-Id: I61c7780a84a1f0cee6975da67d08417cf6aa4ea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b389de3677624189710bda33632d13f38e0c18d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b389de3677624189710bda33632d13f38e0c18d7"}]},"branch":"refs/heads/master"},"78ab12dc0684630798d023aed51bebb2c64a07d2":{"kind":"REWORK","_number":3,"created":"2023-06-01 20:38:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/884954/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/884954/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/884954/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/884954/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/884954/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/884954/3"}}},"commit":{"parents":[{"commit":"e6179e98211a9488a5581f50427df9bdd647c783","subject":"Encode header in latin-1 with wsgi_to_bytes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6179e98211a9488a5581f50427df9bdd647c783"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 23:04:06.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-01 20:38:14.000000000","tz":-420},"subject":"Add non-ascii meta to ssync probe test","message":"Add non-ascii meta to ssync probe test\n\nChange-Id: I61c7780a84a1f0cee6975da67d08417cf6aa4ea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78ab12dc0684630798d023aed51bebb2c64a07d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78ab12dc0684630798d023aed51bebb2c64a07d2"}]},"branch":"refs/heads/master"},"ce1f636e59758857a5946c03bacc19faf86e5e0d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-06-23 21:23:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/884954/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/884954/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/884954/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/884954/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/884954/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/884954/4"}}},"commit":{"parents":[{"commit":"28172879f193afb96017ebd44a41d1f43cb6e2c2","subject":"Encode header in latin-1 with wsgi_to_bytes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28172879f193afb96017ebd44a41d1f43cb6e2c2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 23:04:06.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-23 21:22:45.000000000","tz":-420},"subject":"Add non-ascii meta to ssync probe test","message":"Add non-ascii meta to ssync probe test\n\nChange-Id: I61c7780a84a1f0cee6975da67d08417cf6aa4ea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce1f636e59758857a5946c03bacc19faf86e5e0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce1f636e59758857a5946c03bacc19faf86e5e0d"}]},"branch":"refs/heads/master"},"cb468840b9c14ec3de748d9721685e507db446e8":{"kind":"REWORK","_number":5,"created":"2023-08-03 19:34:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/884954/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/884954/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/884954/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/884954/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/884954/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/884954/5"}}},"commit":{"parents":[{"commit":"55cd675f8b7b7f1d46d463141f86e242f8e5f7a6","subject":"Merge \"proxy: add new metrics to account/container_info cache for skip/miss\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55cd675f8b7b7f1d46d463141f86e242f8e5f7a6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-31 23:04:06.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-03 19:33:56.000000000","tz":-420},"subject":"Add non-ascii meta values to ssync probe test","message":"Add non-ascii meta values to ssync probe test\n\nChange-Id: I61c7780a84a1f0cee6975da67d08417cf6aa4ea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb468840b9c14ec3de748d9721685e507db446e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb468840b9c14ec3de748d9721685e507db446e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
