)]}'
{"id":"openstack%2Fswift~885165","triplet_id":"openstack%2Fswift~master~Id3920f911a2775513bbc1af8e5e730590771fd87","project":"openstack/swift","branch":"master","topic":"slow-get-info","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-06-03 00:23:00.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"Id3920f911a2775513bbc1af8e5e730590771fd87","subject":"some printf profiling to demonstrate the issue","status":"ABANDONED","created":"2023-06-02 22:52:50.000000000","updated":"2023-06-12 15:29:52.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"5c56888f2ae9578f10a896edd7084305f1e14c9e","_number":885165,"virtual_id_number":885165,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-03 00:07:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2023-06-02 22:54:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2023-06-02 22:54:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-02 23:56:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-03 00:07:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe12c792b07db38efdcee5c280860c278d96b856","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-06-02 22:52:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f6685ef483574ae627ccabd04b6bb30d75f0092","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-06-02 22:54:32.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0e75223970fa9133b0ca29565cdb2ab7b49b34d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 23:56:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abc2068e201c4353981554fe6472c4ef\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f7e4f330b2841fe9cc25feb8c57880a : FAILURE in 16m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ad5581080c642ea9fce246e5077149d : FAILURE in 1h 01m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b0589b367464358b8bcbcd20e34b426 : SUCCESS in 21m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/367c59c4f2244c4fb6d99b8126427d8f : SUCCESS in 17m 18s","accounts_in_message":[],"_revision_number":1},{"id":"4245ef71d5d243f187ed17fd6a6a6a53543b5ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-03 00:07:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a72a1083e57a4c6ea6bf4c64e9a9ee71\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5eeca71dad74024baefd469b5efcaf0 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/38c06e56107d473d89c27c53b23488e3 : SUCCESS in 1h 00m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d3358974dd9b4e3e8ec9117980d4aca3 : SUCCESS in 57m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8126401d9b842fe88325ae95c69dccc : SUCCESS in 1h 09m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db56cca59eab4294a267de1f2cd10860 : SUCCESS in 17m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c72ae6975554d519fd515ca49594cba : SUCCESS in 13m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94fab71c66754d81b6548a128b5916db : SUCCESS in 18m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5728dd70ce814a8585421cffdebb0367 : SUCCESS in 7m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fcd35735d1994b3fab82133b16fd315a : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0afdaf1478f44e7592b739d0b95a13f3 : FAILURE in 16m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0d5b0b7e00a1470e8ac9b97286fa23a3 : FAILURE in 16m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/36ff9d0c95d14b6b8d22e3d3fb7279a5 : FAILURE in 12m 35s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/086a08e7bec84a77ac8d16f02f03be0c : FAILURE in 16m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fd400518159a4d468a18750ee3b45a3c : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/21a318220f904eb1a76682158e431719 : SUCCESS in 18m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8761c61551284594aeb3b350a9ad1c5f : SUCCESS in 17m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f1b17af1e063450f941bf326dda41674 : SUCCESS in 14m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/467886f18ef0448fb23ac076f27b1892 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a423f06940ef455689a2351371445bb8 : SUCCESS in 15m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a8254ecdbde84e3587f61701739c8acb : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/03fae8c0d0894d96b9da8a2a34bf052d : SUCCESS in 16m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/147759cfd39e41faa5adf2c8c0f719ad : SUCCESS in 6m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0df4e71cbb1f41399c03d5b42460842b : SUCCESS in 39m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e4e4d035f3a41f7921aa255c24de715 : SUCCESS in 59m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/216b90e0e3c641558b6622e4f7a8a761 : SUCCESS in 55m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b8897c08443c4b1a9a50e5bffc70f41f : SUCCESS in 32m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bc6c81fd4044a4c9fa84f2f9ea763e4 : FAILURE in 16m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcb3e491e9554b69b6d85b10b66d7aa8 : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ed0e24ed2fe418d95a7d67150c668ce : SUCCESS in 18m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69b069fc1e6d4213b23624c73af3fe13 : SUCCESS in 40m 16s","accounts_in_message":[],"_revision_number":1},{"id":"cd9024408b395be3357a5dad326d2a914d795911","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-06-03 00:23:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c56888f2ae9578f10a896edd7084305f1e14c9e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-06-12 15:29:52.000000000","message":"Abandoned\n\nthe fix was developed in https://review.opendev.org/c/openstack/swift/+/885205","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9656cbc7bd3846bc0b292e7f084551a58a5e01a","revisions":{"f9656cbc7bd3846bc0b292e7f084551a58a5e01a":{"kind":"REWORK","_number":1,"created":"2023-06-02 22:52:50.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/65/885165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/885165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/885165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/885165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/885165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/885165/1"}}},"commit":{"parents":[{"commit":"776a9d9a121884b2f400ad632c15c9517aab2162","subject":"Use the db for get_shard_usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/776a9d9a121884b2f400ad632c15c9517aab2162"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-06-02 22:48:40.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-06-02 22:49:52.000000000","tz":-300},"subject":"some printf profiling to demonstrate the issue","message":"some printf profiling to demonstrate the issue\n\nChange-Id: Id3920f911a2775513bbc1af8e5e730590771fd87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9656cbc7bd3846bc0b292e7f084551a58a5e01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9656cbc7bd3846bc0b292e7f084551a58a5e01a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
